)]}'
{"id":"openstack%2Fos-brick~749697","triplet_id":"openstack%2Fos-brick~stable%2Fussuri~Ie8011fc83bc41edd2768ade6d42d737b4598ca2d","project":"openstack/os-brick","branch":"stable/ussuri","topic":"bug/1865754","hashtags":[],"change_id":"Ie8011fc83bc41edd2768ade6d42d737b4598ca2d","subject":"Fix for rbd connector to work with ceph octopus","status":"ABANDONED","created":"2020-09-03 09:59:14.000000000","updated":"2020-09-11 11:06:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32216bb1aa7e113b97548e218109e59d5528928f","_number":749697,"virtual_id_number":749697,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-03 10:04:53.000000000","_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"date":"2020-09-03 18:11:44.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-03 11:39:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-09-03 11:03:20.000000000","_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":-2,"date":"2020-09-08 15:05:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-09-03 13:02:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 10:04:53.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 11:03:20.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-09-03 11:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 13:02:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2020-09-03 18:11:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-08 15:05:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4b72526c6a23be8c963c77d3893714ebe548cba1","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-09-03 09:59:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ea944898fdaad9d61651a5f927f99ea6222bb6d","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-03 10:04:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.purestorage.com/97/749697/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/6ef69b5 : FAILURE in 4m 53s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.purestorage.com/97/749697/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/6b9e813 : FAILURE in 5m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"fa9ff4b5781371aa84f01915e6b781c2bd2ac84b","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-09-03 11:03:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/749697/1/iscsi/ : SUCCESS in 1h 03m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/749697/1/smb/ : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ac507a8767ca18ff494d3a56eb72a13bb5a9b5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:39:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9193403e3aeb469fb507219d2acd0956 : SUCCESS in 1h 16m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/680f0bd02bc948aabbf7354414f8601c : FAILURE in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336107b587144ec0b8ed601bae4993a4 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27a46c6751c54e5e979a1173431badb1 : FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b4590b133d240ffbaf9b59f16e84c13 : FAILURE in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9913effa2305473dbe8afdea1dde44d7 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bba26e3b03c4dbba7d56dd08aad9348 : SUCCESS in 4m 59s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/0d48d0d44b1a4b3490d0eb9c32d6fbb2 : SUCCESS in 1h 03m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46bb9922de514d4ead2b37d9624ed7fc : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"73176f1444fd2304cf61f32dd6826f488efe4acd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-09-03 13:02:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"90fc5e1993c603e3ce23f2bb5fe7320862041c0f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-03 18:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/97/749697/1/check/tempest-dsvm-full-bionic/71be6ab/ : SUCCESS in 1h 55m 46s\n- devstack-plugin-ceph-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/97/749697/1/check/devstack-plugin-ceph-dsvm-bionic/90ef901/ : SUCCESS in 1h 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e93c13f6273ee5eb9ad7f88689b64ec1ae5cf36e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-08 15:05:53.000000000","message":"Patch Set 1: Code-Review-2\n\nThis introduces a pretty heavyweight bindep to os-brick.  We\u0027re going to take an alternative path in master \u0026 victoria, and then will backport the alternative instead.","accounts_in_message":[],"_revision_number":1},{"id":"3eb2f2a5bf2abcc1bfd75da2458306e8ae904bcc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-11 11:06:26.000000000","message":"Abandoned\n\nBackport https://review.opendev.org/#/c/750655/ instead.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11c5a37bf5bf67123a08a702b97bbd1ccc8699db","revisions":{"11c5a37bf5bf67123a08a702b97bbd1ccc8699db":{"kind":"REWORK","_number":1,"created":"2020-09-03 09:59:14.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/97/749697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/97/749697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/749697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/749697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/749697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/97/749697/1"}}},"commit":{"parents":[{"commit":"86b2eb3dd4efe14919f003757200aad405dbad04","subject":"Merge \"Leverage the iSCSI mpath to get the WWN\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/86b2eb3dd4efe14919f003757200aad405dbad04"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-05-22 16:30:45.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-09-03 09:58:16.000000000","tz":60},"subject":"Fix for rbd connector to work with ceph octopus","message":"Fix for rbd connector to work with ceph octopus\n\nFound whilst testing cinder-backup with ceph on Ubuntu focal and\nOpenStack Ussuri.  Focal installs Ceph Octopus, and, as the linked bug\nshows, Octopus has slightly changed its requirements to need a\n[global] ini heading at the beginning of the config file that is\ngenerated within the RBDConnector() class.\n\nIncludes code to resolve which version of ceph is installed as a client\nto only change octopus and later.\n\nIt also fixes cinder Ceph backup driver.\n\nIncludes release note for the change.\n\nCo-Authored-By: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nChange-Id: Ie8011fc83bc41edd2768ade6d42d737b4598ca2d\nCloses-Bug: #1865754\n(cherry picked from commit 3e831480c29c07128f36ae67c44c25177b3d9f36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/11c5a37bf5bf67123a08a702b97bbd1ccc8699db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/11c5a37bf5bf67123a08a702b97bbd1ccc8699db"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
