)]}'
{"id":"openstack%2Fos-brick~762776","triplet_id":"openstack%2Fos-brick~master~I66e866700728eee7160f48455258c3974ada55bf","project":"openstack/os-brick","branch":"master","topic":"multipathd_resize","attention_set":{},"removed_from_attention_set":{"30696":{"account":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"last_update":"2022-06-07 09:14:30.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-06-07 09:14:30.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-06-07 09:14:30.000000000","reason":"Change was submitted"},"8543":{"account":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"last_update":"2022-06-07 08:02:28.000000000","reason":"Lucian Petrut replied on the change"}},"hashtags":[],"change_id":"I66e866700728eee7160f48455258c3974ada55bf","subject":"Avoid volume extension errors caused by multipath-tools version","status":"MERGED","created":"2020-11-16 06:23:57.000000000","updated":"2022-06-07 09:16:07.000000000","submitted":"2022-06-07 09:14:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":52,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"762776-multipathd_resize","meta_rev_id":"0bc028c84a3e9443195fae8700542a9ad87dc7f6","_number":762776,"virtual_id_number":762776,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-07 09:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":2,"date":"2022-06-07 07:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":2,"date":"2022-06-03 14:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":1,"date":"2022-06-07 07:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"value":0,"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 07:09:46.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 09:50:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 10:15:33.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 11:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 12:04:27.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 12:16:42.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"REVIEWER"},{"updated":"2020-11-16 12:23:02.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 18:03:24.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 18:25:57.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 18:26:07.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-23 07:44:25.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2020-12-23 19:05:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-16 15:16:07.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-03-31 15:04:08.000000000","updated_by":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"reviewer":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"state":"CC"},{"updated":"2022-03-31 18:56:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-31 19:20:27.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-07 12:17:10.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-04-28 14:00:50.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2022-04-28 14:00:50.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"6ffcc8a9924ff0a47e4170cd60f847b06297e39e","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-11-16 06:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd9b27ba0e4c56863a5c89340599b3c2b9e963ba","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-16 07:09:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/76/762776/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/2649d4c : FAILURE in 20m 53s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/76/762776/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/4992901 : FAILURE in 19m 15s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"5fc069521c7707e8988d01d0438c71c9ba7b0e88","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-11-16 07:23:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/1/iscsi/ : SUCCESS in 58m 49s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/1/smb/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e961a9af73cb4e82b059d4e69dc085ef6c5569f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-16 07:48:08.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"53a08772e8362cb241852a5de6ae5ca4c1289428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 08:26:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a65375dbacd44b23ab11a787c1252ed9 : SUCCESS in 1h 07m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7424f244c5f49a38594594e7d194f15 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f49662c74f6488396b4f942761cbc8a : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd82427e92ab4ad4b958901137e1bd5e : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b24f757d618499bbda18452c21aa716 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5165472a9cff4750aa6bc09623cb5884 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5717d9933fe842848f439314da8efca8 : SUCCESS in 5m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f37146b69f2d4b3689924b1ec15d1d44 : SUCCESS in 4m 12s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/31e400b703134aa3a5942f76628fbd72 : SUCCESS in 1h 18m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7587f68eb52942989067a07e20599c7a : FAILURE in 1h 40m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/20efeaac5e6743b29e65aade991ca273 : SUCCESS in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90b5f204c17be8010b6a2ea23f022ded4f633715","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-11-16 09:16:30.000000000","message":"Patch Set 1:\n\nRecheck!","accounts_in_message":[],"_revision_number":1},{"id":"9c48ad83760b42bd355ca188af9c87aa1cb59e8c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-16 09:50:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/1/check/devstack-plugin-ceph-dsvm-focal-py3/6748d00/ : TIMED_OUT in 2h 40m 16s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/1/check/tempest-dsvm-full-focal-py3/7d5a92d/ : SUCCESS in 1h 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9bd3f94ec0e38f86489c3c69e0e49dfc4b5eb11d","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-11-16 10:15:33.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/1/iscsi/ : SUCCESS in 58m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/1/smb/ : FAILURE in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4a3949c8de4a5f635a9d9ad0107b46a15b8576c1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 10:39:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_FC_os-brick/03d134f/DellEMC_XtremIO_FC_os-brick/1 : SUCCESS in 53m 13s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_iSCSI_os-brick/50ac239/DellEMC_XtremIO_iSCSI_os-brick/1 : SUCCESS in 1h 18m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"5bee4a3e1e05a9bae7aa0fdaccffe787bef94711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 11:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bfaf282be6940e19b9c9c88565cd203 : SUCCESS in 1h 40m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ec4af0ef6934f5b9e7f4f7ed668368e : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c4cf2871a884874abb735defcb46ec5 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2504299a66ff44e492c9d480248e8a5f : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20c099aa0d0a448f8c8643050f2d58e0 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/544b704ecfcd4fb5bc58908ccbf3000a : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae3d606564840559479b4eebcd679d5 : SUCCESS in 7m 29s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7b1f5ccb748f44e09cf5e16d35d7f3ca : SUCCESS in 5m 42s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/57714964e4cb4c209c1aba0e3d6a5455 : SUCCESS in 1h 10m 44s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee455f01dcdf464f805082fbf5b204b6 : SUCCESS in 1h 29m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5b635131832c48b3baa81060327586c5 : SUCCESS in 1h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0826e1451aab70b554fa943ad8afcf917930e29e","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-11-16 12:04:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/76/762776/1/upstream-solidfire-check/cinder-solidfire-iSCSI/59b2374/ : FAILURE in 1h 00m 00s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"38cc7a3d2b2cf32db0836571fd5b7c905ba69185","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-16 12:13:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d4bb7323ac2befdda19f15e296f577f8eb10ea2e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-11-16 12:16:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/76/762776/1/check/DellEMC_PowerStore_FC-ob/574ccbd/DellEMC_PowerStore_FC-ob/5 : SUCCESS in 1h 12m 30s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/76/762776/1/check/DellEMC_PowerStore_iSCSI-ob/8a2d221/DellEMC_PowerStore_iSCSI-ob/5 : SUCCESS in 1h 11m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"fc6f8654ea4287ba121303070021f4ab9593016e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-16 12:23:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-iSCSI/ba5c32b/ : SUCCESS in 38m 10s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-iSCSI-IPv6/513c8ff/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-NFS/1ad3322/ : SUCCESS in 39m 59s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-NFS-IPv6/75f56ed/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-FCP/3538a44/ : SUCCESS in 1h 18m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"84835b975fb2cd2b567ab138ba3ca0b97743df6f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-16 14:08:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ecc14abb59181a7f0c00b5df2980e3d37ddef717","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 18:03:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_VNX_FC_os-brick/d28583b/DellEMC_VNX_FC_os-brick/4 : SUCCESS in 1h 06m 51s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_VNX_iSCSI_os-brick/4627c26/DellEMC_VNX_iSCSI_os-brick/4 : SUCCESS in 1h 55m 45s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1e2c0ef6dba63ad2250989990e178bc6c9f49a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-16 18:03:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"69796c5bcf0fabdd893a4b35d315bfaf9aca9cfa","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-16 18:25:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_FC_os-brick/83f9cae/DellEMC_XtremIO_FC_os-brick/2 : SUCCESS in 55m 22s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_iSCSI_os-brick/6bdd844/DellEMC_XtremIO_iSCSI_os-brick/2 : SUCCESS in 1h 16m 17s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"181088e4e2475847f18483ccf2b553b232aad019","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-16 18:26:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"27a62020038ec1cd7a0066a1439d02bf978403af","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-12-23 03:16:27.000000000","message":"Patch Set 1:\n\nrecheck!","accounts_in_message":[],"_revision_number":1},{"id":"f9f2b2b560455aff759a34727b1121175f4d636f","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-12-23 04:12:40.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/1/iscsi/ : FAILURE in 53m 20s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/1/smb/ : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":1},{"id":"197f79334850c134181b5d4bd72d910413b219c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 04:53:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ced322405b064fb09f372f1c994140c2 : SUCCESS in 1h 28m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4fd8239623e4eb4b2d8496f6952e177 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c55fd906a241eaadbaecb6ed872525 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb90bbb4f964d719d8f4e8861594eed : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a98f877366f74ab2aa7dcea9095475fc : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a91f2d7e56748fabe04d62bec043462 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6713d60f6f94d80b81a81c0dfd8462e : SUCCESS in 5m 38s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/80df6972730e47e18386da018aa33c77 : SUCCESS in 3m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e0cc56b55d084cac83e3796bfcbda142 : SUCCESS in 1h 04m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b984468cd7e343e29018850395af30b3 : SUCCESS in 1h 28m 50s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e7a0219df03840c4a7c7b97e2bc1a0e2 : SUCCESS in 1h 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5d055befcce9f9d5b9cb68ca15d2a45ac361eaa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-23 04:59:48.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/1/check/devstack-plugin-ceph-dsvm-focal-py3/dbfd80c/ : FAILURE in 29m 08s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/1/check/tempest-dsvm-full-focal-py3/a0ed990/ : SUCCESS in 1h 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f3b8b60a16fd0b893345b1bb373b750aea0751b6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-23 05:53:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-iSCSI/22c1dc2/ : SUCCESS in 40m 12s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-iSCSI-IPv6/d7504c0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-NFS/aeacb05/ : SUCCESS in 40m 45s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-NFS-IPv6/ee9c5f9/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/76/762776/1/upstream-check/cinder-cDOT-FCP/205e769/ : SUCCESS in 58m 50s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"df1d513e0ca8ea6af10dfb5f3df52f12a6c9e0b6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-23 05:56:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/76/762776/1/upstream-solidfire-check/cinder-solidfire-iSCSI/0148f62/ : FAILURE in 1h 02m 05s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"397ea1ddca65f71818ce77411fb678639499bc23","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-23 06:07:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_FC_os-brick/a7809c6/DellEMC_XtremIO_FC_os-brick/53 : SUCCESS in 51m 51s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_XtremIO_iSCSI_os-brick/7ed54c3/DellEMC_XtremIO_iSCSI_os-brick/50 : SUCCESS in 1h 13m 25s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"9bb05f9c24d39dea53a90dab2f844d5a8831b086","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-12-23 06:07:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/76/762776/1/check/DellEMC_PowerStore_FC-ob/023323d/DellEMC_PowerStore_FC-ob/13 : SUCCESS in 1h 13m 37s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/76/762776/1/check/DellEMC_PowerStore_iSCSI-ob/d126897/DellEMC_PowerStore_iSCSI-ob/13 : SUCCESS in 1h 09m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"2fef68f87152ffbe9a76e6bf5d3bca98c7be5fc8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-23 07:01:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4d9bcd346a9d4cd244fc4e1049c95f4a2027a2ce","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-12-23 07:44:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_SC_FC-os-brick/d27f9d6/DellEMC_SC_FC-os-brick/12 : SUCCESS in 1h 15m 00s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_SC_iSCSI-os-brick/930e0e1/DellEMC_SC_iSCSI-os-brick/12 : FAILURE in 2h 50m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"a2575392258ea1e10b2e9c5fa560f37678066a09","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-23 07:56:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_VNX_FC_os-brick/c191d85/DellEMC_VNX_FC_os-brick/15 : FAILURE in 1h 07m 04s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_VNX_iSCSI_os-brick/f65bc2d/DellEMC_VNX_iSCSI_os-brick/15 : FAILURE in 1h 55m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"ba45a6a8d196998ff5d1a46a4d3f472262740a2c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-23 08:57:06.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e6f3ff592ae034f071b79a00bbe375200eff8238","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-23 19:05:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/76/762776/1/check/DellEMC_PowerFlex_os-brick/a5e5c2f/DellEMC_PowerFlex_os-brick/13 : ABORTED in 4h 00m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"14d4a0b26b91ec55ec193cbd392297e5e74a2716","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-23 20:28:26.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/76/762776/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f3ba746cf06bac4ca44e74fc4efd080e770faf51","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-04-16 15:16:07.000000000","message":"Patch Set 1: Code-Review-1\n\nI do not think it is safe to put what could be an infinitely loop in the code.  We should find a better solution.  A limited number of retries at a minimum would be good.","accounts_in_message":[],"_revision_number":1},{"id":"83c848d8db1de31fb137f1a36eef21cc23e83ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-03-31 15:01:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"266fe8d8dcefe7798efe836bc60ca735d8978c02","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-03-31 15:04:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ddc87498a8fd2977a347f543ac660fc132af6b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-03-31 16:28:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/2/iscsi/ : SUCCESS in 1h 06m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/2/smb/ : SUCCESS in 1h 26m 29s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/2/ceph/ : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0c2dfc654f0729dc46d2f18e4ea22a090c03f735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-03-31 16:51:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38c87cfb137c4fab3dc57d7d734185608379645f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-03-31 17:48:41.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/3/iscsi/ : FAILURE in 54m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/3/smb/ : FAILURE in 54m 46s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/3/ceph/ : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":3},{"id":"b53cc75b36f3e20e1f282d525a87442a304b5d27","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-03-31 18:06:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"247b30493a900aeb30d0f2b8d654a32f2240a513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 18:38:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85462f6289494fbea8118696b9e042a2 : SUCCESS in 1h 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f965f8baf8474e66a43ec81aeeea9c12 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f3edd40fcf48f1a9040f149ce6fae3 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dbfc272cea34c3dbc835f2e2492ffc4 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa3653dddb145e5a9f1c067f1e38966 : SUCCESS in 8m 04s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/aadcb2de7e534d5895567bcfd63c35df : SUCCESS in 5m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/1c43df93a8ad4aebbda36444b1082456 : SUCCESS in 1h 22m 14s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/faddda8bb2604b70a11146dffe877656 : SUCCESS in 1h 41m 15s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1c043c2313fc46dd861bd29bc04ab9db : SUCCESS in 1h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9d119bbc2b8387f4bd7246f8891a35c8f0161d8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-03-31 18:38:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_PowerFlex_os-brick/f02685e/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"2868caf5db6ffecb98e0fdc41522fb2746219044","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-31 18:38:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/762776/3/upstream-solidfire-check/cinder-solidfire-iSCSI/ec01745/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"56d241ba85ef66e92379c9b1d786f993716fbf23","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-31 18:56:27.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_Unity_FC_os-brick/6512aff/DellEMC_Unity_FC_os-brick/33 : SUCCESS in 1h 10m 24s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_Unity_iSCSI_os-brick/b8e4653/DellEMC_Unity_iSCSI_os-brick/45 : SUCCESS in 2h 04m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e8760affffd9ce13ac5c4608dcc03fcc62341d09","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-03-31 19:20:27.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/31_03_2022_18_58_01\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"116f7bf5ad440f68709ffd24998cc1bb48d1efff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-31 20:09:50.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ad69b936a6c84180ac1165365ce6905e : TIMED_OUT in 3h 15m 18s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/66ceea1a42f645f69197ff80f50cd285 : POST_FAILURE in 1h 57m 34s","accounts_in_message":[],"_revision_number":3},{"id":"bc560164d96c3cd06d9f8fb37d7640f60358246f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-03-31 23:46:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_PowerStore_FC-ob/fab2d33/DellEMC_PowerStore_FC-ob/38 : SUCCESS in 1h 15m 24s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_PowerStore_iSCSI-ob/ee3bdbc/DellEMC_PowerStore_iSCSI-ob/35 : SUCCESS in 1h 21m 28s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"b3efd719b3bba61d172bae624c7662181b9210e9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-01 01:04:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/3/iscsi/ : SUCCESS in 1h 07m 32s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/3/smb/ : SUCCESS in 1h 25m 25s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/3/ceph/ : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":3},{"id":"3c9a47ec5adef855295383395d14b2d9bcb9db2f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-01 03:15:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_SC_FC-os-brick/d8ce7f1/DellEMC_SC_FC-os-brick/34 : SUCCESS in 1h 11m 05s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/76/762776/3/check/DellEMC_SC_iSCSI-os-brick/aced3ed/DellEMC_SC_iSCSI-os-brick/44 : FAILURE in 1h 58m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"d40986f692cffcd4399f52fae8558ad11c54d968","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-04-07 12:17:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6b1db9217845bf2c5723f24fd4323cd6633c6323","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-04-08 06:20:38.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7eed892e31bdc4199c9f54dd758a1fbca72e4207","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-04-08 09:54:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49b04192e1131880e058624d330649e6446d5dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-04-08 10:16:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"39af9988e3417bf434e31ce0a930f688a83a2367","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-08 10:23:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/4/check/DellEMC_Unity_FC_os-brick/42f1b7f/DellEMC_Unity_FC_os-brick/42 : FAILURE in 1m 33s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/4/check/DellEMC_Unity_iSCSI_os-brick/070114c/DellEMC_Unity_iSCSI_os-brick/54 : FAILURE in 6m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b132e7e25f25ae953a1e181066eb6d68c0e98684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-04-08 10:26:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eacb990eea8f2a73cafc7e669dbf1dbbd5ac483c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-08 10:28:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_Unity_FC_os-brick/81207ff/DellEMC_Unity_FC_os-brick/43 : FAILURE in 1m 34s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_Unity_iSCSI_os-brick/a033760/DellEMC_Unity_iSCSI_os-brick/55 : FAILURE in 1m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"10104e241d4f0127b42a987a5562b5dbdc8a1a2b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-08 11:34:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/5/iscsi/ : SUCCESS in 1h 07m 39s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/5/smb/ : SUCCESS in 57m 10s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/5/ceph/ : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":5},{"id":"03089db30af35e0eadda4ae89d43a1a1da54b978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:29:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/061ab849f19f48feb930f560391a57f7 : SUCCESS in 1h 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da147fca551f440b973111855d94aa9b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa3c19e576a2469aab855f9cd6e83767 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55d65b83b3a441fdb7e04f5e8caede24 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cefded67124fc285cb85486976a440 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f8c99e379774a0fb1c9f6b2d75e11a7 : SUCCESS in 4m 27s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0152dc51d2bc4b15b3de4e7909990766 : SUCCESS in 5m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/eea51e356e1f45189f20c5f7458821b4 : SUCCESS in 1h 25m 57s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3dce808ba34149488609241075be9f50 : SUCCESS in 1h 54m 55s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/09a29e9c31ff46e8b0379055cdcc00cf : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8afc34b2a7d0eb04979f7b53c33dae372cc26f7b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-08 12:29:21.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_PowerFlex_os-brick/296c471/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"44eb4b85afebcfbd97877edb75bfbf048d95a8ab","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-04-08 12:29:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/762776/5/upstream-solidfire-check/cinder-solidfire-iSCSI/6d2573d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"b08694528b705c7165f75fd0def8767b067cde1f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-08 12:30:49.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_Unity_FC_os-brick/34ea6d8/DellEMC_Unity_FC_os-brick/45 : FAILURE in 1m 32s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_Unity_iSCSI_os-brick/6cb4536/DellEMC_Unity_iSCSI_os-brick/57 : FAILURE in 1m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"be35916adce77730ada639dffdd62c0a31128525","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-08 12:30:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_SC_FC-os-brick/3f4e2ff/DellEMC_SC_FC-os-brick/44 : FAILURE in 1m 15s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_SC_iSCSI-os-brick/9def19d/DellEMC_SC_iSCSI-os-brick/54 : FAILURE in 1m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"6bd5a7b1c72889a82a899f256926ab9f5681af55","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-08 12:34:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/762776/5/upstream-check/cinder-cDOT-iSCSI/2708f1b/ : FAILURE in 2m 05s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/762776/5/upstream-check/cinder-cDOT-NFS/0f517cf/ : FAILURE in 2m 45s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/762776/5/upstream-check/cinder-cDOT-FCP/6df7bad/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"a0582dc56415248a5e54175d2ac6d47753a7f5b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-08 13:06:23.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/08_04_2022_12_45_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"29635e9a5361f0399df4aa793d57ddd86c0a7b9b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-08 13:46:54.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/653c91c05fc3491389486d9ecd537349 : POST_FAILURE in 2h 32m 38s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f7d80d45f6474ee69a6d5421806ea400 : POST_FAILURE in 1h 48m 08s","accounts_in_message":[],"_revision_number":5},{"id":"baca596be51f19f71908b80797a4d0c87d0bf48b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 12:18:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_XtremIO_FC_os-brick/464156f/DellEMC_XtremIO_FC_os-brick/117 : FAILURE in 2m 12s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/5/check/DellEMC_XtremIO_iSCSI_os-brick/5b67fd8/DellEMC_XtremIO_iSCSI_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"ac3c85897e55017be5b07ec959d3c0c27f69a4b9","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-03 09:16:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c5fd865b313fe948ad879187f549a3b8aea0b13","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-06 12:17:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4f80f91efca886c02e7b754514b4ef85f3000183","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-06 16:18:38.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f501cb2b4d590947142dda6c19bd747892c06467","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-06 16:24:23.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"943eeff5d8f2b8d974c2bc6ef43e58c7641ef316","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-09 14:13:29.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1b2552719b803a22fe8576e6c06d0db8575d4f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-12 07:53:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0da6cf5bcea97d930fc9ec76f703cbce7f6452c0","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-12 08:04:24.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ff5fa314724163c89c24f790bf82c5c0f93690ea","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-12 08:52:01.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/6/iscsi/ : SUCCESS in 54m 26s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/6/smb/ : SUCCESS in 56m 40s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/6/ceph/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":6},{"id":"d8406880ecf811e29415656894d619ce7d47fcc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 10:08:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/298592952bc4469b9d7445d0266eda35 : SUCCESS in 1h 22m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93cd395b6d854bccb486ec5ea619a5b6 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d848e366584f6fa01269e4a8d64779 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d40ebd88fb524e63856348bce1a59bfb : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e04dbd79a5425ba9b47a031f149ae4 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/231b8d6eef8c4e8f97cbfcc928484838 : SUCCESS in 4m 52s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3450c7fb6bfd4b31b1063989c9fc132a : SUCCESS in 4m 29s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/24e50355eaa94a7cacbbc34fc12bba34 : TIMED_OUT in 2h 07m 49s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4a288a21e9d5446e8b5ca9c7e55fbf08 : SUCCESS in 2h 02m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6120e0372b9747d4a04fd289459821f6 : SUCCESS in 1h 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3be9ef1b36eb2725aa37549478c64a78611edd32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-12 10:08:19.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"0579badab7e0591c35e999119a287567061e1d52","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-12 10:08:23.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/762776/6/check/DellEMC_PowerFlex_os-brick/bab498f/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"ebce30f49bd739bf2790cb1558ec01c91ec50b1c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-12 10:15:37.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4ec1c34e85d94bc2a0c92169a62eb9bc : POST_FAILURE in 2h 20m 01s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/76827ed311704e24a1c1f4180d9ab3d5 : POST_FAILURE in 1h 40m 10s","accounts_in_message":[],"_revision_number":6},{"id":"16080b59041fce8132dd5dc7f10c276551f4f26c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-05-12 11:26:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/76/762776/6/check/DellEMC_PowerStore_FC-ob/62f53f5/DellEMC_PowerStore_FC-ob/58 : SUCCESS in 1h 11m 10s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/76/762776/6/check/DellEMC_PowerStore_iSCSI-ob/339b6eb/DellEMC_PowerStore_iSCSI-ob/55 : SUCCESS in 1h 18m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"da4701a6d7058922a930cd6f1537415815718ef0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-12 13:13:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick https://elab-os-logsrv.delllabs.net/76/762776/6/check/DellEMC_VNX_FC_os-brick/25a720b/DellEMC_VNX_FC_os-brick/33 : FAILURE in 14m 10s\n- DellEMC_VNX_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/762776/6/check/DellEMC_VNX_iSCSI_os-brick/e160ea7/DellEMC_VNX_iSCSI_os-brick/35 : SUCCESS in 2h 08m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"25dfc996b40bfde994ff35c71e816e5d89f9aa2b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-05-25 11:49:49.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"64f0a2e0e99a6e71433b7d82834f649251e78d93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-27 07:48:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f5980ea786f8e911a286fc8e3cbb1d5617739702","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-05-27 07:52:51.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"629bde25516a374b599b5dd1199ab4093d0a149d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-27 08:49:15.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/762776/7/iscsi/ : SUCCESS in 58m 31s\n- os-brick-smb http://cloudbase-ci.com/os-brick/762776/7/smb/ : SUCCESS in 57m 30s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/762776/7/ceph/ : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":7},{"id":"de94e4c9a0e197b971c560c0f0f2e42ff699ce37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 09:30:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e406036092343978a0848613ce76ef3 : SUCCESS in 1h 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede162a3ef5d436b89555df8900637df : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/753ab5d4ae3b48e3aa0e51d749fe2d43 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b6b5d38154a4682ab404768af88a503 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d152500a631744deb85647769e5c9f41 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e7b928185049b49f49e8ae11ddcc61 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66bfbfe508f1421783c3043d08ada759 : SUCCESS in 4m 06s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/609a8a580eff476099628fc39fa502b8 : SUCCESS in 5m 13s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/c31441d779c448708f1f35b03a4d96fe : FAILURE in 1h 15m 41s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8574bbe5f2da4147ae2ece2b48c7fd91 : SUCCESS in 1h 39m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/36cf39db04924860a0bc0a8672e18f93 : SUCCESS in 1h 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b613d94850bf83a8039687a39a4b33376f9ed62c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-27 09:30:52.000000000","message":"Patch Set 7:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"f26845fc2ceac21892ec7bf75c5cdde67b8f1ab5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-27 11:08:04.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/7/check/devstack-plugin-ceph-dsvm-focal-py3/a6d5b1e/ : SUCCESS in 2h 57m 50s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/76/762776/7/check/tempest-dsvm-full-focal-py3/9e6a1ba/ : SUCCESS in 2h 15m 30s","accounts_in_message":[],"_revision_number":7},{"id":"44ddd4ba3f791bdb1950b7ddf8353e4d32e54b3b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-03 14:09:18.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"55c9e9df97dd77caaf24ac4577067a218cdea3ae","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-06-07 06:15:13.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"57bdbd2ab89614827741997b60059e95ddb8b850","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-06-07 07:10:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"78f99f0553230826bc5506df718d2209ef5e95f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 07:10:45.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b56493bad1119279ae836ce9b30e71d6270c0e12","author":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"date":"2022-06-07 08:02:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ab7bf22939a3ecc5823b20b309720823d9f7b6e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 09:14:30.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b8eaa8837724861b1d71b10ee244c38 : SUCCESS in 1h 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35cc6f171fa142b9ab2f301924302819 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6ef685ef5e54a34ac5303eb212a2efc : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec53ede5108f4d118abc7bb7b8a26200 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2318187c02ed444b95fbb964cb2fdf72 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5d96282d1f42aaa967128518ee5476 : SUCCESS in 4m 09s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08d8dae210ca4bd387ad75ba66b76d40 : SUCCESS in 1h 58m 02s","accounts_in_message":[],"_revision_number":7},{"id":"659549440d3c8f06ab60a6d8783dd3714face60a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 09:14:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"0bc028c84a3e9443195fae8700542a9ad87dc7f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 09:16:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc9b8f328dd44a6bd18c781ba9b58be : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/143e8a0e25a0423a91e5b084216ed018 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"557f38677a07386807b8d284a20b0ecaa61490f9","revisions":{"d309a3ace2fcbb3979509ae64661e6b509129b4a":{"kind":"REWORK","_number":1,"created":"2020-11-16 06:23:57.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/76/762776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/1"}}},"commit":{"parents":[{"commit":"1425a834ddc06d8b6be50a18408a5aae6c001bf3","subject":"Replace md5 with oslo version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1425a834ddc06d8b6be50a18408a5aae6c001bf3"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:22:07.000000000","tz":480},"subject":"Make sure extend_volume succeed despite multipath-tools version","message":"Make sure extend_volume succeed despite multipath-tools version\n\nThe `multipathd reconfigure` has became a asynchronous command\nsince the 0.6.1 version of multipath-tools. That leads to fail\nto extend in-use fibre channel volume, because `multipathd resize\nmap` will output \u0027timeout\u0027 before `multipathd reconfigure` command\nfinishes when `multipathd resize map` command will be executed as\nsoon as `multipathd reconfigure` command is executed.\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d309a3ace2fcbb3979509ae64661e6b509129b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d309a3ace2fcbb3979509ae64661e6b509129b4a"}]},"branch":"refs/heads/master"},"9378a479fc92a85a445671406b4fe8dc41ba015f":{"kind":"REWORK","_number":2,"created":"2022-03-31 15:01:42.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/2"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-03-31 15:00:47.000000000","tz":180},"subject":"Make sure extend_volume succeed despite multipath-tools version","message":"Make sure extend_volume succeed despite multipath-tools version\n\nThe `multipathd reconfigure` has became a asynchronous command\nsince the 0.6.1 version of multipath-tools. That leads to fail\nto extend in-use fibre channel volume, because `multipathd resize\nmap` will output \u0027timeout\u0027 before `multipathd reconfigure` command\nfinishes when `multipathd resize map` command will be executed as\nsoon as `multipathd reconfigure` command is executed.\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9378a479fc92a85a445671406b4fe8dc41ba015f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9378a479fc92a85a445671406b4fe8dc41ba015f"}]},"branch":"refs/heads/master"},"75579922e5aa2a0349d03b60a8ab0a4df35ec44a":{"kind":"REWORK","_number":3,"created":"2022-03-31 16:51:49.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/3"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-03-31 16:50:52.000000000","tz":180},"subject":"Make sure extend_volume succeed despite multipath-tools version","message":"Make sure extend_volume succeed despite multipath-tools version\n\nThe `multipathd reconfigure` has became a asynchronous command\nsince the 0.6.1 version of multipath-tools. That leads to fail\nto extend in-use fibre channel volume, because `multipathd resize\nmap` will output \u0027timeout\u0027 before `multipathd reconfigure` command\nfinishes when `multipathd resize map` command will be executed as\nsoon as `multipathd reconfigure` command is executed.\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75579922e5aa2a0349d03b60a8ab0a4df35ec44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75579922e5aa2a0349d03b60a8ab0a4df35ec44a"}]},"branch":"refs/heads/master"},"4f00068d73d427c8bdf8dd19a51555feb7eb5795":{"kind":"REWORK","_number":4,"created":"2022-04-08 10:16:51.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/4"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-04-08 10:14:57.000000000","tz":180},"subject":"Make sure extend_volume succeed despite multipath-tools version","message":"Make sure extend_volume succeed despite multipath-tools version\n\nThe `multipathd reconfigure` has became a asynchronous command\nsince the 0.6.1 version of multipath-tools. That leads to fail\nto extend in-use fibre channel volume, because `multipathd resize\nmap` will output \u0027timeout\u0027 before `multipathd reconfigure` command\nfinishes when `multipathd resize map` command will be executed as\nsoon as `multipathd reconfigure` command is executed.\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f00068d73d427c8bdf8dd19a51555feb7eb5795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f00068d73d427c8bdf8dd19a51555feb7eb5795"}]},"branch":"refs/heads/master"},"a35e9f9eadb799dc03398e400897ac171c9eb267":{"kind":"REWORK","_number":5,"created":"2022-04-08 10:26:57.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/5"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-04-08 10:26:07.000000000","tz":180},"subject":"Make sure extend_volume succeed despite multipath-tools version","message":"Make sure extend_volume succeed despite multipath-tools version\n\nThe `multipathd reconfigure` has became a asynchronous command\nsince the 0.6.1 version of multipath-tools. That leads to fail\nto extend in-use fibre channel volume, because `multipathd resize\nmap` will output \u0027timeout\u0027 before `multipathd reconfigure` command\nfinishes when `multipathd resize map` command will be executed as\nsoon as `multipathd reconfigure` command is executed.\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a35e9f9eadb799dc03398e400897ac171c9eb267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a35e9f9eadb799dc03398e400897ac171c9eb267"}]},"branch":"refs/heads/master"},"8c13de493db134ed8fab160f3ff1d061a245441b":{"kind":"REWORK","_number":6,"created":"2022-05-12 07:53:49.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/6"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-05-12 07:53:29.000000000","tz":180},"subject":"Avoid volume extension errors caused by multipath-tools version","message":"Avoid volume extension errors caused by multipath-tools version\n\n`multipathd reconfigure` is an asynchronous command as of\nmultipath-tools 0.6.1 [1][2], potentially even before that [3].\n\nExtending in-use iSCSI or FC volumes can fail because of that\nas `multipathd resize map` will output \"timeout\" while the\n\"multipathd reconfigure\" operation is still in progress.\n\nThis commit will ensure that multipathd errors are handled\naccordingly, retrying in case of timeouts for up to 2 minutes.\n\n[1] https://github.com/openSUSE/multipath-tools/blob/0.6.0/multipathd/main.c#L997\n[2] https://github.com/openSUSE/multipath-tools/blob/0.6.1/multipathd/main.c#L1135\n[3] https://github.com/opensvc/multipath-tools/blob/b21c567961f518810a1ac3b209c8db45f6dbac2c/multipathd/cli_handlers.c#L847-L851\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8c13de493db134ed8fab160f3ff1d061a245441b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8c13de493db134ed8fab160f3ff1d061a245441b"}]},"branch":"refs/heads/master"},"557f38677a07386807b8d284a20b0ecaa61490f9":{"kind":"REWORK","_number":7,"created":"2022-05-27 07:48:03.000000000","uploader":{"_account_id":8543,"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","username":"plucian"},"ref":"refs/changes/76/762776/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/762776/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/762776/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/762776/7"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-11-16 06:08:59.000000000","tz":480},"committer":{"name":"Lucian Petrut","email":"lpetrut@cloudbasesolutions.com","date":"2022-05-27 07:47:54.000000000","tz":180},"subject":"Avoid volume extension errors caused by multipath-tools version","message":"Avoid volume extension errors caused by multipath-tools version\n\n`multipathd reconfigure` is an asynchronous command as of\nmultipath-tools 0.6.1 [1][2], potentially even before that [3].\n\nExtending in-use iSCSI or FC volumes can fail because of that\nas `multipathd resize map` will output \"timeout\" while the\n\"multipathd reconfigure\" operation is still in progress.\n\nThis commit will ensure that multipathd errors are handled\naccordingly, retrying in case of timeouts for up to 2 minutes.\n\n[1] https://github.com/openSUSE/multipath-tools/blob/0.6.0/multipathd/main.c#L997\n[2] https://github.com/openSUSE/multipath-tools/blob/0.6.1/multipathd/main.c#L1135\n[3] https://github.com/opensvc/multipath-tools/blob/b21c567961f518810a1ac3b209c8db45f6dbac2c/multipathd/cli_handlers.c#L847-L851\n\nChange-Id: I66e866700728eee7160f48455258c3974ada55bf\nCloses-Bug: #1888675\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/557f38677a07386807b8d284a20b0ecaa61490f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/557f38677a07386807b8d284a20b0ecaa61490f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
