)]}'
{"id":"openstack%2Fos-brick~763599","triplet_id":"openstack%2Fos-brick~master~I9a1746658474c0f1be7ec29a36767085aaf2ab7f","project":"openstack/os-brick","branch":"master","hashtags":[],"change_id":"I9a1746658474c0f1be7ec29a36767085aaf2ab7f","subject":"Improve error handling on target query","status":"MERGED","created":"2020-11-21 19:30:44.000000000","updated":"2021-01-15 16:37:31.000000000","submitted":"2021-01-08 20:12:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763599","meta_rev_id":"f642069f107bf1892980d3f21fee5efed42c7b4a","_number":763599,"virtual_id_number":763599,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 20:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-01-08 18:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-12-10 13:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2021-01-07 18:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-01-08 18:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-21 19:52:12.000000000","updated_by":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-21 20:43:18.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2020-11-21 20:56:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-21 21:10:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-21 22:22:56.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-23 07:47:23.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 11:03:20.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2020-12-05 13:37:24.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-05 13:48:56.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2020-12-05 13:57:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2020-12-05 14:07:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-05 14:49:27.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2020-12-05 15:16:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-05 19:02:06.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-09 16:55:19.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2020-12-10 13:29:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-01-07 18:34:15.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2021-01-08 18:24:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"1befc5301dbd2334a3b109301e2a5d49cb0d9ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2020-11-21 19:30:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5074bbaf921d938a01764ef33d27c881e25c3ba3","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-21 19:52:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/99/763599/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/e5337b4 : FAILURE in 21m 11s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/99/763599/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/22ba4e9 : FAILURE in 21m 11s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7a2c12270911d2bf97598f1fbf47133213561aa","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-11-21 20:43:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/763599/1/iscsi/ : SUCCESS in 1h 12m 11s\n- os-brick-smb http://cloudbase-ci.com/os-brick/763599/1/smb/ : FAILURE in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5a61cdd4bee031e11d79294a8de2dcc8a37df353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-21 20:56:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1904614521eb408395402a271e5b813b : SUCCESS in 1h 23m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85a4db5ec7354361a751b4b81ce33b51 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7362e79192134e849676d6f1c69715f4 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f186ea7cf3a34c668ce4c36957ea3a4b : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c88826614e5493790314538da305622 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/746bd597984144f4a0e66860d8c4d885 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24eaf0fc3b9d42dc9aa053841c6950b1 : SUCCESS in 5m 24s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/96db6a0b34cc4b5e828ff24771e3509d : SUCCESS in 3m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/f64742ade44b4fcaa4fceadb5414ce9a : SUCCESS in 1h 24m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7736ad02d45248fda27d733e83897559 : FAILURE in 1h 20m 28s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/571f5129917d4a66bd1cfc660de63835 : SUCCESS in 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8af908c3afc5e25462647c5849208919e1948bc6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-21 21:10:42.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/devstack-plugin-ceph-dsvm-focal-py3/c4be937/ : FAILURE in 1h 33m 05s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/tempest-dsvm-full-focal-py3/9c20502/ : FAILURE in 1h 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b598bada6d6c7deeca580d24a1878130d3b28565","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2020-11-21 21:58:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b3175b71d30e714c02b77d789522fdc1e57ce67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-21 22:22:56.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"fc239af32d12585a1de9f43105ccd8de83cc3597","author":{"_account_id":12369,"name":"Pure Storage CI","username":"purestorage-cinder-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-21 22:23:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap http://openstack-logs.cloud-support.purestorage.com/99/763599/1/check/PureISCSIDriver-tempest-dsvm-bionic-aio-multipath-chap/17c5c1c : FAILURE in 3m 09s\n- PureFCDriver-tempest-dsvm-bionic-aio-multipath http://openstack-logs.cloud-support.purestorage.com/99/763599/1/check/PureFCDriver-tempest-dsvm-bionic-aio-multipath/a02308d : FAILURE in 24m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Pure_Storage_CI","accounts_in_message":[],"_revision_number":1},{"id":"1795258863ab0ec47df873fd8415207ecbef16b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-21 23:25:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cffd99e29444983a977037f01d57899 : FAILURE in 1h 16m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee74a13dff044f43866ac1ca453c5686 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd1b3b90d1441c49c83fdbed280dfd5 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec29c3f73304f95b3cd40e5afb7c8a6 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8897df8ed9f4455cbfada6b7be7d1f0c : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f14ec3646f643f79d9173a0229bad59 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843896f001074f6d9d4c526799d3b9fd : SUCCESS in 5m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a8d3d1affeeb49e0b8fdb6460eb69560 : SUCCESS in 3m 25s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6bcfecbd30644f18b9c53a4d0741440d : SUCCESS in 1h 00m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/310159bdcf994a7ba3194a1bbc386f7b : FAILURE in 1h 25m 42s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/35c2ccc9c1de4518b5f197b4e5b5785b : SUCCESS in 1h 06m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4aa29a0be2b01263e7e1aa9b70071589031a856","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-21 23:36:06.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/devstack-plugin-ceph-dsvm-focal-py3/2abae11/ : FAILURE in 1h 27m 09s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/tempest-dsvm-full-focal-py3/5ab6c70/ : FAILURE in 1h 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a4194560e5d3d35f06b08ce72fc5cebb1439a93b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-22 06:29:51.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"adb1d90a939e28f4c4bc66a54d2c471362a94caf","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-11-23 07:47:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_XtremIO_FC_os-brick/bd78497/DellEMC_XtremIO_FC_os-brick/5 : FAILURE in 53m 29s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_XtremIO_iSCSI_os-brick/f07a4a3/DellEMC_XtremIO_iSCSI_os-brick/4 : SUCCESS in 1h 14m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"85396f2ef5f9a486b009f3c60757b811c249e676","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-23 08:01:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f790d3f28da4d9a079bf06906f4cdb1804d9555c","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-11-30 11:03:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a3141a94c1cf912e410c7a9d140a4dba7eb3388","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-11-30 11:54:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/763599/1/iscsi/ : FAILURE in 49m 07s\n- os-brick-smb http://cloudbase-ci.com/os-brick/763599/1/smb/ : FAILURE in 44m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c25758656081ad92457737b230375b623edc4be3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-30 12:13:44.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"489dc52029de7f754f60626c080bc2b0cee1adce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 12:35:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47dd5a0d265c4ce488c668cac27d8135 : SUCCESS in 1h 29m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d265fac801e24824ba67b65ae5c84114 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb00a383cd146ebbad0de3a5bf4a20a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ade8663a0a04a13b7ecb573fe5e1c85 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3964dd0b4c4f49f38c0d50c5ea9682ad : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbb7a13dcef24151a6f23d50bec368fd : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a758b48821545a4878e69efbbfcd9f8 : SUCCESS in 7m 15s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/083bcb0d83134343a4683e2ccf9b9502 : SUCCESS in 4m 11s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/a858c3585e434576b7dc471646d161ab : SUCCESS in 1h 03m 11s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5dab78048863400ea9aa719fbcc39827 : FAILURE in 1h 23m 19s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a5e582a6b36a4e6c917ae944445f29eb : SUCCESS in 1h 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bf9dd79fbf29fee0b7d0e1b1591d9283fcf3d0d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-30 13:56:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/devstack-plugin-ceph-dsvm-focal-py3/60b9b04/ : SUCCESS in 1h 46m 42s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/854be2b9499b4d79afd51f57e8eb1f8d : POST_FAILURE in 1h 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4559f139847823f4df0d2ee81729ca472d9666ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-11-30 14:23:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cf79ff2fe1efd6d08a291e3807c201727e38d562","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2020-12-04 19:45:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6408cec60e4f78a193f235d58ac0031e5e366451","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-04 21:07:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8cb45552bf75815d93981ae671adec4cd5d04a23","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-04 21:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/devstack-plugin-ceph-dsvm-focal-py3/5998881/ : SUCCESS in 1h 33m 28s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/tempest-dsvm-full-focal-py3/0eec77e/ : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0e1374a45b8a70311502950a5147a00579a8df1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 21:55:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56271d82406d435a9300628a15c02c61 : FAILURE in 1h 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c19f41b3c08e4ecb9fedaed350851704 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba35e08f26d6443eaaa875f27b2252cd : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9e64990f6ae4373921fda61d487686a : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c6fcfabdcd45b7a131281f9355b1f9 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb6713cb57014f7dbc0f1fa7fee29c93 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a877864c394aa9be6273b31adffd52 : SUCCESS in 7m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f26851ab5b214866ab39a08ac96b5295 : SUCCESS in 3m 51s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/72fbe60c22884e8a95b2ba753ea90b51 : SUCCESS in 1h 15m 47s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5175a62b3ef347049c6514236865af5e : SUCCESS in 2h 02m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b6be3c0f558244c6a2243206d7e241b2 : SUCCESS in 1h 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c693ca7e864b42256130362bd9a7384435ed59d2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-04 23:44:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8d593ba4c3ba29e93d6a149c20b1b05fab22d189","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-12-05 10:39:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7bb1695861a72515944a30ad43c21accb7554638","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-12-05 11:51:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/763599/1/iscsi/ : FAILURE in 1h 11m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/763599/1/smb/ : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b1bda7f11cf48f8103bdf8cd15140a771fb703fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-05 11:55:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"8d6104eca257265e619b6d7bd139792b35e29821","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-05 12:37:30.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/devstack-plugin-ceph-dsvm-focal-py3/50e7e84/ : SUCCESS in 1h 36m 59s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/99/763599/1/check/tempest-dsvm-full-focal-py3/44f721d/ : SUCCESS in 1h 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1a5a91a6c8875674804251bf6827f3cb78eefb73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 12:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddf93fb2c8fc4acba8c30c8398201b58 : SUCCESS in 1h 22m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f810225823a043ba90cb11fc0b47ef56 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/612d956ed7664945a03319c8fc6c6d14 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2582f0312d7e403baddccbd2669aabd5 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33280cfdc13a4b02954b1dd26df8c535 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/180e3b34462f47e7ad05567c7ca0b75f : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f46516f107c41f2899d20faa5fb1967 : SUCCESS in 5m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/1e18bbd6ffbe423bb82f075a601a2b43 : SUCCESS in 4m 14s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/f2eded53f6e14070b5332bd8c4e3126b : SUCCESS in 1h 12m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73adb7c612a3431499d2df3e84ad2ad6 : SUCCESS in 1h 53m 00s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/76fdde6a1b4442aa9101a4263490011e : SUCCESS in 1h 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"210956e14b4ecaa52aac3e8020e3c2e0683c94b7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-05 13:37:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/99/763599/1/upstream-check/cinder-cDOT-iSCSI/958dd04/ : SUCCESS in 41m 48s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/99/763599/1/upstream-check/cinder-cDOT-iSCSI-IPv6/7be597c/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/99/763599/1/upstream-check/cinder-cDOT-NFS/c2a95bf/ : SUCCESS in 41m 13s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/99/763599/1/upstream-check/cinder-cDOT-NFS-IPv6/b7aafa8/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/99/763599/1/upstream-check/cinder-cDOT-FCP/d7bb8d3/ : SUCCESS in 58m 51s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e75b2ef46e1871b96a67a80308d4b0d29f4e3867","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2020-12-05 13:48:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/99/763599/1/upstream-solidfire-check/cinder-solidfire-iSCSI/b798f71/ : SUCCESS in 1h 10m 19s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"6d1297baa161d07301eecd5c4acc1a654bcde783","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-05 13:55:36.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cfc944f9a77682b5d66de99f9b321d655a6b0912","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2020-12-05 13:57:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/99/763599/1/check/DellEMC_PowerStore_FC-ob/1125cd0/DellEMC_PowerStore_FC-ob/7 : SUCCESS in 1h 19m 00s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/99/763599/1/check/DellEMC_PowerStore_iSCSI-ob/caae63e/DellEMC_PowerStore_iSCSI-ob/7 : SUCCESS in 1h 15m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"d80125e1c8613af99726f21f26aaf8b22df03ad9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-05 14:02:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_XtremIO_FC_os-brick/6128110/DellEMC_XtremIO_FC_os-brick/14 : SUCCESS in 55m 00s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_XtremIO_iSCSI_os-brick/0a4a8d9/DellEMC_XtremIO_iSCSI_os-brick/12 : SUCCESS in 1h 23m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"3399f4d47b17b4790f060e909636708e93fad612","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-05 14:07:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_Unity_FC_os-brick/937a995/DellEMC_Unity_FC_os-brick/8 : SUCCESS in 1h 28m 38s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_Unity_iSCSI_os-brick/2c3bee7/DellEMC_Unity_iSCSI_os-brick/6 : FAILURE in 31m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8102d5081db066b80acd68d8f4f69cb9edda9068","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2020-12-05 14:49:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_SC_FC-os-brick/fb38b10/DellEMC_SC_FC-os-brick/7 : SUCCESS in 1h 12m 02s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_SC_iSCSI-os-brick/87b2129/DellEMC_SC_iSCSI-os-brick/7 : SUCCESS in 2h 10m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"bb7d47e5d8a48cc0bd00f8d1d0812c7775e80f7f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-05 15:16:26.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_PowerFlex_os-brick/8f4f194/DellEMC_PowerFlex_os-brick/7 : SUCCESS in 1h 06m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"6972c3be8e3c7bcf74205cb11dcbb7fe4cb5c18b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-05 15:44:09.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2d04803637b660e8b4372133d7313f8d97001fb1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-05 19:02:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_VNX_FC_os-brick/b4e6e9f/DellEMC_VNX_FC_os-brick/9 : FAILURE in 1h 06m 26s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/99/763599/1/check/DellEMC_VNX_iSCSI_os-brick/742904e/DellEMC_VNX_iSCSI_os-brick/9 : FAILURE in 1h 56m 04s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d80e42e39d4fada216a6eea68744c4f547c8ff03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-05 20:22:55.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/99/763599/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7ef4c9014ceac500e4a8116dbc44ddcb416f9f60","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2020-12-10 13:29:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef1215c8714b1438ba3174c441edb510e61fe696","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-07 18:34:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d652f0da702de3649d1ea8704d3b24642cafa3e8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-08 18:24:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good and has a test, what\u0027s not to like?","accounts_in_message":[],"_revision_number":1},{"id":"f38913d5fc2d27b4eda89165069b02b53bf9ac04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 18:25:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dd2ea881c8267c5f9e2d9ca74cfa256ca688356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:12:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3467b59325fa4f628e40fd3370501351 : SUCCESS in 1h 39m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85da0250888d4bf5b5a97bc360ddd82a : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3204596f2464472090c296b78b5d688c : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e21d068b80e64d3eb043e9ef12773b8a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7e92e9f6e054f13bb751158643fa59c : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019b50783b804cb981c66ce0d7cced13 : SUCCESS in 5m 53s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b48e14b4d36643d5a4f28817295b0b57 : SUCCESS in 1h 38m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1dba477e184a42f7cb1a542e72c2769abb1529c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:12:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63671d65f1c0690d556b1b4387874f58f6150389","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 20:14:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50949c8e28b4ba4818e18b55469718f : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f642069f107bf1892980d3f21fee5efed42c7b4a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-01-15 16:37:31.000000000","message":"Patch Set 1:\n\nTime to start the backports!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fabe1b33d497b5f017f07b4b5d7fce0c0eabd00","revisions":{"4fabe1b33d497b5f017f07b4b5d7fce0c0eabd00":{"kind":"REWORK","_number":1,"created":"2020-11-21 19:30:44.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/99/763599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/763599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/763599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/763599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/763599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/763599/1"}}},"commit":{"parents":[{"commit":"4ac51ae15388a2333e21673389d19d75eaaded83","subject":"Merge \"Stop configuring install_command in tox.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ac51ae15388a2333e21673389d19d75eaaded83"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2020-11-21 19:28:00.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2020-11-21 19:30:44.000000000","tz":-300},"subject":"Improve error handling on target query","message":"Improve error handling on target query\n\nThe command \u0027iscsiadm -m node\u0027 will return entries for corrupt targets\nin the form \u0027[]:port,-1\u0027 instead of the expected format.  This causes an\nIndexError exception during parsing.  This patch skips invalid entries.\n\nCloses-bug: #1886855\nChange-Id: I9a1746658474c0f1be7ec29a36767085aaf2ab7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4fabe1b33d497b5f017f07b4b5d7fce0c0eabd00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4fabe1b33d497b5f017f07b4b5d7fce0c0eabd00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
