)]}'
{"id":"openstack%2Fos-brick~767500","triplet_id":"openstack%2Fos-brick~master~I9162520c0d6c211cf89dcdcab5653d7230deb77a","project":"openstack/os-brick","branch":"master","topic":"bug/1906768","hashtags":[],"change_id":"I9162520c0d6c211cf89dcdcab5653d7230deb77a","subject":"Fixed multipath residue when evacuating the VM","status":"NEW","created":"2020-12-17 08:58:18.000000000","updated":"2021-06-07 14:38:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7b709dec78a4569114fae788d5abae69bf0378bd","_number":767500,"virtual_id_number":767500,"owner":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-04 08:51:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-03 22:57:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":-1,"date":"2021-06-07 14:38:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 09:35:31.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 09:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 09:57:55.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2020-12-17 10:36:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-17 11:04:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 18:21:12.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 09:45:30.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-01-04 10:01:25.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-01-04 10:04:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 10:14:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 10:47:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 22:57:26.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-06-07 14:38:53.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"faca1e5e704c7f970eb5b2f20d7dc5623605aab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"date":"2020-12-17 08:58:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"454d1ffbd121b32804f520346c1a1beca9f796bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-17 09:35:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3de3610e08e99149d7e20359570f5fd0f8a15a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 09:50:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ea807c443d34c35af74aada31c435ec : FAILURE in 22m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9f7379b24dc4b8681b15c8ba62e3f1d : FAILURE in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ec691038b4417c9777f3ee15afc347 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea337fad408467487e151d11a54ef8c : FAILURE in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9697b05d377745e19e6f2e367a614d38 : FAILURE in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cafaa65ab6024feb833b27618626ccba : FAILURE in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8212805e8414c25b21f0fa9fb85c109 : FAILURE in 6m 44s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/019dc687cbc6438cb4a64de637f39664 : FAILURE in 5m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/807296927a2743b5942079d842af0cf0 : FAILURE in 21m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00b584fdd7eb41a0bf9d59a5b2f093c4 : FAILURE in 21m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4121b39244fe4dd7ab43ec240f67781b : FAILURE in 47m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5534661a26ecd73a845e33a6129d811886fa133","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-12-17 09:57:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/767500/1/iscsi/ : FAILURE in 59m 14s\n- os-brick-smb http://cloudbase-ci.com/os-brick/767500/1/smb/ : FAILURE in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e8fccf2c46dada78718e18dde61275faa6a9a5b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-17 10:36:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/00/767500/1/check/DellEMC_Unity_FC_os-brick/329298c/DellEMC_Unity_FC_os-brick/43 : FAILURE in 26m 33s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/00/767500/1/check/DellEMC_Unity_iSCSI_os-brick/c509fc1/DellEMC_Unity_iSCSI_os-brick/41 : FAILURE in 26m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"289a61b04918a8d695af24aa3306b7ccdca600d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-17 11:04:38.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/1/check/devstack-plugin-ceph-dsvm-focal-py3/20ad27e/ : FAILURE in 1h 14m 28s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/1/check/tempest-dsvm-full-focal-py3/f395519/ : FAILURE in 1h 06m 47s","accounts_in_message":[],"_revision_number":1},{"id":"33255c33d7ce351b1017b5c532fb70ea4d9cbbcb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-17 12:17:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"de1adf7c119eb71d0cce22bb50a989ce30c17231","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-17 18:21:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/00/767500/1/check/DellEMC_XtremIO_FC_os-brick/b61ff27/DellEMC_XtremIO_FC_os-brick/48 : FAILURE in 27m 23s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/00/767500/1/check/DellEMC_XtremIO_iSCSI_os-brick/431c282/DellEMC_XtremIO_iSCSI_os-brick/46 : FAILURE in 27m 56s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"fdf0e7eb89bfa1fecec3bb9e2e009a94ae664262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"date":"2020-12-18 03:22:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"805196d041887b77e50cbf3af9326bc9c080e192","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-18 03:22:59.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"7326e59230856caaa99dba9b9ed159c3505d7349","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-12-18 04:20:06.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/767500/2/iscsi/ : FAILURE in 55m 02s\n- os-brick-smb http://cloudbase-ci.com/os-brick/767500/2/smb/ : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":2},{"id":"57552082464825c04f7be997500005c78fb18095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 04:55:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6fb253415b441dbbacab7129d07f042 : SUCCESS in 1h 15m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63a94be18b7b4fb3aa319cbcf10eb72c : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd02679f74d4db3908f2d595a69f53e : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37a3fa35f2d4454ea5367fcc8bb814db : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a878f792e1894451b41cf00295fc043c : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0129eefb6ea4864843beedacdda01fb : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce77266efce46c7bf0778e75b9fd673 : SUCCESS in 6m 54s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6f881f2bd6fe46fe8e40d50a6d3b3080 : SUCCESS in 3m 26s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b48620adeb7641a4815b13b26aeee736 : SUCCESS in 1h 06m 45s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8633c6cde8cb4cafa2af47d633ebd3ff : SUCCESS in 1h 30m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c38a2104cfab4fae8f72bf1393ccae12 : SUCCESS in 1h 18m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2fe3cf638f597faad7d1af81b7e05844909bccce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-18 05:08:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/2/check/devstack-plugin-ceph-dsvm-focal-py3/6d78237/ : SUCCESS in 1h 27m 03s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/2/check/tempest-dsvm-full-focal-py3/4e50813/ : SUCCESS in 1h 36m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7f30608aba032facda4a2feb3ff154a4bddf85df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"date":"2020-12-18 08:21:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a04709142acf8d914bea7044d3930db3ebd39116","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2020-12-18 09:19:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/767500/3/iscsi/ : SUCCESS in 57m 27s\n- os-brick-smb http://cloudbase-ci.com/os-brick/767500/3/smb/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c35b5660f405d7f603bf2a9487da90fb0d81255f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 09:58:59.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0c7dec2b676498fb21695b6463b4dab : SUCCESS in 1h 36m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ef6a889956c406399a6103a1a8f8c6e : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a3730c94c74c50b8a741f919d6fffe : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384f322a29654f17a29eb408e50358cd : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca8fc2d61074627a66fc33c6841d656 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f9cffba5d3d4802aa22a1177d3a4523 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c341c864e3ad4b6093b65ffe726164c6 : SUCCESS in 5m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/736e869cc4394dff91c6a938f74b3ec4 : SUCCESS in 4m 02s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/04feaf0a9ad6406aab7c21da59c740b5 : SUCCESS in 1h 14m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5fbf6946dda6478888bf3e21aed0659a : SUCCESS in 1h 32m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a68648b92fd847fbbc8e0f93411e29f9 : SUCCESS in 1h 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"394ff2ccdbe6d113b481ff3087ca526ed84c00db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-18 10:25:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/3/check/devstack-plugin-ceph-dsvm-focal-py3/f56c4ff/ : SUCCESS in 1h 49m 13s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/3/check/tempest-dsvm-full-focal-py3/eb38f96/ : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":3},{"id":"09462cf19874d3942fa4e8b35e25832822bae654","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-18 12:00:33.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/00/767500/3/check/DellEMC_Unity_FC_os-brick/cf8ee81/DellEMC_Unity_FC_os-brick/47 : SUCCESS in 1h 21m 50s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/00/767500/3/check/DellEMC_Unity_iSCSI_os-brick/ba210bd/DellEMC_Unity_iSCSI_os-brick/44 : SUCCESS in 1h 52m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e241614b7b54d65a2dd8eb19b3938af2041f7aea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-18 16:08:33.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"e61b83f06444112afa3143f38e8b190e45f43f72","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2020-12-19 02:54:18.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/00/767500/3/check/DellEMC_XtremIO_FC_os-brick/04ceecd/DellEMC_XtremIO_FC_os-brick/52 : SUCCESS in 54m 10s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/00/767500/3/check/DellEMC_XtremIO_iSCSI_os-brick/fca1b58/DellEMC_XtremIO_iSCSI_os-brick/49 : FAILURE in 16m 44s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"a45b5c1fe1a7bb908363108dce172fdf28b76364","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-12-19 05:47:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"91b45144f38d87fada514a3c7fb9c3b9f1489fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"date":"2021-01-04 07:11:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0a64412b7935709912c63fa9bde2a5f679c1fac0","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-01-04 07:48:41.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/767500/4/iscsi/ : FAILURE in 37m 08s\n- os-brick-smb http://cloudbase-ci.com/os-brick/767500/4/smb/ : FAILURE in 36m 42s","accounts_in_message":[],"_revision_number":4},{"id":"5f89732f2339a067edf2e60acce0749b1d6295be","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 07:49:03.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"96907a256a64094cb2b8ad9310344b64cc35007f","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 08:42:50.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_XtremIO_FC_os-brick/e3015db/DellEMC_XtremIO_FC_os-brick/60 : FAILURE in 55m 41s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_XtremIO_iSCSI_os-brick/8e8660c/DellEMC_XtremIO_iSCSI_os-brick/57 : SUCCESS in 1h 16m 50s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"1b8c6e5d40b06c54a75f1cecf1d462162a3e8669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 08:51:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a15e1b1a5c23408a88f3684bb044a7ab : SUCCESS in 1h 25m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72832fe6ea00422da24cc0e428d5c4e8 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a88184914e245b4b90e78b791f77058 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564adc8e78d64173b6e54bd023d2342a : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d630bdfb01b44e4fac2ca9cf97418fac : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e0deda2984c4018b2503958e10244e5 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91b1f84d4c5498e840c97e21281cf77 : SUCCESS in 5m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6ab6b259dbc44381b6e7371940d9d3ce : SUCCESS in 3m 47s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6375edfff6b8437b89d2ad5ff9291416 : SUCCESS in 1h 00m 13s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/376464fa5d094bc2a4218b3faff33605 : SUCCESS in 1h 35m 27s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e645e71f61c540be8bf44f6b0440250f : SUCCESS in 1h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c3344f6854bce007c1cc91a453f72f36e2667d8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-04 08:55:43.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/4/check/devstack-plugin-ceph-dsvm-focal-py3/1d97a27/ : SUCCESS in 1h 35m 40s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/00/767500/4/check/tempest-dsvm-full-focal-py3/35dcb0b/ : SUCCESS in 1h 36m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4ac918fd6dedda72445d3af8f0675073e4d51dcf","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-01-04 09:45:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/00/767500/4/upstream-solidfire-check/cinder-solidfire-iSCSI/fb62d85/ : SUCCESS in 53m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"7f9b19cd6254d9fd362b68b30a7d1863a4065eaf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-01-04 10:01:25.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/00/767500/4/check/DellEMC_PowerStore_FC-ob/0c61e3a/DellEMC_PowerStore_FC-ob/15 : SUCCESS in 1h 08m 57s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/00/767500/4/check/DellEMC_PowerStore_iSCSI-ob/01afebf/DellEMC_PowerStore_iSCSI-ob/15 : SUCCESS in 1h 09m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"b6e6610644dc94290e29cece17c22f66c36eb1f9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 10:04:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_PowerFlex_os-brick/b16b4b9/DellEMC_PowerFlex_os-brick/15 : SUCCESS in 1h 12m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"84c21e0e4f8d2db8fd1be45bc9f4293e0883b994","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-04 10:14:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/00/767500/4/upstream-check/cinder-cDOT-iSCSI/9c78918/ : SUCCESS in 36m 26s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/00/767500/4/upstream-check/cinder-cDOT-iSCSI-IPv6/d0042a7/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/00/767500/4/upstream-check/cinder-cDOT-NFS/df8756d/ : SUCCESS in 36m 46s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/00/767500/4/upstream-check/cinder-cDOT-NFS-IPv6/ddd2327/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/00/767500/4/upstream-check/cinder-cDOT-FCP/1b6b796/ : SUCCESS in 1h 22m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"bb1de47443336e2692ebf9270fbbf95492e8f933","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 10:35:04.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"f10fe3cae0747002a3e8bd15e572a8374fd4079b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 10:47:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_VNX_FC_os-brick/d22dd26/DellEMC_VNX_FC_os-brick/17 : SUCCESS in 1h 08m 47s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_VNX_iSCSI_os-brick/6048afd/DellEMC_VNX_iSCSI_os-brick/17 : SUCCESS in 1h 55m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"eec3c0e2d8bd414631f7a844ab910acf6f87cb57","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 11:33:25.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_XtremIO_FC_os-brick/c959153/DellEMC_XtremIO_FC_os-brick/61 : FAILURE in 54m 48s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/00/767500/4/check/DellEMC_XtremIO_iSCSI_os-brick/e4f820b/DellEMC_XtremIO_iSCSI_os-brick/58 : SUCCESS in 1h 13m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"5c933ce859a64bed7fea91d899a1685d580c8229","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-01-04 11:55:35.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/00/767500/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"5e24f10a93a6441feae857265c11d9ef4ca59be1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-03 22:57:26.000000000","message":"Patch Set 4: Code-Review-1\n\nThe bug report is marked as Won\u0027t Fix. Should we abandon this patch? If not, I think we should create a new bug or reopen #1906768 with a current environment and error (since pike is EOL).","accounts_in_message":[],"_revision_number":4},{"id":"7b709dec78a4569114fae788d5abae69bf0378bd","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-07 14:38:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nPlease use the \"force\" parame","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b020b02d214104042d34c12854b3414cb3dd127e","revisions":{"d8e4472ff4ec77a425f04ec702361815a8a03406":{"kind":"REWORK","_number":1,"created":"2020-12-17 08:58:18.000000000","uploader":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"ref":"refs/changes/00/767500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/00/767500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/00/767500/1"}}},"commit":{"parents":[{"commit":"4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22","subject":"Update requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22"}]}],"author":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-17 08:58:04.000000000","tz":480},"committer":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-17 08:58:04.000000000","tz":480},"subject":"Fixed multipath residue when evacuating the VM","message":"Fixed multipath residue when evacuating the VM\n\nEvacuation results in multipath residue when the back-end storage uses FC and NOVA to configure multipath, this is part of patch and the other part is in Nova， its function is to remove local multipath residues。\n\nCloses-Bug: #1906768\nChange-Id: I9162520c0d6c211cf89dcdcab5653d7230deb77a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d8e4472ff4ec77a425f04ec702361815a8a03406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d8e4472ff4ec77a425f04ec702361815a8a03406"}]},"branch":"refs/heads/master"},"a7d65257732b82e85bbd29d72b343ca69face0b3":{"kind":"REWORK","_number":2,"created":"2020-12-18 03:22:25.000000000","uploader":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"ref":"refs/changes/00/767500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/00/767500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/00/767500/2"}}},"commit":{"parents":[{"commit":"4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22","subject":"Update requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22"}]}],"author":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-17 08:58:04.000000000","tz":480},"committer":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-18 03:21:35.000000000","tz":480},"subject":"Fixed multipath residue when evacuating the VM","message":"Fixed multipath residue when evacuating the VM\n\nEvacuation results in multipath residue when the back-end storage uses FC and NOVA to configure multipath, this is part of patch and the other part is in Nova， its function is to remove local multipath residues。\n\nCloses-Bug: #1906768\nChange-Id: I9162520c0d6c211cf89dcdcab5653d7230deb77a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7d65257732b82e85bbd29d72b343ca69face0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7d65257732b82e85bbd29d72b343ca69face0b3"}]},"branch":"refs/heads/master"},"b0bae30d2a6042e1570e9c5b07275cc8049e1270":{"kind":"REWORK","_number":3,"created":"2020-12-18 08:21:59.000000000","uploader":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"ref":"refs/changes/00/767500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/00/767500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/00/767500/3"}}},"commit":{"parents":[{"commit":"4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22","subject":"Update requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22"}]}],"author":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-17 08:58:04.000000000","tz":480},"committer":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-18 08:21:38.000000000","tz":480},"subject":"Fixed multipath residue when evacuating the VM","message":"Fixed multipath residue when evacuating the VM\n\nEvacuation results in multipath residue when the back-end storage uses FC and NOVA to configure multipath, this is part of patch and the other part is in Nova， its function is to remove local multipath residues。\n\nCloses-Bug: #1906768\nChange-Id: I9162520c0d6c211cf89dcdcab5653d7230deb77a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b0bae30d2a6042e1570e9c5b07275cc8049e1270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b0bae30d2a6042e1570e9c5b07275cc8049e1270"}]},"branch":"refs/heads/master"},"b020b02d214104042d34c12854b3414cb3dd127e":{"kind":"REWORK","_number":4,"created":"2021-01-04 07:11:08.000000000","uploader":{"_account_id":20252,"name":"zhulipeng","email":"zhulp0372@163.com","username":"zhulipeng"},"ref":"refs/changes/00/767500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/00/767500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/00/767500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/00/767500/4"}}},"commit":{"parents":[{"commit":"4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22","subject":"Update requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d4c5e82c97fe69c7f8fa4ba3a36a69bf0a27e22"}]}],"author":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2020-12-17 08:58:04.000000000","tz":480},"committer":{"name":"zhulipeng","email":"zhulp0372@163.com","date":"2021-01-04 07:10:39.000000000","tz":480},"subject":"Fixed multipath residue when evacuating the VM","message":"Fixed multipath residue when evacuating the VM\n\nEvacuation results in multipath residue when the back-end storage uses FC and NOVA to configure multipath, this is part of patch and the other part is in Nova， its function is to remove local multipath residues。\n\nCloses-Bug: #1906768\nChange-Id: I9162520c0d6c211cf89dcdcab5653d7230deb77a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b020b02d214104042d34c12854b3414cb3dd127e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b020b02d214104042d34c12854b3414cb3dd127e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
