)]}'
{"id":"openstack%2Fos-brick~775545","triplet_id":"openstack%2Fos-brick~master~I0278c502806b99f8ec65cb146e3852e43031e9b8","project":"openstack/os-brick","branch":"master","topic":"bug/1915678","hashtags":[],"change_id":"I0278c502806b99f8ec65cb146e3852e43031e9b8","subject":"Avoid unhandled exceptions during connecting to iSCSI portals","status":"MERGED","created":"2021-02-15 02:08:21.000000000","updated":"2021-03-07 06:35:43.000000000","submitted":"2021-03-05 22:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"775545-bug/1915678","meta_rev_id":"97f9ccb88a37272c80a2486a7340e7efd62db58d","_number":775545,"virtual_id_number":775545,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 22:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"date":"2021-03-05 18:42:10.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 13:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2021-03-05 15:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"date":"2021-03-05 18:42:10.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2021-03-05 15:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-05 12:09:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 02:10:53.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 02:12:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-15 02:24:53.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 02:35:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-15 03:06:46.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-02-15 04:28:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 14:13:45.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-02-24 14:15:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-24 14:17:27.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-02-25 18:57:02.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-02-27 21:13:16.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-02-27 21:21:21.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-27 21:22:23.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-02-27 21:49:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-27 22:17:24.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-27 23:03:29.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-03-04 13:33:34.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"0bc0c45cda4e683f9983098ed88a575a4c63fd83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-15 02:08:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1aadc168851aaaf1e168f0999e76cc371083661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-15 02:09:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"edb1293434ac5b38c9904cfab2ef5a341ea4f3ca","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 02:10:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"da9c593f4e86126248b5c16d29362b07a069f0d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-15 02:12:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/45/775545/2/check/DellEMC_Unity_FC_os-brick/a901e48/DellEMC_Unity_FC_os-brick/71 : FAILURE in 2m 13s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/45/775545/2/check/DellEMC_Unity_iSCSI_os-brick/8d38695/DellEMC_Unity_iSCSI_os-brick/68 : FAILURE in 1m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5905a5a21d2d97cdc0f7d80cfb579d92526280f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 02:14:43.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"082e7a5d741fb8ec832a20ed804faa62cc5a2b3b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-15 02:24:53.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/45/775545/2/check/DellEMC_XtremIO_FC_os-brick/ade02da/DellEMC_XtremIO_FC_os-brick/86 : FAILURE in 14m 55s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/45/775545/2/check/DellEMC_XtremIO_iSCSI_os-brick/a8665c1/DellEMC_XtremIO_iSCSI_os-brick/84 : FAILURE in 13m 03s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"3dda9dc20a1ba67bea5cf038facf4fc1290fb6c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-15 02:35:44.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/devstack-plugin-ceph-dsvm-focal-py3/98a2839/ : FAILURE in 22m 26s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/tempest-dsvm-full-focal-py3/a778deb/ : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9abedc75cc21d5c061636a6e4b151119e33bca5d","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-02-15 03:06:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/775545/2/iscsi/ : SUCCESS in 57m 09s\n- os-brick-smb http://cloudbase-ci.com/os-brick/775545/2/smb/ : SUCCESS in 56m 16s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/775545/2/ceph/ : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2ea92c61c53eaba9590a3f0a4a5c5086c1e35f94","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 03:27:46.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"89870d14dad8c36e1797fff0c1db4c4ff5a0ea32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 04:28:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98b2166314d24a5ab6e628813584c88a : SUCCESS in 1h 23m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c259b0576c94495fb43f4a8a3f6f8187 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b459fa2abd4768bcad8765b8db099a : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c0f94e230204b7bafaad60fb903bb84 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff21c9f0aba4448594d8f47c7c596473 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9877a033b0842039d05e0384c7cc573 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d8773967be490d8aecfe9eb217402e : SUCCESS in 5m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f9b9eb87cbe54d50b59b728d5e151c5c : SUCCESS in 4m 00s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/2992040bdf3e4f119ebe3c142e0e9fb8 : SUCCESS in 1h 14m 13s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65125df894434e8e8c8f838ec14e92ea : FAILURE in 2h 18m 13s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5dfb367bfddd48889d7d645d7449485f : SUCCESS in 1h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04fa044731d0c1c9919345f2e859d4d214b62a65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-15 05:39:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0b70798e671ebb4fe43f5357cb4a3100bac61a3c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-15 06:05:06.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/devstack-plugin-ceph-dsvm-focal-py3/d83bcc3/ : FAILURE in 21m 10s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/tempest-dsvm-full-focal-py3/655c642/ : FAILURE in 18m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4d466d7f37c7193e6c665a83edfb8bc58f2502ee","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-02-15 06:36:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/775545/2/iscsi/ : SUCCESS in 54m 47s\n- os-brick-smb http://cloudbase-ci.com/os-brick/775545/2/smb/ : SUCCESS in 55m 55s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/775545/2/ceph/ : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9976ee95133c6e1b6c6b98cf75783b587376e41a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-15 06:58:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6853c4f96cee8c40afc059e5a92cd94f66c95b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 07:43:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b778958baca45ba8898238f25d02a8a : SUCCESS in 1h 33m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dabb9a2debbb45f7b433a82c79cbe4c8 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f313e08594e4766a21730f7dd1c1060 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/747da5e439c7467fb06a526441ba9034 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c2eacd87e774fcfa283f0bce4a891dc : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3f92e9acba84d6e815d0294d2c554b7 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d19ad163b841d4bb4a5dc3a6e4e37f : SUCCESS in 5m 47s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8f131116efbd41ac9c06446f1deb0b95 : SUCCESS in 5m 10s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/14b4957133ce4d5fa8b8b54804865ef7 : SUCCESS in 1h 30m 41s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d3758607ae748b7839d1e68bed47c2d : FAILURE in 1h 59m 24s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/bad141f6c26c4b4384dafd78754b1ccd : SUCCESS in 1h 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"203d2ea215ffa8326f15abb4d32694b4a77adc54","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-24 14:13:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e856afb41f31b8595f64a19ba90d9cc966ef281c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 14:15:12.000000000","message":"Patch Set 2: Review-Priority+1\n\nWould be good to get this fixed in the wallaby release next week.","accounts_in_message":[],"_revision_number":2},{"id":"3a2bc7704a358cf3570cae79b9bf610d199486ce","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-02-24 14:17:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"699d53efae3fe4a5b8ba944e1ee6380908fb5905","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-24 14:48:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"003b8d463a885ce9d8b2b69dda0eae34a8a3f835","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 16:39:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a65f77a3916092d71ef0f74d128d95899d7eec37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 17:25:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a817210e2f05415092e0314bca954a62 : SUCCESS in 1h 29m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8ef99d181a48f0864dc8cd3fe1e3d1 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab7e336bb6f43ef990b22dfdeaa114d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd09e824b3840ec828c04facd181713 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/749ce0214e2f41f19b76b2c2254196f0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d325309ff344c3b0735d8abdf74c6c : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961fc3ba766040a19c4d4010696222c0 : SUCCESS in 7m 05s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c297c7f48900408684bfafea9b52d6a8 : SUCCESS in 5m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/442e3584cce6407a9ac39e5c0dfc7312 : SUCCESS in 1h 05m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b7813d9b5bf44312ba0e282798081d42 : FAILURE in 1h 13m 43s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/21c6b2d81e07441996ae75c13b99b46b : SUCCESS in 1h 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22dc844d59392ee1f4ee9c3ddef2258aef34ad0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-24 17:39:44.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4cc68095dfcdbfd3e23ddad78394cd21f3e61582","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-24 20:32:40.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/devstack-plugin-ceph-dsvm-focal-py3/d19e806/ : SUCCESS in 2h 17m 41s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/45/775545/2/check/tempest-dsvm-full-focal-py3/97014cc/ : SUCCESS in 1h 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9f28e9e7eaa262c9be076feb1957867c41d553b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-24 22:10:18.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f702aeceafff82814acf63b15d901b689b9ae88c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-25 08:29:56.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fc147f1db620a63b594b6542eb4f4b02d5297a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 10:26:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfce72e3483d4051938e759ec3621e6c : SUCCESS in 1h 30m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbbff329a2fc40e8b603f941d5672a8c : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8784a16e8143bd8cd39ac76f32b133 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/350e365d6b6745cb9588f5bf09af2fd1 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88104dfa2bf9448ab5075c7859bdb943 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb19fbe46a984ceca3bed65b0aeb4c5b : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b886672b56143bc9bd43581a48b721a : SUCCESS in 6m 58s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f41f8ab27dd348919230b2f51a54046f : SUCCESS in 4m 40s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/dc0afc8d4a5940a3a3a3ee170558382d : SUCCESS in 1h 36m 10s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a5ee238e525e425992ef6cc6480f7631 : FAILURE in 1h 27m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/3b53c0741d35467281e4cd2e158f69e8 : SUCCESS in 1h 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7c330bbbdc146b0971e900ec29ccd7259114489","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-25 10:35:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4c8b0ff557cbd4b601bd1987c65005f2a28c4b39","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 10:36:18.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"23674e19c54d447c00c0e0a99f2560eda96e676c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 12:30:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2f94ac7a02d4feea3c210a0ce75cf18 : SUCCESS in 1h 45m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09f5de6f7f244569b29052131c1597ab : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0b1e0fed904042ad932d8411aa9f07 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf9406e85834b9487988fddbb7ad074 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e01a5f51fef548da955132130d539eba : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/729c3f05d96f4eaba16a2108ee0fb9c8 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ac8a9dad8b468ba49527cd94c1d75d : SUCCESS in 7m 11s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/2bebaaa82b3d47dabed9333b4e148ec5 : SUCCESS in 5m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/648f084a38504313bceb35dc7369168a : SUCCESS in 1h 22m 20s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3df8ea3a626f45f182aaddfe5f3a838f : FAILURE in 1h 28m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cf9a56a69dc0437cba626def93da1dcc : SUCCESS in 1h 04m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b342257f652486b18f1030b6188b98e2bb21e7af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 13:07:56.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"f046bd4277c5692a64561ff786b33b32636a3c41","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-25 15:06:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_XtremIO_FC_os-brick/1cd9ffa/DellEMC_XtremIO_FC_os-brick/102 : FAILURE in 1h 02m 19s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_XtremIO_iSCSI_os-brick/a532c95/DellEMC_XtremIO_iSCSI_os-brick/102 : SUCCESS in 1h 15m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"cd6264791577fbe0038e999e87bdeb9ccc22ee14","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-25 20:37:35.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4db08660d38b4210a0f21299a5f5e399 : TIMED_OUT in 2h 33m 52s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/18a32ae0c96e43e7a9e7c3ed7f85ca50 : POST_FAILURE in 1h 51m 44s","accounts_in_message":[],"_revision_number":3},{"id":"2012f0fa19c36253f6714360ea737b43c2f96c02","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-25 22:33:01.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"6acecc1df5e24dcec697648be87df49bdb9bc391","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-27 14:01:12.000000000","message":"Patch Set 3:\n\nrecheck\n\nhttps://review.opendev.org/c/openstack/cinder-tempest-plugin/+/777685","accounts_in_message":[],"_revision_number":3},{"id":"4f79e4bf5fe0277aeb4f7a8cadced20615277c9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-27 15:21:01.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"326accc57363d3b38238c8698aff23c1eb8696f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-27 15:39:10.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1f8a7b6211784a5299e612f1a904fe78 : POST_FAILURE in 1h 33m 37s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/880d5e4e0484416fbdd9aa3a63cef192 : POST_FAILURE in 1h 25m 38s","accounts_in_message":[],"_revision_number":3},{"id":"434e247fb97a36c82138724449d91e980ee7ba2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 15:40:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26e6910932974a078f6fb5d71efccda1 : SUCCESS in 1h 21m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a29799ad3a04de99ed8f51c61a3c49a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e23e2f323a475ab6e9f36452a11a0c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fa37df0ba3d4891bb149bfa12854a11 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d261ace76f4ecfa9d0a7d3a1bc961c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9d8ace01b864ccbb790e40d2729923a : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012a8c7f32b843d8b094da2d8be2419c : SUCCESS in 6m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b08fcb8e70714e9d934a183bf1cbae61 : SUCCESS in 4m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/c2453acb11dd4cd68858439ebfa69355 : SUCCESS in 1h 15m 30s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/76fb77674506484a864cb99b4bb91fc0 : FAILURE in 1h 33m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8c5edcc17a594386a4bd6595ca984390 : SUCCESS in 1h 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21030182563c2a3a3b53abad006ac1366d398d56","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-27 17:00:32.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"79b70e5aa75dcca35a52398d086bee1b838d87e8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-02-27 18:32:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a9a849df84528d1036d564e9cf755f62f3789d20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-27 19:49:48.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"b57477a58a2eaf3120e1a5aa63d57ffbe8ad1800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 20:13:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31917e1671b549a39bd167e8b72e9977 : SUCCESS in 1h 31m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e16241d7e5e943268deb089a70b38b05 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cab14d2805048d3994ff4cae7c44c90 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4c858855b640d9af26ee8b14955dc3 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c9d94bfdbd4cfdb80f95ad12afea68 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13d0073cb10b44b791ef0e67ac3fb657 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5501eedb0c364700967d354e00645a92 : SUCCESS in 6m 32s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/324f3db7bf294138bedb6b8d94fd1d75 : SUCCESS in 4m 35s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e07c3500b117495a962323396c962390 : SUCCESS in 1h 35m 14s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f150b70afa424675a90347aee9e00ae4 : SUCCESS in 1h 38m 10s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6b93766f249140469633f8c471ba548d : SUCCESS in 1h 13m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f5301f5597147c9b4db74b40bd4b93aeacf4310","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-02-27 20:38:08.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/a57ae9464c5f4557b1f3df5028c8ae0a : POST_FAILURE in 1h 52m 28s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c1408ca642be44b1b7df0e2712b1cb1e : POST_FAILURE in 1h 33m 53s","accounts_in_message":[],"_revision_number":3},{"id":"0be80eb433f51cf72f677eaecb55d84bb3c57394","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-02-27 21:13:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/45/775545/3/upstream-solidfire-check/cinder-solidfire-iSCSI/40e39b0/ : FAILURE in 59m 22s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"c326025124d788ddd954da3b4319b92c19237463","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-27 21:21:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/45/775545/3/upstream-check/cinder-cDOT-iSCSI/333ea2b/ : SUCCESS in 39m 27s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/45/775545/3/upstream-check/cinder-cDOT-iSCSI-IPv6/bc247f3/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/45/775545/3/upstream-check/cinder-cDOT-NFS/72485bc/ : SUCCESS in 38m 19s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/45/775545/3/upstream-check/cinder-cDOT-NFS-IPv6/e0e51d4/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/45/775545/3/upstream-check/cinder-cDOT-FCP/a6eb044/ : SUCCESS in 1h 07m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"6bfdfaf2169bf619fd2ea6379c47d79e53a2a425","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-02-27 21:22:23.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/45/775545/3/check/DellEMC_PowerStore_FC-ob/034f422/DellEMC_PowerStore_FC-ob/31 : SUCCESS in 1h 06m 56s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/45/775545/3/check/DellEMC_PowerStore_iSCSI-ob/1762666/DellEMC_PowerStore_iSCSI-ob/31 : SUCCESS in 1h 08m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"da6d6d15e6cb8b83f9aaca9b9e8a03900389be72","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 21:31:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_XtremIO_FC_os-brick/0c07fad/DellEMC_XtremIO_FC_os-brick/107 : FAILURE in 1h 02m 59s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_XtremIO_iSCSI_os-brick/8615f5d/DellEMC_XtremIO_iSCSI_os-brick/107 : SUCCESS in 1h 18m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"84e1ae778dadb2176e2361f6c003aa7b9182f4c5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 21:49:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_PowerFlex_os-brick/caff546/DellEMC_PowerFlex_os-brick/37 : SUCCESS in 1h 35m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"ab57ceee61aef3a42151b9b7b2caebada59828aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-27 21:56:12.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bcc252962f3897b659feaa90b7717a5d2f0177c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-27 22:14:03.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_Unity_FC_os-brick/7ba994e/DellEMC_Unity_FC_os-brick/92 : SUCCESS in 1h 32m 42s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_Unity_iSCSI_os-brick/66ec632/DellEMC_Unity_iSCSI_os-brick/87 : SUCCESS in 2h 00m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"812d5ef95b9f836399e0b53687407cc70fa53588","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-27 22:17:24.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_VNX_FC_os-brick/3fa54e1/DellEMC_VNX_FC_os-brick/33 : FAILURE in 1h 27m 24s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_VNX_iSCSI_os-brick/700367c/DellEMC_VNX_iSCSI_os-brick/33 : SUCCESS in 2h 03m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"4e67722c6d4d4bc0c56d088bdaa6c60a913b5b9a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-02-27 23:03:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_SC_FC-os-brick/e23efc7/DellEMC_SC_FC-os-brick/30 : SUCCESS in 1h 04m 52s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/45/775545/3/check/DellEMC_SC_iSCSI-os-brick/b101e8a/DellEMC_SC_iSCSI-os-brick/30 : FAILURE in 2h 49m 46s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"3830272faa2893f29425d93270fd2502e60a55ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-02-28 02:37:55.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/3/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/3/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bc19b008b6242d4151cda77829b0701a54a14da7","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-03 09:31:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\n-1: Thank you for the fix.  The code and test look good, but please add a release note, as describe in the contributor docs: https://docs.openstack.org/cinder/latest/contributor/releasenotes.html\n\nAs a side note, do you have any logs of these `iscsiadm -m session` failures?  Because I believe this could be an open-iscsi bug (races in their DB access) or a deployment issue, and I would like to see if there\u0027s anything else we need to do in other projects.","accounts_in_message":[],"_revision_number":3},{"id":"ba567c069e15ba8470fb0d0d708f9d04a289fd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-03 12:11:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cf9b0414bac25fbd0b0f27ca34f4bd5c7fed97c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-03 12:19:02.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (2 comments)\n\u003e \n\u003e -1: Thank you for the fix.  The code and test look good, but please add a release note, as describe in the contributor docs: https://docs.openstack.org/cinder/latest/contributor/releasenotes.html\n\nThank you for your feedback.\nI added a release note pointing the bug and explaining the issue this change resolves.\n\n\u003e As a side note, do you have any logs of these `iscsiadm -m session` failures?  Because I believe this could be an open-iscsi bug (races in their DB access) or a deployment issue, and I would like to see if there\u0027s anything else we need to do in other projects.\n\nThis issue was already reported to the open-iscsi project and there is a fix proposed there.\nhttps://github.com/open-iscsi/open-iscsi/pull/253\n\nIt seems like the issue is caused by a kind of race condition and the following error appears in stderr and the command returns 2.\n\n```\niSCSI ERROR: Specified SID \u003cID\u003e                                            # session.c:iscsi_session_get():141\niscsiadm: Failed to query iSCSI sessions, error 2: Specified iSCSI session not found\n```","accounts_in_message":[],"_revision_number":3},{"id":"ee0dddbccdd86439303c5eb346f9072923c8effa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 14:11:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b5c04c0469244a9990b71dc5182ff1d : SUCCESS in 1h 24m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/644365a196c043ce9ea674cb4512a5f4 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b67f729f4a243fc9a1d25f088252f62 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab0ef00df83141bc9a1943b19a5f4816 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7511bef8e5ca4b519691a7cb828e5808 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/391c25ab24e445ecbfe5ffe68efd7436 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8848b9ec0d7d428380beb5fdc4947281 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72709f511a6344b8a24693d9dd97e7a7 : SUCCESS in 6m 48s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/e6fe83a4e7f64cb0ae652d7a5c5f84ef : SUCCESS in 7m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/bb64ffcaee0b4cc3b7ee727045cf4618 : SUCCESS in 1h 50m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08d7080fbb8b43c3b18f9e3241a935b3 : FAILURE in 1h 46m 59s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4ea4d597eebd4dc982747c6094a89888 : SUCCESS in 1h 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3eb21cc5d8589a74aa9fae3dd7368daa76f31949","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-03 15:19:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/4/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/4/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"c244a7d535b0d101b07e13e6f1713aa8aaa833ce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-03 21:51:12.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/3f5935370d844977b1fa30f72f4a1009 : POST_FAILURE in 35m 58s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7b5973a47a7c49b9a4a0e4656139bb9b : POST_FAILURE in 2h 13m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e02a6df404bb8f6f01dafe46830a571dc0cac355","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-03 23:05:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"489a2a0d40c913bb3ec14541ee35b743c38f0319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 01:00:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ba557b72d7a495e9219c3acc67345bd : SUCCESS in 1h 52m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/474e9f47c7ae45cfb53b4f1cf078ea08 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce0ccc104824aabb4744ed746df51c1 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0c77c84cb4448f926ab918f9d8d4d0 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/809dd37167aa47e2a7a5df3cf5b5bbbe : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d279f2b104439c8141d69fa47980a0 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcfb416b8bdb4c18a78561f31fea9073 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d85157c4a3c4bb0a25e82f29ecbb907 : SUCCESS in 7m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/e6e306a50ef849c3a437883de6b27a14 : SUCCESS in 8m 09s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/28f24af855cb43869463ca014262e58b : SUCCESS in 1h 22m 44s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d195575b06bc42c9b9fa92e775ef83ea : SUCCESS in 1h 31m 04s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b5a15b726f37480dbc02c60abc253e32 : SUCCESS in 1h 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8fdbd848c262b5cc7565e85157438228367c06d2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 01:52:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nSuggestion inline for revising the release note.  Code and tests look fine.","accounts_in_message":[],"_revision_number":4},{"id":"9433f9df23cfa588ece4f90acbb4fd8c4e9c1c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-04 03:01:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ed5d407340036eabfff6a1c2d5ad1291eb1addfb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 03:02:08.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/4/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/4/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"9e944e71cae21dd9ff234b304e4f11dfd46ceb4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-04 03:02:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"af090168ed4380f5a2b51c65684236e1215be463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:08:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3346e8c70a54f20a1ed280751faf480 : SUCCESS in 1h 31m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e00a55964464c33892f5f1d2f75ab82 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deeea1d2c5f34af7b31ba3c3d227eb87 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5faf4ec887af4ac0bc941ea2ae2df323 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b59e687908324cb282a4c4e1412fe323 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c98f13512ecb4afcacec1af293c1ab74 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cf899b1ba1459db74d712be47b24b7 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b25e1d61db1342488d499d1e5bb92dd5 : SUCCESS in 4m 37s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/59ebbf30633845f1b855acf5a356d180 : SUCCESS in 5m 23s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/f4b836fc974f462992489d2008d088ce : TIMED_OUT in 2h 04m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a936a56a4004351b1ed7bcce2179802 : SUCCESS in 1h 39m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d79cd2aaa87d4abe8a0e11ab38470392 : SUCCESS in 1h 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"03118ddef453a1b85a4f59e620b06ca6cdc836ac","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-04 08:58:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/45/775545/5/upstream-solidfire-check/cinder-solidfire-iSCSI/a69e921/ : FAILURE in 1h 28m 06s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"2ba60ea9455e04e40065411b9d7a287c3342f2be","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-04 09:44:46.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerStore_FC-ob/87094ee/DellEMC_PowerStore_FC-ob/45 : SUCCESS in 1h 04m 51s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerStore_iSCSI-ob/a83c235/DellEMC_PowerStore_iSCSI-ob/45 : SUCCESS in 1h 09m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"8e726f3790831f06db7e190dba357a6df029c00a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 09:44:52.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"cfa7de388105d9134efff12249d2940ec4ead5ab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-04 09:53:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-iSCSI/6b94786/ : FAILURE in 56m 22s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-iSCSI-IPv6/d50b4ae/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-NFS/4af9615/ : SUCCESS in 37m 39s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-NFS-IPv6/58fa401/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-FCP/1f54f82/ : SUCCESS in 1h 31m 53s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"ca68563c10a15314829662bdf920351d9985e7e9","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-04 13:33:34.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)\n\nA couple of nits in the commit message if we need to submit another patch.  Otherwise good.","accounts_in_message":[],"_revision_number":5},{"id":"404aa3bc6000b3eb3a74465c49489c8c5f6511f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-04 19:37:07.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"93096f9aaaf69461cd24bccf9805547643956c59","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 20:36:40.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0e64202f56d943c883f71b85f9f8b9f5 : POST_FAILURE in 1h 46m 38s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e2e278aa7f9643279da7ebc9dc819097 : RETRY_LIMIT in 30m 14s","accounts_in_message":[],"_revision_number":5},{"id":"019c8935281e384a4434ed04e21ddb099dd3dad9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 01:19:40.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerFlex_os-brick/f7386d6/DellEMC_PowerFlex_os-brick/48 : SUCCESS in 1h 32m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b215c80620ff30ff9f5a435db711f6f793a4fc5d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 07:47:12.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"14318a20ca365be03f0740d3843ec71a9acfc360","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-05 12:05:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"46eade69076301043cf637eedf4a6b73b6a280d1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 12:07:08.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nIt looks like all the concerns expressed by reviewers have been addressed. LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"6908f4324c79354bf96afc4ccefad604be02da2c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 12:07:37.000000000","message":"Removed Workflow+1 by Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"f34c50b078ed4af0f7b5e7e040cc23f9c5f3caf6","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 12:07:41.000000000","message":"Removed Code-Review+2 by Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"078c9fe3912025fa25391334789ed040a815771a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:07:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"de02df6ae79b538c03805e42440a08e9cd3b4778","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 12:09:22.000000000","message":"Patch Set 5: Review-Priority+1\n\nWas in a race condition with Gorka.  Not all concerns have yet been met!","accounts_in_message":[],"_revision_number":5},{"id":"610230a3644d3e7943a13e7884fd4b59aa440bad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-05 13:00:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"155776bebe88a0d639b2b7f59ac0b016149854dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:08:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/618abe0ee37d427dbbbdda9c6fe37b82 : SUCCESS in 1h 50m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/641fc92c2fc4411f89ac9ba3509fabdc : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c64932263c34c6e9fe69ae570d3871f : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d27d2616df40edb2c7d7eb3c9ae7d4 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96c05df244ca4a91afc063a80fc87794 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ba365729c14a7e83d136b45a035d14 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90bfd5211086422cab33cd5d0d2df197 : SUCCESS in 5m 38s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/900b8a8d6a5d4571baa8e206ee0553ef : SUCCESS in 1h 59m 13s","accounts_in_message":[],"_revision_number":5},{"id":"d9395e901d5f5b389ec863607e696cad5b7bc8a9","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-05 15:31:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"100d3c7f37c431b520798c8706e288eb5acce584","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-03-05 15:53:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"71359390778f6e784c053782690e1ec10513d5aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:02:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3283db20838e81727d804d2948bd8b9bc6d6d651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 18:21:18.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f5fd18571054d73ba4db511791a4df6 : FAILURE in 1h 26m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fdac99371954a6a80700bb77b7f2bc3 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92abd22ea79046e8bef92dab1b074106 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/605d40833bd44195bb30bda4012ff204 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55874d90066441358574a625d4bcbc19 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76006b52dbd848449193af0395f853fe : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46387e7caf4b4aedac0369c619c45277 : SUCCESS in 7m 37s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f0447aea34f4386913d84be5cc6ccf6 : SUCCESS in 2h 13m 57s","accounts_in_message":[],"_revision_number":5},{"id":"87687f38771d9a49c4747b37beaf298d9d5a5807","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-05 18:42:10.000000000","message":"Patch Set 5:\n\nrecheck tempest-full-py3 failure in tearDownClass (tempest.api.volume.admin.test_volumes_list.VolumesListAdminTestJSON) : Volume 1108fa43-223a-4f68-83d3-407b0646f2d6 failed to delete and is in error_deleting status","accounts_in_message":[],"_revision_number":5},{"id":"22e83ec8bb46a8c02e52fdac6955555bfb54c818","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-05 20:07:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_VNX_FC_os-brick/f7793df/DellEMC_VNX_FC_os-brick/43 : FAILURE in 1h 37m 58s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_VNX_iSCSI_os-brick/e828749/DellEMC_VNX_iSCSI_os-brick/41 : SUCCESS in 1h 55m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"15a93a6780a75e7c089d1b451b8134253ab3cc43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 20:25:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5b2f5c449614cd39b92d2156b932ad5 : SUCCESS in 1h 27m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23a7e98fb9ce4f05975f8ad14d713dd1 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd6eeec8a134eb0a28f85a405421074 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b55094e8884b470689bc0ce8d1c0d147 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82ea8092907c43ecb463e8c53e8563ac : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90a91745faec456fa31287fe5bb5e280 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f355f573348f442799e5b5df97c4b15d : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93216f390a1848f1a0e5f889a53463eb : SUCCESS in 5m 38s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ea3079be727f4da7922164c9ca0cb45d : SUCCESS in 4m 55s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/89d4c079e1e94885bd650c1e7c4f65df : SUCCESS in 1h 13m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a2c6e6f736c4ee5a290314994e67d92 : SUCCESS in 1h 30m 36s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b781cbce6ae940a2974955eba80262c5 : SUCCESS in 1h 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"45902da263df7f84e4d644d0a9758320f3afdb12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 20:26:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fdc8fd63b20b3ef36d6462c56d811d6ccbac120b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-05 21:33:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-iSCSI/285e24d/ : SUCCESS in 39m 30s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-iSCSI-IPv6/c6813ef/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-NFS/e2a5394/ : SUCCESS in 41m 11s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-NFS-IPv6/ccb52b7/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/45/775545/5/upstream-check/cinder-cDOT-FCP/bc49163/ : SUCCESS in 1h 07m 04s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"8611f6237f697425d1380b04b57fa8209187ac56","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-05 21:39:06.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerStore_FC-ob/a90aab4/DellEMC_PowerStore_FC-ob/52 : SUCCESS in 1h 08m 19s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerStore_iSCSI-ob/53bf100/DellEMC_PowerStore_iSCSI-ob/52 : SUCCESS in 1h 12m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"f8e1ffe6d26a06b38267e438677925bf8ce7f79a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-05 21:46:30.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"932b915d2ff6a24cbe90cca1c64f13943eefd3f3","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-05 22:14:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/45/775545/5/upstream-solidfire-check/cinder-solidfire-iSCSI/360364f/ : FAILURE in 1h 48m 01s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"533e339f12f2a1c6ad4d6fbc79abf77b0c9fa4f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 22:29:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0d9ae94777d4f9db73511f36f30fc1f : SUCCESS in 1h 23m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3fac8594986464dadeb4694146daf45 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f3d77a9d1d454b90c6fde602941a1a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e37cd59a6f480f93f812dbf1200d3c : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d353a44d60842acb0807500dff5774b : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5117952e6a4c909b71d6284e2591f2 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d4b986082b45fb95f3f5cfeca56397 : SUCCESS in 6m 08s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11ca847ea7c54b7f9706893b62ad07fd : SUCCESS in 1h 54m 31s","accounts_in_message":[],"_revision_number":5},{"id":"15b3702a89b6a013026f4c83c338d98fdfd00dab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 22:29:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"68b3330adf4aea7750c6a2276f7e01b910817288","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 22:31:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12189a53c8da4f9e9126b05e2c2165ed : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04faa378ec1e47e181a9d1505243acc2 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":5},{"id":"93a1eb2dacf6fee70e5f04b0a0396ae3f174b82c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-03-05 22:44:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_SC_FC-os-brick/75f57fc/DellEMC_SC_FC-os-brick/34 : FAILURE in 1h 08m 14s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_SC_iSCSI-os-brick/9318773/DellEMC_SC_iSCSI-os-brick/34 : SUCCESS in 2h 18m 34s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"990100a077edb684f58c917c9b932093c75fc007","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-06 04:24:25.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"692d10422fb0613c4bc3dcee12d6ad23e35a80d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 17:09:48.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_PowerFlex_os-brick/c83951b/DellEMC_PowerFlex_os-brick/53 : SUCCESS in 1h 42m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"fe62c90988a52be6f66ff73febc3998bff3e9e09","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-06 18:26:44.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0a69bfe6a8e441c28986c6477de6ab7f : POST_FAILURE in 1h 58m 44s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9ad56358312144478b0e85d9c4c874a9 : RETRY_LIMIT in 59m 39s","accounts_in_message":[],"_revision_number":5},{"id":"c94f2bf322686a123a187e79b733b24aa79fb01b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-06 20:46:48.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"deb5d27ddaec6752b465cfeb2f756f24a33fb2e2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-07 02:21:21.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_VNX_FC_os-brick/02ceec5/DellEMC_VNX_FC_os-brick/50 : FAILURE in 1h 25m 50s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/45/775545/5/check/DellEMC_VNX_iSCSI_os-brick/dcb375c/DellEMC_VNX_iSCSI_os-brick/45 : FAILURE in 49m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"97f9ccb88a37272c80a2486a7340e7efd62db58d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-07 06:35:43.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/45/775545/5/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/45/775545/5/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4478433550ddab970f106636a4ba5fbba82c5e39","revisions":{"88bcb5e11e45149c83dd1379de424c413b7a37d8":{"kind":"REWORK","_number":1,"created":"2021-02-15 02:08:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/775545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/775545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/775545/1"}}},"commit":{"parents":[{"commit":"ec70b4092f649d933322820e3003269560df7af9","subject":"Merge \"Improve error handling on target query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ec70b4092f649d933322820e3003269560df7af9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 02:08:18.000000000","tz":540},"subject":"Avould unhandled exceptions during connecting to iSCSI portals","message":"Avould unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\"\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vl threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88bcb5e11e45149c83dd1379de424c413b7a37d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88bcb5e11e45149c83dd1379de424c413b7a37d8"}]},"branch":"refs/heads/master"},"6a854b5d97177b2676ee5a89a1485062be68bf42":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-15 02:09:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/775545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/775545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/775545/2"}}},"commit":{"parents":[{"commit":"ec70b4092f649d933322820e3003269560df7af9","subject":"Merge \"Improve error handling on target query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ec70b4092f649d933322820e3003269560df7af9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 02:08:47.000000000","tz":540},"subject":"Avould unhandled exceptions during connecting to iSCSI portals","message":"Avould unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6a854b5d97177b2676ee5a89a1485062be68bf42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6a854b5d97177b2676ee5a89a1485062be68bf42"}]},"branch":"refs/heads/master"},"bf0c82675f1ece78519e343552838f6cb4b966da":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-25 08:29:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/775545/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/775545/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/775545/3"}}},"commit":{"parents":[{"commit":"ec70b4092f649d933322820e3003269560df7af9","subject":"Merge \"Improve error handling on target query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ec70b4092f649d933322820e3003269560df7af9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-25 08:29:50.000000000","tz":0},"subject":"Avoid unhandled exceptions during connecting to iSCSI portals","message":"Avoid unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bf0c82675f1ece78519e343552838f6cb4b966da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bf0c82675f1ece78519e343552838f6cb4b966da"}]},"branch":"refs/heads/master","description":"Edit commit message"},"812ab43b0fd1b214c0b4a3ad67a85ea0161e6a30":{"kind":"REWORK","_number":4,"created":"2021-03-03 12:11:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/775545/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/775545/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/775545/4"}}},"commit":{"parents":[{"commit":"ec70b4092f649d933322820e3003269560df7af9","subject":"Merge \"Improve error handling on target query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ec70b4092f649d933322820e3003269560df7af9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-03 12:11:04.000000000","tz":540},"subject":"Avoid unhandled exceptions during connecting to iSCSI portals","message":"Avoid unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/812ab43b0fd1b214c0b4a3ad67a85ea0161e6a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/812ab43b0fd1b214c0b4a3ad67a85ea0161e6a30"}]},"branch":"refs/heads/master"},"4478433550ddab970f106636a4ba5fbba82c5e39":{"kind":"REWORK","_number":5,"created":"2021-03-04 03:01:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/775545/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/775545/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/775545/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/775545/5"}}},"commit":{"parents":[{"commit":"ec70b4092f649d933322820e3003269560df7af9","subject":"Merge \"Improve error handling on target query\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ec70b4092f649d933322820e3003269560df7af9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-04 03:01:41.000000000","tz":540},"subject":"Avoid unhandled exceptions during connecting to iSCSI portals","message":"Avoid unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4478433550ddab970f106636a4ba5fbba82c5e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4478433550ddab970f106636a4ba5fbba82c5e39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
