)]}'
{"id":"openstack%2Fos-brick~778583","triplet_id":"openstack%2Fos-brick~master~I5cadb10ee418dd1f5e8a7d407a93d95ebcbaedb8","project":"openstack/os-brick","branch":"master","hashtags":[],"change_id":"I5cadb10ee418dd1f5e8a7d407a93d95ebcbaedb8","subject":"Enforce import order with flake8-import-order","status":"MERGED","created":"2021-03-03 21:29:28.000000000","updated":"2021-03-04 21:55:23.000000000","submitted":"2021-03-04 17:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778583","meta_rev_id":"9d6009979e8fe640c6a685821919bc94f05ffa8b","_number":778583,"virtual_id_number":778583,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 17:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-04 15:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-04 15:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-03 23:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 02:29:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 02:52:12.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-03-04 02:57:55.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-04 06:50:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 07:43:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-04 15:44:18.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-04 21:55:23.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"32e0d4d050d0a193f75b850bb798be950957ae15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-03 21:29:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae78ad72ebd63762909de1d754950ebb72159574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 23:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d27969c720cc4314b7e78503b0851241 : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cb1428453aa471db22f603d6aae0819 : SUCCESS in 1h 28m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d449ddb58ef4dca8526d78a0ad9efe7 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a4337283b34e01b89f378e7666d4a0 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9491174b3559436cbecc8ddbff8dd0e3 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f98d632f1656429a96bd9c6a5c31366f : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a70cc4f9ced4c8e8ea10803bb0097d1 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c260d406df64aa28cb2042e83fa57d8 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e9504327cff475683075ae15b0f8c1d : SUCCESS in 7m 55s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/ecde9e708f0e4981b1af5be46d780b59 : TIMED_OUT in 2h 01m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4005b42c765d4765a5ff755a371c000b : SUCCESS in 1h 33m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8350bfd53a4d4b448b8b12eb8a6417c9 : SUCCESS in 1h 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7655051f34d6b8493547914262c5ff0dc8195e4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-04 02:29:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/83/778583/1/upstream-check/cinder-cDOT-iSCSI/aa9b985/ : FAILURE in 39m 17s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/83/778583/1/upstream-check/cinder-cDOT-iSCSI-IPv6/337f6d1/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/83/778583/1/upstream-check/cinder-cDOT-NFS/74e1a69/ : SUCCESS in 40m 48s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/83/778583/1/upstream-check/cinder-cDOT-NFS-IPv6/fcfea10/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/83/778583/1/upstream-check/cinder-cDOT-FCP/04ce25e/ : FAILURE in 2h 41m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2675ce4de323b6a02151e1570305cd45d6fbdaaa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-03-04 02:52:12.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/83/778583/1/check/DellEMC_PowerStore_FC-ob/cd830c2/DellEMC_PowerStore_FC-ob/44 : SUCCESS in 1h 06m 15s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/83/778583/1/check/DellEMC_PowerStore_iSCSI-ob/f7093a9/DellEMC_PowerStore_iSCSI-ob/44 : SUCCESS in 1h 10m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a28d06f4455b5ad16699f40e2d7426007d98824","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-04 02:57:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/83/778583/1/upstream-solidfire-check/cinder-solidfire-iSCSI/bd1a620/ : FAILURE in 3h 09m 41s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"11e60bdd2489cc4ab593a54d89130b39f400a123","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-04 06:50:27.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/dafec411ad964d1b823530a535dc5ba2 : TIMED_OUT in 2h 36m 45s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ef044a132b9b40cb93ab2b1de7104906 : POST_FAILURE in 1h 46m 50s","accounts_in_message":[],"_revision_number":1},{"id":"72d4f81a0139b122f504eb7f15490f9ef4c12b4c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-04 07:43:46.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/83/778583/1/check/DellEMC_PowerFlex_os-brick/5ffa4c0/DellEMC_PowerFlex_os-brick/47 : SUCCESS in 1h 36m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b7754073df55552d58f037980014201c8fbe1eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-03-04 15:06:35.000000000","message":"Topic bp/six-removal removed","accounts_in_message":[],"_revision_number":1},{"id":"f034c6fb692399443b1efa9b6c8342eb6e2d33c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-04 15:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNinja-approving this one.","accounts_in_message":[],"_revision_number":1},{"id":"d0bd00d1475e699eeb4b07353a5eb077fd8d2516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 15:45:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0849ef7295693b06c9b7e18d0fe96237c3508bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:48:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3433d8a8173448fa9bba7fedf40c5b71 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e525a67894dc447dbe5c9ac561159b47 : SUCCESS in 1h 28m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a38dd73d336f42a0938bbd1480886706 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c665d03e4dc4b1ab023349bdefd8ea8 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d33d415bd1f43e68226d53d6d892c0b : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d9a8e0968b84d4cb48f91bb744b1850 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd097e4cf4c04172bfd50733c8ab6898 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02e367f8427542c082a258c8eea399cc : SUCCESS in 7m 53s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f82a8abcd844265a7516c51a54a6db5 : SUCCESS in 1h 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f5f913ed3867263f856d3463c65dbf9d7cb84212","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:48:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"01c848cde82a8954bae02e97a98505a70f934561","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:50:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde5851a14b24ca7b16a825eb6881401 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e248984e243a46088fdffda99b7f4245 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"9d6009979e8fe640c6a685821919bc94f05ffa8b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-04 21:55:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/83/778583/1/check/DellEMC_VNX_FC_os-brick/0f4995f/DellEMC_VNX_FC_os-brick/42 : FAILURE in 1h 10m 55s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/83/778583/1/check/DellEMC_VNX_iSCSI_os-brick/ae7d459/DellEMC_VNX_iSCSI_os-brick/40 : SUCCESS in 2h 10m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55376b85ffa90a343fd90f3980fc58ea2da9d00f","revisions":{"55376b85ffa90a343fd90f3980fc58ea2da9d00f":{"kind":"REWORK","_number":1,"created":"2021-03-03 21:29:28.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/83/778583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/83/778583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/778583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/778583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/778583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/83/778583/1"}}},"commit":{"parents":[{"commit":"9d19bb93f3ecb2afd957fb2f05c4439ae298601d","subject":"Fix import order of libraries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9d19bb93f3ecb2afd957fb2f05c4439ae298601d"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-03 21:27:33.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2021-03-03 21:27:33.000000000","tz":-300},"subject":"Enforce import order with flake8-import-order","message":"Enforce import order with flake8-import-order\n\nUse flake8-import-order to enforce the correct\norder of imports (stdlib, third-party, project).\n\nChange-Id: I5cadb10ee418dd1f5e8a7d407a93d95ebcbaedb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55376b85ffa90a343fd90f3980fc58ea2da9d00f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/55376b85ffa90a343fd90f3980fc58ea2da9d00f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
