)]}'
{"id":"openstack%2Fos-brick~779461","triplet_id":"openstack%2Fos-brick~stable%2Fvictoria~I0278c502806b99f8ec65cb146e3852e43031e9b8","project":"openstack/os-brick","branch":"stable/victoria","topic":"bug/1915678","hashtags":[],"change_id":"I0278c502806b99f8ec65cb146e3852e43031e9b8","subject":"Avoid unhandled exceptions during connecting to iSCSI portals","status":"MERGED","created":"2021-03-09 11:37:38.000000000","updated":"2021-05-19 16:15:55.000000000","submitted":"2021-05-19 16:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779461-bug/1915678","cherry_pick_of_change":775545,"cherry_pick_of_patch_set":5,"meta_rev_id":"8cf13597381d3459580e99cec733040ffb109b4d","_number":779461,"virtual_id_number":779461,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 16:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2021-03-18 15:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":1,"date":"2021-03-12 20:06:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2021-03-16 06:37:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":2,"date":"2021-04-07 15:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2021-05-08 05:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-19 14:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-19 14:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 13:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 14:16:57.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 14:20:49.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-09 15:01:37.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 17:02:02.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-12 20:06:02.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2021-03-16 06:37:27.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2021-03-18 15:12:17.000000000","updated_by":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"reviewer":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"state":"REVIEWER"},{"updated":"2021-04-07 15:08:46.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-05-08 05:57:02.000000000","updated_by":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"reviewer":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"state":"REVIEWER"},{"updated":"2021-05-19 14:59:01.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"63fdcc900d6a2701d5f6506b5df356a4fcdcb9f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-09 11:37:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2de2ef9283ebf656069d4c6869dc0145df756622","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-09 11:42:31.000000000","message":"Topic bug/1915678-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"f91bc098174ee5959fcf4d6844cd00e9ec8a996d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-09 11:42:34.000000000","message":"Topic set to bug/1915678","accounts_in_message":[],"_revision_number":1},{"id":"07d0bba74cdbde80eaab3308fa6fbe322b91af59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 13:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21d5fc112a7d4fb89b59154577cfb5f8 : SUCCESS in 1h 25m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa601992e0b0460ab172f8f031c97c9b : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dcfd3ce2854c32bbda358f2d77d5ca : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b14efc3c29ef47018760ca082addf351 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97b74dd11ca04aa1aeea322efe25ad1e : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec46cc6fc6f240448cb15ed3fba7e7ef : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9471f3a3aa83440996b43557e1e37d9f : SUCCESS in 7m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/25c4c09f1a7b48b7b224cb037e9d14a5 : SUCCESS in 4m 59s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b954233d37e14f248eabb2e84e55b9cb : TIMED_OUT in 2h 04m 45s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bcbe4506321048a9b236e5c5c0e6e6c4 : SUCCESS in 1h 34m 04s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e750d12ec7384d318946c37db14b5362 : SUCCESS in 1h 30m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"568e290e148970daca58f2ea84ecbe37eaeb1fa2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-09 14:16:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/61/779461/1/upstream-check/cinder-cDOT-iSCSI/f8cdd04/ : FAILURE in 19m 03s\n- cinder-cDOT-iSCSI-IPv6 https://logs.openstack.netapp.com/logs/61/779461/1/upstream-check/cinder-cDOT-iSCSI-IPv6/99ff5e0/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/61/779461/1/upstream-check/cinder-cDOT-NFS/fbe7084/ : FAILURE in 18m 52s\n- cinder-cDOT-NFS-IPv6 https://logs.openstack.netapp.com/logs/61/779461/1/upstream-check/cinder-cDOT-NFS-IPv6/9bf3dcf/ : NOT_REGISTERED (non-voting)\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/61/779461/1/upstream-check/cinder-cDOT-FCP/98a1bc5/ : FAILURE in 27m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2cff1fca85e9ba5dcb3e4aa034251d5f0a9249c","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-09 14:20:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/61/779461/1/upstream-solidfire-check/cinder-solidfire-iSCSI/46ac3cb/ : FAILURE in 31m 17s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"b7dbe74ecb7085ee2785ac60705c85ca9b1c3f04","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 15:01:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/61/779461/1/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/61/779461/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ec45b71e969bb62bfd71836a20e3a1e6a5ef8d0b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-09 17:02:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/61/779461/1/check/devstack-plugin-ceph-dsvm-focal-py3/d04ca13/ : FAILURE in 1h 10m 52s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/61/779461/1/check/tempest-dsvm-full-focal-py3/6266794/ : SUCCESS in 1h 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"51e01c9d0c19da682c16c784a0e4544693fee0bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-09 18:40:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/61/779461/1/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/61/779461/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e9a8e386b429169d79589bbad38dc1b8046beec3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2021-03-12 20:06:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ebc2ea23f1f945a2697641b9e847cde44633f3a","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2021-03-16 06:37:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2aa829700b3ab23b5592fd59014734970960dd7c","author":{"_account_id":21570,"name":"pcaruana","email":"pcaruana@redhat.com","username":"pcaruana"},"date":"2021-03-18 15:12:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a865b51a7d00912320af24fb861d94e7f8fa9f4c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-04-07 15:08:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81653fd413ebd9cd902b3539f5f51d240ee1a682","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-05-08 05:57:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"40f27a3dd74c6b9d049b26a5845a7f9581f2eb32","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-05-19 14:59:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is an appropriate backport.","accounts_in_message":[],"_revision_number":1},{"id":"8396add7651d4fadb1ddd066a86a2bba6d01fa1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 14:59:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb306fdc894781044a4b1448fed32053ca3a323e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:14:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/297e7781712848ca90c3f4ecf4c21014 : SUCCESS in 1h 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7c8742c34d453f98ef2759e80dda72 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a13df4a4ef54c3c9a90fcc23bf8a557 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc0e0cade56b465ba6880a4fea1f4612 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897a5a03a99640c7a07f65cd4f15fd06 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d93afefe4af421888fd80191a262760 : SUCCESS in 6m 32s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/abf7c61724dd44bc8687c962ea49b771 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"adcadb7047209302642b9187243d5886bfc27605","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:14:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8cf13597381d3459580e99cec733040ffb109b4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:15:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557785e4cd4a49a19375459914661ef7 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35c8ee7d44c94bce8032a98e5c96b990 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57c8f4334c0e58b151f449f3316bb9e1c8ed6da4","revisions":{"57c8f4334c0e58b151f449f3316bb9e1c8ed6da4":{"kind":"REWORK","_number":1,"created":"2021-03-09 11:37:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/779461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/61/779461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/61/779461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/61/779461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/61/779461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/61/779461/1"}}},"commit":{"parents":[{"commit":"a280d9f771dedd9dfb7770fadc965a46412b69ed","subject":"Merge \"ScaleIO: More connection info backward compatibility\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a280d9f771dedd9dfb7770fadc965a46412b69ed"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-09 11:37:38.000000000","tz":0},"subject":"Avoid unhandled exceptions during connecting to iSCSI portals","message":"Avoid unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n(cherry picked from commit 4478433550ddab970f106636a4ba5fbba82c5e39)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/57c8f4334c0e58b151f449f3316bb9e1c8ed6da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/57c8f4334c0e58b151f449f3316bb9e1c8ed6da4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
