)]}'
{"id":"openstack%2Fos-brick~783207","triplet_id":"openstack%2Fos-brick~stable%2Fwallaby~I066d456fb1fe9159d4be50ffd8abf9a6d8d07901","project":"openstack/os-brick","branch":"stable/wallaby","topic":"cleanup_connection-stable/wallaby","hashtags":[],"change_id":"I066d456fb1fe9159d4be50ffd8abf9a6d8d07901","subject":"iSCSI: Fix flushing after multipath cfg change","status":"MERGED","created":"2021-03-26 16:39:27.000000000","updated":"2021-03-26 20:54:27.000000000","submitted":"2021-03-26 19:40:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783207-cleanup_connection-stable/wallaby","cherry_pick_of_change":782992,"cherry_pick_of_patch_set":3,"meta_rev_id":"907ebf8f67921b0eaf730d5d372580549dcf5c14","_number":783207,"virtual_id_number":783207,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 19:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-26 17:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-26 19:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-26 19:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-26 17:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 16:39:27.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2021-03-26 16:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 17:12:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-26 19:14:42.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-03-26 19:32:02.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-03-26 20:51:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 20:54:27.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"80af638dec3ea3112335916b9f125730cfe695ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-03-26 16:39:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"28d5b0964391b42e6757d9086153ee2e610389b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb85cb7c7f564e8f918213c7151145c7 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6177b0396deb4a2c94d868c13a57949d : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c796c3196e2a4544a43d3fd469acc33f : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2155b04ea624cdc90442197ccf6663c : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71f452f114ae4362bc3465d58ef086c0 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed4ccabf081146d8bc01dfd545f0bc44 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ba4ace76b4c4768987ca001542c8de7 : SUCCESS in 4m 57s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/230fe2bcdf9041929e8f039c55da3ab8 : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f25eec9a4d191c2869fae6baade4618214fff04","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-26 17:12:35.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\nThis patch addresses a potential data loss issue and is an appropriate backport.","accounts_in_message":[],"_revision_number":1},{"id":"d9f6fd90c154e9c44080ce83462d3bc420868849","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-03-26 19:14:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/07/783207/1/upstream-solidfire-check/cinder-solidfire-iSCSI/661778a/ : FAILURE in 2h 22m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"6eaf04af0d22da112779a7b47003fb2bb6ef8119","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-03-26 19:32:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nDataloss is bad...Mmmmmmkay?","accounts_in_message":[],"_revision_number":1},{"id":"ca31d8c4ed43f2e6783edbe67cea12708ce163df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:32:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fe83b16e134fcddbaf7f95a7cf63926e16a97e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:40:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/282aa0aac8bb495981df08f4be3dbee6 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ede6fd1f2824815b94c74f355bb460e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49532b394767430193c1ee5c8b959cd6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ef28d1e4cd4944909b5ddb67d27772 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cefee70f40498289635da2c30c0c30 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ce354ebc4846dea49ef9ba6df56d98 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9a858c4471494f67ffd2827461ff28c21bbc8e37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:40:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa6b890f68885b485263a76548a72b30ce712e78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:42:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7261226b5247f3ba5af33967974b8b : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92dd3dcb5aee408a9595ba249c53dfa9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"bcfa4af60e490b49eba8020d3b21fb621a816dd1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-26 20:51:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/07/783207/1/upstream-check/cinder-cDOT-iSCSI/94bdb87/ : SUCCESS in 36m 36s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/07/783207/1/upstream-check/cinder-cDOT-NFS/eb054bd/ : SUCCESS in 41m 16s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/07/783207/1/upstream-check/cinder-cDOT-FCP/440a23f/ : SUCCESS in 1h 08m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"907ebf8f67921b0eaf730d5d372580549dcf5c14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 20:54:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/07/783207/1/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/07/783207/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c70d70b24056b81a482de36a8e47757210e2e898","revisions":{"c70d70b24056b81a482de36a8e47757210e2e898":{"kind":"REWORK","_number":1,"created":"2021-03-26 16:39:27.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/07/783207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/07/783207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/07/783207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/07/783207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/07/783207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/07/783207/1"}}},"commit":{"parents":[{"commit":"6649b8d8c461528d585000f0a9b8294de176a7d5","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6649b8d8c461528d585000f0a9b8294de176a7d5"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-03-23 12:46:10.000000000","tz":60},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-03-26 16:39:27.000000000","tz":0},"subject":"iSCSI: Fix flushing after multipath cfg change","message":"iSCSI: Fix flushing after multipath cfg change\n\nOS-Brick disconnect_volume code assumes that the use_multipath parameter\nthat is used to instantiate the connector has the same value than the\nconnector that was used on the original connect_volume call.\n\nUnfortunately this is not necessarily true, because Nova can attach a\nvolume, then its multipath configuration can be enabled or disabled, and\nthen a detach can be issued.\n\nThis leads to a series of serious issues such as:\n\n- Not flushing the single path on disconnect_volume (possible data loss)\n  and leaving it as a leftover device on the host when Nova calls\n  terminate-connection on Cinder.\n\n- Not flushing the multipath device (possible data loss) and leaving it\n  as a leftover device similarly to the other case.\n\nThis patch changes how we do disconnects, now we assume we are always\ndisconnecting multipaths, and fallback to doing the single path\ndisconnect we used to do if we can\u0027t go that route.\n\nThe case when we cannot do a multipathed detach is mostly when we did\nthe connect as a single path and the Cinder driver doesn\u0027t provide\nportal_targets and portal_iqns in the connection info for non\nmultipathed initialize-connection calls.\n\nThis changes introduces an additional call when working with single\npaths (checking the discoverydb), but it should be an acceptable\ntrade-off to not lose data.\n\nCloses-Bug: #1921381\nChange-Id: I066d456fb1fe9159d4be50ffd8abf9a6d8d07901\n(cherry picked from commit d4205bd0be60758d40e342d98424488d56e029ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c70d70b24056b81a482de36a8e47757210e2e898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c70d70b24056b81a482de36a8e47757210e2e898"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
