)]}'
{"id":"openstack%2Fos-brick~788795","triplet_id":"openstack%2Fos-brick~stable%2Ftrain~I65204aa7495740dc1545bff2c5c485a8041e7930","project":"openstack/os-brick","branch":"stable/train","topic":"t/1924652","hashtags":[],"change_id":"I65204aa7495740dc1545bff2c5c485a8041e7930","subject":"multipath/iscsi: remove devices from multipath monitoring","status":"MERGED","created":"2021-04-29 16:28:25.000000000","updated":"2021-05-06 10:22:11.000000000","submitted":"2021-05-05 05:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788795-t/1924652","meta_rev_id":"7d838f3267a39c2b492e8c5d93eb15ce5e1f414c","_number":788795,"virtual_id_number":788795,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 05:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-05-04 20:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-05-03 19:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-05-04 07:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-05-04 20:45:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-05-03 19:42:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 16:28:25.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-29 18:29:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 18:33:01.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-04-29 18:36:02.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 18:36:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 14:37:17.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 14:37:23.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-03 19:40:01.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2021-05-03 19:40:01.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-05-04 14:25:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-05-04 14:25:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-05-04 14:25:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-05-04 20:45:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"751b2935435d208b431adbab2ca5714b58678706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-04-29 16:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d48e75650607de2caca787f49ed639569f1cfcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 18:29:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a2483b4c65484eb5a6e6827f7a12d4aa : SUCCESS in 1h 38m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d21e60e93301425da39add12c07049aa : FAILURE in 1h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03597a26844349b18f99e0c72d388e4f : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/360f9232cdaa48b0b9d8c5948b45c707 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a572be7f4d4873bb7fbdb3cb3b992c : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9193b367c59240a9a712a6b060f56ccc : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903db94387944d4da0042f5b3236e1d4 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd235c3899c2459bb7f149f9ef34e9d4 : SUCCESS in 5m 21s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d8532907f19346ba839da88271aefa50 : SUCCESS in 1h 20m 03s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61b4021f277946d2b5f3664b9d2936ab : SUCCESS in 1h 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6efd45d30f5f42cc64fe0a08ad0f698f7752ec13","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-04-29 18:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/788795/1/iscsi/ : SUCCESS in 1h 08m 31s\n- os-brick-smb http://cloudbase-ci.com/os-brick/788795/1/smb/ : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d7c8f1a8c05051164a29dfd84989297757ebe14c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-29 18:36:02.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/ef353fbf2fe84288a49c3ff5a3e53547 : POST_FAILURE in 1h 25m 06s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/3a5ef80b7c9e406681ca819dcb5402ba : POST_FAILURE in 1h 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bb77af621718adb9442e244a12c7fc8524b7ef57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-29 18:36:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/95/788795/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8fd9db53644c17216327990f9d5a2996133d11bc","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-04-30 12:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"630c7923f8bd6c049d5e57d882a17d5f18afee8e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-30 12:24:51.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"7d07598560beb2c8bef53c95c155cb0bf0876e91","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-04-30 13:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/788795/1/iscsi/ : SUCCESS in 1h 06m 08s\n- os-brick-smb http://cloudbase-ci.com/os-brick/788795/1/smb/ : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fe2ef1f1835211f0f71aa0906ac302f39701c99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/14dd8332acfc4a3499e3dc2374f4ea79 : SUCCESS in 1h 55m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/791858c498aa40499445f38f6e58fd29 : SUCCESS in 1h 50m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f2c8f043df4648a29b63d239590260 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f44e918a9a14b2482dc0e6ab148f159 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ad4d317a4844cf4a559c51fe95f8874 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8fa917e016b42a18be43392ffecc200 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8488ba697a204947b85bda339f1931fe : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e6a385cbcc94824ac1b5abb94992541 : SUCCESS in 4m 30s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/6e24ad7db8ec46dbac30dbfd010450bf : SUCCESS in 1h 17m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/077a5b3f8f77445bb0fbc5b131664fe5 : SUCCESS in 2h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0eded8ffc874a06db723b96019758c8f44dbfb5b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-30 14:37:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-iSCSI/a9c2620/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-NFS/3c2360c/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-FCP/6f34231/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"53917817db030fab38d139084b87feb9965a7957","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-04-30 14:37:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/788795/1/upstream-solidfire-check/cinder-solidfire-iSCSI/9f28afd/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"f84197668bc4b50584b44ef2716b5f09111de99b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-30 15:34:57.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/e785ebcc366f4ace9045704abd6857a8 : POST_FAILURE in 1h 42m 16s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/c30e558786f34b45ab71586ee97091f1 : POST_FAILURE in 1h 45m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5985b5ef7aeb6229c9642adab353c310d99486bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-30 15:35:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/95/788795/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"489870c8885efe48de2d5fb624c1a7cc67351ab5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-03 19:42:30.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nImportant to backport.  I don\u0027t see anything in this change that could cause a problem running under py2.7","accounts_in_message":[],"_revision_number":1},{"id":"c30801e7725200419d09008e7a0d0216484dd205","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 07:43:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d913f4d43cf113c8a94aa1d5e07febc82b22adeb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 20:45:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb39a2c223e48b0767794129e99762ffa34b28b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 20:45:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4420521760898c41504e4c9d7e60f02eee2d0db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 22:19:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/84832d181fff45b3a35d972798bdf55c : SUCCESS in 1h 20m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/970c3de106b545d292df32aacba4e300 : SUCCESS in 1h 07m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9abdc892f447c29900388517edc996 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2e863a4cd914c1d9dfc5b9a77dc98f1 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22d459f2321542d7b02d900fb864e02f : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49346a59ffec4454bbd46363c25845d9 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e0fbc16c924229a0b167a4aac83c05 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6415127b2e4e426a9468b67eb4b85fa6 : SUCCESS in 5m 05s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d4554517d0f48d0949df6682e1bedea : FAILURE in 1h 30m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e04e4c91cbbcb10c154bb5b33bc79c62bf3012dd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-05-04 22:54:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"04fa91b6fa099ad288ef33603b432971845f54f0","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-04 23:50:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/788795/1/iscsi/ : SUCCESS in 55m 30s\n- os-brick-smb http://cloudbase-ci.com/os-brick/788795/1/smb/ : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d49286d8f3fae15520319aa898204b6798cad9b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 00:37:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c7690712d24e45438e59fb2cc9ad4c75 : SUCCESS in 1h 28m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdb46fc61d124c40a38e4b2e0b585783 : SUCCESS in 1h 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8e3935bc9b4491aad9dbf87ed24897f : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5faf0e931f34113aaacb765c8e9ef81 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4998cb37aef14687a9f1c42706247993 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76b86af72e3a45e6817c20e00c70f3e9 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9832abcb97504444851989edee96d813 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bffa0fd567084a11918dc6eacab97dbb : SUCCESS in 5m 47s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/ca586440fa1b48d386c0720dab5275fb : SUCCESS in 1h 08m 07s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8caab87acae64962a38dc01e565c890a : FAILURE in 1h 42m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2b596c78b5da98419ecd24dc68930e0b0b83258e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-05 00:46:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d25626dfdcad25ed59e3174dc5f456270a45be7b","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-05 01:43:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/788795/1/iscsi/ : SUCCESS in 56m 14s\n- os-brick-smb http://cloudbase-ci.com/os-brick/788795/1/smb/ : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3c6134e652363f1bf51265260c3e14b0cdec02da","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-05 01:48:43.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/0b30038fd7b94ea0b16fae69a387fa47 : RETRY_LIMIT in 31m 11s\n- devstack-plugin-ceph-dsvm-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/95/788795/1/check/devstack-plugin-ceph-dsvm-bionic-py3/96f901b/ : SUCCESS in 1h 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"691d69baf34de9e028a6f9cdd385092731ca7c0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-05 01:48:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/95/788795/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3589dea32a84c25d3550258d51d4826e88cc31cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 03:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2400742c0318412d9cb9a1789939ca9f : SUCCESS in 1h 29m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a7d051465ef4988897c89026c3aa1e8 : SUCCESS in 1h 02m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078fcb4222fa49218005678fbf26da6a : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e661b335b1d845c0a8436a4874853b6e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e028af7214de486bad956174a1a23c62 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/865460e8cdd145798bc398f8add1f41e : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433311cacfd641a9b4aff7d79eb73649 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb2892ec90a641b3917046f197e7cc54 : SUCCESS in 5m 03s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/af0cdaee14144b90b8571823ba7713d9 : SUCCESS in 1h 11m 58s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/83be256db1fe44d496143097af5a1fb6 : SUCCESS in 2h 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"855bbb4a16e288a820bb3f219fe8ef9f3fda4fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 03:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d49bcd1508549181afa3e55ca2579dfa0b7c081d","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-05 04:07:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/95/788795/1/upstream-solidfire-check/cinder-solidfire-iSCSI/da770b8/ : FAILURE in 1h 00m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8f5e72e7d22d735f437cd3251bb8b010f804d3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 05:07:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/dcf9247a5ec544048bcfd6e6ca243f6d : SUCCESS in 1h 37m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/924efb8b4e494d1f9f9afd36ddfd1413 : SUCCESS in 1h 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ba166260cd410b9f1f6e8b65271b5c : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d844ff754dc44d8a88c47f6aa7d186a1 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32743ece8e047d7a3edb7af39981b3a : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/914b79ffc00e47979120f1da6cee0bf4 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31f37c908fb491dad8085d480ff309e : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5aaa44434cef44d5a040fd0d5a02e5a8 : SUCCESS in 4m 52s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/88f76e67400f46738742b3b5d8af7d4f : SUCCESS in 1h 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7bd5a2e03a9adeb93f3e780ff1d5ce66494a006b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 05:07:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2096cd12fdd3e80cf87662badfe436b3a9a675b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 05:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7000859924b14c9880cd7e98d88e8772 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c6c89f50b68469e95c4b8bd2ad38ded : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"d4cc64befb26a217573e3918ead409ee22050aa5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-05 05:42:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-iSCSI/4118242/ : SUCCESS in 42m 58s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-NFS/5db30f9/ : FAILURE in 52m 36s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/95/788795/1/upstream-check/cinder-cDOT-FCP/77233c1/ : FAILURE in 2h 34m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f67d234290f4c7a44cbce21af847d2d9f617350","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 10:22:07.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1},{"id":"7d838f3267a39c2b492e8c5d93eb15ce5e1f414c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-06 10:22:11.000000000","message":"Patch Set 1:\n\nBuild Failed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30244b175cac6e643915d762a80b8c8f5848f256","revisions":{"30244b175cac6e643915d762a80b8c8f5848f256":{"kind":"REWORK","_number":1,"created":"2021-04-29 16:28:25.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/95/788795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/788795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/788795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/788795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/788795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/788795/1"}}},"commit":{"parents":[{"commit":"0a142dee7c2a04bd36cfa63b532f0a9d3b6f8c19","subject":"Drop lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a142dee7c2a04bd36cfa63b532f0a9d3b6f8c19"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-12 05:07:46.000000000","tz":540},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-04-29 16:27:45.000000000","tz":120},"subject":"multipath/iscsi: remove devices from multipath monitoring","message":"multipath/iscsi: remove devices from multipath monitoring\n\nRecent multipathd doesn\u0027t remove path devices timely when it receives\nburst of udev events but wait for a while to start actual removal.\nBecause os-brick removes path devices in a short time during detaching\na multipath device, it is likely to hit this burst limit and sometimes\npath devices are not removed before a subsequent operation is started.\n\nThis change ensures that os-brick tells mutlipathd to remove path\ndevices when the devices should be deleted, so that orphan paths are\nnot left when starting a subsequent attach operation.\n\nCloses-Bug: #1924652\nChange-Id: I65204aa7495740dc1545bff2c5c485a8041e7930\n(cherry picked from commit 1b2e2295421615847d86508dcd487ec51fa45f25)\n(cherry picked from commit 0cd58a9b0a5520e184a7d3ef45f03ed8cda93732)\n(cherry picked from commit a0e995b1308b9a188e8719fdc6d6bad15508dba6)\n(cherry picked from commit 5c031cb2d9e9b10a33c82586d11139594c96a25f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30244b175cac6e643915d762a80b8c8f5848f256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30244b175cac6e643915d762a80b8c8f5848f256"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
