)]}'
{"id":"openstack%2Fos-brick~791273","triplet_id":"openstack%2Fos-brick~master~I1f26d7a34e96f542d8fcd8d968976fdc21afbc38","project":"openstack/os-brick","branch":"master","topic":"privsep","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-25 09:32:07.000000000","reason":"Change was marked work in progress"},"1736":{"account":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"last_update":"2026-09-25 09:32:07.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I1f26d7a34e96f542d8fcd8d968976fdc21afbc38","subject":"encryptors: Drop support for out-of-tree encryptors","status":"NEW","created":"2021-05-13 15:47:08.000000000","updated":"2026-09-25 09:32:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":6,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"3c4075ebc48e1dcc5a601058f59e6841bc291571","_number":791273,"virtual_id_number":791273,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-23 16:40:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":-1,"date":"2026-09-24 13:53:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 17:47:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-13 17:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-13 18:11:08.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-13 19:16:31.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-13 19:17:02.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-05-13 19:43:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-14 12:55:46.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2022-02-14 11:32:13.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2026-09-22 16:20:47.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-09-22 16:43:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-22 17:53:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-23 00:57:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-09-24 13:53:24.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"}],"messages":[{"id":"a48c102da02789b633798fe6ac10f4c453dce203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-13 15:47:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"964f5a02f470375104562a97cc9052f01da01048","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-13 17:47:55.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/73/791273/1/check/devstack-plugin-ceph-dsvm-focal-py3/9b1daf2/ : FAILURE in 49m 30s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/73/791273/1/check/tempest-dsvm-full-focal-py3/f3e01c5/ : FAILURE in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1f0cd56b5000227b140beded0aa1392bcfc7d3a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 17:56:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/799b9dcf3e50472a919431b8ac49da15 : FAILURE in 1h 31m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/597550795124466cb8b44eb72fa8175a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3e78880e1442dcbdda4d925332a3c6 : FAILURE in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/743cbb48773e41378ae714c8f900a633 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/221914d62dc845599866e061d5e9d828 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98240588ed454566955f4eb867135921 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6aa75df1e94f608e4de781c31aa2e4 : SUCCESS in 7m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/82caa84bf2cc4ff3a78b3881891875ef : SUCCESS in 6m 10s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/1d7ed907220a443bbeb0bcdc264fdb09 : FAILURE in 1h 23m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6b432e4cce445c59187c5a8f0d8592a : FAILURE in 2h 02m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/499ceaf6694b4f419939e888abfe3fc4 : SUCCESS in 1h 35m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/tests/initiator/connectors/test_iscsi.py","accounts_in_message":[],"_revision_number":1},{"id":"2671ef4445a13aad62d6fae64468003fe29f6f82","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-13 18:11:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"6f8653c795622cbe5a585ab7159e3e9004f15684","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-13 19:16:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/73/791273/1/check/DellEMC_XtremIO_FC_os-brick/86b4375/DellEMC_XtremIO_FC_os-brick/None : RETRY_LIMIT\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/73/791273/1/check/DellEMC_XtremIO_iSCSI_os-brick/1bd430f/DellEMC_XtremIO_iSCSI_os-brick/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"0cc625857e5d5e37b1a90d0fa26d0a310a5e364c","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-13 19:17:02.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/791273/1/iscsi/ : FAILURE in 44m 59s\n- os-brick-smb http://cloudbase-ci.com/os-brick/791273/1/smb/ : SUCCESS in 1h 06m 52s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/791273/1/ceph/ : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ba3819860cff8f0ef77067dfaf806f591788509c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-13 19:18:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"300b429174e4986f61d9221215804a80a1d2877f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-13 19:43:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/73/791273/1/check/DellEMC_Unity_FC_os-brick/9bc86d7/DellEMC_Unity_FC_os-brick/28 : FAILURE in 1s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/73/791273/1/check/DellEMC_Unity_iSCSI_os-brick/0479a1d/DellEMC_Unity_iSCSI_os-brick/27 : FAILURE in 25m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a5c4467c9f083e98d1076019de80fae4135200c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-13 20:13:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3a54e5b7c0353836948ad124d3da019d7a9d5a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-14 11:12:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d5e6dd40109f46cb774055b056b48f52b2f5e2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:43:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f767c1e797694138981d991fb9bc4258 : FAILURE in 1h 26m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/780d6603682144dbb4fcf1d1bddf7ff0 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aab56387a9248d5bf3cc3f553ae56fc : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e6564d9808e418c88a4e5a3f758f704 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/538eae37a58e476e95e96f34322df827 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a07a0bb82c64e8c8165591845973019 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792494d1f26344c1a73461aa8c82e59c : SUCCESS in 7m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c8157abf8cf54003a5c27ba70039a77b : SUCCESS in 5m 10s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/4a20fbbee9424329a3b2580eff561f16 : SUCCESS in 1h 23m 15s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/843c4fa872be439dbf6b5c096b57b836 : FAILURE in 1h 14m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a7dd6ead9a104c6fb984209949621102 : SUCCESS in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb8a18ad9eb9a063acff27d41a816f299acda0aa","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-05-14 12:55:46.000000000","message":"Patch Set 2: Code-Review-1\n\nNeeds a releasenote and to do this correctly we need a db migration in cinder to move existing encryption volume types to the enums. This is the reason I\u0027ve procrastinated for so long on sorting this out in os-brick tbh.","accounts_in_message":[],"_revision_number":2},{"id":"c29dac7f43285bf95087401050e112868448d651","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-14 13:17:02.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0ff45d186e7947e0a0508834ab8e4534 : POST_FAILURE in 1h 28m 54s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b7a62027dc24496d9161cb0db9ef8b9d : POST_FAILURE in 1h 23m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5ed8b745d6998f3cdea0b794399db4fb19d29496","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-14 13:17:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/73/791273/2/check/DellEMC_XtremIO_FC_os-brick/45998c4/DellEMC_XtremIO_FC_os-brick/None : RETRY_LIMIT\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/73/791273/2/check/DellEMC_XtremIO_iSCSI_os-brick/2786d18/DellEMC_XtremIO_iSCSI_os-brick/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad59bad3760f156d9f438cebe28cbdd19a28ea71","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-14 13:23:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/791273/2/iscsi/ : SUCCESS in 56m 04s\n- os-brick-smb http://cloudbase-ci.com/os-brick/791273/2/smb/ : SUCCESS in 1h 06m 14s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/791273/2/ceph/ : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":2},{"id":"40ba0bc0a684f26c5e33f882eb7e14e0e9ae2fc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-14 13:56:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fcc307f4f20696d4e216cfaa002d98cdf647e5b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-14 14:51:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e9847601beedf8b791b02a26da4d079bd5444ead","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-05-14 14:52:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/73/791273/3/check/DellEMC_XtremIO_FC_os-brick/6bc8438/DellEMC_XtremIO_FC_os-brick/None : RETRY_LIMIT\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/73/791273/3/check/DellEMC_XtremIO_iSCSI_os-brick/22bc4e3/DellEMC_XtremIO_iSCSI_os-brick/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"704d8e2187b99cd7917d17072ee211ab491882f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-14 14:52:38.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"7e82fd21b8b5ce6f6b5ba63d6d14efdc3d902ec7","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-14 16:17:39.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/791273/3/iscsi/ : SUCCESS in 57m 17s\n- os-brick-smb http://cloudbase-ci.com/os-brick/791273/3/smb/ : FAILURE in 48m 10s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/791273/3/ceph/ : FAILURE in 1h 06m 42s","accounts_in_message":[],"_revision_number":3},{"id":"2260678daec6e1de6b449b21e4787518fe989c20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-14 16:29:54.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"21e278f73a74872fe5f2a4f3bd9bd9d679f54903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 16:34:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce1f89355d2a4ab58ceae1ed65750e06 : FAILURE in 1h 34m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94fcf49e3d744d0cb1aaa59055b2a6c6 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1879c2207d86447281e48e7628583a93 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a1c81f3f0f445649792c7d830872b5c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2487fd189d845b9afc655443a1b2b7d : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eb92c302f4a402bafbed4fb8ce12863 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d666e92af44a7889a6de54a71a7a9a : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c676a98928ac4197971314df3139c8b9 : SUCCESS in 6m 21s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/afc050ccaa5e4830873e1cdc0679b4c0 : SUCCESS in 5m 05s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d72fde6ae2e64c9aaa12a4dcfbc201fd : SUCCESS in 1h 07m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e8938fd22224ed0be4a6074f071bf04 : FAILURE in 1h 41m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/bf829820767f49f685e21d6613eae2da : SUCCESS in 1h 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84a0712a23e71044b76ac6fc4c07d4375c521f3c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-14 18:57:16.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e78dbe2454bb4c888738924544bb8154 : DISK_FULL in 1h 26m 23s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e7546c0d04fa407c9c17600b9331eebe : POST_FAILURE in 1h 20m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2975c7c7aaef9e9b202b590ee7acceae7da9f320","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-14 19:39:23.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/73/791273/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"e6862ec8257a6866305bfe6b276467ab710b11f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-14 10:10:35.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0fdb228cfcc3cb7740ff64426a9553d7c0a0f97f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-02-14 11:32:13.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0fcba66a809a840f9b839d13aeb597cb0c4ef117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 11:50:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/697ca857d43c4119bdc359889522c51a : FAILURE in 1h 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382b1dee2674459e8e9a5f8e6eebafa5 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf53d14caf64f20bd768c40d7013105 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdf77ee239ad42268e24fde81fa7bf7a : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304ef82910254fa281152b0cbe8f9005 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a04af61dd5ed413d9d1eb6a6ce864fb5 : SUCCESS in 5m 11s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/de066ec0827542219509d953209ff581 : SUCCESS in 4m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/86fa7ceddb8148c2bf862128554bc1d5 : SUCCESS in 1h 26m 13s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5b65098c83ae4c6ca8b69e21d3ad2050 : FAILURE in 1h 37m 31s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b0b1ab939c364ff2b951d047e65d0473 : SUCCESS in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0739c33a9a59996ca5ab93e53d07e5c4d4ce8196","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-14 12:05:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/791273/4/iscsi/ : SUCCESS in 51m 44s\n- os-brick-smb http://cloudbase-ci.com/os-brick/791273/4/smb/ : SUCCESS in 49m 32s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/791273/4/ceph/ : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":4},{"id":"c93ecf35e411c5ee8d2a0f9dae9cba83c96bcba2","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-14 13:03:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/73/791273/4/check/DellEMC_XtremIO_FC_os-brick/78f4b24/DellEMC_XtremIO_FC_os-brick/89 : SUCCESS in 53m 38s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/73/791273/4/check/DellEMC_XtremIO_iSCSI_os-brick/23d0679/DellEMC_XtremIO_iSCSI_os-brick/86 : FAILURE in 15m 31s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":4},{"id":"3e2380496d94ea18ccdea5f8907564cf839ea80a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-14 13:11:02.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/73/791273/4/check/devstack-plugin-ceph-dsvm-focal-py3/e68da23/ : SUCCESS in 2h 34m 04s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/73/791273/4/check/tempest-dsvm-full-focal-py3/b501595/ : SUCCESS in 2h 17m 32s","accounts_in_message":[],"_revision_number":4},{"id":"148b1b8a3af73a99d9a6b161e764625c328582d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:50:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"80fcc943e3caac73323895087b79c80264c495e2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:55:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"18cafaff706a23991089d84586c2b2146f81ab53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:55:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"14b31c945300e516574ec87b144b146400f6912c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:55:47.000000000","message":"Topic privsep removed","accounts_in_message":[],"_revision_number":6},{"id":"0d46bdc52311566f24b27c7a040f44f84198ed5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:55:52.000000000","message":"Topic set to deprecations","accounts_in_message":[],"_revision_number":6},{"id":"ed14001a680c819a4dd66225f3e4ef574256f319","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 16:08:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/73/791273/6/check/DellEMC_Unity_FC_os-brick/17e05e4/DellEMC_Unity_FC_os-brick/60 : FAILURE in 1m 44s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/73/791273/6/check/DellEMC_Unity_iSCSI_os-brick/fb28975/DellEMC_Unity_iSCSI_os-brick/73 : FAILURE in 3m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"9be0352f5398e558bb521aa37c4553f033249284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 17:55:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f46fadcffbe642818f4ea3db216d471e : FAILURE in 1h 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef369f4adc6e4ef39aa3898727313029 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78f1f37b97724162913f802c68573976 : FAILURE in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d9b51b9c7bb480e8904dc9c0371f956 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2269fcbcc004727ac57b93080606433 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98579c1897314608935cf9d0c086b8f6 : SUCCESS in 5m 31s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/54b34455463e433fb1d74b028d7dfca4 : FAILURE in 4m 34s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/a7ff9a80681b44c1a86020070c5b6ea6 : SUCCESS in 1h 17m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/39868cf8db11448688220a818284e4b9 : FAILURE in 1h 52m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/fd971a0e80e3444ea3a274ad2e97480b : SUCCESS in 1h 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bfd32dbd63944d261f4de21cbbed07d4b15c46e9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-11 20:54:28.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/791273/6/iscsi/ : SUCCESS in 1h 09m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/791273/6/smb/ : SUCCESS in 1h 10m 49s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/791273/6/ceph/ : FAILURE in 1h 12m 30s","accounts_in_message":[],"_revision_number":6},{"id":"622a466eba3cf4f5db55a662c5b8f2093ce088c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-11 23:51:41.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/bb5d6496765d494d8063aa9f48ef45aa : TIMED_OUT in 3h 07m 53s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"8b2ec8d04552ff6616eaeea0cf0dc4de5e89faab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-22 15:46:24.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"059309952ea2854d208efbf9ef28ceb0fb931685","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-22 16:20:47.000000000","message":"Patch Set 7:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/os-brick-CI/tree/gh-pages/f6887cbcc0cd445ba7b1a805ebad490a/","accounts_in_message":[],"_revision_number":7},{"id":"0566c27d5e2f1ed915383e7eb0b2f80ff7fdb374","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-22 16:43:26.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1f26d7a34e96f542d8fcd8d968976fdc21afbc38 .","accounts_in_message":[],"_revision_number":7},{"id":"53792393280f7db681e48e84b357b3422ecbd0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 17:50:16.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37f98c0f28cc4c8082585da232660b2d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb501a8b087d42c1bc901d8ab772794b : FAILURE in 1h 29m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cbf3c2dbfb64b1d8bd77109c868cc8d : FAILURE in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f60db32a6544ce8b6ed9bc1d8568d54 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bea9f29fe4174d9f991eb4139194adf6 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a3b6c27e9ff44fda69b931d64418071 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a87df6166b432786e2f2b3c6de9cd8 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0b476430b534b5ebde1e08dd6f0a5fc : SUCCESS in 7m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c1efad8aae394a78940cd52b941ee822 : SUCCESS in 4m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/882b66e52fed4eb1b0c1a7d8585dd425 : FAILURE in 1h 37m 55s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59b4154df3984bbeba6976462bd446d2 : FAILURE in 1h 55m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a836a7aa37c24349b2ed469293120023 : SUCCESS in 1h 37m 29s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c2878e62887b4d08b1585f67c106405b : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1253f1d496b2372ca67f771e2ad8dd7a70213ec9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-22 17:53:22.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest build/d859d02b03014dbc9e5f914e911ac177 : FAILURE in 53m 38s","accounts_in_message":[],"_revision_number":7},{"id":"2ace667371e762ec5f18cf4a0fb81663626cf263","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-22 18:55:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f6887cbcc0cd445ba7b1a805ebad490a\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/99651c0c19244201a28f5f86fab37437 : SUCCESS in 38m 55s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eb5857ed81464e3884415b38f7508e13 : TIMED_OUT in 3h 01m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6a5957c8d7af48edbf6d39bfa0ce0970 : SUCCESS in 46m 29s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f3cf00e032b24228bab2f3f89cb117a9 : SUCCESS in 31m 40s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f56339f282848c39022b7357c3f5dd7 : FAILURE in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c290514e7bfeb4e27224ec853090395228fbff09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-23 00:57:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5639/.","accounts_in_message":[],"_revision_number":7},{"id":"9895ca48cf1930e74863f2c87c8226ad33b1de00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-23 01:39:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5462/.","accounts_in_message":[],"_revision_number":7},{"id":"a2aa055abb7a860249966e5918cda20b395dfc42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-23 14:09:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"18a1001a58fbf57cdebb63e75c27ede07c84bbf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-23 14:17:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c876c7340b377ababe3456e6c0f13efb9cbd6979","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-23 14:20:36.000000000","message":"Topic deprecations removed","accounts_in_message":[],"_revision_number":9},{"id":"04bf13b0d830955598256ad5702c2d5c791144ed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-23 14:20:38.000000000","message":"Topic set to privsep","accounts_in_message":[],"_revision_number":9},{"id":"ff9218a69922d612e44417b82ac3252a36fe9328","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-23 14:33:35.000000000","message":"Patch Set 9:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/os-brick-CI/tree/gh-pages/c90ad62baaf543ccbffe190f1c2b9b38/","accounts_in_message":[],"_revision_number":9},{"id":"8facc76018c90971c402cc3b4b365aaebb952619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:40:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dc46779acc9436aace92cd53bafb4e9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/610e4bf311ab4346a4a9b5b22d69c5a1 : FAILURE in 1h 28m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95b6490ce6b48a6954778e651cd44e0 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e57cdcfc8232436b9702612716e7fc97 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aba70cc228384d88ae686da3eee1355f : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eaadf4b7e544beab6dfb13df19503ec : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c98518e9d20444bafc722bd356e20a8 : SUCCESS in 6m 55s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f33c9871ec3a4560b2d8f409494e8506 : SUCCESS in 4m 12s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/61ebbd19b2ea424695c5fadf4bd34e7a : FAILURE in 1h 49m 36s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ebf0a99560b0499a8118a06af2ad67d5 : FAILURE in 1h 40m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5356abf3c7bb46478e3ba8efa0a3970d : SUCCESS in 2h 14m 47s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/16bed06751a8490fa76793cfae0f81a8 : SUCCESS in 2m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ebc9a8ea0a84af0f94dd5ac94119bf807032a4a7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-09-23 17:46:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest build/8cad368561d041f5894ae46b4f9591f9 : FAILURE in 55m 20s","accounts_in_message":[],"_revision_number":9},{"id":"151d3bfab50bc9b7efaba0c1e0ae00575ea17272","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-09-24 01:20:35.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c90ad62baaf543ccbffe190f1c2b9b38\n\n- pure-devstack-cinder-tempest-iscsi-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7e5b5028ce7b4431be1359d08ceb1421 : SUCCESS in 37m 23s\n- pure-devstack-cinder-tempest-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d509cca115c742d988eae038118c68ea : NODE_FAILURE Node request 200-0001389745 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/91aecef6090040c2ae32b2aafd0881ea : SUCCESS in 42m 39s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/844ea1e5bcd44161b80084da047c56df : SUCCESS in 27m 48s\n- pure-devstack-cinder-tempest-nvme-fc-aio-resolute https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/23f7d0c609e9403b82c6dea636a07b12 : FAILURE in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2d9584c80b64e3dda0267335c2afaef838592c57","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 04:05:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5664/.","accounts_in_message":[],"_revision_number":8},{"id":"9ba967990feea11225806b6c3f7c941b512cd44d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 04:30:40.000000000","message":"Patch Set 9:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5665/.","accounts_in_message":[],"_revision_number":9},{"id":"ef0aa483d380fcdf1dbc4ea790c9b89e659d4b56","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 04:47:59.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5487/.","accounts_in_message":[],"_revision_number":8},{"id":"ea01fa7fafc6491e7d67061fc9f99ab413e5bcdf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-09-24 06:21:20.000000000","message":"Patch Set 9:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/5488/.","accounts_in_message":[],"_revision_number":9},{"id":"6f994c7f1b7fcc9186419187e0ab5fa4dbb0edc5","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-09-24 13:53:24.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b19d1e18682c11bdb07ca91ce5c3e90d8853399c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 07:37:54.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1f26d7a34e96f542d8fcd8d968976fdc21afbc38 .","accounts_in_message":[],"_revision_number":8},{"id":"a6088f8bb3d956f681ee5359275fa4a4caf7cde8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 07:37:58.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1f26d7a34e96f542d8fcd8d968976fdc21afbc38 .","accounts_in_message":[],"_revision_number":9},{"id":"3c4075ebc48e1dcc5a601058f59e6841bc291571","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-25 09:32:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"771c6bfefbdc8abcd3f5a7bd5a7eb435da18a2b0","revisions":{"468b53876b187e0bef5ee814a7b9b7d234d6a641":{"kind":"REWORK","_number":1,"created":"2021-05-13 15:47:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/1"}}},"commit":{"parents":[{"commit":"5e1719dfd6fd6cc1e3f9ff8c4c4e7bb2ce2b04d2","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5e1719dfd6fd6cc1e3f9ff8c4c4e7bb2ce2b04d2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 15:46:59.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/468b53876b187e0bef5ee814a7b9b7d234d6a641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/468b53876b187e0bef5ee814a7b9b7d234d6a641"}]},"branch":"refs/heads/master"},"6aafc7c12b4738f5716e78517a09d2aa5fd6609d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-14 11:12:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/2"}}},"commit":{"parents":[{"commit":"bb68d1a349ae51a0567b3dfe62ec8064e9ee8dbd","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bb68d1a349ae51a0567b3dfe62ec8064e9ee8dbd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-14 09:33:00.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6aafc7c12b4738f5716e78517a09d2aa5fd6609d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6aafc7c12b4738f5716e78517a09d2aa5fd6609d"}]},"branch":"refs/heads/master"},"df01bee1ca0b171aafd80ec947bb4b534bd8bf85":{"kind":"REWORK","_number":3,"created":"2021-05-14 14:51:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/3"}}},"commit":{"parents":[{"commit":"b6d7a4257f5dcd04a0601810770fc28a688372bd","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b6d7a4257f5dcd04a0601810770fc28a688372bd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-14 14:47:09.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/df01bee1ca0b171aafd80ec947bb4b534bd8bf85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/df01bee1ca0b171aafd80ec947bb4b534bd8bf85"}]},"branch":"refs/heads/master"},"09df39f401b7a4c3c06c16bffb2b994f755f5e56":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-14 10:10:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/4"}}},"commit":{"parents":[{"commit":"e3575725b32a2adbdb42c2a0befdb79575e01245","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e3575725b32a2adbdb42c2a0befdb79575e01245"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-14 10:06:31.000000000","tz":0},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/09df39f401b7a4c3c06c16bffb2b994f755f5e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/09df39f401b7a4c3c06c16bffb2b994f755f5e56"}]},"branch":"refs/heads/master"},"735c0a3ff9f88c2919b239d2fcdd1afab2639e7c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-04-11 15:50:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/5"}}},"commit":{"parents":[{"commit":"237f51584a632b774b419bc9b01d53deee2906f5","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/237f51584a632b774b419bc9b01d53deee2906f5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-04-11 15:50:13.000000000","tz":0},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/735c0a3ff9f88c2919b239d2fcdd1afab2639e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/735c0a3ff9f88c2919b239d2fcdd1afab2639e7c"}]},"branch":"refs/heads/master","description":"Rebase"},"1d60b72af17d680fd664e619de72159b2f896600":{"kind":"REWORK","_number":6,"created":"2022-04-11 15:55:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/6"}}},"commit":{"parents":[{"commit":"237f51584a632b774b419bc9b01d53deee2906f5","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/237f51584a632b774b419bc9b01d53deee2906f5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 15:54:53.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d60b72af17d680fd664e619de72159b2f896600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d60b72af17d680fd664e619de72159b2f896600"}]},"branch":"refs/heads/master"},"3a1d6a8dba1e49c600687e956d320cb6e4cd220a":{"kind":"REWORK","_number":7,"created":"2026-09-22 15:46:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/7"}}},"commit":{"parents":[{"commit":"18e7d2c8a9bbb2cd1290a35de3d460babdfab33c","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/18e7d2c8a9bbb2cd1290a35de3d460babdfab33c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-09-22 15:46:06.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3a1d6a8dba1e49c600687e956d320cb6e4cd220a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3a1d6a8dba1e49c600687e956d320cb6e4cd220a"}]},"branch":"refs/heads/master"},"3b61677944bc86cc341cf373fdc5be3e5501af02":{"kind":"REWORK","_number":8,"created":"2026-09-23 14:09:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/8"}}},"commit":{"parents":[{"commit":"18e7d2c8a9bbb2cd1290a35de3d460babdfab33c","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/18e7d2c8a9bbb2cd1290a35de3d460babdfab33c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-23 14:09:04.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3b61677944bc86cc341cf373fdc5be3e5501af02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3b61677944bc86cc341cf373fdc5be3e5501af02"}]},"branch":"refs/heads/master"},"771c6bfefbdc8abcd3f5a7bd5a7eb435da18a2b0":{"kind":"REWORK","_number":9,"created":"2026-09-23 14:17:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/791273/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/73/791273/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/73/791273/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/73/791273/9"}}},"commit":{"parents":[{"commit":"18e7d2c8a9bbb2cd1290a35de3d460babdfab33c","subject":"encryptors: Remove \u0027CryptsetupEncryptor\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/18e7d2c8a9bbb2cd1290a35de3d460babdfab33c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-05-13 13:14:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-23 14:17:42.000000000","tz":60},"subject":"encryptors: Drop support for out-of-tree encryptors","message":"encryptors: Drop support for out-of-tree encryptors\n\nThis should have been done in Queens. Do it now.\n\nChange-Id: I1f26d7a34e96f542d8fcd8d968976fdc21afbc38\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/771c6bfefbdc8abcd3f5a7bd5a7eb435da18a2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/771c6bfefbdc8abcd3f5a7bd5a7eb435da18a2b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
