)]}'
{"id":"openstack%2Fos-brick~792237","triplet_id":"openstack%2Fos-brick~master~I252dd958767dcdd4f9a2767b362aaf675edb79c4","project":"openstack/os-brick","branch":"master","topic":"nvme-hostnqn","hashtags":[],"change_id":"I252dd958767dcdd4f9a2767b362aaf675edb79c4","subject":"NVMe-oF: Return right nqn when missing hostnqn","status":"MERGED","created":"2021-05-19 18:31:57.000000000","updated":"2021-06-30 19:21:25.000000000","submitted":"2021-06-30 19:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"792237-nvme-hostnqn","meta_rev_id":"2a56e68dbfee845df1beae7defef9779669aa7ca","_number":792237,"virtual_id_number":792237,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 19:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-06-30 17:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 12:43:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-06-30 17:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 19:23:28.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-05-19 19:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 20:04:34.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-19 20:32:31.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 20:47:33.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 20:47:38.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 22:43:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 22:53:01.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-05-19 22:53:06.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-06-03 12:47:53.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-06-03 13:10:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-07 16:22:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-07 20:55:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-07 22:33:24.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 08:50:59.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2021-06-08 12:43:01.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-06-30 17:44:12.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"}],"messages":[{"id":"f1bc80f2f10fd10287c0cce894e6e7bc0f96c366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-05-19 18:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"041e589de5c2fcd7ccf757aa5f7b687828e13644","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-19 19:23:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/792237/1/iscsi/ : SUCCESS in 49m 16s\n- os-brick-smb http://cloudbase-ci.com/os-brick/792237/1/smb/ : SUCCESS in 50m 53s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/792237/1/ceph/ : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":1},{"id":"26cf590b0e31b50ba1abdef9cba6968d3ed5a0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 19:52:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74f391f0aebb40e08173014f8adfc304 : SUCCESS in 1h 19m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6699b41742a843d3a285c6b9f966cebd : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994a75da31b7458e8783442a2b9ee0d0 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bef244d12c64cf4acd0dceb5692ebaf : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e59626204b8e47b491aedfb084d75b48 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d4b07f67e6467fbf89c2d48fabeec4 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c6b0eee8ba46c181c7590223acb4d6 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61e1c5f1af974bb3af2df83cfe177f71 : SUCCESS in 6m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/2b306f509c254aaf87bea14876ac0d64 : SUCCESS in 5m 02s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/ec82f615a0034ae2ac5228af4d4c14c1 : SUCCESS in 1h 07m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/844772b9391847f390cf66df46d4833d : SUCCESS in 1h 13m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/010de16272324cd8b87998d90c91c63b : SUCCESS in 1h 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4cadc176e0e438893e5f2278c4433aa2fae0a58","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-19 20:04:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/792237/1/upstream-solidfire-check/cinder-solidfire-iSCSI/6030ad9/ : FAILURE in 11m 29s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"9399e4ede1322489455ebc4371173b6d3c9016fa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-19 20:32:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/37/792237/1/upstream-check/cinder-cDOT-iSCSI/ac7d7bf/ : FAILURE in 38m 51s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/37/792237/1/upstream-check/cinder-cDOT-NFS/56fc74e/ : FAILURE in 39m 29s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/37/792237/1/upstream-check/cinder-cDOT-FCP/a37e8c1/ : FAILURE in 13m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"cfffc94792a251f045ccd8bc47032c9eade7528f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-19 20:47:33.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/37/792237/1/check/devstack-plugin-ceph-dsvm-focal-py3/33bcb1f/ : FAILURE in 2h 10m 59s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/37/792237/1/check/tempest-dsvm-full-focal-py3/5ec354e/ : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"75b47c617c3c08dcde0bac2e8410650ecda26ad1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-19 20:47:38.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"37f3a45c9558225cd205b87196dd392ff1a318e6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-05-19 22:43:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/37/792237/1/check/DellEMC_PowerFlex_os-brick/db5e5f3/DellEMC_PowerFlex_os-brick/18 : FAILURE in 2h 50m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"4a2b91cb39a832422f8f2caf321b40902f4c5486","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-19 22:43:34.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"814bb36246e84518cd9f2e5f2f4ecce4d5e12a4f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-05-19 22:53:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/37/792237/1/check/DellEMC_PowerStore_FC-ob/f836d4d/DellEMC_PowerStore_FC-ob/15 : ABORTED in 3h 00m 03s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/37/792237/1/check/DellEMC_PowerStore_iSCSI-ob/faf285c/DellEMC_PowerStore_iSCSI-ob/15 : ABORTED in 3h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0d157abad2f9d39b0653474799ed845704dca3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-19 22:53:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a879b274d61fa59f8df1b46de4e652741aced037","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-05-19 22:53:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/37/792237/1/check/DellEMC_SC_FC-os-brick/422b058/DellEMC_SC_FC-os-brick/20 : ABORTED in 3h 00m 03s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/37/792237/1/check/DellEMC_SC_iSCSI-os-brick/d20184e/DellEMC_SC_iSCSI-os-brick/20 : FAILURE in 1h 03m 58s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ae73e2fbd0b82697f3bc101a47a2faa2c0c3d76","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-19 22:53:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fb8e852d21070deeea4bfd53105a3db71be014bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-25 11:09:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"33df09bf5f4db817d45cdc860ea943ef81935cd4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-03 13:10:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nCode and tests look good, just need to fix the merge conflict.","accounts_in_message":[],"_revision_number":1},{"id":"94304efab305de33cc5ef08b224ee6baa4c9453f","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-07 14:05:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f2b4373cdbe747cb506c738501810d9f8e1f457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-07 14:05:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6ab78d43a22f543dba027db7fd26eb7ba638651","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-06-07 15:00:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/792237/2/iscsi/ : SUCCESS in 53m 22s\n- os-brick-smb http://cloudbase-ci.com/os-brick/792237/2/smb/ : SUCCESS in 50m 29s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/792237/2/ceph/ : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a3052f5e853d69fa2ce76d027aa26631d7117a06","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 15:35:34.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"aa77d404bf6281069790f6187b1603f68641787b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/524d1a6408834fcfad92625c110845cd : SUCCESS in 1h 29m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76150e350cee485f90e84858ed9776cd : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822903eb4bfe4cbe812a9871aaaab4c7 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84719d924783412fb0c9643aceef0e3c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7864ae7c00424f9ac4772a6360df85 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53b9b8f37a8d46088628372fca282981 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fba995215d4419d9b4b27a60cd0313e : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1863fc7c737e4e66b33005b53cc84e28 : SUCCESS in 5m 36s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c2adbad7eda5466eaaa263e3ab9d1ca4 : SUCCESS in 5m 10s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/f8fb3a8ba3fe4a23ae73129e6868a963 : SUCCESS in 1h 18m 43s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/36ea199e5d194101b4059701637ef618 : SUCCESS in 1h 37m 54s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8fe2e19fffea445fb6ab16869836273d : SUCCESS in 1h 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d462a19259132526c0bd2a8791ae8f1f440a8d03","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-06-07 15:48:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/37/792237/2/upstream-solidfire-check/cinder-solidfire-iSCSI/9e608de/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"d34a33c33f503f6dc9c0fd1f2d9cb692083dacaa","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 16:22:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_VNX_FC_os-brick/0d1e7b4/DellEMC_VNX_FC_os-brick/28 : FAILURE in 33m 37s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_VNX_iSCSI_os-brick/54e0513/DellEMC_VNX_iSCSI_os-brick/28 : FAILURE in 32m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"cb0033c7322acaa3f481769ddda185431eedd37f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-07 16:22:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/37/792237/2/check/DellEMC_PowerStore_FC-ob/0d2f7f6/DellEMC_PowerStore_FC-ob/23 : FAILURE in 31m 29s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/37/792237/2/check/DellEMC_PowerStore_iSCSI-ob/804f2f5/DellEMC_PowerStore_iSCSI-ob/23 : FAILURE in 33m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"2d4b07c3736cccc47584acbe3830732c35d297e0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 16:26:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_PowerFlex_os-brick/b8d6196/DellEMC_PowerFlex_os-brick/23 : FAILURE in 37m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b84419d8539291bea596f122504494ff3e0db075","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 16:28:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"c0229d56194d27fbb59b202aa9b134e4fc3f0344","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-07 17:15:47.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/37/792237/2/check/devstack-plugin-ceph-dsvm-focal-py3/d8ec800/ : FAILURE in 48m 12s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/37/792237/2/check/tempest-dsvm-full-focal-py3/eb98dc7/ : SUCCESS in 1h 42m 04s","accounts_in_message":[],"_revision_number":2},{"id":"53449ec8cbf7fabf9bd53311165ab57ff1fc97f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 17:16:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/2/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"97cabf8fda84a599ab60fa28341bf5bdc78abef8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 19:13:00.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b518251e336b2b8d42ac246837a3d3e15b55dd8e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-06-07 20:44:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_SC_FC-os-brick/4f5c3bc/DellEMC_SC_FC-os-brick/27 : FAILURE in 31m 26s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_SC_iSCSI-os-brick/2063d0f/DellEMC_SC_iSCSI-os-brick/26 : FAILURE in 23m 43s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"e0cf95dfa9d9b3a783f5f2eeca5af3a9f9cf022a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-07 20:55:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_Unity_FC_os-brick/c04b550/DellEMC_Unity_FC_os-brick/44 : FAILURE in 31m 54s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_Unity_iSCSI_os-brick/e0f44e3/DellEMC_Unity_iSCSI_os-brick/43 : FAILURE in 23m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"441ed862ad082dfeab946b94bc6bccdc807d3552","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 20:56:10.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"52860c0f79176e821cac3b5b08d15a44ca66f7ca","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-06-07 22:33:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_XtremIO_FC_os-brick/b4f4095/DellEMC_XtremIO_FC_os-brick/15 : FAILURE in 23m 53s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/37/792237/2/check/DellEMC_XtremIO_iSCSI_os-brick/5064928/DellEMC_XtremIO_iSCSI_os-brick/14 : FAILURE in 21m 14s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd07bb24020b2b186b43a9b3c58e2745a0571944","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-07 22:34:11.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/2/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"33d8f961bb998d5a7f774e50c17dd6dbc8a7129e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-08 02:47:03.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/37/792237/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/37/792237/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"d692884d6aa33dac7027ee07a9b641fa24713ecb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-08 08:50:59.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for fixing this Gorka.\nI\u0027m not very familiar with nvme but i tried a case mentioned inline.","accounts_in_message":[],"_revision_number":2},{"id":"ad7eae9a4f73c786b158c2efb87ccaaf4130df25","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-08 10:34:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63bfdf2752b51aa89aebfff94fd0d851aa6018de","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-08 12:43:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nLGTM, thanks Gorka.","accounts_in_message":[],"_revision_number":2},{"id":"7d56abe0d14396cb61735e1b9d632e01096c6d72","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2021-06-09 19:28:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb635a98ed8f3128885f9f7532674138d6d5697b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-06-30 17:44:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ea3ccfb9134d126b808912e5f0d32d65715f7172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 17:44:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ed81d9390606dfe209e0bfb9242a24fbfeb1048c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 19:20:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a958d1c58b2e4c118cf225577c0ae338 : SUCCESS in 1h 20m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8831c8c2dd0d4c7c98797bb65ca521bf : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772426e7726143a28ac0f95264cb772a : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c240916404e34bdb983274a90a5ab0d7 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1276d99036b44485abd47b2a6e532deb : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3a3113d7cd4c8ebe7c1fbaf6b7e7e3 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3aaefada8e145b4a7cf47e1015e25e8 : SUCCESS in 5m 23s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3588d053ae2c4e8988fea5236fd54115 : SUCCESS in 1h 31m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9e43f4835b443521f6170a590a09db62e172577a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 19:20:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a56e68dbfee845df1beae7defef9779669aa7ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 19:21:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ff635fd8b34caeb20da8f26cbbb5e2 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c9bc8ecf46c435da270a918ce17f291 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37d57c4306acc59dfe8f025040b7701dea2d304c","revisions":{"174e18da854dfabfed9c4cd9641c2453e52f33a9":{"kind":"REWORK","_number":1,"created":"2021-05-19 18:31:57.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/37/792237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/37/792237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/37/792237/1"}}},"commit":{"parents":[{"commit":"1b2e2295421615847d86508dcd487ec51fa45f25","subject":"multipath/iscsi: remove devices from multipath monitoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1b2e2295421615847d86508dcd487ec51fa45f25"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-05-19 14:57:30.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-05-19 18:31:25.000000000","tz":120},"subject":"NVMe-oF: Return right nqn when missing hostnqn","message":"NVMe-oF: Return right nqn when missing hostnqn\n\nThis patch fixes a problem where we don\u0027t return the right nqn value\nwhen the /etc/nvme/hostnqn file doesn\u0027t exist.\n\nCurrently get_connector_properties returns a new nqn value every time it\ngets called if the hostnqn file is not present in the system.\n\nThe reason for this is that we are calling \"nvme gen-hostnqn\" and piping\nit to tee, but for some reason tee is failing to write the file, so on\neach call we\u0027ll generate a new nqn.\n\nEven if we were successfully creating the hostnqn file, that could\ncreate problems if we have already attached some volumes, since those\nwould have used the existing nqn.\n\nThis patch replaces the series of command executions (mkdir, nvme,\nchmod) with a privsep method that then leverages the equivalent python\nmethods when possible and also tries to use the show-hostnqn nvme\nsubcommand first to get the existing nqn before defaulting to generating\na new one with gen-hostnqn.  This way we don\u0027t replace the current nqn\nvalue that may be already being used by other nvme connections.\n\nSubcommand show-hostnqn returns the host NQN configured for the system.\nFirst looks for /etc/nvme/hostnqn, if it\u0027s not present (our case when we\ncall the new create_hostnqn method) tries to construct it from dmi\n(/sys/firmware/dmi/entries) or from systemd\u0027s application-specific\nmachine IDs for the system [1].\n\nIt\u0027s important to differentiate between the ENOENT returned as OSError\nand the same value returned by nvme [2] when show-hostnqn cannot return\nanything.\n\n[1]: https://github.com/linux-nvme/nvme-cli/blob/ed9538622ac6fc9e0094dd6804fc3b2ab46477b9/fabrics.c#L858-L875\n[2]: https://github.com/linux-nvme/nvme-cli/blob/5b8b065b1d036d8e9050adc49ffeb3b7adad1dbf/nvme.c#L5642-L5643\n\nCloses-Bug: #1928944\nChange-Id: I252dd958767dcdd4f9a2767b362aaf675edb79c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/174e18da854dfabfed9c4cd9641c2453e52f33a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/174e18da854dfabfed9c4cd9641c2453e52f33a9"}]},"branch":"refs/heads/master"},"37d57c4306acc59dfe8f025040b7701dea2d304c":{"kind":"REWORK","_number":2,"created":"2021-06-07 14:05:40.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/37/792237/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/37/792237/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/37/792237/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/37/792237/2"}}},"commit":{"parents":[{"commit":"7d3af72cc2c5b068dc73878c2a6bda431e1c8af5","subject":"Merge \"NVMe-oF: Flush on disconnect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7d3af72cc2c5b068dc73878c2a6bda431e1c8af5"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-05-19 14:57:30.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2021-06-07 14:04:28.000000000","tz":120},"subject":"NVMe-oF: Return right nqn when missing hostnqn","message":"NVMe-oF: Return right nqn when missing hostnqn\n\nThis patch fixes a problem where we don\u0027t return the right nqn value\nwhen the /etc/nvme/hostnqn file doesn\u0027t exist.\n\nCurrently get_connector_properties returns a new nqn value every time it\ngets called if the hostnqn file is not present in the system.\n\nThe reason for this is that we are calling \"nvme gen-hostnqn\" and piping\nit to tee, but for some reason tee is failing to write the file, so on\neach call we\u0027ll generate a new nqn.\n\nEven if we were successfully creating the hostnqn file, that could\ncreate problems if we have already attached some volumes, since those\nwould have used the existing nqn.\n\nThis patch replaces the series of command executions (mkdir, nvme,\nchmod) with a privsep method that then leverages the equivalent python\nmethods when possible and also tries to use the show-hostnqn nvme\nsubcommand first to get the existing nqn before defaulting to generating\na new one with gen-hostnqn.  This way we don\u0027t replace the current nqn\nvalue that may be already being used by other nvme connections.\n\nSubcommand show-hostnqn returns the host NQN configured for the system.\nFirst looks for /etc/nvme/hostnqn, if it\u0027s not present (our case when we\ncall the new create_hostnqn method) tries to construct it from dmi\n(/sys/firmware/dmi/entries) or from systemd\u0027s application-specific\nmachine IDs for the system [1].\n\nIt\u0027s important to differentiate between the ENOENT returned as OSError\nand the same value returned by nvme [2] when show-hostnqn cannot return\nanything.\n\n[1]: https://github.com/linux-nvme/nvme-cli/blob/ed9538622ac6fc9e0094dd6804fc3b2ab46477b9/fabrics.c#L858-L875\n[2]: https://github.com/linux-nvme/nvme-cli/blob/5b8b065b1d036d8e9050adc49ffeb3b7adad1dbf/nvme.c#L5642-L5643\n\nCloses-Bug: #1928944\nChange-Id: I252dd958767dcdd4f9a2767b362aaf675edb79c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/37d57c4306acc59dfe8f025040b7701dea2d304c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/37d57c4306acc59dfe8f025040b7701dea2d304c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
