)]}'
{"id":"openstack%2Fos-brick~792359","triplet_id":"openstack%2Fos-brick~stable%2Fussuri~I0278c502806b99f8ec65cb146e3852e43031e9b8","project":"openstack/os-brick","branch":"stable/ussuri","topic":"bug/1915678","hashtags":[],"change_id":"I0278c502806b99f8ec65cb146e3852e43031e9b8","subject":"Avoid unhandled exceptions during connecting to iSCSI portals","status":"MERGED","created":"2021-05-20 11:31:37.000000000","updated":"2021-06-17 02:08:02.000000000","submitted":"2021-06-17 02:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792359-bug/1915678","meta_rev_id":"a7f41c3b220bdb24ff603bd345ba52d155ade990","_number":792359,"virtual_id_number":792359,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 02:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-06-17 00:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-05-25 10:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":1,"date":"2021-05-20 13:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"value":2,"date":"2021-06-15 14:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-01 14:09:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-06-17 00:22:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 12:42:18.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-05-20 13:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 13:52:45.000000000","updated_by":{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},"reviewer":{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},"state":"REVIEWER"},{"updated":"2021-05-20 14:02:15.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-05-20 16:01:26.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 16:28:42.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 16:28:48.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 10:57:50.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2021-06-01 14:09:58.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2021-06-15 14:40:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-06-17 00:22:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"cb786f138bd1178c27fbb96c9ef23709bff019d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-20 11:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"247595aced10efb00086eb0c5cb6707c7281bcf1","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-05-20 12:42:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/792359/1/iscsi/ : SUCCESS in 1h 10m 16s\n- os-brick-smb http://cloudbase-ci.com/os-brick/792359/1/smb/ : SUCCESS in 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"29bd7553ce34c0e517fcce3037902ac3875a4086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:03:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e9ca559003146179b25f1940123d14e : SUCCESS in 1h 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5a11f1af934042a035e8976098c01e : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/452eaedb089842aa9ab924d0db53cb77 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c671d28ef9d47daae6922e175452593 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96c5e0c44ae48bdbd94d3a0c6a1cb76 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04427c511ead4d3db156de8c8926ff9a : SUCCESS in 5m 03s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/fc02cada348d48c6b2959cb87e4fde9a : SUCCESS in 1h 15m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f80d70ca60424554baed1cf1240724e1 : SUCCESS in 1h 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"97f95ca515ec3cb51684a4560d72582bf0f247cc","author":{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},"date":"2021-05-20 13:52:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f46fc9ed76f92dcfee84d69083800ec7602ffa45","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-05-20 14:02:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/59/792359/1/upstream-solidfire-check/cinder-solidfire-iSCSI/d436b3c/ : FAILURE in 58m 12s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"3891f9f31d63862db53d706f91b1a8c8d5c7bd5c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-20 16:01:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/59/792359/1/upstream-check/cinder-cDOT-iSCSI/4dc5621/ : FAILURE in 2h 57m 25s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/59/792359/1/upstream-check/cinder-cDOT-NFS/00b9275/ : SUCCESS in 1h 25m 38s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/59/792359/1/upstream-check/cinder-cDOT-FCP/a1fcf8e/ : SUCCESS in 1h 01m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"159ded65d1978bf5093b5c036b962a508c48bf5a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-05-20 16:28:42.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/9fd79bdac94a49beb78efa6417724de7 : TIMED_OUT in 3h 09m 18s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/61f5260751a54a3a892a29c456d73206 : POST_FAILURE in 1h 56m 25s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/096699cb2af8460f9f5792c63e24ea0e : POST_FAILURE in 2h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cd1f3afa8dae7d9b81a16febf781792c442ce923","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-20 16:28:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/59/792359/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"443c2f30cf2a921d1287a7e786766a266f493df5","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-05-25 10:57:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c4966a709faf16d794d09c93c02e50901f9f32d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-05-25 11:18:12.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/59/792359/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"73c306a9fde07ce52eabb8be54402c797ec826b7","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-01 14:09:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nMerged in victoria, patch looks good to backport for ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"22daff60ef30e8f93e93a5d8d861203b573503e8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-06-15 14:40:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e82b0818ac06c57ad9c9c5d21394312d8e9cdd8b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-17 00:22:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nAgree that this is a good backport.","accounts_in_message":[],"_revision_number":1},{"id":"8334955ae4321b73589de23e15fc3d669c4336d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 00:22:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dce5902cc8d7c75d7687ee12f79cf33bf6b996d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:06:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b18b1665795d4be4934954c4b2eb85de : SUCCESS in 1h 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb9a387b26b4e7585a0a9839148818b : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a88fa4c151f04439a315ed97e611ad3d : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fd5346969d14447a13f14aae77fb766 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7e18690264450eab32856da10acd9c : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca6fb37a3b8442b9a3c33003917aab5e : SUCCESS in 4m 45s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d6716ef8b82a40c4b8eca1a1aec063a5 : SUCCESS in 1h 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2ef57256f1b7f3e30cd0f0acc31d480df94cba4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:06:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7f41c3b220bdb24ff603bd345ba52d155ade990","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:08:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdb5d3346db40f0bbbdf9c6a06727a8 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2854d68312fd448f980052874f5f2a03 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c336cb76b318f987d43ff908a5b7c9d40b5c4dd6","revisions":{"c336cb76b318f987d43ff908a5b7c9d40b5c4dd6":{"kind":"REWORK","_number":1,"created":"2021-05-20 11:31:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/792359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/792359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/792359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/792359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/792359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/792359/1"}}},"commit":{"parents":[{"commit":"5c031cb2d9e9b10a33c82586d11139594c96a25f","subject":"multipath/iscsi: remove devices from multipath monitoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5c031cb2d9e9b10a33c82586d11139594c96a25f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-15 01:58:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-20 11:31:27.000000000","tz":540},"subject":"Avoid unhandled exceptions during connecting to iSCSI portals","message":"Avoid unhandled exceptions during connecting to iSCSI portals\n\nCurrently we don\u0027t properly catch some possible exceptions during\nconnectiing to iSCSI portals, like failures in \"iscsiadm -m session\".\nBecause of this _connect_vol threads can abort unexpectedly in some\nfailure patterns, and this abort causes hung in subsequent steps\nwaiting for results from _connct_vol threads.\n\nThis change ensures that any exceptions during connecting to iSCSI\nportals are handled in the _connect_vol method corectly, to avoid\nunexpected abort without updating thread results.\n\nConflicts:\n        os_brick/tests/initiator/connectors/test_iscsi.py\n\nCloses-Bug: #1915678\nChange-Id: I0278c502806b99f8ec65cb146e3852e43031e9b8\n(cherry picked from commit 4478433550ddab970f106636a4ba5fbba82c5e39)\n(cherry picked from commit 57c8f4334c0e58b151f449f3316bb9e1c8ed6da4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c336cb76b318f987d43ff908a5b7c9d40b5c4dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c336cb76b318f987d43ff908a5b7c9d40b5c4dd6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
