)]}'
{"id":"openstack%2Fos-brick~798743","triplet_id":"openstack%2Fos-brick~master~I62fadd401591025ecf9f83fa095249008f75bbaa","project":"openstack/os-brick","branch":"master","topic":"change-798743-2","hashtags":[],"change_id":"I62fadd401591025ecf9f83fa095249008f75bbaa","subject":"LVM: Retry lv calls on segfault due to lvs failure","status":"MERGED","created":"2021-06-29 20:06:07.000000000","updated":"2021-08-09 18:30:12.000000000","submitted":"2021-08-09 18:28:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"798743-change-798743-2","meta_rev_id":"f2e9d34319fecef287e3ae26a82aa65c68b25e30","_number":798743,"virtual_id_number":798743,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-09 18:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-08-07 16:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2021-08-09 17:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2021-08-09 17:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-08-07 16:59:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 20:07:07.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2021-06-29 20:07:07.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2021-06-29 20:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 20:53:10.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-06-29 20:53:15.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-29 21:23:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 20:02:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-01 00:23:45.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 20:13:06.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2021-07-26 17:11:03.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-07-26 17:13:05.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-07-26 17:15:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-26 18:00:36.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-26 18:35:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-08-07 16:59:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-09 17:07:07.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"}],"messages":[{"id":"6b6b2abe75ee1c0d1830a6d338ab46643aa0049c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-29 20:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f88175f874b8edc04624a9a34b7bbde3029736c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 20:50:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9dcd572c6b1466abfdefc3ef3e66b5b : FAILURE in 30m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01d1255904364a048e9d4f47d9194952 : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68a117c07a6485dacc6d88e49810c36 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5decb47962114d0ebf19daa5afea1f48 : FAILURE in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17fa572cabc49a58db84736ce129321 : FAILURE in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17b63b643e9149599a28f41ab25b7af8 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c2dc0b268845ae8a9e389d0aabbc35 : FAILURE in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/439b17fdeb0141ac831696700b82f9a2 : SUCCESS in 7m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a4eaa2bfc35448c6b1ac2ca97d7a734c : FAILURE in 6m 02s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d9a0fcc258c94641a34feb10e8032bda : FAILURE in 29m 14s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f73e692cba14bc599696de96d8cc7d9 : FAILURE in 31m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a6727a05c47342fb87e62b91606495a4 : FAILURE in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"320412290486de62a8ca0b01b0f70268e9d08007","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-06-29 20:53:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798743/1/iscsi/ : FAILURE in 46m 32s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798743/1/smb/ : FAILURE in 46m 39s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798743/1/ceph/ : FAILURE in 43m 17s","accounts_in_message":[],"_revision_number":1},{"id":"759326a5d1ca0d14bf846c781e73bbefbdb6d443","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-29 20:53:15.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d54f2974bb3d2a6267330939c7792062e5c81c3b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-29 21:23:07.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/43/798743/1/check/devstack-plugin-ceph-dsvm-focal-py3/78b3c1a/ : FAILURE in 1h 01m 41s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/43/798743/1/check/tempest-dsvm-full-focal-py3/57a5e2f/ : FAILURE in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5a5d7d9c1f5a3b4694751dfc3bf6c95b40868c51","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-29 21:23:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"32316840bbacba334ed54b48f97b14e349a80bfc","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-30 14:23:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"988047d12f2dea7cfc924b6be6feead1f9cb732d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-06-30 17:46:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e4d59c500dbb6429e558cfd2f7b9071650a2c5d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-30 17:46:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/1/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"61db99252aae63d7c0df94a71f38843173324e15","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-06-30 19:01:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798743/2/iscsi/ : SUCCESS in 51m 10s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798743/2/smb/ : SUCCESS in 1h 14m 16s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798743/2/ceph/ : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5bc199fa894f95e36562e600c3aa6fa827fa2168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 19:29:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef25f2e021af48cf886a6fd4164dab13 : SUCCESS in 1h 35m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b53a7b0e932d4835be70ff9f4d86dfc1 : FAILURE in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29184a30c3714cf9ae4af4aacc3aaa6f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/170b9bb1805b41af889de945b89deddc : FAILURE in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bdf04ace783480583fb564602f1168d : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ea88cbdd66f43baa8c2aa927966edc2 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59926d870ca43289ae1bdca6ffd7566 : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a4daf2c90d34f0694d8eaa6fa4b46e1 : SUCCESS in 5m 22s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/44935d0defff46e9895d98a24f520b2d : FAILURE in 4m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/13a48ec1b93148a98b3757cddd349755 : SUCCESS in 1h 36m 07s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/18ede3490cc44b28a6e66c5e189a90eb : SUCCESS in 1h 37m 40s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/695fa7976ad04a768cfee0d69c3a117c : SUCCESS in 1h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2dadf2452a5e06f691747dc930bc1c733f7f12af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-30 19:48:34.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/baea90fe11204b2ea1451ef11a520ea7 : POST_FAILURE in 57m 23s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/31ef77231c054ca3b234e7614ed8f5c8 : POST_FAILURE in 1h 20m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a097f896adec034efb82c2574db7d7564637a775","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-30 19:48:39.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"fd80a9f1d2333f49a827667891d50d403d1dde52","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-30 20:02:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/43/798743/2/check/DellEMC_Unity_FC_os-brick/ec53740/DellEMC_Unity_FC_os-brick/46 : FAILURE in 42m 11s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/43/798743/2/check/DellEMC_Unity_iSCSI_os-brick/1badf42/DellEMC_Unity_iSCSI_os-brick/45 : FAILURE in 32m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"9e43af9cb2e5363d2b45dae3425cd6dd75d14ed6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-06-30 20:02:52.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"cb00f41e86207d4fcd7fef19f92073fc506664a1","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-01 00:23:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/43/798743/2/check/DellEMC_XtremIO_FC_os-brick/6af11e2/DellEMC_XtremIO_FC_os-brick/17 : FAILURE in 37m 47s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/43/798743/2/check/DellEMC_XtremIO_iSCSI_os-brick/bf1d197/DellEMC_XtremIO_iSCSI_os-brick/16 : FAILURE in 24m 52s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"bbc116cf25383241c3dfa74ba06985dd2fc41e7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 00:23:49.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"5b2b165fdeead5cb1c556813b41c74806e7aa1ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-01 10:56:18.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/2/Cinder-tgtadm : SUCCESS\n\n* SPDK http://13.74.249.42/refs/changes/43/798743/2/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ed25171f7c7a772fb2a467a387c459e143fd8dcb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2021-07-13 20:13:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f12135478585d3f4f22f903e1324b828dcb063e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-23 16:53:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5a4d3c30973a9b2d148e26cb6fa85ed247e43af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 18:31:09.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68ed14afa099434a8340516ad72fc55f : SUCCESS in 1h 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c32ecc45ce94a3aa3abf4495d24dfab : FAILURE in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07b2f5add47d414c94e4483331aa64e1 : FAILURE in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be1d807882484111bdbb6412438a0b85 : FAILURE in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55df78f39ce04150a08b75e0a2b6089e : FAILURE in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3640a0fcaf04c1790650e9f71c70559 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a0d3f5d18f44eb9aedf024243fc598e : SUCCESS in 5m 52s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/83d12e674a174844a7a45bce970ae793 : FAILURE in 5m 58s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/7b502d7fb72e4f2baddffedf1f6f1b0f : SUCCESS in 1h 17m 20s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b8f11fdd9c849afa9d9b349e0bf6ace : SUCCESS in 1h 27m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4ad3ea4ffe4f45c09b6accd2fc98eb82 : SUCCESS in 1h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"258fb9d5db3872bcdfdcfe4b578c01de4fe53179","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-23 20:30:34.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/3/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ee1206873693528e7f4f7e349290317ad0b713ee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-23 21:46:20.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/43/798743/3/check/devstack-plugin-ceph-dsvm-focal-py3/673d93d/ : FAILURE in 1h 02m 52s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/43/798743/3/check/tempest-dsvm-full-focal-py3/381674a/ : SUCCESS in 2h 28m 19s","accounts_in_message":[],"_revision_number":3},{"id":"2990c1cd1338324221ee6ef8291f61d48edebe83","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-07-23 23:17:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798743/3/iscsi/ : SUCCESS in 51m 03s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798743/3/smb/ : SUCCESS in 55m 31s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798743/3/ceph/ : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":3},{"id":"e3c4720b66f875dd3ae209ef63ba587873eb44d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-24 01:45:11.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/3/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"c7ddc65fde95f5b11ffe7ec7580946ef9b1c4144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-07-26 15:19:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"70a784a8a428a7bc661f6e6053034ad2a4d63225","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-07-26 16:16:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798743/4/iscsi/ : SUCCESS in 55m 47s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798743/4/smb/ : SUCCESS in 52m 36s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798743/4/ceph/ : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e799e5076d2c35f9351f201c74b1ad6b3481e5b4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-26 16:43:55.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/4/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"06a900a7afecafdd97430f39eb083c6c1c8a68fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 16:52:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3bffdce80ed40558e0ef1a649c3be0a : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756edc20387c41a0ba611558f4867e8e : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e889fadcd11e4d64a52f8bd7b4d93386 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba538a2cee4e4e1a87a0dcaf0b8e1f83 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b119bb34466444158410ea4f6144eef9 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8317fd0c63f4d2eab52ed4af32f7d59 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b2d8d72e9449dc9965705e4e4c4874 : SUCCESS in 5m 01s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9905bd9be5a340f99d62641e79f1a189 : SUCCESS in 4m 53s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/3db9c165a86748efa76a051a17039793 : SUCCESS in 1h 15m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/944184d44e24456cbf9f6fb2c2c543d1 : SUCCESS in 1h 28m 07s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1b0a4ba7944343e6958a971524b56ac1 : SUCCESS in 1h 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5852ff9efc55259e98bde853291c8cd4ac7c3361","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-26 17:05:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/43/798743/4/check/DellEMC_Unity_FC_os-brick/e913be5/DellEMC_Unity_FC_os-brick/56 : FAILURE in 27m 39s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/43/798743/4/check/DellEMC_Unity_iSCSI_os-brick/635ed2d/DellEMC_Unity_iSCSI_os-brick/53 : FAILURE in 20m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"faa1d494287ebaf351916296632d06dacab311d6","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-26 17:11:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/43/798743/4/check/DellEMC_SC_FC-os-brick/2b2f8e4/DellEMC_SC_FC-os-brick/28 : FAILURE in 18m 02s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/43/798743/4/check/DellEMC_SC_iSCSI-os-brick/cf0e2e0/DellEMC_SC_iSCSI-os-brick/27 : FAILURE in 17m 53s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"8cf6e7ec912122276ced18b0b2d077384acdf28d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-26 17:12:37.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/4/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"9ccfee6102568e23894d259c03edf86a8dec6dd9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-26 17:13:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/43/798743/4/check/DellEMC_PowerStore_FC-ob/833fd56/DellEMC_PowerStore_FC-ob/31 : FAILURE in 18m 56s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/43/798743/4/check/DellEMC_PowerStore_iSCSI-ob/cf41bdd/DellEMC_PowerStore_iSCSI-ob/29 : FAILURE in 20m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"8f8760fca083b2682e8ecd678635985cb47ceadd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-26 17:14:52.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/4/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"7d792eed7d057082a2ab26f2d63aa03612bfa4b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-26 17:15:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/43/798743/4/check/DellEMC_PowerFlex_os-brick/1c0f14b/DellEMC_PowerFlex_os-brick/28 : FAILURE in 22m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b375e51e8a3d5f12712a577d30e8869f1c564fb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-26 17:17:28.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/4/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"40f1a1eeace9defc25b3adfcae0b1e4fe25b9831","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-26 18:00:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/43/798743/4/upstream-check/cinder-cDOT-iSCSI/e1a7acf/ : SUCCESS in 35m 12s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/43/798743/4/upstream-check/cinder-cDOT-NFS/8557749/ : SUCCESS in 34m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/43/798743/4/upstream-check/cinder-cDOT-FCP/13e33fd/ : FAILURE in 1h 07m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"9cc0d9834263d4f5201a7595e52a7b803dba48b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-26 18:32:39.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/43/798743/4/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/43/798743/4/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"8cf4b4595a249e1af6c6e14757ae6c3f1fad8bf9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-26 18:35:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/43/798743/4/upstream-solidfire-check/cinder-solidfire-iSCSI/40629d5/ : FAILURE in 1h 42m 32s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5a15e9de0739520bc33be1ee466d4b923d76492","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-07 16:59:10.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\nLooks good.  Thanks, Sofia.","accounts_in_message":[],"_revision_number":4},{"id":"5e5f286f55340b6158a8376542530f4ccfed1ca7","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2021-08-09 17:07:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"89666659a40c320fb7921224744f40f1c944a676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 17:07:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5e097a9be1d1b204c35a3e352623600a328c97d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 18:28:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a5895f7f9954fe0ad39af26fc64ef93 : SUCCESS in 1h 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7504a4a6df094a73b6e02b01cf68cf24 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1011880682a47548f41f7eddb8c73df : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6995cf6c654345f0954d785b1d30eb4a : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a8be1bc4194c908906eb2dbf7dbb75 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db6838a47c06434caa6ab29184c6ffbe : SUCCESS in 5m 28s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/966347239cba4ac086ddcc617dacd58f : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ec83e3d1ac476d6d85aea925882cde6338f0eec2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 18:28:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f2e9d34319fecef287e3ae26a82aa65c68b25e30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 18:30:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71b5458a34745b0a8e2fa344030de28 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56d2f6455fdd4a56b4e3f09cc2b05cb9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"80f0bc4f5d0a1be0871331b9fbaf6657403a5e05","revisions":{"1c9940758e015724aa681c3af0497de0486db685":{"kind":"REWORK","_number":1,"created":"2021-06-29 20:06:07.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/43/798743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/798743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/798743/1"}}},"commit":{"parents":[{"commit":"1d444f0842f9261a55ddd9d5e5429b9fecc27845","subject":"Merge \"Ussuri+ is python3 only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d444f0842f9261a55ddd9d5e5429b9fecc27845"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-29 19:57:54.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-29 20:05:43.000000000","tz":0},"subject":"LVM: Retry lv calls on segfault due to lvs failure","message":"LVM: Retry lv calls on segfault due to lvs failure\n\nFollow-up to 0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247 and I6824ba4f.\n\nSometimes we get an unexpected failure on the lvs command where it exits\nwith code 139, which is not one of the 5 predefined ones in the code\nECMD_PROCESSED, ENO_SUCH_CMD, EINVALID_CMD_LINE, EINIT_FAILED, or\nECMD_FAILED.\n\nSince the LVM code indicates that a retry can help, we\u0027ll retry on error\n139 when calling ``get_lv_info``. Retry lvdisplay, lvs and lvcreate calls\non segfault.\n\nIn addition, we only need readonly mode for lvdisplay.\n\nCo-Authored-By: Gorka Eguileor \u003cgeguileo@redhat.com\u003e\nCo-Authored-By: Eric Harney \u003ceharney@redhat.com\u003e\nCloses-Bug: #1901783\nChange-Id: I62fadd401591025ecf9f83fa095249008f75bbaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1c9940758e015724aa681c3af0497de0486db685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1c9940758e015724aa681c3af0497de0486db685"}]},"branch":"refs/heads/master"},"cf7e43808d9fa985ec4e50f404de64a18f73b7ff":{"kind":"REWORK","_number":2,"created":"2021-06-30 17:46:01.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/43/798743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/798743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/798743/2"}}},"commit":{"parents":[{"commit":"1d444f0842f9261a55ddd9d5e5429b9fecc27845","subject":"Merge \"Ussuri+ is python3 only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d444f0842f9261a55ddd9d5e5429b9fecc27845"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-29 19:57:54.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-30 17:45:42.000000000","tz":0},"subject":"LVM: Retry lv calls on segfault due to lvs failure","message":"LVM: Retry lv calls on segfault due to lvs failure\n\nFollow-up to 0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247 and I6824ba4f.\n\nSometimes we get an unexpected failure on the lvs command where it exits\nwith code 139, which is not one of the 5 predefined ones in the code\nECMD_PROCESSED, ENO_SUCH_CMD, EINVALID_CMD_LINE, EINIT_FAILED, or\nECMD_FAILED.\n\nSince the LVM code indicates that a retry can help, we\u0027ll retry on error\n139 when calling ``get_lv_info``. Retry lvdisplay, lvs and lvcreate calls\non segfault.\n\nIn addition, we only need readonly mode for lvdisplay.\n\nCo-Authored-By: Gorka Eguileor \u003cgeguileo@redhat.com\u003e\nCo-Authored-By: Eric Harney \u003ceharney@redhat.com\u003e\nCloses-Bug: #1901783\nChange-Id: I62fadd401591025ecf9f83fa095249008f75bbaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cf7e43808d9fa985ec4e50f404de64a18f73b7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cf7e43808d9fa985ec4e50f404de64a18f73b7ff"}]},"branch":"refs/heads/master"},"acba029822fe9089ea9392b225a1fe929748c93d":{"kind":"REWORK","_number":3,"created":"2021-07-23 16:53:16.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/43/798743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/798743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/798743/3"}}},"commit":{"parents":[{"commit":"1d444f0842f9261a55ddd9d5e5429b9fecc27845","subject":"Merge \"Ussuri+ is python3 only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d444f0842f9261a55ddd9d5e5429b9fecc27845"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-29 19:57:54.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-07-23 16:53:08.000000000","tz":0},"subject":"LVM: Retry lv calls on segfault due to lvs failure","message":"LVM: Retry lv calls on segfault due to lvs failure\n\nFollow-up to 0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247 and I6824ba4f.\n\nSometimes we get an unexpected failure on the lvs command where it exits\nwith code 139, which is not one of the 5 predefined ones in the code\nECMD_PROCESSED, ENO_SUCH_CMD, EINVALID_CMD_LINE, EINIT_FAILED, or\nECMD_FAILED.\n\nSince the LVM code indicates that a retry can help, we\u0027ll retry on error\n139 when calling ``get_lv_info``. Retry lvdisplay, lvs and lvcreate calls\non segfault.\n\nIn addition, we only need readonly mode for lvdisplay.\n\nCo-Authored-By: Gorka Eguileor \u003cgeguileo@redhat.com\u003e\nCo-Authored-By: Eric Harney \u003ceharney@redhat.com\u003e\nCloses-Bug: #1901783\nChange-Id: I62fadd401591025ecf9f83fa095249008f75bbaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/acba029822fe9089ea9392b225a1fe929748c93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/acba029822fe9089ea9392b225a1fe929748c93d"}]},"branch":"refs/heads/master"},"80f0bc4f5d0a1be0871331b9fbaf6657403a5e05":{"kind":"REWORK","_number":4,"created":"2021-07-26 15:19:40.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/43/798743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/798743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/798743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/798743/4"}}},"commit":{"parents":[{"commit":"1d444f0842f9261a55ddd9d5e5429b9fecc27845","subject":"Merge \"Ussuri+ is python3 only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d444f0842f9261a55ddd9d5e5429b9fecc27845"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-06-29 19:57:54.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2021-07-26 15:19:16.000000000","tz":0},"subject":"LVM: Retry lv calls on segfault due to lvs failure","message":"LVM: Retry lv calls on segfault due to lvs failure\n\nFollow-up to 0b2c0d9238b6f84a02fcd82f97bc5a2d9f681247 and I6824ba4f.\n\nSometimes we get an unexpected failure on the lvs command where it exits\nwith code 139, which is not one of the 5 predefined ones in the code\nECMD_PROCESSED, ENO_SUCH_CMD, EINVALID_CMD_LINE, EINIT_FAILED, or\nECMD_FAILED.\n\nSince the LVM code indicates that a retry can help, we\u0027ll retry on error\n139 when calling ``get_lv_info``. Retry lvdisplay, lvs and lvcreate calls\non segfault.\n\nIn addition, we only need readonly mode for lvdisplay.\n\nCo-Authored-By: Gorka Eguileor \u003cgeguileo@redhat.com\u003e\nCo-Authored-By: Eric Harney \u003ceharney@redhat.com\u003e\nCloses-Bug: #1901783\nChange-Id: I62fadd401591025ecf9f83fa095249008f75bbaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80f0bc4f5d0a1be0871331b9fbaf6657403a5e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80f0bc4f5d0a1be0871331b9fbaf6657403a5e05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
