)]}'
{"id":"openstack%2Fos-brick~798956","triplet_id":"openstack%2Fos-brick~master~I4be4c4167adf01b70980269b338fbb720c49aeab","project":"openstack/os-brick","branch":"master","topic":"skip-missing-interp","hashtags":[],"change_id":"I4be4c4167adf01b70980269b338fbb720c49aeab","subject":"Remove skip_missing_interpreters","status":"MERGED","created":"2021-06-30 20:37:34.000000000","updated":"2021-07-13 14:19:17.000000000","submitted":"2021-07-13 14:17:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798956-skip-missing-interp","meta_rev_id":"3f58abb382d3e4d2a8f32a5b3066bdcfe6557f65","_number":798956,"virtual_id_number":798956,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-13 14:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-30 20:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-13 02:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-13 02:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-07-02 12:42:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 20:39:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-06-30 21:48:25.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-06-30 22:23:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-30 22:42:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-06-30 22:47:31.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 23:00:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-01 01:25:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-01 06:37:44.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-02 06:02:24.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-13 02:37:40.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"69fff434209ca4df7175cf249ca9604d66636a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-30 20:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aa68ddc678b295e9e54ac1284a92ad96c8c050d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-30 20:39:34.000000000","message":"Patch Set 1: Code-Review+2\n\nGood call!","accounts_in_message":[],"_revision_number":1},{"id":"5c2a1b6f56d742d40c2d55889834702abce5a27c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-06-30 21:48:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798956/1/iscsi/ : SUCCESS in 1h 10m 26s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798956/1/smb/ : SUCCESS in 1h 00m 29s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798956/1/ceph/ : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":1},{"id":"16d10d540f26aa78b0f5989b129f8790ab37d434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 22:23:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/222d2b448db04211a02f8886c287ca73 : SUCCESS in 1h 29m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9e3f23e0fe43b3a6efc196ff3c0426 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7408d9cf4cf46a98730400555a28a13 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cdb0c8d0aa243d5af02909ae15cf6e3 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9beaca648c34023a1d121c30431427a : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a01c3c675d44ffb9401004736d267bc : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3d098a887441e884facf3948581791 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2885e9f81f4a97b615c1e28396e3ee : SUCCESS in 6m 32s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d2e489c64604612a9786e51fe4c2f16 : SUCCESS in 1h 44m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0b96599d77bb4bba83401bf5fcb784ba : SUCCESS in 1h 07m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ab1deb59130eed30f44ea350a90d0d221df0537","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-06-30 22:42:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/56/798956/1/check/DellEMC_PowerStore_FC-ob/4154af5/DellEMC_PowerStore_FC-ob/25 : FAILURE in 18m 55s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/56/798956/1/check/DellEMC_PowerStore_iSCSI-ob/1157040/DellEMC_PowerStore_iSCSI-ob/25 : FAILURE in 18m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"df75ee2da8eee72439d5a26c9eab28a7523d34af","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-06-30 22:47:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_PowerFlex_os-brick/9d314cc/DellEMC_PowerFlex_os-brick/24 : FAILURE in 23m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2e2fe09f6d748704f77f9a19bad09dd9199d0f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-30 23:00:25.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/407c507edb0642c2be62757f60219bd4 : POST_FAILURE in 50m 03s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cba4ce0e9efa4c8aa24918ab280d09a6 : POST_FAILURE in 1h 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c00285f4b054103ff640302540608a0c967c807b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-01 01:25:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_Unity_FC_os-brick/514c325/DellEMC_Unity_FC_os-brick/47 : FAILURE in 27m 56s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_Unity_iSCSI_os-brick/2e4f3e0/DellEMC_Unity_iSCSI_os-brick/46 : FAILURE in 20m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"939d6aa3e4a723af28569795ec4d65d5997dd746","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-01 06:37:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_XtremIO_FC_os-brick/a681965/DellEMC_XtremIO_FC_os-brick/18 : FAILURE in 21m 09s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_XtremIO_iSCSI_os-brick/6ea2515/DellEMC_XtremIO_iSCSI_os-brick/17 : FAILURE in 22m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"d338eba46729159700f4291118d34d92d4557ed8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-02 06:02:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_VNX_FC_os-brick/9451401/DellEMC_VNX_FC_os-brick/29 : FAILURE in 19m 37s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_VNX_iSCSI_os-brick/e0fcb48/DellEMC_VNX_iSCSI_os-brick/29 : FAILURE in 17m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8c3808aa79dbfa9f2ec87267fd66d29802692765","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-02 12:42:56.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"7f7055ae20ca584ced0c95be03518d48cda35319","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2021-07-13 02:37:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good.  Thank you!","accounts_in_message":[],"_revision_number":1},{"id":"3a1c4abb85ec9368b746fe2927ca79d71612efe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 02:37:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe5619997df6e01cea19ab35ce39c4b30fdbe67f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 04:29:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e443eb885bce4e8bae6e3dcd6d944c58 : FAILURE in 1h 51m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c43963d257994e79aab3750cf745118a : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad4422cbd4f45bdbf7c153ad8a822b6 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48170250cd0b4540821d28351543b043 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af4d745ef38e4c3a98a2909fc1450835 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ddd869c5f84834817d1298f50856ef : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b31c898097884a74b86d600218f67e46 : SUCCESS in 7m 19s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f8a899798e524b4aa35eea3fb413114a : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c20718cb935bfe035079d44feb65bd59db75a31f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-13 11:11:00.000000000","message":"Patch Set 1:\n\nrecheck tempest-full-py3 - failure is tempest.scenario.test_server_basic_ops.TestServerBasicOps.test_server_basic_ops, looks like the lvm problem we\u0027ve addressed in cinder, should be solved by https://review.opendev.org/c/openstack/os-brick/+/798743","accounts_in_message":[],"_revision_number":1},{"id":"4911d1cffa9683b3cf4ba69eae1670d1636a345b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-07-13 12:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/798956/1/iscsi/ : SUCCESS in 49m 45s\n- os-brick-smb http://cloudbase-ci.com/os-brick/798956/1/smb/ : SUCCESS in 49m 34s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/798956/1/ceph/ : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5f0aa0d0cd2f79400485bc50373ce6013dfb99de","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-13 12:26:35.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/56/798956/1/check/devstack-plugin-ceph-dsvm-focal-py3/ddf5282/ : FAILURE in 47m 40s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/56/798956/1/check/tempest-dsvm-full-focal-py3/7c1df2b/ : FAILURE in 1h 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9a9e093be186758c947897e9b7cce0cbfacde53a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 12:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/641768a46ca84324ac0e55d6a2b39d18 : SUCCESS in 1h 14m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/675d3559ead24f608ee1ee6321317370 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d26455fc3264da2a54c3dd7ee0b6a36 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d864d6224bf4c73ad36a8893f2439f8 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f449d8adec3f4b2b8b3ae94e773593d8 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d456ddab17f4793b21d5bdc300e5b01 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0de995526b4a7ea742d220a76e9ec7 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/461cae0938034aff80edc21a596ede06 : SUCCESS in 5m 07s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/486f7cb2eb484f84a13fe4eed9cbf034 : SUCCESS in 1h 37m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/15db99229fc2493d98ef699178034377 : SUCCESS in 1h 08m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e8785851b30526fd54c9df9672934cf1fe77623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 12:51:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be627ad08c690a5b762af479b9d694d2f758c5ca","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 13:13:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_XtremIO_FC_os-brick/5ad80c8/DellEMC_XtremIO_FC_os-brick/26 : FAILURE in 21m 02s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_XtremIO_iSCSI_os-brick/574909f/DellEMC_XtremIO_iSCSI_os-brick/25 : FAILURE in 22m 36s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"1a6ea0ed38b4ac9c4bd22078ccfdf8bba744e293","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 13:41:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_Unity_FC_os-brick/d7e4bdb/DellEMC_Unity_FC_os-brick/53 : FAILURE in 23m 16s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/56/798956/1/check/DellEMC_Unity_iSCSI_os-brick/87cf19e/DellEMC_Unity_iSCSI_os-brick/50 : FAILURE in 21m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1790d0ac32b8765afcb74caf1cf6b9d5a76c7c19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 14:17:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c818d5006504a2e9f399fa40cebd1b0 : SUCCESS in 1h 22m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0890b5e3e9804e318a454935459ad0ae : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8618c28aafcd4dabad5e28a3ff5a379b : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea88208da29a471db65be35a8279bd40 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcc1f31f11784cb18a324b82250dd449 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52cc35d25cd4c5b8f411cf77216a7f4 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff61114448414c2bb455f23e7214aaea : SUCCESS in 4m 52s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/45d242f919b24989ac55ce47a8ed264d : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c08debaef887713f5518a4678b0b348d3e9107fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 14:17:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f58abb382d3e4d2a8f32a5b3066bdcfe6557f65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 14:19:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa1a59ad37a40b39524b5f65866bd63 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43d2be1bfbcd450bbbfe38d212b439e5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6f1a93c138c94a33889c72d792b83d3e0101d80","revisions":{"a6f1a93c138c94a33889c72d792b83d3e0101d80":{"kind":"REWORK","_number":1,"created":"2021-06-30 20:37:34.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/56/798956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/56/798956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/798956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/798956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/798956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/56/798956/1"}}},"commit":{"parents":[{"commit":"736454730fd66946bd46a95f612388189cbf3cfb","subject":"Merge \"NVMe-oF: Return right nqn when missing hostnqn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/736454730fd66946bd46a95f612388189cbf3cfb"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-06-30 20:05:22.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2021-06-30 20:14:27.000000000","tz":-240},"subject":"Remove skip_missing_interpreters","message":"Remove skip_missing_interpreters\n\nThis prevents a job reporting \u0027success\u0027 when the appropriate python\ninterpreter cannot be found, when actually it didn\u0027t run at all.\nAlso change the default envlist to use generic \u0027py3\u0027 instead of a\nspecific version which might not be present.\n\nJeremy Stanley has given a more thorough explanation of why this\nis a good change:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014810.html\n\nChange-Id: I4be4c4167adf01b70980269b338fbb720c49aeab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a6f1a93c138c94a33889c72d792b83d3e0101d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a6f1a93c138c94a33889c72d792b83d3e0101d80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
