)]}'
{"id":"openstack%2Fos-brick~802691","triplet_id":"openstack%2Fos-brick~master~I34e475f9a2a07e19413d40371d62fc809df2105a","project":"openstack/os-brick","branch":"master","topic":"bp/nvmeof-connection-agent","attention_set":{"16721":{"account":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"last_update":"2022-08-23 01:36:51.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I34e475f9a2a07e19413d40371d62fc809df2105a","subject":"NVMeOF Connection Agent","status":"NEW","created":"2021-07-28 12:10:38.000000000","updated":"2022-08-23 01:36:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":38,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"8f39ab852088e5121dc01ce53fc23468996aacf7","_number":802691,"virtual_id_number":802691,"owner":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-19 18:21:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 13:02:58.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-07-28 13:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 14:13:32.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2021-07-28 14:13:36.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-28 15:02:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-28 15:20:28.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2021-07-28 15:34:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-28 15:39:30.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2021-07-28 19:00:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-29 02:45:53.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-04 14:51:57.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2021-08-05 12:26:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-06 09:56:30.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-09 19:32:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-08-09 22:26:13.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-12 11:05:18.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2021-08-12 11:39:16.000000000","updated_by":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2021-08-13 11:55:49.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-08-23 01:36:51.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"38b1fb2f218a12cd43a95b1b1535b51ade051407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-07-28 12:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a2500828a72e5f8318d9630c815fded255a2fb2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-07-28 13:02:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/1/iscsi/ : SUCCESS in 52m 00s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/1/smb/ : SUCCESS in 50m 01s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/1/ceph/ : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"00b01490c04aa25b52d47b69187a9d5afcd7abe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 13:54:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7ebc86ac1fc4e6ab97baa0a3757ab9d : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5688f8435984be7b72a52eabb3a4e02 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c613de08b359450db4c50d264954b45d : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22aab22af99a4c7b99f6048ca7a14e46 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10de9cd5c4eb46409c9036ae2634cdc1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43da56a48f542529cdf198e66891b8b : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb3706da71f74e41be52f1c303a47f63 : SUCCESS in 5m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/535bd99a60ff4e5999dcb58ec2f58169 : SUCCESS in 4m 53s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/7adf0739a4994b599036a0a63d371919 : SUCCESS in 1h 12m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c57839bd0f1d4ba7baf6988a3ee86287 : SUCCESS in 1h 39m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/123f31c9de3c46ec8012b24f8684761a : SUCCESS in 1h 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"968edb89924dbacdd5a19ba28233b0b78ca25586","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-07-28 14:13:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/1/check/DellEMC_PowerStore_FC-ob/ed689bc/DellEMC_PowerStore_FC-ob/32 : FAILURE in 18m 23s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/1/check/DellEMC_PowerStore_iSCSI-ob/825c926/DellEMC_PowerStore_iSCSI-ob/30 : FAILURE in 18m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"867b8dc4b2251301fb72393ea03a329c0c7d479e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-28 14:13:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e4d08b4c17cb012d99e48b4ebae05dfa3afe2887","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-28 15:02:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/1/upstream-check/cinder-cDOT-iSCSI/0c2d056/ : SUCCESS in 39m 17s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/1/upstream-check/cinder-cDOT-NFS/d636233/ : SUCCESS in 36m 22s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/1/upstream-check/cinder-cDOT-FCP/7a18711/ : SUCCESS in 1h 07m 39s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2321e8ef1dcdeabd31d41d6b43edcd85afef90f1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-07-28 15:20:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_SC_FC-os-brick/025771d/DellEMC_SC_FC-os-brick/29 : SUCCESS in 47m 55s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_SC_iSCSI-os-brick/00efe47/DellEMC_SC_iSCSI-os-brick/28 : SUCCESS in 1h 25m 53s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f116aeeb9d6b7d2c9f7f080d8f8e9e50cfa720a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-28 15:20:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a5d7c3326451b10f310060964f567385d2f992dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-28 15:34:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_PowerFlex_os-brick/b3af7e7/DellEMC_PowerFlex_os-brick/29 : FAILURE in 22m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"53366a666b846039626cd5c6c549e4c8c83c10c6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-07-28 15:39:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/1/upstream-solidfire-check/cinder-solidfire-iSCSI/2ab85d9/ : FAILURE in 1h 44m 48s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"887081b23332d48e6f694a39c0f13856f751edf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-28 15:41:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"512844937bb5b05aa751d0fa8f49fdf2a4a97aba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-28 19:00:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_Unity_FC_os-brick/d4f5c37/DellEMC_Unity_FC_os-brick/57 : FAILURE in 26m 50s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_Unity_iSCSI_os-brick/1f977de/DellEMC_Unity_iSCSI_os-brick/54 : FAILURE in 21m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ae132560984561e503f989f31ad2ea7b63cb374d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-28 19:00:20.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f382712a43134c5f39a12ae20890b737e12b1ab3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-28 23:19:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"89f2586d1bc222dae6e39846f326902815647754","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 02:45:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_VNX_FC_os-brick/aa102db/DellEMC_VNX_FC_os-brick/33 : FAILURE in 19m 12s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/91/802691/1/check/DellEMC_VNX_iSCSI_os-brick/c9ea677/DellEMC_VNX_iSCSI_os-brick/33 : FAILURE in 25m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"162f193af6491a94996102daccb452bdc99a716c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-07-29 05:51:36.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/1/Cinder-tgtadm : FAILURE\n\n* SPDK http://13.74.249.42/refs/changes/91/802691/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"cb883905275e3bb043175bcbb8ef322d9d0c20f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-05 11:58:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4246f7dba2f9509083fc1c6d4148db89cd331f55","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-05 12:01:44.000000000","message":"Patch Set 2:\n\nTriggering Kioxia CI\n\n(temporarily it is using my gerrit account until we get a key updated in the official Kioxia CI account)\n\nrun-KIOXIA CI","accounts_in_message":[],"_revision_number":2},{"id":"5b470876974e24830bebb43110b9082db5483db2","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-05 12:09:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-91-802691-2/1628164904 : FAILURE in 07m 25s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"2df2811355257296c43fc0f3dc25ce0c93c38eee","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-05 12:26:23.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"2316bab8ae96e700b04a54487ad0e62ed3f8fbed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-05 12:26:57.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"6930b54c8d6097ed320f54623f2f7d80e131320e","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-05 12:38:29.000000000","message":"Patch Set 2:\n\nLooks like devstack is currently running into stack.sh failure - it currently reproduces every time running stack.sh manually\n\nError at end of log file: http://104.254.65.37/refs-changes-91-802691-2/1628164904/devstack/stack.sh.log","accounts_in_message":[],"_revision_number":2},{"id":"fffa0ac32b503ef75eb2ee6bd3e32aa6654575d6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-05 12:54:43.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/2/iscsi/ : SUCCESS in 53m 04s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/2/smb/ : SUCCESS in 55m 39s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/2/ceph/ : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":2},{"id":"95fcb1ac460405283460ab8054bcf71a38c20845","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-05 13:41:12.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"b202fcc4428f12ac4c32fb291b02962b6684ff02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 13:42:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2eb1c7eb84f47b6b85729ab688ee263 : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c9bfc255fce4e638829764e4dce0ed0 : SUCCESS in 1h 23m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287cb7acead641c3a41fc5277a2abab2 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f979b7bd0eae43b9b9962ef2359a3baf : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f98e7313332e4015b025f48ee68295b4 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffcbadc483644fa7905a56f94c9317b4 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/591b3445f36b4f4f9da219faaf67a5cd : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea434ce69084e8daf4e441b0127e633 : SUCCESS in 4m 57s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/fc229ff33ed443638b9ae4ff0202653a : SUCCESS in 6m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/789135e8af324efea1be3bff66ff435b : SUCCESS in 1h 13m 36s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1372d39a68564a648a86ebd3f3bb2f90 : SUCCESS in 1h 42m 09s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/82f8602e567843be8d054e47e25653f2 : SUCCESS in 1h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"310299a958518e42eb4daa7bdd78f4bce2cf496f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-05 14:20:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/2/upstream-check/cinder-cDOT-iSCSI/dabb723/ : SUCCESS in 36m 34s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/2/upstream-check/cinder-cDOT-NFS/13d858e/ : SUCCESS in 37m 52s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/2/upstream-check/cinder-cDOT-FCP/f8d2d6f/ : FAILURE in 5m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d17d6ea987f37747f51852a2cd60c2525151ed3e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-05 14:49:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/2/check/DellEMC_PowerStore_FC-ob/5452987/DellEMC_PowerStore_FC-ob/34 : SUCCESS in 48m 09s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/2/check/DellEMC_PowerStore_iSCSI-ob/3cd8f4f/DellEMC_PowerStore_iSCSI-ob/32 : SUCCESS in 51m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"2560a6d0b4afd1386b740b20124b35cd2a1fbd57","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-05 15:28:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/2/upstream-solidfire-check/cinder-solidfire-iSCSI/d185e91/ : FAILURE in 1h 45m 30s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"39ae41b84a0e9b390a7d707fbdc209ee38b2b962","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-05 16:12:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7e5edfbf249013b54e093fb139c68efc68f21ae4","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-05 17:20:47.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/2/check/DellEMC_SC_FC-os-brick/e9afccc/DellEMC_SC_FC-os-brick/31 : FAILURE in 59m 09s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/2/check/DellEMC_SC_iSCSI-os-brick/077fb35/DellEMC_SC_iSCSI-os-brick/30 : FAILURE in 1h 57m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"6fffb4fdf3cc81683b5a78aa31114dbec1844b5b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-05 20:18:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/91/802691/2/check/DellEMC_PowerFlex_os-brick/7fa536d/DellEMC_PowerFlex_os-brick/31 : FAILURE in 3h 16m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"18949756adaba2045904d9d190dc7c04ae97e662","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-05 20:20:14.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"6b3cd9da544c9e3a3c03bf296037a4b953d43a4e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-06 00:08:51.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8509e0acee25acee2b7fffe86565a2b25ff3d0a8","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-06 07:05:08.000000000","message":"Patch Set 2:\n\nrun-KIOXIA CI","accounts_in_message":[],"_revision_number":2},{"id":"6f755e283654078c48b4971b955e44423f215695","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-06 07:34:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-91-802691-2/1628233508 : SUCCESS in 29m 36s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"c7f2742ea42e255ef71a65ba48adaba0550ed668","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-06 08:47:26.000000000","message":"Patch Set 2:\n\nrun-KIOXIA CI","accounts_in_message":[],"_revision_number":2},{"id":"4db978e7e3d7f9c02119879f63a36256befc6202","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-06 09:07:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-91-802691-2/1628239647 : SUCCESS in 19m 33s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"e0d647b61307c2d233a3cec7ec4b59366a649cfe","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-06 09:56:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/91/802691/2/check/DellEMC_XtremIO_FC_os-brick/23d8f90/DellEMC_XtremIO_FC_os-brick/29 : FAILURE in 1h 32m 39s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/91/802691/2/check/DellEMC_XtremIO_iSCSI_os-brick/c037a11/DellEMC_XtremIO_iSCSI_os-brick/28 : FAILURE in 3h 10m 00s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"d0e3d912fc1ef9de63f2112d54ef53f14e5c5605","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-06 13:48:01.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/2/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/2/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"8dd2e9d561ad538ecf58f73bae44930e02a4610d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-09 19:32:23.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a58a90148c78e71762d34c6037e5b87dca87add9","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-09 22:26:13.000000000","message":"Patch Set 2: Code-Review-1\n\nI think this patch should be rebase or something because of the merge conflict before any review.","accounts_in_message":[],"_revision_number":2},{"id":"2a3c083936b7b6d5d0b6e3ea3fa776362c7bdf55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-10 03:24:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"783b9136e53bf596fb244d05f9c39f64b730ae8a","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-10 03:40:04.000000000","message":"Patch Set 3:\n\nrun-KIOXIA CI","accounts_in_message":[],"_revision_number":3},{"id":"2d3d0d23354caf5563ed5bbe50bd868641c34d66","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-10 03:58:01.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-91-802691-3/1628566804 : SUCCESS in 17m 55s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"8fef6eb574d388245b2ebc760427e7edfffef68c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-10 04:20:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/3/iscsi/ : SUCCESS in 49m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/3/smb/ : SUCCESS in 55m 11s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/3/ceph/ : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":3},{"id":"4f2fef65f28fed00f3c3d2577837558d3ef9211d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 05:03:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/00fa9febe315441f8c682decc71a7ede : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25d0a928304f401aacfa424945232ea6 : SUCCESS in 1h 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f30dad67374a9daf26df56245f9e50 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff2cf1ced74b4f03a622413a94bb911e : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c1b302c6f7c49e28f7f4ac9ea761cbe : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84024ce6bf424369bc39bd932ba5d349 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74464813cf04a0cb3745afaf6d0efce : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc6200bf72cd45cbb8259b19ca513eb7 : SUCCESS in 5m 38s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/44c7f4a073d94961b54356f4d94ef531 : SUCCESS in 4m 33s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/20c7314cca144373a95babe4d31847e4 : SUCCESS in 1h 18m 29s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e837f8f981884450955bf567b64f41a1 : SUCCESS in 1h 31m 36s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f2bb5801ef50410da199d747b60d5690 : SUCCESS in 1h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ae68c371aa7c54edc63e32e559f2b33246bcdb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-10 05:38:17.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/3/check/tempest-dsvm-full-focal-py3/d7ccbaf/ : FAILURE in 2h 00m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1b3848fe5987f9f2d3196227fdd091cdfbe1a2b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 05:39:52.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/3/SPDK : ABORTED\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/3/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"a71919ea75fe34b91740df914a03cb4418a450b1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-10 06:48:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/3/upstream-solidfire-check/cinder-solidfire-iSCSI/b699886/ : FAILURE in 1h 44m 36s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"a8a14e7c9b97fe158eb3fd6f4010f0554aa40c92","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-10 07:10:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/3/upstream-check/cinder-cDOT-iSCSI/88c1675/ : SUCCESS in 36m 23s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/3/upstream-check/cinder-cDOT-NFS/fc832e5/ : SUCCESS in 37m 27s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/3/upstream-check/cinder-cDOT-FCP/9c60303/ : FAILURE in 2h 06m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"78252d1ee1abeea71bbba0611ac30cbf324cb08e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 08:06:39.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/3/SPDK : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"81c7d584fc874c12283ad633784e5f423a3333ec","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-10 11:59:01.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"e254ed801fe80208b817334a3408c53d26998409","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-10 16:53:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_VNX_FC_os-brick/2bf40e6/DellEMC_VNX_FC_os-brick/34 : SUCCESS in 1h 53m 04s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_VNX_iSCSI_os-brick/cc317e1/DellEMC_VNX_iSCSI_os-brick/34 : FAILURE in 4h 05m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"fd09be0e7b8ec3d36ad7342b70e7fa7c8e52e5be","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-10 19:50:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_SC_FC-os-brick/916aae6/DellEMC_SC_FC-os-brick/32 : FAILURE in 54m 32s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_SC_iSCSI-os-brick/b6f7748/DellEMC_SC_iSCSI-os-brick/31 : FAILURE in 2h 01m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"61be8be975e33296ccd91fc7c707f3dd736e04c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 19:52:20.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"737d54ea4f04a8386fec65f757de13bcb317f861","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-10 23:46:10.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/3/SPDK : FAILURE\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"409850459f87b5385ec7c6fb04e68d77305e8eb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-11 03:08:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_Unity_FC_os-brick/3edcc3f/DellEMC_Unity_FC_os-brick/59 : SUCCESS in 1h 16m 28s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_Unity_iSCSI_os-brick/8edc26c/DellEMC_Unity_iSCSI_os-brick/56 : FAILURE in 2h 38m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c34b4872590ad58c4395b3b5ae0129cb6b03405f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-11 05:46:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_PowerFlex_os-brick/1a653a4/DellEMC_PowerFlex_os-brick/32 : ABORTED in 16m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"258ad7ac5882fb5daa8db1542c28a416a7bff52a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-11 05:49:05.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/3/SPDK : ABORTED","accounts_in_message":[],"_revision_number":3},{"id":"115c8868a13f2466a1e353970721e4ce70ce430b","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-11 09:59:59.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_XtremIO_FC_os-brick/46c9533/DellEMC_XtremIO_FC_os-brick/30 : FAILURE in 2h 56m 11s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/91/802691/3/check/DellEMC_XtremIO_iSCSI_os-brick/69d6f9c/DellEMC_XtremIO_iSCSI_os-brick/29 : FAILURE in 3h 08m 11s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"459cbf23f5c6e476924bcace6c4623a2cd0787ed","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-12 11:05:18.000000000","message":"Patch Set 3: Code-Review-1\n\nI think the changes are not generic, also some functions should go to Cinder\u0027s volume/drivers/...\nCan the changes relate to NVMe/TCP? Can they facilitate this type of NVMe fabric?","accounts_in_message":[],"_revision_number":3},{"id":"b3d7296a62ce31a17158a8a1abf292850952ed56","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-12 11:16:45.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e I think the changes are not generic, also some functions should go to Cinder\u0027s volume/drivers/...\n\u003e Can the changes relate to NVMe/TCP? Can they facilitate this type of NVMe fabric?\n\nThank you for the comment Vladislav\n\nCan you please provide some detail?\n\nWhat needs to be generic and in what way? (if you are talking specifically about the agent console script implementation, this could be improved over time as other backends start using it)\n\nWhich functions should go to Cinder\u0027s volume/drivers/... that you mentioned?\n\nAnd yes, the changes are related to all NVMeOF protocols, specifically using NVMeoF (TCP, or RDMA, etc..) in both replicated and un-replicated modes (feature exposed by KumoScale backend) - but more specifically, these changes introduce a console script connection agent, which monitors and self heals the connections and (initiator side) replicated volumes. If your concern is about the changed NVMeOFConnector methods - they are actually made more generic, not requiring an \"executor\" anymore (and this allows other contexts such as the agent to use them - rather than needing an executor, all they need is a (sometimes priviliged) rootwrap, which happens either way in both implementations)\n\nFinally, if there is also concern regarding the other nvmeof change / SPDK driver, these changes don\u0027t affect any of these methods (the only method affected is a simple changed which I commented on your patch: https://review.opendev.org/c/openstack/os-brick/+/800014/3/os_brick/initiator/connectors/nvmeof.py#422\n\nThank you!","accounts_in_message":[],"_revision_number":3},{"id":"b320adc643befe990ce656e05b4632dbf668fe2a","author":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"date":"2021-08-12 11:39:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nmostly nits.","accounts_in_message":[],"_revision_number":3},{"id":"16a9f90a17a15df34bf104903b849547e20d9a78","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-12 11:42:13.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \n\u003e \u003e I think the changes are not generic, also some functions should go to Cinder\u0027s volume/drivers/...\n\u003e \u003e Can the changes relate to NVMe/TCP? Can they facilitate this type of NVMe fabric?\n\u003e \n\u003e Thank you for the comment Vladislav\n\u003e \n\u003e Can you please provide some detail?\n\u003e \n\u003e What needs to be generic and in what way? (if you are talking specifically about the agent console script implementation, this could be improved over time as other backends start using it)\n\u003e \n\u003e Which functions should go to Cinder\u0027s volume/drivers/... that you mentioned?\n\u003e \n\u003e And yes, the changes are related to all NVMeOF protocols, specifically using NVMeoF (TCP, or RDMA, etc..) in both replicated and un-replicated modes (feature exposed by KumoScale backend) - but more specifically, these changes introduce a console script connection agent, which monitors and self heals the connections and (initiator side) replicated volumes. If your concern is about the changed NVMeOFConnector methods - they are actually made more generic, not requiring an \"executor\" anymore (and this allows other contexts such as the agent to use them - rather than needing an executor, all they need is a (sometimes priviliged) rootwrap, which happens either way in both implementations)\n\u003e \n\u003e Finally, if there is also concern regarding the other nvmeof change / SPDK driver, these changes don\u0027t affect any of these methods (the only method affected is a simple changed which I commented on your patch: https://review.opendev.org/c/openstack/os-brick/+/800014/3/os_brick/initiator/connectors/nvmeof.py#422\n\u003e \n\u003e Thank you!\n\nThanks a lot for clarification, Zohar! I think I understood the idea.","accounts_in_message":[],"_revision_number":3},{"id":"35d66e23911b4d024c1ac19f80b5585508c0eae9","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-13 05:37:33.000000000","message":"Patch Set 3:\n\n(12 comments)\n\nThank you for the review and useful comments!\n\nSubmitting followup patchset 4.","accounts_in_message":[],"_revision_number":3},{"id":"8aec2248fe6edb84c46b53646ac9f82f38113c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-13 05:38:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f85136317883009460d28553c3ed7100574fce7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-13 05:38:53.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/3/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"37d8fbe501cca177dbf3013de8bc5d264643aa3d","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-13 06:05:56.000000000","message":"Patch Set 4:\n\nrun-KIOXIA CI","accounts_in_message":[],"_revision_number":4},{"id":"4807740694dec92f4b2fc6e270345e0e445cf3fb","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-13 06:24:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-91-802691-4/1628834756 : SUCCESS in 18m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e3e3efe070a8649f6ab69d7fa5c0a1c9c2f0caa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-13 06:31:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/4/iscsi/ : SUCCESS in 52m 43s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/4/smb/ : SUCCESS in 50m 13s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/4/ceph/ : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":4},{"id":"89359aa5e1c2ef61f9bae9771f75d27d80e90bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 07:11:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5626ef13b5d44c9b786e524826f757b : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/916750e4ec424901aae91afe3a773658 : SUCCESS in 1h 30m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85583ec74e2a460e8f4ec2d7fccf02e3 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56d1b412f80148d58030566c62baa2a5 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cedbf6762d684d0d9fcda1fa7a6e8f38 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f00b8d53ff2479db1b848814e5da4a3 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d9d812ebbf4aa381d445f9504db7c2 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096cd51d830d4fb5ad8f54624ea61121 : SUCCESS in 5m 29s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/5781f1c7f5e24a818d92b7bf1ece341a : SUCCESS in 6m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/3f2ed5011d084535b0097b4f7dcdfde6 : SUCCESS in 1h 24m 42s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/320fdd33706d49da87d12820c24ad02a : SUCCESS in 1h 29m 14s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f4d8d9da24cc428e945587652cc0e36b : SUCCESS in 1h 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"92f7bb8db666d71bf29fca6f253cd6b03962c16b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-13 07:16:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/4/upstream-solidfire-check/cinder-solidfire-iSCSI/09cf1e9/ : FAILURE in 3m 10s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"63c24fb94618dd46f68eca0ba4f6a05972696507","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-13 07:34:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/4/upstream-check/cinder-cDOT-iSCSI/8b8d851/ : FAILURE in 4m 23s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/4/upstream-check/cinder-cDOT-NFS/8bd1ac7/ : FAILURE in 2m 28s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/4/upstream-check/cinder-cDOT-FCP/d9042e6/ : FAILURE in 19m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"93d1a3524cfd9b0881dc72a45192ab8e66343776","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-13 08:15:03.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/4/check/devstack-plugin-ceph-dsvm-focal-py3/5011cca/ : SUCCESS in 2h 23m 20s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/4/check/tempest-dsvm-full-focal-py3/bc4a01d/ : FAILURE in 1h 58m 22s","accounts_in_message":[],"_revision_number":4},{"id":"9a7b43623ea20f14b6ed0b6ed0232d1e791e812a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-13 08:15:07.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"39cf6f3dfbf44b2d5cffb6e33678514de7fd6a4a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-13 10:17:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/4/check/DellEMC_SC_FC-os-brick/0add33b/DellEMC_SC_FC-os-brick/33 : SUCCESS in 58m 23s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/4/check/DellEMC_SC_iSCSI-os-brick/09f20e7/DellEMC_SC_iSCSI-os-brick/32 : SUCCESS in 1h 45m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"757c4afe23d22c886591ce27c886468edea43cd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-13 10:18:32.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/4/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":4},{"id":"2c061bf01b4b132f6f68dbdaea4e9acf4081e403","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-08-13 10:48:51.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"50357de52956248de3737ea14cebdb8f1541ca74","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-13 11:55:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2a14eca676e520f6aca4170814a21b73e21379fa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-13 14:23:29.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"f6eb7ff66a3afe6f58b3d8e2954e878bb5f54673","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-15 23:23:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nA few things noted inline.  This will require some documentation, which could be done in a followup.  We\u0027ll need two types of docs: (1) dev docs describing the agent interface, and (2) operator docs explaining how this can be deployed, what goes in the conf file, etc.","accounts_in_message":[],"_revision_number":4},{"id":"21aa9da57e156bce8f470504eed60da2c4734601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-16 15:40:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c37089a5245d5c2151036190320e5c0912393c2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 15:40:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/4/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"31c7c148166aa33354e675d21cf62d603d306324","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-16 15:46:08.000000000","message":"Patch Set 5:\n\n(3 comments)\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e A few things noted inline.  This will require some documentation, which could be done in a followup.  We\u0027ll need two types of docs: (1) dev docs describing the agent interface, and (2) operator docs explaining how this can be deployed, what goes in the conf file, etc.\n\nThank you for the review.\n\nRegarding docs, sounds good, I will work on a follow up patch this week. I could use some pointers to location and format that both these dev and operator docs should go to.\n\nThank you!","accounts_in_message":[],"_revision_number":5},{"id":"c90efe8961153fc0ec061a895c009f111e749e99","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-16 16:49:24.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/5/iscsi/ : SUCCESS in 1h 08m 46s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/5/smb/ : SUCCESS in 1h 07m 17s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/5/ceph/ : FAILURE in 1h 08m 52s","accounts_in_message":[],"_revision_number":5},{"id":"151a51c08716eaf2d1a05e2ef8ed52ce38e6d61c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 16:50:31.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"a5a551d610be3ca8dc335cfb8669b34fd7dc4a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:02:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ae8e094da0341a6a5dc08806d9c1461 : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f1be3012aa84d2986b955d7a4a11f6e : SUCCESS in 1h 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fcef5e0df444718d4ad107601cad67 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b06d3303beda43f385743d45623aae30 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3926eeb4fa4f15b26ed6034de0205e : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/973f8c99d3ad4e72ab1ec379cd551e28 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24713a7181c4445ba573440b07cc66d : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce0f56dfe920423e95de05d87ccd77df : SUCCESS in 5m 50s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6e39fcc59006480aac3ed5551ac2a80a : SUCCESS in 5m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/bc99b2b592f544e49836bf80e1e91fee : SUCCESS in 1h 13m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3d180244e78c4e7cbff635b5b0846c29 : SUCCESS in 1h 17m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/49fd4c6f55634eb795187ebf00015d5b : SUCCESS in 1h 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"10763d242f71393638decb7c7a9a7e0856375516","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-16 18:12:03.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/5/check/tempest-dsvm-full-focal-py3/483f3cd/ : FAILURE in 2h 18m 41s","accounts_in_message":[],"_revision_number":5},{"id":"289e0bd9e1adccc2939b2b089ef9a0336450c758","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 18:12:07.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"497fc6691da998fb3f03711a8c438f1f4aa81c45","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-16 18:39:12.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/5/check/DellEMC_PowerStore_FC-ob/223cb09/DellEMC_PowerStore_FC-ob/37 : SUCCESS in 1h 11m 09s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/5/check/DellEMC_PowerStore_iSCSI-ob/abbf00a/DellEMC_PowerStore_iSCSI-ob/34 : FAILURE in 1h 26m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"4fdde53ff60d831550ec4680d022d303fb0f8bc1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 18:39:52.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"8643f4926b352104888f0bab8f10f4d8d5ae9adb","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-16 18:44:37.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/5/upstream-solidfire-check/cinder-solidfire-iSCSI/102b52e/ : FAILURE in 1h 42m 09s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"a7877c2827b9b678debc2d8cf6a3398bdf1e80fe","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-16 19:23:05.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/5/check/DellEMC_SC_FC-os-brick/3aa3ed7/DellEMC_SC_FC-os-brick/35 : SUCCESS in 52m 40s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/5/check/DellEMC_SC_iSCSI-os-brick/b088d7a/DellEMC_SC_iSCSI-os-brick/34 : SUCCESS in 1h 50m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"26bbd15e4a86c38b42efb63e8dab6ffb8bc9576b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 19:24:02.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"8bde5d64ea1f9a5426f8aa134f4fa596450c6d81","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-16 20:19:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/5/upstream-check/cinder-cDOT-iSCSI/f2c0dd0/ : SUCCESS in 37m 22s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/5/upstream-check/cinder-cDOT-NFS/f0f77a1/ : SUCCESS in 35m 27s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/5/upstream-check/cinder-cDOT-FCP/ca1be75/ : FAILURE in 3h 17m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"7219934d9a7b3cb1eb6baa6345f2dd1199eea177","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-16 20:33:34.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick http://publiclogs.emc.com/91/802691/5/check/DellEMC_PowerFlex_os-brick/05ac18e/DellEMC_PowerFlex_os-brick/33 : SUCCESS in 2h 36m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"52b652d9f2b2de07d0456d4b7e8b9da1c495ecf6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-16 20:34:46.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":5},{"id":"0ebeb9b3e4815956898af228f518fd65d884ceca","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-17 03:11:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/91/802691/5/check/DellEMC_XtremIO_FC_os-brick/f550866/DellEMC_XtremIO_FC_os-brick/31 : FAILURE in 1h 20m 38s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/91/802691/5/check/DellEMC_XtremIO_iSCSI_os-brick/33ff82b/DellEMC_XtremIO_iSCSI_os-brick/30 : SUCCESS in 2h 01m 07s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":5},{"id":"e2d4e9f26cb2231ad0871d4b6f8c564fa1ef4ef9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-17 03:11:28.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/91/802691/5/Cinder-tgtadm : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"bc2a837986917d788315a5e523fb1b93535fd3e2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-17 04:21:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nThe generic agent/vendor implementation separation needs to be stronger.  One way to see this is to look at how the configuration would work.  I put up a prototype based on this patch that you can look at to see what I mean: https://review.opendev.org/c/openstack/os-brick/+/804813.  Feel free to use as much of it as is helpful.","accounts_in_message":[],"_revision_number":5},{"id":"892ad9b1c01b6ee767e9eb5667a4bbf087ffb186","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-18 10:41:51.000000000","message":"Patch Set 5:\n\n(1 comment)\n\n\u003e Patch Set 5: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e The generic agent/vendor implementation separation needs to be stronger.  One way to see this is to look at how the configuration would work.  I put up a prototype based on this patch that you can look at to see what I mean: https://review.opendev.org/c/openstack/os-brick/+/804813.  Feel free to use as much of it as is helpful.\n\nMakes sense, there is still much to be improved, but I am liking this direction. I\u0027ve incorporated most of your suggestions from your prototype patch, with one main design difference (and also dropped a bunch of discussion comments there)\n\nMy main point being that I think that instead of having a \"provisioner rest client interface\" - it would be better to have an \"agent driver interface\" - which will also wrap provisioner communication methods/logic, but will not force new driver developers to follow the same \"provisioner rest client\" structure that kioxia kumoscale uses in this case.\n\nFollowup patchset coming up.\n\nThank you!","accounts_in_message":[],"_revision_number":5},{"id":"141fb7e1dc5f4c7eb97a0b6c2e9d9b545e3186e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-18 10:47:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2a52d08d7c725e879a47c48abb61aeb7ba5b24b6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-18 11:54:01.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/6/iscsi/ : SUCCESS in 51m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/6/smb/ : SUCCESS in 48m 14s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/6/ceph/ : FAILURE in 1h 06m 31s","accounts_in_message":[],"_revision_number":6},{"id":"2cee8a191ecf34e99bc7977b45ae17bf49e9973f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:12:41.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d740ba119d71435f98403fa8b7768259 : SUCCESS in 2m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23e9de68750c4890aadeda5b4005ca5e : SUCCESS in 1h 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb260d39c494563bb48748d29594fbe : FAILURE in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab1c8c84aa5346b7b134cea9ae34c44b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ada0015aa9447e8adbd470b20a87502 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e20f90d3b40468a9eb78b53897a2824 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7f2121d1be46e9ae6c4c1723698e33 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e78803ddf1a4f43bfce6740ca7a3de7 : SUCCESS in 6m 06s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/cc81a40a6c304b5a8ca5a42a251b0e7b : SUCCESS in 5m 42s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/d74ab3c27ccb49d6ba75e173ed263e6e : SUCCESS in 1h 14m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/00a14112316347829fc410090ab259cf : SUCCESS in 1h 23m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/9001aaa1a31b4743ac40cebd57c4ca06 : SUCCESS in 1h 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9766c97bf7467a9865f32169a42eba28dd30bcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-18 12:41:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6b6dd61196e83a23149f975d2ba8d2efdc7d5ba6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-18 13:47:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/7/iscsi/ : SUCCESS in 1h 05m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/7/smb/ : SUCCESS in 53m 21s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/7/ceph/ : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":7},{"id":"a2d0945058576fa6fdd49b7d72943917257ba293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:10:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d35ca60ed462491fa59c7187b6226a6b : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e6ec485bf6347728e8df2b73e72e1be : SUCCESS in 1h 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be4aa43a9709457d89e8589068fd540e : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/343bf4c04eb74b8081b983fdd194fe32 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebaf35337de9448ba69265f6965f4103 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/563eea1f442d40bf8d834e3a172dd22e : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b4407908f14e3dac110dbd57036a2e : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/431f96a7a0034822925ba0e00e05d2f2 : SUCCESS in 5m 46s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9c62a035f5d04637b7f7698ab18312e1 : SUCCESS in 5m 55s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/fc7aeb2ac4154a34bfaf632b246a9182 : SUCCESS in 1h 16m 32s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ae6319c1bbb34a80a74732aca02cf19d : SUCCESS in 1h 21m 11s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/28e54e91300d4d6a9882359b7955176c : SUCCESS in 1h 09m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a7e456eeab62413c7fa8b0679b17438819946ba1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-18 15:28:40.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/7/upstream-check/cinder-cDOT-iSCSI/c946b88/ : SUCCESS in 33m 12s\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/7/upstream-check/cinder-cDOT-NFS/2dacd3e/ : SUCCESS in 33m 57s\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/7/upstream-check/cinder-cDOT-FCP/14d3222/ : FAILURE in 1h 17m 25s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"2ddde8dd8d6e94092f55807f18e58ab230c314c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-18 15:37:55.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/7/check/devstack-plugin-ceph-dsvm-focal-py3/610622c/ : SUCCESS in 2h 51m 54s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/7/check/tempest-dsvm-full-focal-py3/16f989b/ : FAILURE in 2h 23m 24s","accounts_in_message":[],"_revision_number":7},{"id":"031b6e1656cd9d5b38d0a375b72592cf067131a2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-18 16:15:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/7/check/DellEMC_PowerStore_FC-ob/c935e3b/DellEMC_PowerStore_FC-ob/40 : SUCCESS in 1h 06m 20s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/7/check/DellEMC_PowerStore_iSCSI-ob/b2250f6/DellEMC_PowerStore_iSCSI-ob/37 : FAILURE in 1h 34m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"ee5093b661f256148ccdbc76be32d08f91df206c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-18 16:55:07.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/7/check/DellEMC_SC_FC-os-brick/dfe7670/DellEMC_SC_FC-os-brick/38 : SUCCESS in 52m 56s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/7/check/DellEMC_SC_iSCSI-os-brick/ca1c16d/DellEMC_SC_iSCSI-os-brick/37 : SUCCESS in 1h 53m 51s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"a3f347ccc4548885b3d48dfaa7e2f1c1206c9286","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-18 23:45:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nI think the architectural direction makes sense; will be back to look some more after dinner.\n\nFor now, noticed a minor problem with the config file generation (see inline comment).","accounts_in_message":[],"_revision_number":7},{"id":"0190af246f9f524e116a85d333311db09b908cb2","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-19 03:08:51.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f43e4a824b970f528d311cdc981ccd42c596b152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-19 03:11:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"45cce55a6e0e10cb8b601ee9f19a0e2d3912d3f4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-19 04:13:48.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/8/iscsi/ : SUCCESS in 49m 25s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/8/smb/ : SUCCESS in 48m 05s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/8/ceph/ : FAILURE in 1h 01m 28s","accounts_in_message":[],"_revision_number":8},{"id":"4822b162f10c9636d3a02a96b75c1787f804bd90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 04:34:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3dffd09f7db84b3a954aaf3142fcdb56 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebd198ff6d9248b28d03ad937d8d5961 : SUCCESS in 1h 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074a9857da0c4f80a313807c80271fb7 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e305f7f3114bc48bb3d1f9b4f093fa : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b8ea2b7d6a842ea989f36d387ce2cff : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/799b6bb9216347e2845856200ad1c32c : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c79cb1edb8480590a0512a9bc0bea1 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/675383cb8fd84cc8930a6fa26df357b7 : SUCCESS in 4m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/dd16e9cdea644a0e9c46ef38d5769ab1 : SUCCESS in 5m 21s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/1bf9beb3694e41f09a0203b0a723f11c : SUCCESS in 1h 12m 38s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f11639b1d7e245a5bf041654dc04b2ae : SUCCESS in 1h 21m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ffbf6ef2431d4f25b28ea397b94f6a4d : SUCCESS in 1h 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"98971edd5482f321ce1ef7f9c4a7c2ba9fbadfa8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-19 04:34:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://logs.openstack.netapp.com/logs/91/802691/8/upstream-check/cinder-cDOT-iSCSI/91f7640/ : NOT_REGISTERED\n- cinder-cDOT-NFS https://logs.openstack.netapp.com/logs/91/802691/8/upstream-check/cinder-cDOT-NFS/ed8e40f/ : NOT_REGISTERED\n- cinder-cDOT-FCP https://logs.openstack.netapp.com/logs/91/802691/8/upstream-check/cinder-cDOT-FCP/492b038/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f72d879a82d00778748f51211413bce74fcf8904","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-19 04:34:58.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/8/upstream-solidfire-check/cinder-solidfire-iSCSI/19d0cba/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"7e6066d2a84093a8e24b0fd48e184a42b5515a76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-19 05:11:44.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/91/802691/8/check/tempest-dsvm-full-focal-py3/31feae8/ : FAILURE in 1h 54m 54s","accounts_in_message":[],"_revision_number":8},{"id":"c53e3d84a2869dd4d42610147f4fd4faed730e7a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-19 09:24:28.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/8/check/DellEMC_PowerStore_FC-ob/ae238e1/DellEMC_PowerStore_FC-ob/41 : SUCCESS in 1h 06m 29s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/8/check/DellEMC_PowerStore_iSCSI-ob/07f8624/DellEMC_PowerStore_iSCSI-ob/38 : FAILURE in 1h 29m 35s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"fa2e35fb50e77396a83a068188ff5c1ca943ed5f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-19 12:55:49.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/8/check/DellEMC_SC_FC-os-brick/735ba13/DellEMC_SC_FC-os-brick/39 : SUCCESS in 48m 21s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/8/check/DellEMC_SC_iSCSI-os-brick/02f792d/DellEMC_SC_iSCSI-os-brick/38 : SUCCESS in 1h 41m 32s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"2dd03959eec242a9175efa2e5a6757c3bc40e8db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-19 16:48:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c78ae065734238121f6ec342a73e04a1a25cd51c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-19 17:54:40.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/802691/9/iscsi/ : SUCCESS in 51m 58s\n- os-brick-smb http://cloudbase-ci.com/os-brick/802691/9/smb/ : SUCCESS in 51m 45s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/802691/9/ceph/ : FAILURE in 1h 05m 38s","accounts_in_message":[],"_revision_number":9},{"id":"cfbdec19b2659299553aeed28816b08240777725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:21:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7eb8a0ac469449fb506909b9f04e9dd : SUCCESS in 2m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a35c50734fbe40439a98d3f8afece46a : SUCCESS in 1h 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f212b5c1e5f84a1e9275a5dce765835a : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf3983d6183b4fa3a300707a5d19dcf5 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c8dfefcd99d477bbef7e7c96b8ea3fc : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8314b18cc8c04c1b8b0ae013451a1850 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f37008046ba486384fa951c7c0e197e : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7998857c412040bcbb7e5369f106cacf : SUCCESS in 6m 09s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9d9c1248e50842b6baa181ca68e4cc8d : SUCCESS in 5m 22s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/362cd91d31794592abb8b2914d6b072e : SUCCESS in 1h 27m 18s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dbc19409719c4b308cbf8eeaa5cb37ff : SUCCESS in 1h 24m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cb1f465715d9491aaa8aecd89ed87ee3 : SUCCESS in 1h 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"85a73b27bb0c63f527427546c9ad153d2c2873e4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-19 19:31:10.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/54783275b91c4115a4dae75ae9d86b01 : DISK_FULL in 2h 25m 08s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b75d77399bf2494e93a334446288f1f8 : DISK_FULL in 2h 11m 41s","accounts_in_message":[],"_revision_number":9},{"id":"513d9fb3fc12e10667e8e9df57bfdea7e8c9ccff","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2021-08-19 19:51:33.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob http://publiclogs.emc.com/91/802691/9/check/DellEMC_PowerStore_FC-ob/069dc0f/DellEMC_PowerStore_FC-ob/42 : SUCCESS in 1h 10m 19s\n- DellEMC_PowerStore_iSCSI-ob http://publiclogs.emc.com/91/802691/9/check/DellEMC_PowerStore_iSCSI-ob/a6f7c09/DellEMC_PowerStore_iSCSI-ob/39 : FAILURE in 1h 29m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"02363d31049f3b69d4592978aee921af65b0d6c9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2021-08-19 19:54:22.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://logs.openstack.netapp.com/logs/91/802691/9/upstream-solidfire-check/cinder-solidfire-iSCSI/aecbba7/ : FAILURE in 1h 32m 35s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":9},{"id":"79f215c626f1997ad0ef301def7357bdb8656b42","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2021-08-19 21:52:33.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick http://publiclogs.emc.com/91/802691/9/check/DellEMC_SC_FC-os-brick/98a79bf/DellEMC_SC_FC-os-brick/40 : SUCCESS in 53m 38s\n- DellEMC_SC_iSCSI-os-brick http://publiclogs.emc.com/91/802691/9/check/DellEMC_SC_iSCSI-os-brick/4fa3eba/DellEMC_SC_iSCSI-os-brick/39 : SUCCESS in 1h 48m 38s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":9},{"id":"f09dcfd94c5808e5ea5da1f7c6d2b39b56b3183d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-20 12:50:41.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick http://publiclogs.emc.com/91/802691/9/check/DellEMC_VNX_FC_os-brick/6ed9abd/DellEMC_VNX_FC_os-brick/36 : SUCCESS in 1h 36m 49s\n- DellEMC_VNX_iSCSI_os-brick http://publiclogs.emc.com/91/802691/9/check/DellEMC_VNX_iSCSI_os-brick/8af5339/DellEMC_VNX_iSCSI_os-brick/37 : FAILURE in 3h 59m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"66f2f4cf603add6e5a90d8b007210045360ea944","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-08-21 12:08:05.000000000","message":"Patch Set 9:\n\nIt would be helpful to have some info (could be on an etherpad) about how you\u0027re deploying the agent and testing it.  I\u0027m having trouble getting it to start, but that could be for some stupid reason.  In any case, I won\u0027t be able to actually test it (I just want to verify that it can be deployed and the config is working properly), so it would be good to have some assurance that it\u0027s working in a deployed KumoScale environment.","accounts_in_message":[],"_revision_number":9},{"id":"c3187f06a252a24657e7be83055dba3ca3a45f4e","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-22 12:43:34.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e It would be helpful to have some info (could be on an etherpad) about how you\u0027re deploying the agent and testing it.  I\u0027m having trouble getting it to start, but that could be for some stupid reason.  In any case, I won\u0027t be able to actually test it (I just want to verify that it can be deployed and the config is working properly), so it would be good to have some assurance that it\u0027s working in a deployed KumoScale environment.\n\nMy apologies for the late response.\n\nHere is an example config file /etc/kioxia/provisioner.conf\n\n[DEFAULT]\nagent_driver\u003dos_brick.cmd.nvmeof_agent.kioxia.driver.KumoScaleAgentDriver\nkioxia_prov_ip\u003d70.0.0.217\nkioxia_prov_port\u003d30100\nkioxia_token\u003deyJhb...\nkioxia_cert_file\u003d/etc/ssl/ssdtoolbox.pem\n\nThe way I run the agent is:\n\n/usr/bin/python3.8 agent_entrypoint.py --config-file /etc/kioxia/provisioner.conf\n\n(where the agent_entrypoint.py looks like this:)\n\n#!/usr/bin/python3.8\nimport sys\nfrom os_brick.cmd.nvmeof_agent.agent import main\nif __name__ \u003d\u003d \"__main__\":\n    sys.exit(main())\n\n\nThis should be the equivalent of building and installing os-brick with this code and running:\n\n/usr/bin/nvmeof-agent --config-file /etc/kioxia/provisioner.conf","accounts_in_message":[],"_revision_number":9},{"id":"e6fc4060a72712608796fd3360ae62b5dc15473f","author":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"date":"2021-08-22 15:11:09.000000000","message":"Patch Set 9:\n\nMade a quick initial etherpad for our jotting down needs\nhttps://etherpad.opendev.org/p/nvmeof-agent","accounts_in_message":[],"_revision_number":9},{"id":"c5a868571f31f7569dd5bca1604d71c7ae715545","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-05 14:13:04.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/91/802691/5/SPDK : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8f39ab852088e5121dc01ce53fc23468996aacf7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-08-23 01:36:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b790ea40ac10ff968f3cd38fbef44873416ae1da","revisions":{"b03014b2971abc68dec2e08691487fba81013c13":{"kind":"REWORK","_number":1,"created":"2021-07-28 12:10:38.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/1"}}},"commit":{"parents":[{"commit":"98c0916336f2e1ba50fc1798ee507961f420e600","subject":"Update mypy infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/98c0916336f2e1ba50fc1798ee507961f420e600"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 12:10:34.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b03014b2971abc68dec2e08691487fba81013c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b03014b2971abc68dec2e08691487fba81013c13"}]},"branch":"refs/heads/master"},"44b50e3f53f56b3fadb7ed8b91e170d625e472a3":{"kind":"REWORK","_number":2,"created":"2021-08-05 11:58:32.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/2"}}},"commit":{"parents":[{"commit":"98c0916336f2e1ba50fc1798ee507961f420e600","subject":"Update mypy infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/98c0916336f2e1ba50fc1798ee507961f420e600"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-05 11:58:05.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/44b50e3f53f56b3fadb7ed8b91e170d625e472a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/44b50e3f53f56b3fadb7ed8b91e170d625e472a3"}]},"branch":"refs/heads/master"},"ecae9415580693ad5a8e5132d493f6944bfb4dd0":{"kind":"REWORK","_number":3,"created":"2021-08-10 03:24:40.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/3"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-10 03:24:07.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ecae9415580693ad5a8e5132d493f6944bfb4dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ecae9415580693ad5a8e5132d493f6944bfb4dd0"}]},"branch":"refs/heads/master"},"1a513ca13d182c769e6ba1d640568afa6efc40fc":{"kind":"REWORK","_number":4,"created":"2021-08-13 05:38:49.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/4"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-13 05:34:00.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1a513ca13d182c769e6ba1d640568afa6efc40fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1a513ca13d182c769e6ba1d640568afa6efc40fc"}]},"branch":"refs/heads/master"},"6ad181acf97ad628c203e1f9f2fa873e63be670f":{"kind":"REWORK","_number":5,"created":"2021-08-16 15:40:11.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/5"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-16 15:39:34.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6ad181acf97ad628c203e1f9f2fa873e63be670f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6ad181acf97ad628c203e1f9f2fa873e63be670f"}]},"branch":"refs/heads/master"},"665b7ac222ac367780c5c3458b6ae18d5207f3cb":{"kind":"REWORK","_number":6,"created":"2021-08-18 10:47:09.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/6"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-18 10:45:42.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/665b7ac222ac367780c5c3458b6ae18d5207f3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/665b7ac222ac367780c5c3458b6ae18d5207f3cb"}]},"branch":"refs/heads/master"},"5a45671f4f636a0c8b6bae7b6fe74c0e4df045ca":{"kind":"REWORK","_number":7,"created":"2021-08-18 12:41:16.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/7"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-18 12:40:05.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5a45671f4f636a0c8b6bae7b6fe74c0e4df045ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5a45671f4f636a0c8b6bae7b6fe74c0e4df045ca"}]},"branch":"refs/heads/master"},"7be7caf2712cf8d74ad1028e58e985dbc815c798":{"kind":"REWORK","_number":8,"created":"2021-08-19 03:11:59.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/8"}}},"commit":{"parents":[{"commit":"13516073d54058807432cd7dda88fb6ffc367c6e","subject":"Merge \"linuxscsi: Only raise and log multipathd errors when required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/13516073d54058807432cd7dda88fb6ffc367c6e"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-19 03:11:43.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7be7caf2712cf8d74ad1028e58e985dbc815c798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7be7caf2712cf8d74ad1028e58e985dbc815c798"}]},"branch":"refs/heads/master"},"b790ea40ac10ff968f3cd38fbef44873416ae1da":{"kind":"REWORK","_number":9,"created":"2021-08-19 16:48:32.000000000","uploader":{"_account_id":16721,"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","username":"zohar"},"ref":"refs/changes/91/802691/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/91/802691/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/91/802691/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/91/802691/9"}}},"commit":{"parents":[{"commit":"f4931f592c7ab31f111d6cfc480ba5221d428e6c","subject":"Merge \"rbd windows: log the device number\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f4931f592c7ab31f111d6cfc480ba5221d428e6c"}]}],"author":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-07-28 11:57:20.000000000","tz":0},"committer":{"name":"Zohar Mamedov","email":"zohar.cloud@gmail.com","date":"2021-08-19 16:48:09.000000000","tz":0},"subject":"NVMeOF Connection Agent","message":"NVMeOF Connection Agent\n\nThis is a console script responsible for initiator-side monitoring and self healing of NVMeoF connections and replicated mdraid volumes.\n\nIt initially supports block storage exposed by KumoScale volume driver used by NVMeoF connector.\n\nThe functions of the agent are:\n* Probing of storage orchestrator\n* Monitor NVMeoF connections and replicated volumes\u0027 mdraid states\n* Report initiator-side connection and volume events to storage orchestrator\n* Connection and replication state re-conciliation\n\nSince this agent relies on methods defined in the NVMeOFConnector \"Replicated+\" mode, such methods were converted to be usable by the console script.\nThis was done such that the top level privileged methods (for running nvme and mdadm cli tools) are no longer class methods, and are kept in privileged/nvmeof.py\nAll other methods now either use these privileged methods, or run their own rootwrap execute.\n\nImplements: Blueprint nvmeof-connection-agent\nChange-Id: I34e475f9a2a07e19413d40371d62fc809df2105a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b790ea40ac10ff968f3cd38fbef44873416ae1da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b790ea40ac10ff968f3cd38fbef44873416ae1da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
