)]}'
{"id":"openstack%2Fos-brick~806009","triplet_id":"openstack%2Fos-brick~master~I2f096c7342c3f9867175357cdc5db618dc878a93","project":"openstack/os-brick","branch":"master","topic":"bug1940608","hashtags":[],"change_id":"I2f096c7342c3f9867175357cdc5db618dc878a93","subject":"flush failed when both host and vm use lvm the multipath","status":"ABANDONED","created":"2021-08-25 15:14:46.000000000","updated":"2022-08-23 17:51:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c92dad6bbdb9f3fe7bedd36df47f672925de7db","_number":806009,"virtual_id_number":806009,"owner":{"_account_id":33897,"name":"efineshi","display_name":"efineshi","email":"675727327@qq.com","username":"efineshi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-25 16:42:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-08-25 21:55:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 15:40:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 16:23:40.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-08-25 16:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 21:04:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-08-25 21:06:02.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 21:55:39.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2021-08-26 07:37:09.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"6f47b96bd991d8224cd851489d06fb12b93fcab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33897,"name":"efineshi","display_name":"efineshi","email":"675727327@qq.com","username":"efineshi"},"date":"2021-08-25 15:14:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e372b711fd5545b16ae1c6b1424f6014e4b54de","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-25 15:40:48.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7e59f5f378ef455f9a54efb14895006a : RETRY_LIMIT in 1m 00s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/bc02156e1dad4eec953daa37b76b2691 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f11820deb9f147443bf6e0411f1717983a21c457","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-08-25 16:23:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/806009/1/iscsi/ : SUCCESS in 1h 07m 07s\n- os-brick-smb http://cloudbase-ci.com/os-brick/806009/1/smb/ : SUCCESS in 1h 03m 03s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/806009/1/ceph/ : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fc48ceacd849ecb22a7243a39ed02eefde1fe7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:42:38.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8eafdd60db243aebb635a28aeebb29c : SUCCESS in 1h 23m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862eb0b7ecff49b798440c0b577c1a0a : FAILURE in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b257f5479d4739ac447ed5a852887b : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5bc00921a14af8a81d647cc3b842c8 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1704b95e2e1043d0810b6e6271e871a2 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c5b3c90c22843e5bb402802029c0402 : SUCCESS in 7m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4d9c9d38c3f44c3f98d8fafce7fe9060 : SUCCESS in 5m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/7588d0202a06441698a345fd8dbc555c : SUCCESS in 1h 16m 55s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/80b682ddcf5f4d4cbaa461dcc2b4b445 : SUCCESS in 1h 24m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/df59090c32f549f08476933122b464e7 : SUCCESS in 1h 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bc2f7047c8df67f71749f08a5ed7bcbc278d16f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-25 21:04:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick http://publiclogs.emc.com/09/806009/1/check/DellEMC_Unity_FC_os-brick/5c6ce2e/DellEMC_Unity_FC_os-brick/67 : SUCCESS in 1h 10m 51s\n- DellEMC_Unity_iSCSI_os-brick http://publiclogs.emc.com/09/806009/1/check/DellEMC_Unity_iSCSI_os-brick/f1cf478/DellEMC_Unity_iSCSI_os-brick/63 : FAILURE in 2h 11m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f5aa123b1072c97406662428c43609d07e09218","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-25 21:06:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/09/806009/1/Cinder-tgtadm : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"4d9d86bd3b9619c074519ad134173defc989a1c4","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2021-08-25 21:55:39.000000000","message":"Patch Set 1: Code-Review-1\n\nFrom Gorkas\u0027s comment https://bugs.launchpad.net/os-brick/+bug/1940608/comments/1 I think this isn\u0027t the right way to go.","accounts_in_message":[],"_revision_number":1},{"id":"a86521575e258ca9da66a2588080ea6a5d6c2c6c","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-08-26 07:37:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick http://publiclogs.emc.com/09/806009/1/check/DellEMC_XtremIO_FC_os-brick/3b743ea/DellEMC_XtremIO_FC_os-brick/42 : FAILURE in 1h 24m 45s\n- DellEMC_XtremIO_iSCSI_os-brick http://publiclogs.emc.com/09/806009/1/check/DellEMC_XtremIO_iSCSI_os-brick/2ca9aed/DellEMC_XtremIO_iSCSI_os-brick/39 : FAILURE in 2h 57m 37s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"4b723b8f4e7af9bd9aea9e9a2db0c64e790af5e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-08-26 07:38:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/09/806009/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"0f2046a3fd70c20d853a5b7afdde13aa582c0b54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-09-05 15:56:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/09/806009/1/SPDK : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7c92dad6bbdb9f3fe7bedd36df47f672925de7db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33897,"name":"efineshi","display_name":"efineshi","email":"675727327@qq.com","username":"efineshi"},"date":"2022-08-23 17:51:49.000000000","message":"Abandoned\n\nit is not a bug","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62a2020547bd6b9ca6f6110f10566f718c7f1697","revisions":{"62a2020547bd6b9ca6f6110f10566f718c7f1697":{"kind":"REWORK","_number":1,"created":"2021-08-25 15:14:46.000000000","uploader":{"_account_id":33897,"name":"efineshi","display_name":"efineshi","email":"675727327@qq.com","username":"efineshi"},"ref":"refs/changes/09/806009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/09/806009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/09/806009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/09/806009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/09/806009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/09/806009/1"}}},"commit":{"parents":[{"commit":"4f63dd045ca07f66003d1ccbbb19423ee91cf926","subject":"Add Python3 yoga unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f63dd045ca07f66003d1ccbbb19423ee91cf926"}]}],"author":{"name":"efineshi","email":"675727327@qq.com","date":"2021-08-25 14:45:53.000000000","tz":480},"committer":{"name":"efineshi","email":"675727327@qq.com","date":"2021-08-25 14:45:53.000000000","tz":480},"subject":"flush failed when both host and vm use lvm the multipath","message":"flush failed when both host and vm use lvm the multipath\n\nwhen the compute host used lvm and not reject the other block devices，if the vm are running with multipath，and then used lvm in the vm，nova-compute will flush multipath device failed because of the compute host lvmd will scan the vg and active the vg on the compute host。\nwhen we try to detach the multipath volume or cold migrate it，novac-compute log will report that:\nmultipath -f xxxxx failed , map in use。\nwe can manual\nvgchange -a n vgname\nand then it will flush success，but i think maybe we can try to control it in the code\n\nCloses-Bug: #1940608\nChange-Id: I2f096c7342c3f9867175357cdc5db618dc878a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/62a2020547bd6b9ca6f6110f10566f718c7f1697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/62a2020547bd6b9ca6f6110f10566f718c7f1697"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
