)]}'
{"id":"openstack%2Fos-brick~817005","triplet_id":"openstack%2Fos-brick~master~I8e14a55c3e376d427ce39b1ae1feb906326a13c9","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2021-12-03 08:12:02.000000000","reason":"Change was abandoned"},"28050":{"account":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"last_update":"2021-11-30 11:11:02.000000000","reason":"Maciej Szwed replied on the change"},"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2021-12-03 08:12:02.000000000","reason":"Change was abandoned"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2021-12-03 08:12:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8e14a55c3e376d427ce39b1ae1feb906326a13c9","subject":"Actually disconnect NVMe drive on disconnect_volume()","status":"ABANDONED","created":"2021-11-08 13:50:05.000000000","updated":"2021-12-03 08:12:02.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"811f7790045840928c6172700380501894b603a7","_number":817005,"virtual_id_number":817005,"owner":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 14:15:57.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-08 14:16:44.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-08 14:54:08.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2021-11-08 15:27:41.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-23 07:53:33.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-11-23 07:53:33.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2021-11-30 08:13:30.000000000","updated_by":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"}],"messages":[{"id":"fd374658abcbe0837acaf8047a3f107b5e40e248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-11-08 13:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c44b8ffedd90b52e554060b0a06819127153d38","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-11-08 14:15:57.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e553d4163f2bc361202d33538aa5812ea3181ac9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 14:16:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SPDK http://13.74.249.42/refs/changes/05/817005/1/SPDK : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"289a4e1d2918222db0b8f8b02d283bb4804a33a4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2021-11-08 14:54:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/817005/1/iscsi/ : SUCCESS in 51m 02s\n- os-brick-smb http://cloudbase-ci.com/os-brick/817005/1/smb/ : SUCCESS in 1h 03m 41s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/817005/1/ceph/ : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d9a77257265bb75a016faa4137607fc689148f81","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2021-11-08 15:27:41.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/05/817005/1/check/DellEMC_XtremIO_FC_os-brick/bff8fb0/DellEMC_XtremIO_FC_os-brick/1 : SUCCESS in 1h 08m 06s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/05/817005/1/check/DellEMC_XtremIO_iSCSI_os-brick/53cd4a2/DellEMC_XtremIO_iSCSI_os-brick/2 : SUCCESS in 1h 37m 19s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"762e3953bc03d57848beac53deb932b8d40f9c99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 15:28:40.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/817005/1/Cinder-tgtadm : ABORTED\n\n* SPDK http://13.74.249.42/refs/changes/05/817005/1/SPDK : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"05e3183059feda7f504d606a275f92035642e04e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-11-08 18:11:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SPDK http://13.74.249.42/refs/changes/05/817005/1/SPDK : SUCCESS\n\n* Cinder-tgtadm http://13.74.249.42/refs/changes/05/817005/1/Cinder-tgtadm : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a105de1da162aa0e85e6bcdbb26022f82244d364","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-11-18 08:49:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"351873fe258456d9a816b5f2b2549d69b792b67c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-11-18 08:49:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"dbf15baa89abc57f8d4b6fc972b9dc0afce60ab8","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-11-30 08:20:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b147de7078246437df779ceab77a06d18ced9ca3","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2021-11-30 08:32:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f88c0033f33f52937188a6b665456bf8f58880c8","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-11-30 11:11:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"811f7790045840928c6172700380501894b603a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"date":"2021-12-03 08:12:02.000000000","message":"Abandoned\n\nIntroduces new issue.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d3c7638e511f25f66eea40c9eae3f51595b7845","revisions":{"8d3c7638e511f25f66eea40c9eae3f51595b7845":{"kind":"REWORK","_number":1,"created":"2021-11-08 13:50:05.000000000","uploader":{"_account_id":28050,"name":"Maciej Szwed","email":"maciej.szwed@intel.com","username":"maciejszwed"},"ref":"refs/changes/05/817005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/05/817005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/817005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/817005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/817005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/05/817005/1"}}},"commit":{"parents":[{"commit":"d0e762299143e0a2dbc4688f763e801df349472e","subject":"Merge \"multipath/iscsi: iSCSI connections are not reinitiated after reboot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d0e762299143e0a2dbc4688f763e801df349472e"}]}],"author":{"name":"Maciej Szwed","email":"maciej.szwed@intel.com","date":"2021-11-08 12:56:12.000000000","tz":60},"committer":{"name":"Maciej Szwed","email":"maciej.szwed@intel.com","date":"2021-11-08 13:14:33.000000000","tz":60},"subject":"Actually disconnect NVMe drive on disconnect_volume()","message":"Actually disconnect NVMe drive on disconnect_volume()\n\ndisconnect_volume() is missing actual NVMe-oF\ninitiator disconnect command. This causes\nunusable ghost drives to be present in a system\nafter disconnecting volume from node.\n\nSigned-off-by: Maciej Szwed \u003cmaciej.szwed@intel.com\u003e\nChange-Id: I8e14a55c3e376d427ce39b1ae1feb906326a13c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8d3c7638e511f25f66eea40c9eae3f51595b7845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8d3c7638e511f25f66eea40c9eae3f51595b7845"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
