)]}'
{"id":"openstack%2Fos-brick~827423","triplet_id":"openstack%2Fos-brick~master~Idb9c9b711bf729e8f8dfd83db2e81f6f95a4f789","project":"openstack/os-brick","branch":"master","topic":"muli/lightosrelnote1","attention_set":{},"removed_from_attention_set":{"34459":{"account":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"last_update":"2022-02-03 21:33:38.000000000","reason":"Change was submitted"},"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2022-02-03 10:06:35.000000000","reason":"yuval replied on the change"}},"hashtags":[],"change_id":"Idb9c9b711bf729e8f8dfd83db2e81f6f95a4f789","subject":"releasenotes: add Lightbits LightOS connector release note","status":"MERGED","created":"2022-02-02 08:30:56.000000000","updated":"2022-02-03 21:35:43.000000000","submitted":"2022-02-03 21:33:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827423-muli/lightosrelnote1","meta_rev_id":"f404e7bfa05e9b1badbd22a3cacbc1e5c16ee51f","_number":827423,"virtual_id_number":827423,"owner":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 21:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-03 21:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-02-03 10:06:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-03 21:23:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 08:33:04.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"REVIEWER"},{"updated":"2022-02-02 09:02:48.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 09:41:13.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-02-02 09:55:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 09:55:34.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 10:06:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-02 10:09:24.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-02-02 11:20:28.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-02 11:33:12.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-03 21:23:46.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"8996796e06d0eeadb0417d6fae269f15d89ca2df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-02-02 08:30:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a79246c86ccf3a16ba65d9a4dec4a1503139115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-02-02 08:45:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"08e8d967daa25a7e08d9ee91b987c90df14816c0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-02 09:02:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_XtremIO_FC_os-brick/2435235/DellEMC_XtremIO_FC_os-brick/52 : FAILURE in 15m 12s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_XtremIO_iSCSI_os-brick/b6eb7c3/DellEMC_XtremIO_iSCSI_os-brick/49 : FAILURE in 17m 04s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"6c73d26dd7945e76adb7ddfd0796afe2c97578e4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-02 09:41:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/827423/2/iscsi/ : SUCCESS in 49m 55s\n- os-brick-smb http://cloudbase-ci.com/os-brick/827423/2/smb/ : SUCCESS in 49m 37s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/827423/2/ceph/ : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":2},{"id":"22791deeed850c252d33a0fe63b0b998989787a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 09:55:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9e27c84a4e406e8fead965897764d4 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f88c17d0b74b83b5c6efb2f8d0a8f8 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/756be856b1af439d934dae19ce2e2cb0 : SUCCESS in 4m 35s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/a075c66e66a24790a806ecb286748571 : SUCCESS in 1h 08m 01s (non-voting)\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b2374fe9efc147de97185feef97403b3 : SUCCESS in 1h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf750a710e7efe4df404d0c9eebd0e9a89e112eb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-02 09:55:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_PowerFlex_os-brick/a07afd0/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"ca2582a93a5364c0f77ab6da1ac6ec7a360e2e99","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-02 10:06:47.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/02_02_2022_09_58_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"c888a8d85509382fe1b616fc0be9e8118b835525","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-02 10:09:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_PowerStore_FC-ob/926d043/DellEMC_PowerStore_FC-ob/21 : FAILURE in 13m 50s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_PowerStore_iSCSI-ob/aa7ebd2/DellEMC_PowerStore_iSCSI-ob/18 : FAILURE in 12m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"48b1a304096d8b364bd04747b9ecf27cb194c82e","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-02 10:25:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_XtremIO_FC_os-brick/176f270/DellEMC_XtremIO_FC_os-brick/53 : FAILURE in 14m 51s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_XtremIO_iSCSI_os-brick/855b045/DellEMC_XtremIO_iSCSI_os-brick/50 : FAILURE in 16m 27s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"c23bb5c02165798cae84e77edecc780ff20d822e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-02 11:20:28.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/23/827423/2/check/devstack-plugin-ceph-dsvm-focal-py3/2f88c9d/ : SUCCESS in 2h 30m 45s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/23/827423/2/check/tempest-dsvm-full-focal-py3/8f5d8d1/ : SUCCESS in 2h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"17baff96287ab7212826442f4fdc5165cab261de","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-02 11:33:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_SC_FC-os-brick/57f4b4c/DellEMC_SC_FC-os-brick/18 : FAILURE in 59m 30s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/23/827423/2/check/DellEMC_SC_iSCSI-os-brick/b92348c/DellEMC_SC_iSCSI-os-brick/18 : FAILURE in 1h 37m 38s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"fd6b3a7318e677f203ff0881d94428c2c2c5bf53","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-03 10:06:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1cc0dab873b0c83f7917f36fd4cb401f30d5352","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-03 21:23:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01e7f8d56d8f993f6578fde3d4ea22ee5a1ddf87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:24:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a059e850420c397bb22435fae0449645f436524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:33:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd4a229ada744fd38b7864a46b94f6c8 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09c9aeea12ea4869bbee613297b92cde : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/624952dc287745a2a79ac26179e2c0f3 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":2},{"id":"de0dc2126186fa21e5bfa9ef6850c790d80152a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:33:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f404e7bfa05e9b1badbd22a3cacbc1e5c16ee51f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:35:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0b4bebb10a4020809b2fe960bf6b7b : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa9f6f222434d62b01c8df9848f0975 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"593870338e48dec94a0fcd1e32edc9b40bafef8c","revisions":{"0516f74071fbd1d67e32d7a2e4eb7171057d312e":{"kind":"REWORK","_number":1,"created":"2022-02-02 08:30:56.000000000","uploader":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"ref":"refs/changes/23/827423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/23/827423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/23/827423/1"}}},"commit":{"parents":[{"commit":"627da6a40b4679a41245502b12e06b8340f1ca05","subject":"Lightbits LightOS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/627da6a40b4679a41245502b12e06b8340f1ca05"}]}],"author":{"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","date":"2022-02-02 08:03:42.000000000","tz":120},"committer":{"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","date":"2022-02-02 08:30:44.000000000","tz":120},"subject":"releasenotes: add Lightbits LightOS connector release note","message":"releasenotes: add Lightbits LightOS connector release note\n\nSigned-off-by: Muli Ben-Yehuda \u003cmuli@lightbitslabs.com\u003e\nChange-Id: Idb9c9b711bf729e8f8dfd83db2e81f6f95a4f789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0516f74071fbd1d67e32d7a2e4eb7171057d312e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0516f74071fbd1d67e32d7a2e4eb7171057d312e"}]},"branch":"refs/heads/master"},"593870338e48dec94a0fcd1e32edc9b40bafef8c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-02 08:45:25.000000000","uploader":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"ref":"refs/changes/23/827423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/23/827423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/23/827423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/23/827423/2"}}},"commit":{"parents":[{"commit":"627da6a40b4679a41245502b12e06b8340f1ca05","subject":"Lightbits LightOS driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/627da6a40b4679a41245502b12e06b8340f1ca05"}]}],"author":{"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","date":"2022-02-02 08:03:42.000000000","tz":120},"committer":{"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","date":"2022-02-02 08:45:25.000000000","tz":0},"subject":"releasenotes: add Lightbits LightOS connector release note","message":"releasenotes: add Lightbits LightOS connector release note\n\nSigned-Off-By: Muli Ben-Yehuda \u003cmuli@lightbitslabs.com\u003e\nChange-Id: Idb9c9b711bf729e8f8dfd83db2e81f6f95a4f789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/593870338e48dec94a0fcd1e32edc9b40bafef8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/593870338e48dec94a0fcd1e32edc9b40bafef8c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
