)]}'
{"id":"openstack%2Fos-brick~828087","triplet_id":"openstack%2Fos-brick~master~Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12","project":"openstack/os-brick","branch":"master","topic":"yuval/followup-reuse_get_host_nqn","attention_set":{"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2022-02-15 04:01:22.000000000","reason":"Rajat Dhasmana replied on the change"}},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-09 13:40:52.000000000","reason":"Brian Rosmaita replied on the change"},"33431":{"account":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"last_update":"2022-02-14 23:21:31.000000000","reason":"Change was submitted"},"34459":{"account":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"last_update":"2022-02-09 10:26:59.000000000","reason":"Muli Ben-Yehuda replied on the change"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-02-14 23:21:31.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-02-10 11:35:56.000000000","reason":"Gorka Eguileor replied on the change"}},"hashtags":[],"change_id":"Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12","subject":"Reuse get_host_nqn code","status":"MERGED","created":"2022-02-07 08:15:05.000000000","updated":"2022-02-15 04:01:22.000000000","submitted":"2022-02-14 23:21:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":55,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"828087-yuval/followup-reuse_get_host_nqn","meta_rev_id":"ad7bf653085d1770bc254989e6ac2baa71cd9736","_number":828087,"virtual_id_number":828087,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 23:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-02-09 13:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-02-09 10:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-14 15:28:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-02-10 11:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":1,"date":"2022-02-15 04:01:22.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-14 21:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-02-09 13:40:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 08:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 08:16:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 08:26:28.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 08:26:41.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-02-07 08:41:00.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 09:15:13.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"CC"},{"updated":"2022-02-07 09:31:12.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-02-07 10:00:16.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 10:13:55.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-02-07 10:25:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-07 11:48:35.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-02-07 13:33:40.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-02-07 13:39:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-07 23:51:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-02-08 15:00:45.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2022-02-08 19:13:57.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-08 19:28:08.000000000","updated_by":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"reviewer":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"state":"CC"},{"updated":"2022-02-09 10:26:59.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"REVIEWER"},{"updated":"2022-02-10 11:35:56.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-02-14 15:28:07.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-02-14 15:54:14.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"b5d42c378d31404c515d3ab623959052c05e4008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 08:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c537edbbce19b8dc228ce46e14c5e092448275b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 08:16:50.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a7cc9a82c03ba8953ffab12ca1a8b34224caa413","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 08:16:55.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6283b7191ce83a53711a7658fa4743b98788ee8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 08:26:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d1fcff6e47225f3431dafa19fcab413c2100dad","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 08:26:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"8cfe94cf375a10d72ebdaf8fb22d1cdb142665f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 08:26:26.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"bd2331a6002653acf087a9e8255cbe97f5ce105f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-02-07 08:26:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"24abbab1edaecbbb9b98968fa5ca45ec60c2f4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 08:26:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"6ee96a22531363779775d8b84f3c40ef11997a55","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-07 08:26:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":2},{"id":"52ba5cbe793d8242c68d9d4528d6f973ff9c10db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 08:27:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"74b9e3781f39434ca2c64c9ff5b15b62841c084a","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 08:41:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_XtremIO_FC_os-brick/16a04eb/DellEMC_XtremIO_FC_os-brick/70 : FAILURE in 13m 14s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_XtremIO_iSCSI_os-brick/5946876/DellEMC_XtremIO_iSCSI_os-brick/67 : FAILURE in 12m 49s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"a055158cd2991339abe6e2844eb63df7116dc6aa","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 08:55:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0cbdd93d391802fe5dd0ec6ab40f3ee1b1e30642","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-02-07 09:15:13.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a588786563d0759f83ecdf2d25fd714eb7c5c9d","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 09:30:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c0c720af732ec9fb320d0fc2d7d467f6c895f1fc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-07 09:31:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828087/3/iscsi/ : SUCCESS in 1h 02m 56s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828087/3/smb/ : SUCCESS in 1h 03m 20s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828087/3/ceph/ : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c638794d47195050902b6c7ff0c9243e989a613b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 09:59:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba69a479535a4814b527ea377361606b : SUCCESS in 1h 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90dca0211a38444fb9b62ca9fa345c4c : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd22466c72b34ada87ef1198f6745707 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b51c5a368c804150b581218d66f933c8 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6675931313e8465e8d8579726a3898cb : SUCCESS in 5m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b8573ff4c78a496abbab7af7c0080029 : SUCCESS in 4m 14s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/20a462cd639f48a78199af7de950185b : SUCCESS in 1h 00m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68413da545b943f4b1adda4709d97af9 : SUCCESS in 1h 23m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/faa24ef6b4d6423eb67153877853bc0c : SUCCESS in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a31d243360a37cf091e0aa5f12180a9d93e2512f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 10:00:16.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_PowerFlex_os-brick/8e67aec/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"1624a5eb6410ec0c8f70d7fe3b3d859f32b2ca29","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-07 10:13:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_PowerStore_FC-ob/f502b74/DellEMC_PowerStore_FC-ob/29 : FAILURE in 13m 42s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_PowerStore_iSCSI-ob/dd0a11b/DellEMC_PowerStore_iSCSI-ob/26 : FAILURE in 13m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"d5ec2f0b3fbbabb1de3f87a25a0e5e6dad7d25f9","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-07 10:15:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_XtremIO_FC_os-brick/431042e/DellEMC_XtremIO_FC_os-brick/71 : FAILURE in 15m 01s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_XtremIO_iSCSI_os-brick/dfd1b86/DellEMC_XtremIO_iSCSI_os-brick/68 : FAILURE in 15m 40s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3},{"id":"af9001ac044cffb42db3f10311f012a2f3ec7175","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-07 10:25:57.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/07_02_2022_10_00_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"409ebb236d399b812c16be90fdc9d7c8c9fdaae2","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-07 10:51:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc26232729c1a0dc78fb4e9c2bf6364bb74f8009","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-07 10:59:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/828087/3/check/devstack-plugin-ceph-dsvm-focal-py3/019c210/ : SUCCESS in 2h 28m 47s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/828087/3/check/tempest-dsvm-full-focal-py3/46be2a4/ : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4e6a1f0ab55f741cfcd3dc758800099ace868285","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-07 11:48:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_SC_FC-os-brick/1b97f5d/DellEMC_SC_FC-os-brick/26 : FAILURE in 58m 38s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/87/828087/3/check/DellEMC_SC_iSCSI-os-brick/cbfc4a1/DellEMC_SC_iSCSI-os-brick/26 : FAILURE in 1h 48m 22s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"36cd3126c49296e935893a05ed02032457f6421a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-07 12:18:08.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/07_02_2022_12_03_20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"d44aa76b163c20c66e5cb94f6d45e93037e0271b","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-07 13:33:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/3/upstream-solidfire-check/cinder-solidfire-iSCSI/ed0a881/ : FAILURE in 3h 33m 18s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"58b59f4eec9a4b54c27b6675dedbaaf5f35e606b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-07 13:37:35.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/07_02_2022_13_26_51\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"c72d2a1c65c3f01d044d175e908ab256d3036d0b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-07 13:39:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/3/upstream-check/cinder-cDOT-iSCSI/63a0ff7/ : FAILURE in 28m 10s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/3/upstream-check/cinder-cDOT-NFS/3351a49/ : FAILURE in 3h 39m 20s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/3/upstream-check/cinder-cDOT-FCP/527ab58/ : SUCCESS in 54m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4d24231b4b087e1b123ffe132a15044eb4ea1b1e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-07 23:51:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6d079c4b91aa344534586bfe15da94a427350ec1","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 07:53:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5d3d9b7fb88d62c1b1d248b13c6525d8e0ac959","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-08 14:25:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e850ddcd86d5e1f47c7e42ce0ecd6c7a6484b1b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 14:38:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a124b2b0b398c636b525e6828c4f192d88322f4d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-02-08 14:38:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"aeff2295c68c588a26773062c20770831bf1b64f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-08 14:38:50.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"5a6cd473fa1bd9dd31ed8651b63ac65ed21ebe46","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-08 14:38:51.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"6e8500cfe0331784f054f9bbd080f13a1b902371","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-02-08 14:38:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"3454ee75560550e8b2fd6c81324bce76aff7ccc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 14:40:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"a45d11aa011af956be2ba357bf67e0f8335a90aa","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-08 15:00:45.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9c31b1d14d602ba2460e0fc8a1bc81f7519c1849","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-08 15:49:39.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828087/4/iscsi/ : SUCCESS in 1h 10m 51s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828087/4/smb/ : SUCCESS in 1h 04m 49s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828087/4/ceph/ : FAILURE in 1h 10m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5f21d301c056a30a030d4f336a3b67c9744cfdb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 15:57:09.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"7d54d87debc5d0838840888e3a1be6f90153af97","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 16:00:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3aaf3938db3a3fef05aa1a760236e06eb9dd9df","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 16:05:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"40beb07996537b4bb3c8054b015574a751435500","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 16:10:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f853417d77444a3783c5f8b4d4d3f9c49ccbf747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-08 16:36:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"01d5c5f8862ee55143eb2230753006932cda9075","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-08 16:46:13.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"31a35663b7e463bdb4bbdb26f50b9bcec83a8195","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-08 17:52:43.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828087/6/iscsi/ : FAILURE in 1h 16m 08s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828087/6/smb/ : SUCCESS in 1h 10m 08s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828087/6/ceph/ : SUCCESS in 1h 15m 03s","accounts_in_message":[],"_revision_number":6},{"id":"0a3010a4d4390d8761e9457523fc7fd3f077765b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 18:45:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4f353b2494b4713ad78cc0ecfa85f6b : SUCCESS in 1h 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215dc1da42c442439aee37c72d15a038 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5384b90825cb4447a9c9ce1feffb2869 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e40230b0909145669fd8cf483ae58f87 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c718058a6a1b4a88a5eb6c36f8b07b74 : SUCCESS in 8m 22s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/1e324bc12ff549d09b97f70c69f1dde7 : SUCCESS in 4m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/ba75abfa50a24dc4808b9776d2eae485 : SUCCESS in 1h 16m 02s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/97f362111b9a41e7b14b6096cb507c52 : SUCCESS in 1h 56m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6c41c84df0494e3e86e799b21c2e10d4 : SUCCESS in 1h 18m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"49d5cc4e3a5defa45fdde02e5502cf0919edc381","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-08 18:54:25.000000000","message":"Patch Set 6:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/08_02_2022_18_45_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"361e521480712639cea430913c2a84dd5cd7638b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-08 19:00:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/828087/6/check/DellEMC_PowerStore_FC-ob/f00520e/DellEMC_PowerStore_FC-ob/30 : FAILURE in 14m 38s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/87/828087/6/check/DellEMC_PowerStore_iSCSI-ob/c4b4f14/DellEMC_PowerStore_iSCSI-ob/27 : FAILURE in 14m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"77b637213cac4bf3b4f089b6201b3650f738b3f4","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-08 19:13:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-87-828087-6/1644345936 : SUCCESS in 28m 19s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":6},{"id":"91a637bac0926808f14e114b3769a42afad35460","author":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"date":"2022-02-08 19:28:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"472f74f580ad9b43a1b08491671b5f40121e258a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-08 19:59:19.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/828087/6/check/devstack-plugin-ceph-dsvm-focal-py3/382a4e7/ : SUCCESS in 2h 34m 56s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/828087/6/check/tempest-dsvm-full-focal-py3/2edd0ec/ : SUCCESS in 1h 51m 45s","accounts_in_message":[],"_revision_number":6},{"id":"501de043f96c52a6b3524cc9aac74d43ed7b7fb1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-08 20:30:24.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/87/828087/6/check/DellEMC_SC_FC-os-brick/f3a8562/DellEMC_SC_FC-os-brick/27 : FAILURE in 1h 02m 59s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/87/828087/6/check/DellEMC_SC_iSCSI-os-brick/9177cf0/DellEMC_SC_iSCSI-os-brick/27 : SUCCESS in 1h 44m 34s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"b2c2cf819ec621c86f1dd9dccae0fb093c5f8ec6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-08 23:10:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c222f9c9d9adc3e5e7d4159c1a3d8973847016a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-09 09:41:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"57d52a917574cc0b25cffe098d19c8b5c24451bd","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-09 09:42:56.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4f1b30379d5b7aa848fa11123f39e1fefb642ba4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 09:57:16.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/87/828087/7/check/DellEMC_XtremIO_FC_os-brick/a8fe0eb/DellEMC_XtremIO_FC_os-brick/73 : FAILURE in 13m 59s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/828087/7/check/DellEMC_XtremIO_iSCSI_os-brick/ee9f067/DellEMC_XtremIO_iSCSI_os-brick/70 : FAILURE in 13m 02s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":7},{"id":"11d14847d9f9b4040fc8f0513db13217f5ff2381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-09 10:01:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5c63efd46b03ed19fe742336baec60b8c424902f","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-09 10:02:54.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"409b5e3f47d7cd7665ceec1a21b3410d90b3ce00","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-09 10:04:22.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ad4e706d701bbf61f39ff479f260126d968ed544","author":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"date":"2022-02-09 10:26:59.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"f127389874cdab2593050f5bf01063b35f8fd615","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 10:28:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_XtremIO_FC_os-brick/b96b959/DellEMC_XtremIO_FC_os-brick/75 : FAILURE in 14m 47s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_XtremIO_iSCSI_os-brick/720240d/DellEMC_XtremIO_iSCSI_os-brick/72 : FAILURE in 14m 59s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"26823fb668a797935f02886013cd966d809a4f7f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-09 10:57:24.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828087/8/iscsi/ : SUCCESS in 55m 08s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828087/8/smb/ : SUCCESS in 53m 49s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828087/8/ceph/ : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":8},{"id":"76240853d0f3116208378c2c3aa8a39868f5db3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 12:15:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2964fc8fcb654ad4930e6217924df5ef : SUCCESS in 1h 43m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41c0cc11dca459b90742edc55ed2e9d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/779c85712cef471db3adba907318056a : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/220793f3648a4876b835dd6ca21241ce : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e546128a46304e24b7c1f2c9a49998ba : SUCCESS in 6m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/15a63817ec9f4176a0ef87b1ce5b2f43 : SUCCESS in 5m 47s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/353f1f15610e444f8d6bcdd3914a5e79 : TIMED_OUT in 2h 06m 01s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c75bc869ad144478b8010453f1574a64 : SUCCESS in 1h 45m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8c139bfb61104ff1bf231350136a5647 : SUCCESS in 1h 16m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"061f697058b4e401cabd77af40c94c59089ca549","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 12:20:12.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_PowerFlex_os-brick/cdfc1d2/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"9be750cdaa6929e235fe9bd6b5a9e0a8117d051e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-09 12:23:47.000000000","message":"Patch Set 8:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/09_02_2022_12_15_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"b9c5511e8cd4c3129777edac7638815157143f8c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-02-09 12:29:20.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_PowerStore_FC-ob/c0acbb8/DellEMC_PowerStore_FC-ob/32 : FAILURE in 13m 59s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_PowerStore_iSCSI-ob/e69a8b5/DellEMC_PowerStore_iSCSI-ob/29 : FAILURE in 13m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"82d91c4be0dd7ec366c8d79f92ce136fd5c1ffad","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-09 12:31:50.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-87-828087-8/1644408907 : SUCCESS in 16m 41s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":8},{"id":"42d3d3167a81f3f942df10e81e01d79fd4b307fc","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-09 12:45:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_XtremIO_FC_os-brick/150d637/DellEMC_XtremIO_FC_os-brick/76 : FAILURE in 15m 17s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_XtremIO_iSCSI_os-brick/cf5a0bf/DellEMC_XtremIO_iSCSI_os-brick/73 : FAILURE in 15m 10s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":8},{"id":"c6356d0938ae173910168f5b201c3746f636f816","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-09 13:12:29.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/828087/8/check/devstack-plugin-ceph-dsvm-focal-py3/737c85a/ : SUCCESS in 2h 30m 40s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2e37c7c828334a4aac771afdb3d1a705 : POST_FAILURE in 2h 07m 11s","accounts_in_message":[],"_revision_number":8},{"id":"c1f4df1cb06eb0cab59592b053d3330f06db058c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-09 13:40:52.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6431251d52858267e22456c36097ee39df54015b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-09 13:41:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"df0342f840ae8f96b0326b7b24b65a5cc8d2f91b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-09 14:20:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/8/upstream-check/cinder-cDOT-iSCSI/bda0262/ : SUCCESS in 35m 39s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/8/upstream-check/cinder-cDOT-NFS/f559185/ : SUCCESS in 34m 47s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/8/upstream-check/cinder-cDOT-FCP/8549589/ : SUCCESS in 51m 30s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"0a023d74996176c7ac7d840f674c47ed6a3428c6","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-09 14:20:37.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/828087/8/upstream-solidfire-check/cinder-solidfire-iSCSI/8eb54ec/ : SUCCESS in 1h 03m 39s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":8},{"id":"1acf34f5da32a62b6d9bf2913d80e54205fa520b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-02-09 16:25:47.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_SC_FC-os-brick/e472226/DellEMC_SC_FC-os-brick/29 : FAILURE in 1h 00m 38s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/87/828087/8/check/DellEMC_SC_iSCSI-os-brick/8058a20/DellEMC_SC_iSCSI-os-brick/29 : FAILURE in 1h 49m 09s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":8},{"id":"f9b423f68e23bf1ae3b372dc5c551756c3684f06","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-02-10 11:35:56.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1ba7070e1b87d67900952c7ff3b2f088b84de19c","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-14 15:28:07.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9e3f386553ebfe90be02744cbb72d94cb8e8e91f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-14 15:54:14.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5ef69e4fd6425ea736363b3f5f073430547c5c7b","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-14 19:37:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"40128013ca51aa30b81df5ebeff798f0bd234ca3","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2022-02-14 19:38:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bc0948a1a316847005ee3e318b0c8f5fc06b6f2d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-14 21:51:34.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7094a5d94437f868685903adb8ab73dd522bc52b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:52:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c762883e5ed2cd16e8ae366c36149fc03682e8bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 23:21:31.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cf590915bf54dc7a70fa09bdb0fffd5 : SUCCESS in 1h 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4add3a823af4d8d863d569934ba1e68 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b845aeaf8194441faa1eab2b096d7b54 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d723c1b403b6461fbef1ee1d20d9c471 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b72a07e9a0f4b4eb88bc22996c669de : SUCCESS in 7m 25s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e7c9b30af5c4fd59a09ba00cafc2b0d : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":8},{"id":"444a53847e30ffaf26c9809924210966e1c05783","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 23:21:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"8cb87efd06460a0018845de073b3ef5e7de73d8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 23:23:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c5dfba3cb1c4a879d7ec8d303c48522 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":8},{"id":"ad7bf653085d1770bc254989e6ac2baa71cd9736","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-15 04:01:22.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"71a6ea1f4df8b789ca0133a06c5b8dee7fdfa853","revisions":{"d2a5193275bd33b788f3833dd23fffbf5ea5bc7c":{"kind":"REWORK","_number":1,"created":"2022-02-07 08:15:05.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/1"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 08:14:31.000000000","tz":120},"subject":"Followup - Reuse get_host_nqn code","message":"Followup - Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nSigned-off-by: Yuval Brave  yuval@lightbitslabs.com\nDepends-On: I68f6cb9a7da56966827bb8683da44e6852549cc4\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d2a5193275bd33b788f3833dd23fffbf5ea5bc7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d2a5193275bd33b788f3833dd23fffbf5ea5bc7c"}]},"branch":"refs/heads/master"},"6fe39c3aeac5ea2b66eb50b230173fb5d548b51c":{"kind":"REWORK","_number":2,"created":"2022-02-07 08:26:13.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/2"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 08:26:05.000000000","tz":120},"subject":"Followup - Reuse get_host_nqn code","message":"Followup - Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nSigned-off-by: Yuval Brave  yuval@lightbitslabs.com\nDepends-On: I68f6cb9a7da56966827bb8683da44e6852549cc4\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6fe39c3aeac5ea2b66eb50b230173fb5d548b51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6fe39c3aeac5ea2b66eb50b230173fb5d548b51c"}]},"branch":"refs/heads/master"},"df2874daec486818986e77a56314192aa9ffa819":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-07 08:27:30.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/3"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2022-02-07 08:27:30.000000000","tz":0},"subject":"Followup - Reuse get_host_nqn code","message":"Followup - Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nSigned-off-by: Yuval Brave  yuval@lightbitslabs.com\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/df2874daec486818986e77a56314192aa9ffa819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/df2874daec486818986e77a56314192aa9ffa819"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5ee1197bba03066bd8bb0c3e19b15d5e6591f330":{"kind":"REWORK","_number":4,"created":"2022-02-08 14:38:25.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/4"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-08 14:35:31.000000000","tz":120},"subject":"Reuse get_host_nqn code","message":"Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nThis patch is a followup to change I2e86fa840490.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nDepends-On: I68f6cb9a7da56966827bb8683da44e6852549cc4\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5ee1197bba03066bd8bb0c3e19b15d5e6591f330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5ee1197bba03066bd8bb0c3e19b15d5e6591f330"}]},"branch":"refs/heads/master"},"77b4ea3a41dcf04177d345daa2d18ed7937d13ff":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-02-08 15:57:09.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/5"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-08 15:56:48.000000000","tz":120},"subject":"Reuse get_host_nqn code","message":"Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nThis patch is a followup to change I2e86fa840490.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/77b4ea3a41dcf04177d345daa2d18ed7937d13ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/77b4ea3a41dcf04177d345daa2d18ed7937d13ff"}]},"branch":"refs/heads/master"},"29659bcd5614f949235071642a7decd5f3f071ee":{"kind":"REWORK","_number":6,"created":"2022-02-08 16:36:14.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/6"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-08 16:36:03.000000000","tz":120},"subject":"Reuse get_host_nqn code","message":"Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nThis patch is a followup to change I2e86fa840490.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/29659bcd5614f949235071642a7decd5f3f071ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/29659bcd5614f949235071642a7decd5f3f071ee"}]},"branch":"refs/heads/master"},"ed1cec95127cabacac298afd308363b604796f6c":{"kind":"REWORK","_number":7,"created":"2022-02-09 09:41:12.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/7"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-09 09:40:59.000000000","tz":120},"subject":"Reuse get_host_nqn code","message":"Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nThis patch is a followup to change I2e86fa840490.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ed1cec95127cabacac298afd308363b604796f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ed1cec95127cabacac298afd308363b604796f6c"}]},"branch":"refs/heads/master"},"71a6ea1f4df8b789ca0133a06c5b8dee7fdfa853":{"kind":"REWORK","_number":8,"created":"2022-02-09 10:01:54.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/87/828087/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/828087/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/828087/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/828087/8"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-07 07:49:41.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2022-02-09 10:01:39.000000000","tz":120},"subject":"Reuse get_host_nqn code","message":"Reuse get_host_nqn code\n\n- The lightos connectors is making use of the get_host_nqn function in\n  nvmeof, in order to reduce code duplication I extracted the function\n  out of the nvmeof class.\n\n- The connection property \"hostnqn\" removed to only use single\n  key \"nqn\".\n  In the lightos connector there was a use of the key \"nqn\" in a\n  different context - so that was changed to \"subsysnqn\".\n\n- Lightos tests updated according the changes.\n\nThis patch is a followup to change I2e86fa840490.\n\nSigned-off-by: Yuval Brave  \u003cyuval@lightbitslabs.com\u003e\nChange-Id: Ia4d8c6fda875b4f4d32d511ca9282ca0fb5d6f12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/71a6ea1f4df8b789ca0133a06c5b8dee7fdfa853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/71a6ea1f4df8b789ca0133a06c5b8dee7fdfa853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
