)]}'
{"id":"openstack%2Fos-brick~828802","triplet_id":"openstack%2Fos-brick~master~Iec14833ab502b4bb342938d5477c57742068f1b8","project":"openstack/os-brick","branch":"master","topic":"update-min-req","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-02-14 18:33:22.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-02-14 15:35:08.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"Iec14833ab502b4bb342938d5477c57742068f1b8","subject":"Update requirements minima for Yoga release","status":"MERGED","created":"2022-02-10 22:17:32.000000000","updated":"2022-02-15 02:04:46.000000000","submitted":"2022-02-14 18:33:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828802-update-min-req","meta_rev_id":"eae8fb6055fc794ae788e765057a0f3b30923c20","_number":828802,"virtual_id_number":828802,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 18:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2022-02-14 16:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-02-14 15:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2022-02-14 16:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 22:33:10.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-10 23:12:59.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-02-11 00:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 01:50:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-11 05:40:05.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-11 05:47:50.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-02-11 05:57:10.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-02-11 07:35:31.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-02-11 08:34:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-11 13:17:14.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2022-02-11 14:18:17.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-02-14 13:52:12.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-02-14 16:24:13.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"5333e187f06a45bc0a3a00c177708df8aefc4cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-10 22:17:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"575ffc7af7ebc215390d10fc530f92b365fcdd89","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-10 22:33:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/02/828802/1/check/DellEMC_XtremIO_FC_os-brick/095e341/DellEMC_XtremIO_FC_os-brick/80 : FAILURE in 15m 05s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/02/828802/1/check/DellEMC_XtremIO_iSCSI_os-brick/cc58600/DellEMC_XtremIO_iSCSI_os-brick/77 : FAILURE in 15m 21s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"79b18c9ba195d792e8f03c4388217533fdeb4a37","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-10 23:12:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828802/1/iscsi/ : SUCCESS in 54m 17s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828802/1/smb/ : SUCCESS in 54m 55s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828802/1/ceph/ : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0bea772ce50a7c778c8fe4898e2cb1480b543da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 00:30:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ec43574a64a4a94a0cf19498e75303b : FAILURE in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1545820dfd9d44578e8fac7c0b6594d7 : SUCCESS in 1h 43m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce50f3731a542f0817eef075d337827 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21aa224d12c34ba983f2972627aceace : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e36aae37bb6a4bd4b687119b53f75839 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564026bad4384f20800a7378627d34c8 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89275070356c48a7afff8e3d124b7161 : SUCCESS in 4m 39s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/5e39d070528b46588b4a438d69a9375d : SUCCESS in 1h 40m 55s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5a374d731754690aa6c6fd2f464d866 : SUCCESS in 2h 06m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/95d4b45e587c4453b093fbe4c01b3fd7 : SUCCESS in 1h 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb35b5953dc4adcf95000e2df83abd33f0def47f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-11 01:50:27.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/1/check/devstack-plugin-ceph-dsvm-focal-py3/23a5edb/ : SUCCESS in 3h 06m 41s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/1/check/tempest-dsvm-full-focal-py3/e1489c8/ : SUCCESS in 2h 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"542a11e9b8689771f9becb258013c4f13b34d83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-11 04:03:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aeed02df54b71fd090066d28df922c8f7e93b1a0","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 04:55:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/02/828802/2/check/DellEMC_XtremIO_FC_os-brick/8d98484/DellEMC_XtremIO_FC_os-brick/81 : SUCCESS in 52m 11s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/02/828802/2/check/DellEMC_XtremIO_iSCSI_os-brick/fedb52a/DellEMC_XtremIO_iSCSI_os-brick/78 : FAILURE in 15m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"bfff10d42cf572bbd5f0084aa712bce229cbb1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 05:39:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da94f64238cd4fcf984091dcda4d25b4 : SUCCESS in 3m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15d66c8950454dec982b3bd50f6ac322 : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6799ce05534a648714775f4bfec837 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bee69a8abef543b8ab35a0caf520738c : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf6321bba9e4058b82bfe5510d74dac : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5483c827a3c40be9bd0d865efa47476 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ea958816e040e3a049e86b3f0cbc7a : SUCCESS in 5m 54s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/3b1933c6cc6946eab7b06ea969da9d61 : SUCCESS in 1h 10m 28s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02c717eb2a7e468ab8615caee815f40d : SUCCESS in 1h 33m 27s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5b9db26564b34057afa59c439862bb15 : SUCCESS in 1h 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"821a57cfefe7b9c84df5d374bdeea556abc1abcf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 05:40:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/02/828802/2/check/DellEMC_PowerFlex_os-brick/66e261f/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"4583f994013226fbfe99ed7719fc0a96acf2579c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-11 05:47:50.000000000","message":"Patch Set 2:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/11_02_2022_05_39_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"b8f910acf071d20c6ed06a8c2d908922db2936f3","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-11 05:57:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-02-828802-2/1644557992 : SUCCESS in 17m 16s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"dde790b1f329d23147165eeb560c949962bcf0aa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-11 06:15:05.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828802/2/iscsi/ : SUCCESS in 1h 14m 56s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828802/2/smb/ : SUCCESS in 1h 16m 10s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828802/2/ceph/ : FAILURE in 1h 08m 30s","accounts_in_message":[],"_revision_number":2},{"id":"de903831fa2ef1166e8db4e91f9424f2f8a458ed","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-11 06:32:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/02/828802/2/check/DellEMC_XtremIO_FC_os-brick/d5a2309/DellEMC_XtremIO_FC_os-brick/82 : SUCCESS in 52m 46s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/02/828802/2/check/DellEMC_XtremIO_iSCSI_os-brick/3545889/DellEMC_XtremIO_iSCSI_os-brick/79 : FAILURE in 15m 42s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":2},{"id":"abd28df1f0c0d622ab2254244dc2e818d42fcb2c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-11 07:20:45.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/2/check/devstack-plugin-ceph-dsvm-focal-py3/ea6f8d0/ : TIMED_OUT in 3h 09m 24s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/2/check/tempest-dsvm-full-focal-py3/5c78fcc/ : SUCCESS in 2h 20m 14s","accounts_in_message":[],"_revision_number":2},{"id":"38af32e6a8fd475188b67fb6cc0b5945a660c3ec","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-11 07:35:31.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/2/upstream-solidfire-check/cinder-solidfire-iSCSI/ce4ed3a/ : SUCCESS in 1h 00m 07s\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"def8b63bad3060c9a5db43a8eb63db01e0eb67f9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-11 08:34:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/2/upstream-check/cinder-cDOT-iSCSI/9085a17/ : SUCCESS in 35m 38s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/2/upstream-check/cinder-cDOT-NFS/f6b9896/ : FAILURE in 7m 19s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/2/upstream-check/cinder-cDOT-FCP/6649f7e/ : SUCCESS in 55m 26s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"9c66db913404838525b02399e4a49999c25d4d7f","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-02-11 13:17:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"543873b49f49a626a155879f183c20537a8163ce","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-11 14:18:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a55bff0e40237c0a20930607ab762b0a39dec60","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-14 13:52:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11558023ab44d676280756fb236d6fa762f1c532","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-14 14:33:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"87cbb661ebdc83d1bdf1f89899672058c600ff5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-02-14 14:34:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5b426e1ee8724ec02f35298b46fa18fcef8102e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-02-14 15:35:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"693afadef8bcc4abb3732919c9ed218156a757a9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-02-14 15:46:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/828802/3/iscsi/ : SUCCESS in 1h 00m 25s\n- os-brick-smb http://cloudbase-ci.com/os-brick/828802/3/smb/ : SUCCESS in 1h 12m 10s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/828802/3/ceph/ : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":3},{"id":"40a4959e081d22019e4af62e95b19851a5e707b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:18:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/634179d727634fffa57ffebc6b97a4be : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62c6c8208def4a039e84ac78290a87ff : SUCCESS in 1h 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b904e36ee13548e99f91350b40b7b4ad : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a9e65c280bb400e9be2a37d3db6af99 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27f4b597492c44588ff2c3c3b8139e83 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237247acd07b4b0db8b488baaddd2478 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/962be3bdf5f94d5396579d559eb44d47 : SUCCESS in 5m 21s\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/ed17ad938505401080bae274bd22a328 : SUCCESS in 1h 13m 39s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/539951c2977d429293689f273225c6bb : SUCCESS in 1h 43m 02s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/663f123c2930483688d134f13ab9350e : SUCCESS in 1h 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"00304563fa4ab77daffeb74cf6b1b88ca91bf490","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-14 16:19:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/02/828802/3/check/DellEMC_PowerFlex_os-brick/0c2edaa/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"84691255d67ae1f9514499211c3cc8fa42941420","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-02-14 16:19:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/3/upstream-solidfire-check/cinder-solidfire-iSCSI/e15df3e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"8ef0ee93af6f1eecd6b930aeeb6bb61b0f9e622a","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-02-14 16:24:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"213d11549ab4e638b12f1e6ea2686184297a8742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 16:25:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0af729407c93c6b97f1c430bca256a119e9aecb8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-02-14 16:26:56.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/14_02_2022_16_19_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"dd9be751fd408effd41a689dafd5925623c959dc","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-02-14 16:35:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-02-828802-3/1644855532 : SUCCESS in 16m 37s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"89a6bfa336bc07c4ce47284a6e719ee1fcc85f95","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-14 18:01:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-cDOT-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/3/upstream-check/cinder-cDOT-iSCSI/817d57b/ : FAILURE in 25m 14s\n- cinder-cDOT-NFS https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/3/upstream-check/cinder-cDOT-NFS/6e0622d/ : SUCCESS in 1h 36m 19s\n- cinder-cDOT-FCP https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/02/828802/3/upstream-check/cinder-cDOT-FCP/d5ab594/ : FAILURE in 1h 41m 33s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b787e642422c16bd5ab25e5e1e6eb88e71356601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:33:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0a6a0575f6940f8a81ce25ec7dc8319 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79d7d897b9134e9fab463e8c6d8adc00 : SUCCESS in 1h 09m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512e4ddecaf04b088342654e91819c27 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc0b002d8b294c78ab01c01911f37f1d : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb534ae87a6d48578f035567801e03cc : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c15cbfb6464faa9728620d69247294 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37d2cc0b47b8446a96ec0df2b10c0fa5 : SUCCESS in 4m 54s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5f6b781a74549579d079fc46c5221bb : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a799f3fcff53d60b30f7b3566228e531c8ec1678","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:33:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0284aa4a403216d8e58ed4ea7dbe167a1a404a79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 18:37:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f766fe3a21c244a3842353cab6e19273 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a578cefd367424b89a8e77928888b53 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a47baa7504fe870e0a4e61e7b0554d369a9d50fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-14 19:04:44.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/3/check/devstack-plugin-ceph-dsvm-focal-py3/a3fddb4/ : SUCCESS in 3h 03m 53s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/02/828802/3/check/tempest-dsvm-full-focal-py3/a4409cb/ : SUCCESS in 2h 20m 44s","accounts_in_message":[],"_revision_number":3},{"id":"eae8fb6055fc794ae788e765057a0f3b30923c20","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-02-15 02:04:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/02/828802/3/check/DellEMC_XtremIO_FC_os-brick/fe96bd2/DellEMC_XtremIO_FC_os-brick/94 : SUCCESS in 53m 46s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/02/828802/3/check/DellEMC_XtremIO_iSCSI_os-brick/8d8813f/DellEMC_XtremIO_iSCSI_os-brick/91 : FAILURE in 15m 45s\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8fb0282630d096902114c933a52c1feeac06fdd9","revisions":{"2e48134df02b41c7dc5572b2f41e310d7675def7":{"kind":"REWORK","_number":1,"created":"2022-02-10 22:17:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/02/828802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/02/828802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/02/828802/1"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-10 22:14:31.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-10 22:14:31.000000000","tz":-300},"subject":"Update requirements minima for Yoga release","message":"Update requirements minima for Yoga release\n\nHad to raise min oslo.vmware to 3.10.0 or can\u0027t run on py3.9.\nSo while I was in there, looked at the other requirements.\nUpdated most of the openstack libraries, left the third party\nones alone (see notes below).\n\nrequirements\n------------\neventlet\u003e\u003d0.30.1          -\u003e \u003e\u003d0.30.1,!\u003d0.32.0    using 0.33.0\n                                                         for 3 months\npbr\u003e\u003d5.5.1                -\u003e \u003e\u003d5.8.0              in u-c for 3 months\noslo.concurrency\u003e\u003d4.4.0   -\u003e \u003e\u003d4.5.0              in u-c for 4 months\noslo.context\u003e\u003d3.1.1       -\u003e \u003e\u003d3.4.0              in u-c for 4 months\noslo.log\u003e\u003d4.4.0           -\u003e \u003e\u003d4.6.1              in u-c for 4 months\noslo.i18n\u003e\u003d5.0.1          -\u003e \u003e\u003d5.1.0              in u-c for 5 months\noslo.privsep\u003e\u003d2.4.0       -\u003e \u003e\u003d2.6.2              in u-c for 6 months\noslo.serialization\u003e\u003d4.1.0 -\u003e \u003e\u003d4.2.0              in u-c for 6 months\noslo.service\u003e\u003d2.5.0       -\u003e \u003e\u003d2.8.0              in u-c for 1 month\noslo.utils\u003e\u003d4.8.0         -\u003e \u003e\u003d4.12.1             in u-c for 3 weeks\nos-win\u003e\u003d5.4.0             -\u003e \u003e\u003d5.5.0              in u-c for 6 months\n\neventlet \u003e\u003d0.30.1  -\u003e  \u003e\u003d0.30.1,!\u003d0.32.0\n- 0.33.0 has been in u-c for about 3 months\n- 0.32.0 breaks ironic\n- no reason to raise the min right now\n\nrequests \u003e\u003d2.25.1  -\u003e  unchanged\n- 2.27.1 has been in u-c for 1 month\n- 2.26.0 drops py3.5 support\n- no reason to raise the min right now\n\ntest requirements\n-----------------\ncastellan\u003e\u003d3.7.0          -\u003e \u003e\u003d3.10.0             in u-c for 4 months\noslo.vmware\u003e\u003d3.8.0        -\u003e \u003e\u003d3.10.0             in u-c for 4 months\noslotest\u003e\u003d4.4.1           -\u003e \u003e\u003d4.5.0              in u-c for 5 months\nstestr\u003e\u003d3.1.0             -\u003e \u003e\u003d3.2.1              in u-c for 4 months\n\ncoverage\u003e\u003d5.5  -\u003e  coverage\u003e\u003d5.5,\u003c6.3;python_version\u003d\u003d\u00273.6\u0027\n                   coverage\u003e\u003d5.5\n- 6.3: stopped supporting py3.6\n- current is 6.3.1\n- no reason to raise the min right now\n\nChange-Id: Iec14833ab502b4bb342938d5477c57742068f1b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2e48134df02b41c7dc5572b2f41e310d7675def7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2e48134df02b41c7dc5572b2f41e310d7675def7"}]},"branch":"refs/heads/master"},"45ef02d3136bee7a10a7dbc4ea3dc03f5287d5d3":{"kind":"REWORK","_number":2,"created":"2022-02-11 04:03:21.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/02/828802/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/02/828802/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/02/828802/2"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-10 22:14:31.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-11 04:02:50.000000000","tz":-300},"subject":"Update requirements minima for Yoga release","message":"Update requirements minima for Yoga release\n\nHad to raise min oslo.vmware to 3.10.0 or can\u0027t run on py3.9.\nSo while I was in there, looked at the other requirements.\nUpdated most of the openstack libraries, left the third party\nones alone (see notes below).\n\nrequirements\n------------\npbr\u003e\u003d5.5.1                -\u003e \u003e\u003d5.8.0      in u-c for 3 months\noslo.concurrency\u003e\u003d4.4.0   -\u003e \u003e\u003d4.5.0      in u-c for 4 months\noslo.context\u003e\u003d3.1.1       -\u003e \u003e\u003d3.4.0      in u-c for 4 months\noslo.log\u003e\u003d4.4.0           -\u003e \u003e\u003d4.6.1      in u-c for 4 months\noslo.i18n\u003e\u003d5.0.1          -\u003e \u003e\u003d5.1.0      in u-c for 5 months\noslo.privsep\u003e\u003d2.4.0       -\u003e \u003e\u003d2.6.2      in u-c for 6 months\noslo.serialization\u003e\u003d4.1.0 -\u003e \u003e\u003d4.2.0      in u-c for 6 months\noslo.service\u003e\u003d2.5.0       -\u003e \u003e\u003d2.8.0      in u-c for 1 month\noslo.utils\u003e\u003d4.8.0         -\u003e \u003e\u003d4.12.1     in u-c for 3 weeks\nos-win\u003e\u003d5.4.0             -\u003e \u003e\u003d5.5.0      in u-c for 6 months\n\neventlet \u003e\u003d0.30.1  -\u003e  \u003e\u003d0.30.1,!\u003d0.32.0\n- 0.33.0 has been in u-c for about 3 months\n- 0.32.0 breaks ironic\n- no reason to raise the min right now\n\nrequests \u003e\u003d2.25.1  -\u003e  unchanged\n- 2.27.1 has been in u-c for 1 month\n- 2.26.0 drops py3.5 support\n- no reason to raise the min right now\n\ntest requirements\n-----------------\ncastellan\u003e\u003d3.7.0          -\u003e \u003e\u003d3.10.0     in u-c for 4 months\noslo.vmware\u003e\u003d3.8.0        -\u003e \u003e\u003d3.10.0     in u-c for 4 months\noslotest\u003e\u003d4.4.1           -\u003e \u003e\u003d4.5.0      in u-c for 5 months\nstestr\u003e\u003d3.1.0             -\u003e \u003e\u003d3.2.1      in u-c for 4 months\n\ncoverage\u003e\u003d5.5  -\u003e  unchanged\n- 6.3: stopped supporting py3.6; u-c caps py3.6 at 6.2\n- current is 6.3.1 which is also the u-c cap for py3.8\n- no reason to raise the min right now\n\nChange-Id: Iec14833ab502b4bb342938d5477c57742068f1b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/45ef02d3136bee7a10a7dbc4ea3dc03f5287d5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/45ef02d3136bee7a10a7dbc4ea3dc03f5287d5d3"}]},"branch":"refs/heads/master"},"8fb0282630d096902114c933a52c1feeac06fdd9":{"kind":"REWORK","_number":3,"created":"2022-02-14 14:34:13.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/02/828802/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/02/828802/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/02/828802/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/02/828802/3"}}},"commit":{"parents":[{"commit":"e761ab64230faf9b210a1457ceca592a51202fd6","subject":"Merge \"Prevent CI from running on irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e761ab64230faf9b210a1457ceca592a51202fd6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-10 22:14:31.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-02-14 14:34:06.000000000","tz":-300},"subject":"Update requirements minima for Yoga release","message":"Update requirements minima for Yoga release\n\nHad to raise min oslo.vmware to 3.10.0 or can\u0027t run on py3.9.\nSo while I was in there, looked at the other requirements.\nUpdated most of the openstack libraries, left the third party\nones alone (see notes below).\n\nrequirements\n------------\npbr\u003e\u003d5.5.1                -\u003e \u003e\u003d5.8.0      in u-c for 3 months\noslo.concurrency\u003e\u003d4.4.0   -\u003e \u003e\u003d4.5.0      in u-c for 4 months\noslo.context\u003e\u003d3.1.1       -\u003e \u003e\u003d3.4.0      in u-c for 4 months\noslo.log\u003e\u003d4.4.0           -\u003e \u003e\u003d4.6.1      in u-c for 4 months\noslo.i18n\u003e\u003d5.0.1          -\u003e \u003e\u003d5.1.0      in u-c for 5 months\noslo.privsep\u003e\u003d2.4.0       -\u003e \u003e\u003d2.6.2      in u-c for 6 months\noslo.serialization\u003e\u003d4.1.0 -\u003e \u003e\u003d4.2.0      in u-c for 6 months\noslo.service\u003e\u003d2.5.0       -\u003e \u003e\u003d2.8.0      in u-c for 1 month\noslo.utils\u003e\u003d4.8.0         -\u003e \u003e\u003d4.12.1     in u-c for 3 weeks\nos-win\u003e\u003d5.4.0             -\u003e \u003e\u003d5.5.0      in u-c for 6 months\n\neventlet \u003e\u003d0.30.1  -\u003e  \u003e\u003d0.30.1,!\u003d0.32.0\n- 0.33.0 has been in u-c for about 3 months\n- 0.32.0 breaks ironic\n- no reason to raise the min right now\n\nrequests \u003e\u003d2.25.1  -\u003e  unchanged\n- 2.27.1 has been in u-c for 1 month\n- 2.26.0 drops py3.5 support\n- no reason to raise the min right now\n\ntest requirements\n-----------------\ncastellan\u003e\u003d3.7.0          -\u003e \u003e\u003d3.10.0     in u-c for 4 months\noslo.vmware\u003e\u003d3.8.0        -\u003e \u003e\u003d3.10.0     in u-c for 4 months\noslotest\u003e\u003d4.4.1           -\u003e \u003e\u003d4.5.0      in u-c for 5 months\nstestr\u003e\u003d3.1.0             -\u003e \u003e\u003d3.2.1      in u-c for 4 months\n\ncoverage\u003e\u003d5.5  -\u003e  unchanged\n- 6.3: stopped supporting py3.6; u-c caps py3.6 at 6.2\n- current is 6.3.1 which is also the u-c cap for py3.8\n- no reason to raise the min right now\n\nChange-Id: Iec14833ab502b4bb342938d5477c57742068f1b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8fb0282630d096902114c933a52c1feeac06fdd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8fb0282630d096902114c933a52c1feeac06fdd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
