)]}'
{"id":"openstack%2Fos-brick~836056","triplet_id":"openstack%2Fos-brick~master~Idbba6f9a6d928aa94d2920fc1bf8b6e2a0626a07","project":"openstack/os-brick","branch":"master","topic":"nvme-2","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-08-23 14:21:34.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-08-23 23:21:14.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-08-23 23:21:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbba6f9a6d928aa94d2920fc1bf8b6e2a0626a07","subject":"NVMe-oF: read mdstat in Python","status":"MERGED","created":"2022-03-31 17:58:21.000000000","updated":"2022-08-26 14:14:03.000000000","submitted":"2022-08-23 23:21:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836056-nvme-2","meta_rev_id":"3a63648655f5f4b4098f9cf206b77e928abe39c3","_number":836056,"virtual_id_number":836056,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 23:21:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-23 21:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-23 14:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-23 15:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-08-23 21:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 19:19:56.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-03-31 20:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 20:12:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-31 20:12:35.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-03-31 22:16:17.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-31 23:06:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-01 00:03:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-01 06:59:05.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-04-01 19:33:06.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-04-11 16:56:20.000000000","updated_by":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-23 05:15:30.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-08-17 14:39:42.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-17 17:20:25.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-08-17 17:38:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-18 01:17:58.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-08-23 15:26:44.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-08-23 21:18:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-23 23:21:15.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"bd1ec5983e98398a161d9a060b57af3e669f1b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-03-31 17:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27b8edb9ea4ee0482e02b3603f20fb253e6ce225","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-03-31 19:19:56.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/836056/1/iscsi/ : SUCCESS in 1h 19m 57s\n- os-brick-smb http://cloudbase-ci.com/os-brick/836056/1/smb/ : SUCCESS in 1h 19m 35s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/836056/1/ceph/ : FAILURE in 1h 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6bf732209358497ca3f57d9e162e4cdcb80bd73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 20:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/180e55c634c04a82bb53e4ba15383660 : SUCCESS in 1h 52m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2370994434b84311a3ae2b36aa643aab : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b66ca75c9941268313567d211866e7 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eec9a91cbf3409daf3f6c651548559a : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c65ce2fa674ec08c47ea3d7a97f61f : SUCCESS in 6m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a249421a294243bb81d64093d0c26602 : SUCCESS in 4m 53s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/2b5bfba677d6491e97b45540f600583c : SUCCESS in 1h 21m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0849723c505e4da9966d42fb99112532 : SUCCESS in 1h 58m 31s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1c6b963c767a4d729ac435b4b8c6f3d0 : SUCCESS in 1h 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d996592ac2e0a6b92a5a6d1365f60092dd31bf7e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-03-31 20:12:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_PowerFlex_os-brick/091533e/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"56d237dfd9d94e5a9b5fc9dd527ed81aa2bb6978","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-03-31 20:12:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-solidfire-iSCSI https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/836056/1/upstream-solidfire-check/cinder-solidfire-iSCSI/85fa404/ : NOT_REGISTERED\n\nComment \u0027run-NetApp Solidfire CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"8edfcd14bccddfd3d8f969d4efed4c2059eabf21","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-03-31 22:16:17.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5b30a68e70a34138bd65f2fefd923c34 : TIMED_OUT in 3h 10m 46s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b30471f7e40545268d2f12640ed9f414 : POST_FAILURE in 2h 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6c3ae7973a53cb056b5c9f38e7e3c0d1abe1415d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-31 23:06:35.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_Unity_FC_os-brick/d6d0689/DellEMC_Unity_FC_os-brick/35 : SUCCESS in 1h 09m 09s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_Unity_iSCSI_os-brick/d81f1fd/DellEMC_Unity_iSCSI_os-brick/47 : SUCCESS in 2h 02m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"af99d7119075306bbddcedf73895b3975d183535","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-01 00:03:40.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/31_03_2022_23_34_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"7fdb88f461d8984f1ef78b0f335ea0546c5afd95","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-04-01 06:59:05.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_PowerStore_FC-ob/d053926/DellEMC_PowerStore_FC-ob/45 : FAILURE in 46m 27s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_PowerStore_iSCSI-ob/e705bcb/DellEMC_PowerStore_iSCSI-ob/42 : FAILURE in 48m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9cf5d7d67504ba784c17a63b176a70315ea6ceb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-04-01 19:33:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_SC_FC-os-brick/29ec120/DellEMC_SC_FC-os-brick/41 : SUCCESS in 59m 40s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_SC_iSCSI-os-brick/7fcce2c/DellEMC_SC_iSCSI-os-brick/51 : FAILURE in 1h 59m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"c56f745f9ee1be63a59ee47f23893b61a073fbb4","author":{"_account_id":12033,"name":"DellEMC XtremIO CI","email":"emc.xio.ci@emc.com","username":"emc-xio-ci","tags":["SERVICE_USER"]},"date":"2022-04-11 16:56:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_XtremIO_FC_os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_XtremIO_FC_os-brick/83263dd/DellEMC_XtremIO_FC_os-brick/107 : SUCCESS in 1h 11m 58s\n- DellEMC_XtremIO_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/56/836056/1/check/DellEMC_XtremIO_iSCSI_os-brick/2b069d1/DellEMC_XtremIO_iSCSI_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC XtremIO CI\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_XtremIO_CI","accounts_in_message":[],"_revision_number":1},{"id":"646bd7c48c31a972d172ef9b4fed6a29eae5fde8","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-05-23 05:15:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12df48f2b3ee2fdc68d26b7f49358dd8825d272c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-14 10:42:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e01a9d673edb74c7d9b4ef69ea4874debd26480a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-17 14:39:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18c3d8b8003d3d7e8b09cb5ab41b1f05743d0724","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-17 14:39:42.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660747179770 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"af7e1d7cb4b44b53c54fbe8a96e13ba0d89d6096","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-17 15:45:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/836056/2/iscsi/ : SUCCESS in 1h 05m 37s\n- os-brick-smb http://cloudbase-ci.com/os-brick/836056/2/smb/ : SUCCESS in 1h 04m 39s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/836056/2/ceph/ : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b7d866edd75866fd544a05a83d6776d40383ddb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 16:28:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a35c4e6c704a3bbebfd46da1a06eb4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c061c2ede004c2f82f09ddb53b4c75d : SUCCESS in 1h 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91d8c4fea204769bf0c77ee02726a3e : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7338294c0784a2dbd77940048c82d86 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22af9cd50d3b499ab4ff2e386ac459f4 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe3c16d3d6db444e9e3bc7ef798afd7f : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7191a350ab4940b4b28763d323edb5c0 : SUCCESS in 5m 41s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4ad3d48eb6a642b18cc99521da171e03 : SUCCESS in 4m 11s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/c3af3773b04443b2919d3d58e6f1f49d : SUCCESS in 1h 28m 48s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/270f4e32637a418e8de6e29cbc90ff42 : SUCCESS in 1h 43m 02s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0a7606bf82a442b289bd5788ee72b5c7 : SUCCESS in 1h 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"111ba60917613050c16fb7de212f056b0c4e10c1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-17 16:28:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_PowerFlex_os-brick/26a2660/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8ddd631324a71f77d55e1f4dbb7dd3ee4d8d83b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-17 16:50:45.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_08_2022_16_43_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a13be83908686b013a0ddfc4b26d0a68336e9b8","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-08-17 17:20:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-56-836056-2/1660753709 : SUCCESS in 30m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"a9466c8b8efec57a1c9e2d90fd23dc1520843de9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-17 17:38:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/836056/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8d2b0a7/ : SUCCESS in 1h 36m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/836056/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/562033d/ : SUCCESS in 1h 36m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/56/836056/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1ed69f0/ : SUCCESS in 1h 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"632499e198c0b6c863ea92dd41593cc697a1097e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-17 21:16:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_PowerStore_FC-ob/a61993b/DellEMC_PowerStore_FC-ob/150 : SUCCESS in 1h 28m 15s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_PowerStore_NFS-ob/2dd2159/DellEMC_PowerStore_NFS-ob/60 : SUCCESS in 1h 22m 01s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_PowerStore_NVMeTCP-ob/1966b2a/DellEMC_PowerStore_NVMeTCP-ob/27 : SUCCESS in 1h 47m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"b29e265b49a9847451131d21e3c52af990aee45c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-08-18 01:17:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dc36d377cdaf8091318c3fed7c72ecfbdfe0c93f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-18 20:09:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_Unity_FC_os-brick/62de536/DellEMC_Unity_FC_os-brick/179 : FAILURE in 18m 39s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_Unity_iSCSI_os-brick/686366f/DellEMC_Unity_iSCSI_os-brick/197 : FAILURE in 1h 57m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"53e33413a012231dd90987d8fd83ea1b68262310","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-21 17:23:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_SC_FC-os-brick/4350a3e/DellEMC_SC_FC-os-brick/129 : SUCCESS in 1h 01m 37s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/56/836056/2/check/DellEMC_SC_iSCSI-os-brick/3fd4f65/DellEMC_SC_iSCSI-os-brick/141 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"a0bdf3333f84dcbc55a3a63c2415ce9f752f3b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-23 10:18:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c824ae2ca6fd890e0d93c22a7cc704a6cf13aa53","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-23 10:18:59.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8ebd8754da032b419fd19e3af5bde351193f9d36","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-23 10:53:27.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661249930086 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2395baf489502d47f72f38b1963eb8ac55db60fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 11:53:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee2527e68a14c57a090e79764505adb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82f5c3dea5da49049bd006492aaba8ef : SUCCESS in 1h 27m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26af8f10e15d44ce9d7a14ae96355868 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd086d2cb50d4d9ea24850305a0c1bc2 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90272fb1588b4dfd8deebecd053bc0fe : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/976c4d727020406fb76e9b71ee454816 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b8c5ce3a724eaeb9178e9cd195d461 : SUCCESS in 7m 49s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/77c048b6d6f642c1b3313eec8e4af86b : SUCCESS in 5m 11s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/12951ffb644a4ebbbf54410247366f76 : SUCCESS in 1h 20m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0a3d662d0b44ea5b148f7c59b132c75 : SUCCESS in 1h 33m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cc435b81fff147d7aa7323fd74390b2a : SUCCESS in 1h 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7028e84fe36dff32742b4bb3a0445faf9ef9f90","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-23 11:54:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_PowerFlex_os-brick/0e4fd7f/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"82afeab7efab59c5abb64f9f4099a98aa44407ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-23 12:55:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/56/836056/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a49ae89/ : SUCCESS in 1h 29m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/56/836056/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3268f01/ : SUCCESS in 1h 51m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/56/836056/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/0debf7c/ : SUCCESS in 1h 35m 33s","accounts_in_message":[],"_revision_number":3},{"id":"026d2d4fd93fafb868129aab2bc56aec5a0edaf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-23 12:59:30.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/23_08_2022_12_23_06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"2956cf715bfc44930f9a5564df459956f734e906","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-08-23 14:21:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23cea990e4f377d494827898184475930fa2ec5c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-23 15:04:30.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/56/836056/3/check/devstack-plugin-ceph-dsvm-focal-py3/598c586/ : TIMED_OUT in 3h 11m 17s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/56/836056/3/check/tempest-dsvm-full-focal-py3/1605345/ : SUCCESS in 2h 04m 17s","accounts_in_message":[],"_revision_number":3},{"id":"658bf0c24abb0c892e88c7f43ba140745fb59b5f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-23 15:26:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30284404e1ede4d77c3b53eb60dc19738996b86a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-23 15:42:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_PowerStore_FC-ob/77cf95f/DellEMC_PowerStore_FC-ob/169 : SUCCESS in 1h 16m 22s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_PowerStore_NFS-ob/792c9fa/DellEMC_PowerStore_NFS-ob/65 : SUCCESS in 1h 31m 48s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_PowerStore_NVMeTCP-ob/6b5eff2/DellEMC_PowerStore_NVMeTCP-ob/46 : FAILURE in 1h 41m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"bd10052bbb42891312c7b858cbfb4c3801c9343c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-23 16:39:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/836056/3/iscsi/ : SUCCESS in 1h 04m 47s\n- os-brick-smb http://cloudbase-ci.com/os-brick/836056/3/smb/ : SUCCESS in 1h 09m 24s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/836056/3/ceph/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":3},{"id":"44e58c4f7a801cd84b7fa32d3774948a924d5459","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-23 21:18:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"57ea3103da4fa7efa2e09a0fe94fc5ee5f6d6351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 21:19:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1895c503aefadafca20e812c5ae05722ea59badf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-23 22:59:32.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_Unity_FC_os-brick/0f35a11/DellEMC_Unity_FC_os-brick/215 : FAILURE in 14m 26s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/56/836056/3/check/DellEMC_Unity_iSCSI_os-brick/788f539/DellEMC_Unity_iSCSI_os-brick/230 : SUCCESS in 1h 45m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d28de045f8a37a080e4c1a253c238183d417bd70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 23:21:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7cc8d40eaf464187556ad3064dba2c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37b331445e624640827ab6b4140c3d47 : SUCCESS in 2h 00m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485e1af5448c456ba14586df9678a5dc : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0b5a0cf586d414a818ece348b4b744e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a6171dd71e147d394d5e4cdc02bc9cd : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a5d37373c741f7b73caade6a968752 : SUCCESS in 6m 31s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7e8e8b41c00444fe9c4d2d212be95bac : SUCCESS in 1h 54m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ce3460acdad36d45788297e2392170a18fec7fa2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 23:21:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b2f49ae71662d43de739a4c7a889c617dac49363","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-23 23:21:15.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"18b892130eba43293bf5c784bd751cbbc76bbb81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 23:22:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af3bf874ec6147b18167a5ef366e83c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96516300f1484ab082eefb86d2c0fd21 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3},{"id":"3a63648655f5f4b4098f9cf206b77e928abe39c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-26 14:14:03.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/482 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da2bfa41f5418cfe3845aed92f2de2c3cfe1947f","revisions":{"9aecf31d9ca835718b6313dd4c850d76b6304a0e":{"kind":"REWORK","_number":1,"created":"2022-03-31 17:58:21.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/56/836056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/56/836056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/56/836056/1"}}},"commit":{"parents":[{"commit":"56bf0272b55dcbbc7f5b03150973a80af1407f4f","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/56bf0272b55dcbbc7f5b03150973a80af1407f4f"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-03-17 13:21:24.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-03-17 16:09:57.000000000","tz":60},"subject":"NVMe-oF: read mdstat in Python","message":"NVMe-oF: read mdstat in Python\n\nThe get_md_name uses the contents of mdstat using cat and piping it to\ngrep and then to awk using a privileged shell, but there\u0027s no need to do\nthat.\n\nActually the mdstat file ha 0444 access, so we can read it in Python\ndirectly instead of having to call the privsep daemon to create the\nsubprocess with the shell.\n\nThis patch changes the current implementation of the method to do the\nreading and searching in Python.\n\nChange-Id: Idbba6f9a6d928aa94d2920fc1bf8b6e2a0626a07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9aecf31d9ca835718b6313dd4c850d76b6304a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9aecf31d9ca835718b6313dd4c850d76b6304a0e"}]},"branch":"refs/heads/master"},"763d8cd1de1da0dec2538f8de1e9bf28159483e7":{"kind":"REWORK","_number":2,"created":"2022-08-17 14:39:31.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/56/836056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/56/836056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/56/836056/2"}}},"commit":{"parents":[{"commit":"b41b3c8814b2a557222d3bbbabcb9e6dc342c970","subject":"Merge \"Add flake8-logging-format extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b41b3c8814b2a557222d3bbbabcb9e6dc342c970"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-03-17 13:21:24.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-17 14:22:40.000000000","tz":120},"subject":"NVMe-oF: read mdstat in Python","message":"NVMe-oF: read mdstat in Python\n\nThe get_md_name uses the contents of mdstat using cat and piping it to\ngrep and then to awk using a privileged shell, but there\u0027s no need to do\nthat.\n\nActually the mdstat file has 0444 access, so we can read it in Python\ndirectly instead of having to call the privsep daemon to create the\nsubprocess with the shell.\n\nThis patch changes the current implementation of the method to do the\nreading and searching in Python.\n\nChange-Id: Idbba6f9a6d928aa94d2920fc1bf8b6e2a0626a07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/763d8cd1de1da0dec2538f8de1e9bf28159483e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/763d8cd1de1da0dec2538f8de1e9bf28159483e7"}]},"branch":"refs/heads/master"},"da2bfa41f5418cfe3845aed92f2de2c3cfe1947f":{"kind":"REWORK","_number":3,"created":"2022-08-23 10:18:39.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/56/836056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/56/836056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/56/836056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/56/836056/3"}}},"commit":{"parents":[{"commit":"b41b3c8814b2a557222d3bbbabcb9e6dc342c970","subject":"Merge \"Add flake8-logging-format extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b41b3c8814b2a557222d3bbbabcb9e6dc342c970"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-03-17 13:21:24.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 10:18:17.000000000","tz":120},"subject":"NVMe-oF: read mdstat in Python","message":"NVMe-oF: read mdstat in Python\n\nThe get_md_name uses the contents of mdstat using cat and piping it to\ngrep and then to awk using a privileged shell, but there\u0027s no need to do\nthat.\n\nActually the mdstat file has 0444 access, so we can read it in Python\ndirectly instead of having to call the privsep daemon to create the\nsubprocess with the shell.\n\nThis patch changes the current implementation of the method to do the\nreading and searching in Python.\n\nChange-Id: Idbba6f9a6d928aa94d2920fc1bf8b6e2a0626a07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/da2bfa41f5418cfe3845aed92f2de2c3cfe1947f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/da2bfa41f5418cfe3845aed92f2de2c3cfe1947f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
