)]}'
{"id":"openstack%2Fos-brick~837358","triplet_id":"openstack%2Fos-brick~master~I0d5f7b19f4077546e2e7a46af07a791315450597","project":"openstack/os-brick","branch":"master","topic":"privsep","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-07 17:07:38.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I0d5f7b19f4077546e2e7a46af07a791315450597","subject":"WIP: privsep: Move linuxscsi initiator to privsep","status":"ABANDONED","created":"2022-04-11 15:37:25.000000000","updated":"2025-12-12 21:42:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"9d130c65050504d1236fa2ebc3d126525c9c1d06","_number":837358,"virtual_id_number":837358,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-23 12:32:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 15:56:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-11 18:31:53.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-04-11 18:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 03:01:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"807d1ad76ccd00a60bb91b604e25fb21b42d6a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f14a4e7524e8909b399098af4c80e634a3ae91","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 15:56:18.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/58/837358/1/check/DellEMC_Unity_FC_os-brick/e118664/DellEMC_Unity_FC_os-brick/56 : FAILURE in 1m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/58/837358/1/check/DellEMC_Unity_iSCSI_os-brick/4a20d7a/DellEMC_Unity_iSCSI_os-brick/68 : FAILURE in 1m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2591d3d5df8cfa51f9c633767eccd047ee938552","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-11 18:31:53.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837358/1/iscsi/ : SUCCESS in 1h 12m 27s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837358/1/smb/ : SUCCESS in 1h 17m 00s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837358/1/ceph/ : FAILURE in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c358e0a98b158cf97e3bca1efe6ac3e0cc63a211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:32:54.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6a5080792ea94fb89d427118ca386636 : SUCCESS in 3m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b206eb33e73f4dfaaf3cbea0a7f6d74d : FAILURE in 1h 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c318bf00b674f0586e7903dfe85505b : FAILURE in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21068df0a4c5446a9b8859b03ea5621d : FAILURE in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbaca094faf3417bb3827167b062e280 : FAILURE in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac76970d3b2245ff98401a90f0d860f9 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a367df852f34b6aa5acea3901650013 : SUCCESS in 5m 24s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0d1c2709ec68450494d1645382dc956f : FAILURE in 6m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b80af06ae7104c1aaf70bf0a90bc41f0 : SUCCESS in 1h 47m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf4e352d8ca843e0bb672fd2a0c17eaa : FAILURE in 2h 38m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8fb6aba786534afbb17dffb5c33fdc13 : SUCCESS in 1h 22m 31s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/encryptors/base.py","accounts_in_message":[],"_revision_number":1},{"id":"294ae1ebbb55c4808d8026032ec123d5496cbbf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 20:17:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c96ba17b7794f29f4fa68fab7b7d3108542f9a70","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 20:23:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/58/837358/2/check/DellEMC_Unity_FC_os-brick/ad7fdd0/DellEMC_Unity_FC_os-brick/65 : FAILURE in 1m 44s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/58/837358/2/check/DellEMC_Unity_iSCSI_os-brick/077a4cf/DellEMC_Unity_iSCSI_os-brick/78 : FAILURE in 1m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f8ff1f8ce3f5b8195e3bfdbf2c4e4495858fccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 22:19:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af53752398644e7d841b251f55705464 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc05fdfa103248dc9819dea73347782f : FAILURE in 1h 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34ff5ebf594e453ab6a877e87c32f3ef : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681ed646ddd348cf83e13fee8a4c6718 : FAILURE in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59f260418bd84bddb1f06a98193d6022 : FAILURE in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87004d0954024426ac1cc74e27973f9b : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c25dfd809074a4988d93033cefd7ee7 : SUCCESS in 5m 46s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d8a2dfac0e3448c8b5c76021b72f6aa9 : FAILURE in 4m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/05374c04095c44f2b452f5f9495e07b6 : SUCCESS in 1h 10m 30s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df510d5fc5f54326a3be3ebdd621eb96 : FAILURE in 1h 59m 07s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/773ea93c92c44f3da880524fc56c7e25 : SUCCESS in 1h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2197c9508579ed89dd0e02f0f8b33962bd969b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-11 22:43:26.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837358/2/iscsi/ : SUCCESS in 1h 09m 33s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837358/2/smb/ : SUCCESS in 1h 09m 01s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837358/2/ceph/ : FAILURE in 1h 09m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8c3a45c58b967bf71caaa11e76fe3cd022f877ff","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-12 03:01:19.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/d4860bb5dcb84b0893744730cb8d6775 : POST_FAILURE in 3h 06m 22s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"c27eb2e360aa8cb078818c430f7ec7dc7d809524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-23 09:22:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"635b05bb61b9ce2088e4fddb5ef7484aac550407","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-23 09:51:05.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837358/3/iscsi/ : FAILURE in 13m 54s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837358/3/smb/ : FAILURE in 14m 07s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837358/3/ceph/ : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5603f330d472b585f12d737806d91767d5ad49ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 12:32:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a7c1ca3e5564dfd9f9c13151a7d5a6a : SUCCESS in 2m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82dd1ea9069c424db0dfac33d31e8aa7 : FAILURE in 2h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a869fb002914765972753bef9d15ebe : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c407582f5154129ae0920d15d57e600 : FAILURE in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3c2467532c84dfc8bfb637a04253e80 : FAILURE in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb5f7efc53054d329e46d962601e2138 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793dfc28a2c54f7a96bbfd1d3d4c1370 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e25b62fe584f78a31203dbc58de437 : SUCCESS in 5m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/984a55a0ef6a47d2adb9d5222debc5e1 : FAILURE in 4m 05s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/968b8716c4254f1bad9ff7937312e905 : SUCCESS in 1h 24m 31s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4159a85ee6234fe88ee4d4fd38520a1f : TIMED_OUT in 3h 07m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0b1b1d44196d47e3b0d723a64a89320e : SUCCESS in 1h 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eea5b32f4b3b9fda7aace3ce1156e4ad17d52af5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 14:17:38.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/58/837358/3/check/tempest-dsvm-full-focal-py3/5a0a03e/ : FAILURE in 2h 49m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c929dc1a2f08a2a23aaaa1417f7ab9e8c509f071","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-23 15:53:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/58/837358/3/check/DellEMC_Unity_FC_os-brick/ef4cf46/DellEMC_Unity_FC_os-brick/83 : SUCCESS in 1h 08m 15s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/58/837358/3/check/DellEMC_Unity_iSCSI_os-brick/31eb9d2/DellEMC_Unity_iSCSI_os-brick/111 : SUCCESS in 1h 36m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"50a97b1a20a4d89d1ad4a71346e99072cf1126e8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 17:07:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"9d130c65050504d1236fa2ebc3d126525c9c1d06","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 21:42:50.000000000","message":"Abandoned\n\nGuessing if this is still a WIP after so long it\u0027s not actually needed, so closing this out. Feel free to reopen and update or submit something new, if still needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18d4468e3418176e60cf15d218b895168aed3bb8","revisions":{"474db4d0c5b577578e19d2a4463c49757ed9943f":{"kind":"REWORK","_number":1,"created":"2022-04-11 15:37:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/837358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/837358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/837358/1"}}},"commit":{"parents":[{"commit":"d4bfd6e7262238570f2c9f39ba6bf90dec2f4016","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d4bfd6e7262238570f2c9f39ba6bf90dec2f4016"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 10:33:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 10:46:04.000000000","tz":60},"subject":"WIP: privsep: Move linuxscsi initiator to privsep","message":"WIP: privsep: Move linuxscsi initiator to privsep\n\nWIP cos tests.\n\nChange-Id: I0d5f7b19f4077546e2e7a46af07a791315450597\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/474db4d0c5b577578e19d2a4463c49757ed9943f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/474db4d0c5b577578e19d2a4463c49757ed9943f"}]},"branch":"refs/heads/master"},"96a98c26567d81934b526440e2556fedfceb9f13":{"kind":"REWORK","_number":2,"created":"2022-04-11 20:17:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/837358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/837358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/837358/2"}}},"commit":{"parents":[{"commit":"6610443d5332b135b96672d7889599e4121d0cca","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6610443d5332b135b96672d7889599e4121d0cca"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 10:33:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 20:16:59.000000000","tz":60},"subject":"WIP: privsep: Move linuxscsi initiator to privsep","message":"WIP: privsep: Move linuxscsi initiator to privsep\n\nWIP cos tests.\n\nChange-Id: I0d5f7b19f4077546e2e7a46af07a791315450597\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/96a98c26567d81934b526440e2556fedfceb9f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/96a98c26567d81934b526440e2556fedfceb9f13"}]},"branch":"refs/heads/master"},"18d4468e3418176e60cf15d218b895168aed3bb8":{"kind":"REWORK","_number":3,"created":"2022-05-23 09:22:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/837358/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/837358/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/837358/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/837358/3"}}},"commit":{"parents":[{"commit":"d005653c110abe57237b9b65742c9a47c160d87c","subject":"privsep: Convert \u0027os_brick.caches.opencas\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d005653c110abe57237b9b65742c9a47c160d87c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 10:33:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-23 09:22:03.000000000","tz":60},"subject":"WIP: privsep: Move linuxscsi initiator to privsep","message":"WIP: privsep: Move linuxscsi initiator to privsep\n\nWIP cos tests.\n\nChange-Id: I0d5f7b19f4077546e2e7a46af07a791315450597\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/18d4468e3418176e60cf15d218b895168aed3bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/18d4468e3418176e60cf15d218b895168aed3bb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
