)]}'
{"id":"openstack%2Fos-brick~837360","triplet_id":"openstack%2Fos-brick~master~I9e398e72470f3eb077178a43fc075cb0f7a2fc40","project":"openstack/os-brick","branch":"master","topic":"privsep","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-07 17:07:49.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I9e398e72470f3eb077178a43fc075cb0f7a2fc40","subject":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027","status":"ABANDONED","created":"2022-04-11 15:37:25.000000000","updated":"2025-12-12 21:42:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"4aefb4ade1690a422b227a0014280a8510bc53ac","_number":837360,"virtual_id_number":837360,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-23 12:32:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 15:59:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-11 18:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 19:41:58.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-04-12 03:14:48.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7a238697293bbdf3094e110850ba932aac5c5b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 15:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc95fc2fb69dbf0ade7471ddcd4ccc64c94d311e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 15:59:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/60/837360/1/check/DellEMC_Unity_FC_os-brick/f80ace8/DellEMC_Unity_FC_os-brick/58 : FAILURE in 1m 43s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/60/837360/1/check/DellEMC_Unity_iSCSI_os-brick/5a8c603/DellEMC_Unity_iSCSI_os-brick/70 : FAILURE in 1m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a035b9003e712160806902474ff41cd1019c9f6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:31:36.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9831e08d0a614ee4a8215ae251458e68 : FAILURE in 1h 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7852141ffcd34fdab19d36c3b46159c3 : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af3ef2265c8e4c339210b377ecfd5531 : FAILURE in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a631f64b4384357813e764ae40ad45f : FAILURE in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090c6604e4e64c958ca3d45e2bb3f197 : SUCCESS in 6m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/e3da9828f9e34a4288c8d7e1047b552c : FAILURE in 5m 40s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/80c1b5c2875848269c78579bb6820baf : FAILURE in 1h 06m 12s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/605df2383cb94f8faa6735e7297735c0 : FAILURE in 1h 52m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f28d29651abf4c5fa42303c7c38cba17 : SUCCESS in 1h 34m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/encryptors/base.py\n  Comments left for invalid file os_brick/initiator/connectors/base.py\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":1},{"id":"574e4d5c4165534d6fb29c1bb79b2ca8469d0329","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-11 19:41:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837360/1/iscsi/ : SUCCESS in 1h 11m 13s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837360/1/smb/ : SUCCESS in 55m 39s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837360/1/ceph/ : FAILURE in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e618446fe36383a5045c294b75a25f250839ab83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-11 20:17:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"38b79cb320f0ae5333fa3fe52d59e564a3197749","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-11 20:26:45.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/60/837360/2/check/DellEMC_Unity_FC_os-brick/8cb6cef/DellEMC_Unity_FC_os-brick/67 : FAILURE in 1m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/60/837360/2/check/DellEMC_Unity_iSCSI_os-brick/def3f3b/DellEMC_Unity_iSCSI_os-brick/80 : FAILURE in 1m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e47add05b4dc252f876e9e91bf01fae980906334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 22:13:01.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/617aaa0fce80497f8677b4acc131dcfe : FAILURE in 1h 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ee218c07fe42899741e596d727cd19 : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/357e7c89df7848bb945a0f6d22696309 : FAILURE in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d75688ef7d6848b4a25406085bba8a26 : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3f9fa0d6394a6b83efde24fdebba4f : SUCCESS in 7m 27s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9179d8eaddf54e348452480ce30e99a3 : FAILURE in 4m 51s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/8caa3f5e5749478594c2fcfe0e7310ba : FAILURE in 1h 10m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/281e2d89adcc4ee694f3319b5faf1075 : FAILURE in 1h 52m 21s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b2e289e156f048f68c06361606ab1baa : SUCCESS in 1h 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfc0e4d900ec856f1e5109f21f46032467aa64b7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-04-12 00:11:24.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837360/2/iscsi/ : SUCCESS in 1h 14m 04s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837360/2/smb/ : SUCCESS in 1h 10m 28s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837360/2/ceph/ : FAILURE in 1h 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"18d25adedf4ad2deab0c3e05647aca168e5b7772","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-12 03:14:48.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/c1e9361f78fa4efa8bbe16b901efd842 : TIMED_OUT in 3h 11m 29s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f663e4ebcd454ab798141ee31d543704 : POST_FAILURE in 3h 06m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f9389eeed2a38c951e20b2117d08964b7f262f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-23 09:22:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4bb683dff5c0a68d405a5123ef795176f9fa8487","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-23 10:05:15.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/837360/3/iscsi/ : FAILURE in 14m 00s\n- os-brick-smb http://cloudbase-ci.com/os-brick/837360/3/smb/ : FAILURE in 14m 09s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/837360/3/ceph/ : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e84f4082e66145e3a39c9bdfeca9626c225a243a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 12:32:50.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cd8b622f3a64f5d81fa99a3b70c319b : FAILURE in 1h 55m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4749d2cb9fb34c29910f5581f0fa6d82 : FAILURE in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53b450ebc8604af19c5e0a8e75fa1888 : FAILURE in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40438c6ef887494a9e1ca07542cff535 : FAILURE in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/165e3cdd84084911bfe769bb9f0ad74f : FAILURE in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61927eb2b39d4d26a2f52a964ef57e6b : SUCCESS in 8m 05s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d00460ea4ec54997a94ae07769b9796d : FAILURE in 3m 59s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/0ab5c6cf80c04b7682897d1aafd2573c : FAILURE in 1h 22m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a6d37684a6404b3ea3ca870eb1e0e30d : TIMED_OUT in 3h 07m 45s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b85c92db0a6545c0a32f956bc09f67fe : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34d7cb6807f7bfe10f879fb54d851b2591d5ed72","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-23 13:27:45.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/680e72f6d67f475baa35fd538f142c8c : POST_FAILURE in 2h 48m 54s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/eb584c3afc31421984264a2e0f3af584 : POST_FAILURE in 2h 44m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f557d1c3199f1b5acf2b2f07f8810f94d8f05a0f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-23 19:06:58.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/60/837360/3/check/DellEMC_Unity_FC_os-brick/818f0ca/DellEMC_Unity_FC_os-brick/85 : SUCCESS in 1h 11m 47s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/60/837360/3/check/DellEMC_Unity_iSCSI_os-brick/7cf42b0/DellEMC_Unity_iSCSI_os-brick/113 : SUCCESS in 1h 35m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5a7dd5eac5ebd9031fbd4afb9b34df5aef0af1db","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 17:07:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"4aefb4ade1690a422b227a0014280a8510bc53ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-12-12 21:42:29.000000000","message":"Abandoned\n\nGuessing if this is still a WIP after so long it\u0027s not actually needed, so closing this out. Feel free to reopen and update or submit something new, if still needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aeea339385751f89650770d9e90ecaf081647ba9","revisions":{"48a52f7c2cd82ac808dcdf1fd601d0e86c217a42":{"kind":"REWORK","_number":1,"created":"2022-04-11 15:37:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/837360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/60/837360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/60/837360/1"}}},"commit":{"parents":[{"commit":"7b3eef77677c1b5f2b2686c62da737ce80834216","subject":"WIP: privsep: Add a privsep-compatible base connector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7b3eef77677c1b5f2b2686c62da737ce80834216"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 09:06:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 15:37:14.000000000","tz":60},"subject":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027","message":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027\n\nWIP cos tests. Also, this should come after the base linux connector.\n\nChange-Id: I9e398e72470f3eb077178a43fc075cb0f7a2fc40\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/48a52f7c2cd82ac808dcdf1fd601d0e86c217a42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/48a52f7c2cd82ac808dcdf1fd601d0e86c217a42"}]},"branch":"refs/heads/master"},"640134afad13bcb199a8e9765cbd071f7202fa9c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-11 20:17:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/837360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/60/837360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/60/837360/2"}}},"commit":{"parents":[{"commit":"4fff6e6b389402993ba8cf1adf7d37f40c2eda91","subject":"WIP: privsep: Add a privsep-compatible base connector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4fff6e6b389402993ba8cf1adf7d37f40c2eda91"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 09:06:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 20:17:29.000000000","tz":60},"subject":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027","message":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027\n\nWIP cos tests. Also, this should come after the base linux connector.\n\nChange-Id: I9e398e72470f3eb077178a43fc075cb0f7a2fc40\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/640134afad13bcb199a8e9765cbd071f7202fa9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/640134afad13bcb199a8e9765cbd071f7202fa9c"}]},"branch":"refs/heads/master"},"aeea339385751f89650770d9e90ecaf081647ba9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-23 09:22:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/837360/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/60/837360/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/60/837360/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/60/837360/3"}}},"commit":{"parents":[{"commit":"9f01fd767975ba37c408aa9f5e548b07d61a352d","subject":"WIP: privsep: Add a privsep-compatible base connector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9f01fd767975ba37c408aa9f5e548b07d61a352d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-11 09:06:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-23 09:22:14.000000000","tz":60},"subject":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027","message":"WIP: privsep: Convert \u0027os_brick.initiator.connectors.rbd\u0027\n\nWIP cos tests. Also, this should come after the base linux connector.\n\nChange-Id: I9e398e72470f3eb077178a43fc075cb0f7a2fc40\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/aeea339385751f89650770d9e90ecaf081647ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/aeea339385751f89650770d9e90ecaf081647ba9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
