)]}'
{"id":"openstack%2Fos-brick~842017","triplet_id":"openstack%2Fos-brick~master~Iad7fc4ca66da54e9b60e770236736626820283ef","project":"openstack/os-brick","branch":"master","topic":"jobs","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2022-06-13 10:53:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iad7fc4ca66da54e9b60e770236736626820283ef","subject":"Updating tests with Zed testing runtime","status":"ABANDONED","created":"2022-05-17 01:22:10.000000000","updated":"2022-06-13 10:53:47.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c26bd28b482be65cb122497419251a228a9aa97c","_number":842017,"virtual_id_number":842017,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-17 10:27:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":-1,"date":"2022-05-25 21:50:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2022-05-19 09:30:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 03:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 03:33:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-17 03:34:02.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-05-17 10:27:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-05-17 10:27:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-17 12:01:34.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-05-17 12:33:21.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-17 15:58:03.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-19 09:30:43.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-05-25 21:50:53.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"5cbf71d8e473c391a7936fd0984964da9f42c1b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-17 01:22:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9afa814ee7c3e56d7b167f64bc0c12ffcf3fb606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 03:18:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/858bd511af5d47f28beba442c15dfd52 : SUCCESS in 1h 42m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66508eef855e4b71b771ebf5d3b713ea : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ac25c92e554457cadbf0e51bc778204 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc60837929648408893f19d584422e3 : FAILURE in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45d4d58956da406bb4f6869a0812b729 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede8a94f9a9448c5a86ca5478bbabf70 : SUCCESS in 5m 29s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/42185cb7affc4cdea2618803788d5484 : SUCCESS in 4m 01s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/4196d98f6ea94932853cb7a4b6302cba : SUCCESS in 1h 14m 42s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9e86e969c1b54e4c82ed780e08c700cc : SUCCESS in 1h 51m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/454d804eb0f74194ba85aa37d3115573 : SUCCESS in 1h 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a26886f1c2c5b5981207c5e5b996c6229d7b6ed3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-17 03:33:41.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/50999529d41a40eab3b459d3ffa21174 : POST_FAILURE in 2h 08m 00s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ec2322916b9e489389243068a4ddc709 : POST_FAILURE in 1h 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d2ee31d971324817376d4a07d61f81bfb1522c9d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-17 03:34:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/842017/1/iscsi/ : SUCCESS in 2h 11m 31s\n- os-brick-smb http://cloudbase-ci.com/os-brick/842017/1/smb/ : SUCCESS in 1h 21m 50s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/842017/1/ceph/ : SUCCESS in 1h 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"52fd8e0fb649dad239341409c413397fbca84210","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-17 08:36:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a8523ff66b502efab1af103ee9b65f09fa8bd8d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-05-17 09:48:49.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/842017/1/iscsi/ : SUCCESS in 1h 07m 36s\n- os-brick-smb http://cloudbase-ci.com/os-brick/842017/1/smb/ : SUCCESS in 54m 54s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/842017/1/ceph/ : FAILURE in 1h 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"17722195479dccbc95d9d6a7f28dd7504f5f6d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 10:27:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/232761e0c731479d9b7a8debb8d2979b : SUCCESS in 1h 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9dfef056ad4a329cbbce245dd84382 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c85b99639014092ac77ef64cd53084a : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2853433b774cadb1753028cd78e3e7 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3ae348678a24ddb8f8bba9fcee27893 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4074b339b7e4e21b0e7bfd4756267d3 : SUCCESS in 5m 45s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a3b0a1c8f5a04be3ad688fb7ab65bf5f : SUCCESS in 3m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/e88b8bcc3371492d8411ae55f266d953 : FAILURE in 1h 15m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f64b13e04fa04006be50fabedc379764 : SUCCESS in 1h 33m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5af0ac08f6614f1680096abf823bbbd2 : SUCCESS in 1h 05m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c033637d588346ffe744f946f8d75032e21c22f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-05-17 10:27:22.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"26e924c8a31ba16a14d89507f58076f543dd76d9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-05-17 10:27:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_PowerFlex_os-brick/3416035/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c55289eb8f9b1a28c5f44d720a31836cb428260a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-05-17 11:05:03.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/96e4eeb10fb74ad9ac7fc797555d31c4 : POST_FAILURE in 2h 26m 40s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/237a9e2014584c20965f77379765bce9 : POST_FAILURE in 1h 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c56a2a9a160923599c3068b79f38f96bed23bdd8","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-05-17 12:01:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_SC_FC-os-brick/d36b661/DellEMC_SC_FC-os-brick/67 : SUCCESS in 42m 15s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_SC_iSCSI-os-brick/8cd4fdc/DellEMC_SC_iSCSI-os-brick/77 : FAILURE in 1h 34m 06s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"588eaed8cb37f0abf901402677b6def034474d8e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-17 12:33:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_FC_os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_VNX_FC_os-brick/ccee757/DellEMC_VNX_FC_os-brick/38 : FAILURE in 9m 47s\n- DellEMC_VNX_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_VNX_iSCSI_os-brick/2a34da1/DellEMC_VNX_iSCSI_os-brick/40 : SUCCESS in 2h 05m 54s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a9fb1ed6070d7ff3233369ab63e393534caa739","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-17 15:58:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_Unity_FC_os-brick/54c21c2/DellEMC_Unity_FC_os-brick/79 : SUCCESS in 1h 05m 31s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/17/842017/1/check/DellEMC_Unity_iSCSI_os-brick/82d2fcf/DellEMC_Unity_iSCSI_os-brick/107 : SUCCESS in 1h 40m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"70e1ec8f7c9c1758f3ae604c863492fdc83446eb","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-05-19 09:30:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f16874e3fc9744ed0172c872d553a40faac33a66","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-05-25 21:50:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c26bd28b482be65cb122497419251a228a9aa97c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-06-13 10:53:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"953e3b661bf049f4fb2dc813cfdef1932c525d27","revisions":{"953e3b661bf049f4fb2dc813cfdef1932c525d27":{"kind":"REWORK","_number":1,"created":"2022-05-17 01:22:10.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/17/842017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/17/842017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/842017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/842017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/17/842017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/17/842017/1"}}},"commit":{"parents":[{"commit":"2ec42175f49dc61c006fc5f646e29e32449be59b","subject":"Merge \"Bump mypy version to 0.942\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2ec42175f49dc61c006fc5f646e29e32449be59b"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-05-17 01:20:51.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-05-17 01:22:07.000000000","tz":480},"subject":"Updating tests with Zed testing runtime","message":"Updating tests with Zed testing runtime\n\nChange-Id: Iad7fc4ca66da54e9b60e770236736626820283ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/953e3b661bf049f4fb2dc813cfdef1932c525d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/953e3b661bf049f4fb2dc813cfdef1932c525d27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
