)]}'
{"id":"openstack%2Fos-brick~845149","triplet_id":"openstack%2Fos-brick~master~I7d6fff66c5cc27bbd1f5caa737f01bc7784af710","project":"openstack/os-brick","branch":"master","topic":"nvme-4","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-06-15 14:02:20.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-06-15 02:31:01.000000000","reason":"Rajat Dhasmana replied on the change"}},"hashtags":[],"change_id":"I7d6fff66c5cc27bbd1f5caa737f01bc7784af710","subject":"Fix Unit Test failures due to host\u0027s ANA support","status":"MERGED","created":"2022-06-08 15:13:09.000000000","updated":"2022-06-15 16:44:03.000000000","submitted":"2022-06-15 14:02:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845149-nvme-4","meta_rev_id":"c6756a8196daffa74f82d39b2f5d7a9a74efff54","_number":845149,"virtual_id_number":845149,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-15 14:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-14 21:16:45.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2022-06-15 02:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2022-06-15 02:31:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 17:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-08 17:00:54.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-08 17:00:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-08 17:29:03.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-06-08 17:40:49.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-06-08 18:35:43.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-06-08 19:49:15.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-09 05:15:48.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2022-06-09 14:01:53.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2022-06-10 18:59:39.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-06-11 00:37:56.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-14 21:16:45.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-06-15 02:31:01.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-06-15 11:56:35.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"}],"messages":[{"id":"236491f303d1ae14b2428d1a7998aa0b559c2f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-08 15:13:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7889010f6fd1afb32f86cb241c98754ee35062b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a03e227964d64c0ea3af6d7fc5143e32 : SUCCESS in 1h 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d753b3b877b34d32b8a0344add639b10 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9364b15edd33422186f29fcaaec2db68 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd45c760e0af42f38d59e2cb1e072fa1 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c725c9daf4448cebecc935c6e3615b3 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a97b8dfec98455b89f236094b515068 : SUCCESS in 5m 46s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6b500ff7f6ea4a758e7df86eca49afda : SUCCESS in 4m 15s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/0031f88d51604b909c0358663d29add4 : FAILURE in 1h 25m 55s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32892186e32149438bdbda70d7c64dcb : SUCCESS in 1h 40m 00s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8f49633c13a246639d53c50d19e25149 : SUCCESS in 1h 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bf4548d991064196ad61b1246036a21519ecc4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-08 17:00:54.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2ef4faa0210c68bba172631557170ec20b509e6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-08 17:00:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/49/845149/1/check/DellEMC_PowerFlex_os-brick/3a436f5/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"7bb4cb50de341037d7c8c2bc8f1254e84e9d9d4b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-08 17:29:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654701197281 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4b1aa3ce88a28c5ed2605e406a25ddb4299aefd0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-06-08 17:40:49.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/845149/1/iscsi/ : FAILURE in 50m 55s\n- os-brick-smb http://cloudbase-ci.com/os-brick/845149/1/smb/ : FAILURE in 39m 46s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/845149/1/ceph/ : FAILURE in 51m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0a37e99304ac5dbd682b6930a1397808d36b79da","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-08 18:35:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/49/845149/1/check/DellEMC_SC_FC-os-brick/dd82995/DellEMC_SC_FC-os-brick/76 : SUCCESS in 45m 44s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/49/845149/1/check/DellEMC_SC_iSCSI-os-brick/017c45e/DellEMC_SC_iSCSI-os-brick/86 : FAILURE in 1h 34m 47s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"211bd04ef6abce4b378a6b8b4f5336b3fb84ff96","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-08 19:49:15.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/49/845149/1/check/devstack-plugin-ceph-dsvm-focal-py3/b84864f/ : TIMED_OUT in 3h 29m 19s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/49/845149/1/check/tempest-dsvm-full-focal-py3/0fc54ee/ : SUCCESS in 2h 02m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b389d81268518a80b13794c1388560657a13b658","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-06-09 05:15:48.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"55b4e218f41ad6fc5d10dd49ef77fa40bffeb828","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-06-09 14:01:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6fd7a32b4d324d5620390fb1182e2232a08b3a9","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-10 14:00:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d379f5a7c20dfc98529d1bb5c16895373fc18ffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-10 14:16:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"479d77a2ce4fd129cb14d2ac1b670b2cef585279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 15:59:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c64582f29e24c34b1daaf1456304433 : SUCCESS in 1h 34m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec5cac2bb854d91a47e6eca0392472a : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d7688627fec4bc1ac98f5351fd1e1ca : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdab25b467484760b0062143218a28dd : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b1cd59515f549cca14964b4dea9b87f : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056128ce7d14454b90067a75b7fab1d2 : SUCCESS in 6m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/11e2ac4268f54fa89f1506eace119411 : FAILURE in 3m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/b806747865c2421391f5fdece99a2752 : POST_FAILURE in 9m 32s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b6ad7873d1444719500895980b5f6d6 : SUCCESS in 1h 38m 00s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/335b1e52e9c6447e99806464ce67a307 : FAILURE in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9052b229e559889df2d0ae61b98f1a75898bbd9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-10 15:59:29.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"0daf797d00535a34d230b5cc47e314bfac1d8541","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-10 15:59:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerFlex_os-brick/1223662/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"9f065a53c229fbd88196c157e1c5138caeb366e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-06-10 16:06:19.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1654870587385 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"190501a34d5343cdb488994830a22e24c0816041","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-10 18:59:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_SC_FC-os-brick/1b5dcf7/DellEMC_SC_FC-os-brick/80 : ABORTED in 3h 00m 00s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_SC_iSCSI-os-brick/b374d7c/DellEMC_SC_iSCSI-os-brick/90 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3138c3494bbb42c21282dfbd6eeeb72b1738fad","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-10 18:59:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerStore_FC-ob/869f5ed/DellEMC_PowerStore_FC-ob/75 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerStore_iSCSI-ob/90156e8/DellEMC_PowerStore_iSCSI-ob/68 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ff0f2faa82bcb909f7ad9b237602a01cedd7a020","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-11 00:37:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_Unity_FC_os-brick/c1c98e2/DellEMC_Unity_FC_os-brick/106 : ABORTED in 3h 00m 01s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_Unity_iSCSI_os-brick/dbd44cb/DellEMC_Unity_iSCSI_os-brick/122 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a15f33ca064be13b938584c00a667c9d5298c398","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-06-14 21:16:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b96830b656af914a7f84cf20aec6071ee7521e09","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-06-15 02:31:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3a7c8496fb5290cc2d025c5fa31582b2bfac6048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 02:32:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c115813de2d5e4eb34352c9cce9da45f8a116651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 04:06:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6699cc48bb1441c2b5803ea16c693d6d : SUCCESS in 1h 33m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1dbf9263c1b43aea39e9bdca53ba8aa : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e17d08919e834e2aa8d4e41443ceeae0 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05121f86d50b4baba264679cbdb7ed70 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a4e74c056284778be94807c7052318c : SUCCESS in 7m 33s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/079ea58e6a8045b991a089aff0f5df42 : FAILURE in 26m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7d87013c343f55c36b8fc13906e17ae18f41481a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-15 09:20:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9465b885e5d9d0a1a45bdbd8c86273575fddfb38","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-06-15 10:03:31.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/c3efce9d8e464286a966788fda9dd044 : POST_FAILURE in 41m 05s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/71ffa7daa6944d6e834e8c51a9be5fba : POST_FAILURE in 39m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3d46a802f34d5e35530bc0d52f4c07cac30037ad","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-06-15 10:40:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/845149/2/iscsi/ : SUCCESS in 1h 19m 34s\n- os-brick-smb http://cloudbase-ci.com/os-brick/845149/2/smb/ : SUCCESS in 1h 10m 49s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/845149/2/ceph/ : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f41428561b4702f8fd72459979c1c6a99e3658dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 11:24:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd113521c9584ece9cd24abe519c9a17 : SUCCESS in 1h 27m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c38ea48fc046f99c295d95f162e91f : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d8d2dc1736f48179432664f4cc61e05 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3598ca09ef4b47fa8ac000bffee7229a : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/738fcea1366f42d4bb838f72c017e801 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a153d7048dd4ab2b037be4b1ae33c05 : SUCCESS in 5m 51s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3bbb10904ae74f5b857ef5bdc775b8da : SUCCESS in 4m 31s (non-voting)\n- os-brick-src-devstack-plugin-ceph https://zuul.opendev.org/t/openstack/build/204df0406d5240a8a5f8b5b4b483b4a6 : FAILURE in 1h 22m 34s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/385bae9f303640f2b06c3c35df410c26 : SUCCESS in 1h 25m 13s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/aaf4612655504a6bae7affc21ca8770b : SUCCESS in 1h 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"583c29f5238eb83bd9237b87cdb4bab401bb2163","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-15 11:24:34.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"7842f72665fd47799748ecbe48baa30d411c7e42","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-06-15 11:24:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerFlex_os-brick/a5988b6/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"40a39f3188d70c96889a4b4b76829c7e0c3e7626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 11:25:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21c55fd769f40dd9671b1abe88c610d19a3626dd","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-06-15 11:56:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-49-845149-2/1655292264 : SUCCESS in 17m 24s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"1c74173eb37b5cb2ad8aeb3be6c5a462f84b2c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:02:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d705acd78e4b4292c7124c2bcfc23d : SUCCESS in 2h 00m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f40e88070aa34416973e39f27bbae055 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b245ed73b1741f7934e1e930a97bd18 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53481985307148db9ab277e477a09530 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c839d08e634998b4c5ac84b747af1d : SUCCESS in 6m 27s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0c5f6748ba6d44e0bb2a43064fe10810 : SUCCESS in 2h 25m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ef3b3fe25691e0264c5479ded0b12b9e807a219d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:02:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f603175f52223ee4dd61ebe9dec2634a1273b43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:03:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b078999836e2420889afe83ae1794b96 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b1f255789532b9dee54d824c91f3624659546217","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-06-15 16:43:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerStore_FC-ob/58880b4/DellEMC_PowerStore_FC-ob/84 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_PowerStore_iSCSI-ob/a04f0d5/DellEMC_PowerStore_iSCSI-ob/77 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"c6756a8196daffa74f82d39b2f5d7a9a74efff54","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-06-15 16:44:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_SC_FC-os-brick/c6266ec/DellEMC_SC_FC-os-brick/89 : ABORTED in 3h 00m 00s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/49/845149/2/check/DellEMC_SC_iSCSI-os-brick/274dadd/DellEMC_SC_iSCSI-os-brick/99 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"924af884db5797092e16e6176e9a70feddc9c892","revisions":{"a87a6a5278a59e81d550c855c8f63ed9c53efdc0":{"kind":"REWORK","_number":1,"created":"2022-06-08 15:13:09.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/49/845149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/49/845149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/49/845149/1"}}},"commit":{"parents":[{"commit":"7f747456dce8845a9e4982586729ade99d2bb14e","subject":"Merge \"Avoid volume extension errors caused by multipath-tools version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7f747456dce8845a9e4982586729ade99d2bb14e"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-08 10:15:21.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-08 10:21:35.000000000","tz":120},"subject":"Fix Unit Test failures due to host\u0027s ANA support","message":"Fix Unit Test failures due to host\u0027s ANA support\n\nWe recently added NVMe-oF native multipathing support but the unit tests\nassume that the host they are run on doesn\u0027t support it, so when running\nthem on a system that does support it then 6 tests fail.\n\nThis patch just forces this assumption on the class attribute changing\nthe value of native_multipath_supported from None to False and thus\ndisabling the detection mechanism.\n\nChange-Id: I7d6fff66c5cc27bbd1f5caa737f01bc7784af710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a87a6a5278a59e81d550c855c8f63ed9c53efdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a87a6a5278a59e81d550c855c8f63ed9c53efdc0"}]},"branch":"refs/heads/master"},"924af884db5797092e16e6176e9a70feddc9c892":{"kind":"REWORK","_number":2,"created":"2022-06-10 14:16:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/49/845149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/49/845149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/49/845149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/49/845149/2"}}},"commit":{"parents":[{"commit":"7f747456dce8845a9e4982586729ade99d2bb14e","subject":"Merge \"Avoid volume extension errors caused by multipath-tools version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7f747456dce8845a9e4982586729ade99d2bb14e"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-08 10:15:21.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-10 13:57:39.000000000","tz":120},"subject":"Fix Unit Test failures due to host\u0027s ANA support","message":"Fix Unit Test failures due to host\u0027s ANA support\n\nWe recently added NVMe-oF native multipathing support (Change-Id\nI3af33c5e43cfb104e436fb785b08fb28b50a031a) but the unit tests assume\nthat the host they are run on doesn\u0027t support it, so when running them\non a system that does support it then 6 tests fail.\n\nThis patch just forces this assumption on the class attribute changing\nthe value of native_multipath_supported from None to False and thus\ndisabling the detection mechanism.\n\nIt also adds a couple of tests for the _set_native_multipath_supported\nmethod that had no tests.\n\nChange-Id: I7d6fff66c5cc27bbd1f5caa737f01bc7784af710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/924af884db5797092e16e6176e9a70feddc9c892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/924af884db5797092e16e6176e9a70feddc9c892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
