)]}'
{"id":"openstack%2Fos-brick~852610","triplet_id":"openstack%2Fos-brick~master~Ica954cffd760d7a9d48ba38d01e0e01e84553ff8","project":"openstack/os-brick","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-08-18 22:08:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ica954cffd760d7a9d48ba38d01e0e01e84553ff8","subject":"mypy: work around mypy bug #13214","status":"MERGED","created":"2022-08-09 15:13:07.000000000","updated":"2022-08-21 17:23:51.000000000","submitted":"2022-08-18 22:08:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852610-mypy","meta_rev_id":"a522a2a90082318a5e725226d2a9f6c2e3ac7632","_number":852610,"virtual_id_number":852610,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 22:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-18 20:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-08-18 20:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 15:13:17.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-09 16:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 17:08:31.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-08-10 02:35:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-10 04:28:17.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-08-10 07:11:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-10 11:45:46.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-18 19:55:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 20:17:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-18 20:27:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-18 20:32:38.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"}],"messages":[{"id":"2f88481b85e6b992d26255093a47be30acf461da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-09 15:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e7adcd88947ccb06c8481dff201416aa3a923d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-09 15:13:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660057996342 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e31bd1335a820ec45a71292a0fa0cc24e98b5974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 16:58:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b8a26f55a034e0381bf30a6f0b3356c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d6020003a5c49d0a735188fba05ea57 : SUCCESS in 1h 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312b02ee16474ef08d5dc6f007af6c27 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae907e194ed54a60bfadc0b330caca98 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e358324e6a5430d8f3dc78234e176b7 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2213e633f4a43ca9ed799d8d1d83f9f : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0900d5c405f1487a92a0373f0f706bc8 : SUCCESS in 6m 39s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ec92425cd1ca4beeb2914cb6769a1916 : SUCCESS in 1h 26m 59s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7d6c320746ba44dfa303545e622cfb22 : SUCCESS in 1h 38m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/96e494a617ae460b8dc73903b82d0a68 : SUCCESS in 1h 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1acaffeaa537da911c474d1c8a0beb592edcd4f7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-09 17:08:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/852610/1/iscsi/ : SUCCESS in 1h 17m 52s\n- os-brick-smb http://cloudbase-ci.com/os-brick/852610/1/smb/ : SUCCESS in 1h 40m 37s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/852610/1/ceph/ : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7df92ab4d049793dbd277bc1f0989bd6e9708cd4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-10 02:35:43.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/7af35bb6b9b44357bc8ffc3ebe3ab8b2 : POST_FAILURE in 1h 44m 19s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/971a01e2ed8f4a1b81960dd32ecad63a : TIMED_OUT in 3h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3651972e40fe1215e4558cde3cd88060d71d317b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-10 04:28:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/10/852610/1/check/DellEMC_SC_FC-os-brick/13905b2/DellEMC_SC_FC-os-brick/124 : SUCCESS in 1h 03m 09s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/10/852610/1/check/DellEMC_SC_iSCSI-os-brick/f955a5a/DellEMC_SC_iSCSI-os-brick/134 : FAILURE in 2h 05m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"ff8260d5fd5d47a91d2f33c23da209e657701daf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-10 07:11:46.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/413","accounts_in_message":[],"_revision_number":1},{"id":"098d18ed2d04bf46aa58f4050321ea43d83a46f5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-10 11:45:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/10/852610/1/check/DellEMC_Unity_FC_os-brick/c6d9359/DellEMC_Unity_FC_os-brick/175 : FAILURE in 20m 47s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/10/852610/1/check/DellEMC_Unity_iSCSI_os-brick/77c54a0/DellEMC_Unity_iSCSI_os-brick/189 : FAILURE in 2h 15m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"11e7edd19949e0bfd29b5cb39efa144e01731c21","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-18 14:39:07.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":1},{"id":"e91d544693458f7afd9c92b1388cf03fd43bb768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-18 18:05:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"aafe6c9e7f94d241a07df7f20c9e4aaa90c21a1f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-18 18:23:33.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660845952593 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b9b0231645bd29517e2fbc26d6a5b7b63be3fa3d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-18 18:34:32.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"d9881c7f0612a36e981ebca0c1edba5b2d271761","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-18 19:05:56.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/10/852610/2/check/devstack-plugin-ceph-dsvm-focal-py3/0152240/ : FAILURE in 58m 06s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/10/852610/2/check/tempest-dsvm-full-focal-py3/2ee3b6f/ : FAILURE in 56m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8ff2ef8e9df58366b3960b302e6993046e993a44","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-18 19:35:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/852610/2/iscsi/ : SUCCESS in 1h 15m 44s\n- os-brick-smb http://cloudbase-ci.com/os-brick/852610/2/smb/ : SUCCESS in 1h 18m 42s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/852610/2/ceph/ : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4bc8b89af2bb4dac718d8eff173533bad3aa58cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 19:55:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db7f046c5ed449ca947f6c740025ab3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/abc86c0ee12a4bbfa4b64b9a85117fbe : SUCCESS in 1h 27m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67788d63f4b145ad91868a5a6a2d39a0 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55237ba8b65548239926dfe625ea8075 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c44e768c4fe64b18b4e66e7301377ae9 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e710390b2568428492e82ab8f2beeb89 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e003a1363cde446083d4e78e0b654853 : SUCCESS in 6m 21s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/33c5c8da34354f918cb12c114ae81dc8 : SUCCESS in 1h 37m 37s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/813b695c77254c00b9806a794c95a932 : SUCCESS in 1h 43m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7232752ce1ef49a08c4aeca86a3518fd : SUCCESS in 1h 39m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dd74bedbdf31da347aa43b2d349cf222937dc0c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-18 19:55:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/10/852610/2/check/DellEMC_PowerFlex_os-brick/cafa7ff/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"246207beeaa76c4af08436cf4e61c7370f7816d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-18 20:17:56.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/18_08_2022_19_55_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"cafb2da7a33badb48229a9bf8fc81c213e77400c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-18 20:27:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44bdd0215be1472a44ad8c15b8d50639982ce076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 20:27:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3d05e2aafca4819d06480643043baaefeb4f86d","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-08-18 20:32:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-10-852610-2/1660852545 : SUCCESS in 36m 50s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"4b5e70ce9e54d34b0a328c66738f40716f652b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 22:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b259f5f592455ea8e501f433e1bd6b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fdf4b5e77244a55bf7d9c397275aee4 : SUCCESS in 1h 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a6ef6dc26d41b0aa574ffacd289261 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/159d200f07cd4bb4968cf0d9748351ef : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db54e29f9b234a87b6f52049d0b5b207 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41a0d7c1a4941aa990be99217f126cb : SUCCESS in 5m 50s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c3db9a093af64ed8b8dd5267b3ff265b : SUCCESS in 1h 36m 50s","accounts_in_message":[],"_revision_number":2},{"id":"6ecd8d4326fd7e67abdd339ff6355029d72ce708","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 22:08:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9b6808f4d673a4110e6296b020553ce243d510a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 22:09:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02b2f7e166f6425985fcbc5178b3a26f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2968c2ea9751484093105b69d18bbb3e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2},{"id":"f9295d41c06e0abd86ac8e0619c13aa9c79e85a8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 05:23:00.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/10/852610/2/check/DellEMC_Unity_FC_os-brick/062e90c/DellEMC_Unity_FC_os-brick/184 : FAILURE in 21m 54s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/10/852610/2/check/DellEMC_Unity_iSCSI_os-brick/391c615/DellEMC_Unity_iSCSI_os-brick/198 : FAILURE in 1h 52m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"527e7f09be3db995a58a39c0ba8288a8e489d7ac","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-19 14:05:45.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"a522a2a90082318a5e725226d2a9f6c2e3ac7632","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-21 17:23:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/10/852610/2/check/DellEMC_SC_FC-os-brick/7c35fc7/DellEMC_SC_FC-os-brick/133 : FAILURE in 58m 55s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/10/852610/2/check/DellEMC_SC_iSCSI-os-brick/2ef10a6/DellEMC_SC_iSCSI-os-brick/143 : FAILURE in 2s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5820f531901fd87b692b225c155c4e262731375","revisions":{"92ad0cfc239327a08ba6adf9af125b488995df72":{"kind":"REWORK","_number":1,"created":"2022-08-09 15:13:07.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/10/852610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/10/852610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/10/852610/1"}}},"commit":{"parents":[{"commit":"91e4f4015250d8d033ce699678eba617e58a91c2","subject":"Move mypy job to check queue (non-voting)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/91e4f4015250d8d033ce699678eba617e58a91c2"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-09 15:12:10.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-09 15:12:10.000000000","tz":-240},"subject":"mypy: work around mypy bug #13214","message":"mypy: work around mypy bug #13214\n\nTurn off in-project stubs for now until this bug\nis fixed and in a mypy release we can consume.\n\nhttps://github.com/python/mypy/issues/13214\n\nChange-Id: Ica954cffd760d7a9d48ba38d01e0e01e84553ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/92ad0cfc239327a08ba6adf9af125b488995df72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/92ad0cfc239327a08ba6adf9af125b488995df72"}]},"branch":"refs/heads/master"},"d5820f531901fd87b692b225c155c4e262731375":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-18 18:05:44.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/10/852610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/10/852610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/10/852610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/10/852610/2"}}},"commit":{"parents":[{"commit":"b41b3c8814b2a557222d3bbbabcb9e6dc342c970","subject":"Merge \"Add flake8-logging-format extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b41b3c8814b2a557222d3bbbabcb9e6dc342c970"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-09 15:12:10.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-18 18:04:26.000000000","tz":-240},"subject":"mypy: work around mypy bug #13214","message":"mypy: work around mypy bug #13214\n\nTurn off in-project stubs for now until this bug\nis fixed and in a mypy release we can consume.\n\nhttps://github.com/python/mypy/issues/13214\n\nChange-Id: Ica954cffd760d7a9d48ba38d01e0e01e84553ff8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d5820f531901fd87b692b225c155c4e262731375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d5820f531901fd87b692b225c155c4e262731375"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
