)]}'
{"id":"openstack%2Fos-brick~852672","triplet_id":"openstack%2Fos-brick~master~I650b92bde5509b654d16301ffe50056e1e5ba579","project":"openstack/os-brick","branch":"master","topic":"flake8-logging","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-08-11 18:45:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I650b92bde5509b654d16301ffe50056e1e5ba579","subject":"Add flake8-logging-format extension","status":"MERGED","created":"2022-08-09 20:25:05.000000000","updated":"2022-08-11 18:54:34.000000000","submitted":"2022-08-11 18:45:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852672-flake8-logging","meta_rev_id":"d1b6a3dc1a03f74016ccde46cdf5eee6512c6bd0","_number":852672,"virtual_id_number":852672,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 18:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2022-08-11 16:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-10 14:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2022-08-11 16:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 20:25:14.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-09 21:45:31.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-08-09 22:07:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 22:20:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-09 23:53:15.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-10 02:49:58.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-10 08:12:28.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-08-10 14:49:46.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-08-10 17:32:07.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-10 20:59:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-11 16:45:40.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"0c97eee3ac9c41e4421f8c5742bc94be4a925961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-09 20:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f50c2811a4cea2e414c098783dcc4b8117292f82","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-09 20:25:14.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660076711819 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"49827c3a47e307829d58cf15d92952aa37e0dfac","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-09 21:45:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/852672/1/iscsi/ : SUCCESS in 1h 20m 05s\n- os-brick-smb http://cloudbase-ci.com/os-brick/852672/1/smb/ : SUCCESS in 1h 00m 43s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/852672/1/ceph/ : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aa427f69d1a9362a3b36d1d5803297eae69a4daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 22:07:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d8b88670fb4140b9da11024336e09b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/af7c9d95acfa45eda6a1782190ed8967 : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d8aae24d0294f33884baa6a76460f4d : SUCCESS in 1h 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdf28ff16b943e381a80411324584b4 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99cbfc19b8944f019be2b007ce1418fe : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533c3151c17a4819a5fe23bfa538c35f : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41d53e9c7b8f470e8683bbe751782c00 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40cf833e77004ccf9e54d40cacfba9b0 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12fabf337596405384725c0e3dce2472 : SUCCESS in 5m 04s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b5d56b7da9604674bf9ff20d11608590 : SUCCESS in 3m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/399a3448e2b04e5f9984bd06f7ca06f1 : SUCCESS in 1h 24m 38s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64bc1cc726524da7b984d517079aeeec : SUCCESS in 1h 35m 52s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/41505a27a20946a9a514a58f5240e69b : SUCCESS in 1h 37m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ccc0e73c65803a122a5118be07d5b5dc50b9938","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-09 22:20:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/72/852672/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1f836e7/ : POST_FAILURE in 1h 35m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/72/852672/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f9e4f2a/ : SUCCESS in 1h 47m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/72/852672/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/01eff46/ : SUCCESS in 1h 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63e40e77569828b357e931469b9be3067f2f076b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-09 23:53:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_PowerStore_FC-ob/3adecf6/DellEMC_PowerStore_FC-ob/146 : SUCCESS in 1h 19m 59s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_PowerStore_NFS-ob/0d0f610/DellEMC_PowerStore_NFS-ob/56 : SUCCESS in 1h 29m 58s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_PowerStore_NVMeTCP-ob/caea72c/DellEMC_PowerStore_NVMeTCP-ob/23 : FAILURE in 1h 45m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c743f00cfee846a5014ab390149af5ebcfe3c70","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-10 02:49:58.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/58615bb3c720419cb57328b364250da1 : POST_FAILURE in 2h 28m 25s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/55c02257f83d45789246d9618866f50c : POST_FAILURE in 2h 06m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dcd7347ebc935b4a0854f747bb0611dd5152565c","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-10 08:12:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_SC_FC-os-brick/11f0bdf/DellEMC_SC_FC-os-brick/125 : SUCCESS in 48m 10s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_SC_iSCSI-os-brick/c28772a/DellEMC_SC_iSCSI-os-brick/135 : FAILURE in 1h 58m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1800ef21fa5f25823d0da47268850576e741a5a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-10 14:49:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec88bd76e29b8f52e2413906a7aeeddf177d9050","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-10 17:32:07.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/416","accounts_in_message":[],"_revision_number":1},{"id":"93e45dd13df760a58f7ed0f2f4d3ee73a7c8ced4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-10 20:59:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_Unity_FC_os-brick/85d445a/DellEMC_Unity_FC_os-brick/176 : FAILURE in 22m 02s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/72/852672/1/check/DellEMC_Unity_iSCSI_os-brick/65e4b46/DellEMC_Unity_iSCSI_os-brick/190 : FAILURE in 2h 20m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"73d62556ee708602b309558e722b9b735dff4dfe","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-11 16:45:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86e9c3d8410a02a30f5d0baf6dece42880a76703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:46:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b97dee6b05c2e9f7d918fc76a2f20b8b0c4fa797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:45:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4f13329f6c4e75a24b739fd2ff1ab5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a528ddd67924b6ea132ddb9dba6b2bb : SUCCESS in 2m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e68799dedb8d41ac9c006121bcb86850 : SUCCESS in 1h 52m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4941134df94748dc80d775eb6a7e4151 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f5a8e782d444d7699c0800d1bc64576 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0660ba882a684ff5b243b3b42faceb5e : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77ab8b89531486f97c5d83557b7da8e : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f67b81fa0af7495a8ea90a5c1f68bd72 : SUCCESS in 4m 47s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/acd8c4c13c9f497ebcea4ed048331639 : SUCCESS in 1h 50m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bbf221098afdfcd485b70954033138f7cda4a1a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:45:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"99b92d73ab26c47d9e18ec34012a1642c59890e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:46:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f836dcc7374277a3dedd5c134050aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25eca504891c4076ab98fb283ec94080 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c614b088d634c35b2aa72a1d4126e67 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"d1b6a3dc1a03f74016ccde46cdf5eee6512c6bd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-11 18:54:34.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/440 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e768fb5ca67f1332c24ffa761c75bd0383cca4a5","revisions":{"e768fb5ca67f1332c24ffa761c75bd0383cca4a5":{"kind":"REWORK","_number":1,"created":"2022-08-09 20:25:05.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/72/852672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/72/852672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/72/852672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/72/852672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/72/852672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/72/852672/1"}}},"commit":{"parents":[{"commit":"871c28f536924c5ac7f642d8f5b28dae9b6ceaae","subject":"Merge \"Support shared_targets tristate value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/871c28f536924c5ac7f642d8f5b28dae9b6ceaae"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-09 20:08:40.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-09 20:19:32.000000000","tz":-240},"subject":"Add flake8-logging-format extension","message":"Add flake8-logging-format extension\n\nAutomated logging format checking was added to cinder in ussuri\nby change I1dedc0b31f78f518c, but it wasn\u0027t added to os-brick and\na few infelicities have crept into the code.  Enable the extension\nand correct the issues.\n\nTo be consistent with cinder, we don\u0027t enable G200 (\"Logging\nstatement uses exception in arguments\").\n\nChange-Id: I650b92bde5509b654d16301ffe50056e1e5ba579\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e768fb5ca67f1332c24ffa761c75bd0383cca4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e768fb5ca67f1332c24ffa761c75bd0383cca4a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
