)]}'
{"id":"openstack%2Fos-brick~853787","triplet_id":"openstack%2Fos-brick~master~I08440acb4582e29e4c6398be524e790957c9c6d2","project":"openstack/os-brick","branch":"master","topic":"bp/add-lustre-driver","attention_set":{},"removed_from_attention_set":{"10848":{"account":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"last_update":"2026-02-18 15:12:25.000000000","reason":"Change was abandoned"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-12-22 01:52:28.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-02-18 15:12:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I08440acb4582e29e4c6398be524e790957c9c6d2","subject":"Add Lustre support in remote fs","status":"ABANDONED","created":"2022-08-22 16:41:15.000000000","updated":"2026-02-18 15:12:25.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c21217dab799674a19c0098c6dd33da1b10562e0","_number":853787,"virtual_id_number":853787,"owner":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-02 15:20:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"all":[{"value":1,"date":"2023-02-02 13:14:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},{"value":1,"date":"2023-11-08 22:55:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-03-16 10:01:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 16:54:16.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-08-22 17:10:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-22 17:52:43.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-22 18:08:17.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-08-22 18:52:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 18:52:32.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 19:46:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-22 20:15:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-22 20:38:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 22:47:06.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-08-23 06:11:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-23 06:36:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-21 15:45:01.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 14:16:40.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-02 16:20:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-16 10:01:08.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-03-30 11:50:09.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"CC"},{"updated":"2023-03-30 16:39:09.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-09 21:00:10.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"da298646037fb07560b83486b8de4a478a139930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2022-08-22 16:41:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f64ffa90eee2f6c610e9e6f55a2ef26081cf321f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-22 17:10:42.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"3515abd1d0f73fb8e9b381d74ea94b46fc896a4a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-22 17:52:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661186483045 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1edef3a146f1937e461cfc434561ce093c0f3505","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-22 18:08:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/853787/1/iscsi/ : SUCCESS in 1h 26m 41s\n- os-brick-smb http://cloudbase-ci.com/os-brick/853787/1/smb/ : SUCCESS in 1h 19m 29s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/853787/1/ceph/ : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a22e2062e900df9af580c2a9805790eb38d7f986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:52:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5bf02c3b0146138421c5c5b246d0db\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d2b01d8eb1747f39cde95119facd4ca : SUCCESS in 2h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0272a346b044479b5ea33131939a388 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f3f05d1a26446e49671d66b8c4c5e4f : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3c4224b23fc4f5fbf270896e282e613 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbbbcd2f9741474482c5dfcfc4e45a98 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aeeac730e874e45afe75189835b6000 : SUCCESS in 7m 38s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/96eb44cc9ffa40b1b3d3abb26e8da9ae : SUCCESS in 5m 19s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/84c59f28adb24f20bae1f9b6dc607fc1 : SUCCESS in 1h 22m 10s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/38ebaca1dda643bab134d9209e47b0d4 : SUCCESS in 1h 26m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/08a5422ab812467fb6fd2e1a19c82a4f : SUCCESS in 1h 50m 11s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a4012d9e1f1d4c7382df858992de4b4b : SUCCESS in 1h 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff284734bf9970acd1b12a7b1cef9f430f8b0c95","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-22 18:52:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_PowerFlex_os-brick/42a6d8e/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1905d22a3be07bf81fb70645b96277fdeeec9a4e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-22 19:46:24.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2022_19_29_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"39a14468068f317793723984c27a99a12c77154d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-22 20:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/853787/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c19c446/ : SUCCESS in 1h 40m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/853787/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c0e7f6f/ : SUCCESS in 1h 33m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/853787/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2ce56ba/ : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c90d92486e4ef0f4870b1951e9c981a83c9d81ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-22 20:38:36.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/853787/1/check/devstack-plugin-ceph-dsvm-focal-py3/a613c17/ : TIMED_OUT in 3h 14m 02s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/87/853787/1/check/tempest-dsvm-full-focal-py3/124e0f3/ : SUCCESS in 2h 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7c9e3c04dcd6d09886fbe19aad79cd39c21cc1ff","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-08-22 22:47:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a4b209ab6d745e4b0a260b651c9052260a66d00","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-23 06:11:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_PowerStore_FC-ob/8676a59/DellEMC_PowerStore_FC-ob/166 : FAILURE in 1h 17m 01s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_PowerStore_NFS-ob/9ae5f78/DellEMC_PowerStore_NFS-ob/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_PowerStore_NVMeTCP-ob/8c02d93/DellEMC_PowerStore_NVMeTCP-ob/43 : FAILURE in 1h 47m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"3fa522cff865bcb55b142aa12a32fb2a53528847","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-23 06:36:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_Unity_FC_os-brick/9b535d4/DellEMC_Unity_FC_os-brick/210 : FAILURE in 16m 50s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/853787/1/check/DellEMC_Unity_iSCSI_os-brick/b57c425/DellEMC_Unity_iSCSI_os-brick/225 : SUCCESS in 1h 46m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc089ee49f842ee980dae006eba7757c1c20e25f","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2022-12-21 14:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdccc145c703e717d1fc047e9c307156e08fe44a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2022-12-21 14:38:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"69a400fa6cc0eb56943ca027fcd23e830e565451","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-21 15:45:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/853787 : FAILURE in 1h 05m 12s","accounts_in_message":[],"_revision_number":2},{"id":"46cf4e77b94c4570ab30628ad8a2a1b109f119dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 16:18:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf6840ea73b440ca58e34e6d21a7e07\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a41e928d4407477ba5e1ff186e7de8a4 : SUCCESS in 1h 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5565feceb6944db0ad85112a1e7e30ff : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f98f1e47eb0b4cfb83989ba038a43d59 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16cd7d0094c94a75b17ada156b29aa73 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a368c51a2a43e09ae37d80bfea3fb2 : SUCCESS in 5m 00s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b1210f4e9d9343bfb10626d3442b1554 : SUCCESS in 3m 34s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/d5ffd11e0ad14ac7808ee3cad32719eb : SUCCESS in 1h 12m 01s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/5113e151f3df4f4382e8a5554beb25b6 : SUCCESS in 1h 09m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8430837879414b5081f42989af294fc3 : SUCCESS in 1h 31m 15s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8b5298967c22440d8b65c6bb597d49ce : SUCCESS in 1h 25m 53s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4cb311b2666c4cbe81c6e443062bfca0 : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d401af21bd78fcc8f3eab99a803d9c4dad94d89a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-21 16:18:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/87/853787/2/check/DellEMC_PowerFlex_os-brick/001fb2f/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8f9b086c57cbd5a097ca1b1779d76aa3a3422ee5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-21 16:59:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/853787/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/760b66d/ : SUCCESS in 1h 53m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/853787/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/4ac03f7/ : SUCCESS in 1h 59m 31s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/853787/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ed4dbb7/ : FAILURE in 2h 13m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/87/853787/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/557f8f5/ : SUCCESS in 2h 12m 20s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"acb08d5c925aab2561be44a0c51fd02b1eb1c6ef","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-12-21 20:04:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/853787/2/check/DellEMC_PowerStore_FC-ob/523a5cf/DellEMC_PowerStore_FC-ob/199 : SUCCESS in 1h 25m 55s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/87/853787/2/check/DellEMC_PowerStore_iSCSI-ob/cb6d9d5/DellEMC_PowerStore_iSCSI-ob/120 : SUCCESS in 1h 27m 46s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/87/853787/2/check/DellEMC_PowerStore_NFS-ob/d6bc67b/DellEMC_PowerStore_NFS-ob/92 : SUCCESS in 1h 17m 17s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/87/853787/2/check/DellEMC_PowerStore_NVMeTCP-ob/225db01/DellEMC_PowerStore_NVMeTCP-ob/72 : SUCCESS in 1h 26m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"ec4fe17cbe0e7f9c611c1e6df029833428b9def4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-22 01:08:07.000000000","message":"Patch Set 2:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/901/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"5c0dc4d381d7038e1b72a78e3c16526480b2c293","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-22 01:52:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8987929aea2da6a3abf4e869ed3b483752a9052a","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2022-12-29 14:39:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae97fb4b0917621983f56a0c56f247015d74e075","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2022-12-30 08:57:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"11181840395cbc1780d9c41ae044c5fd463e1408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-02-01 14:11:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6e9a77b5633478aad89014680c8232a385cce696","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-01 14:16:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8de619b1ea214906a93462e989602ca4 : POST_FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"11d0c0212553ca407d9bf2892e01421c6bf66d91","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-01 14:40:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/853787 : FAILURE in 23m 50s","accounts_in_message":[],"_revision_number":3},{"id":"44a2bd29550e47e733c423ddbd92eb8263d722cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 16:05:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/982c28fec4af4641a5f45f5f532a4603\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/346d712dd5c443419439a95879397f61 : SUCCESS in 1h 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeaa652984a84331ab654ee28c177ca7 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03be7f02063b48bf8a584ef62dbb691e : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/689ebe12b3de428581492309efbc959c : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfa7673c5fb4b4ea722194ee18e8226 : SUCCESS in 7m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9b028e9f2c7e433b8e1668438c6a7532 : SUCCESS in 4m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/80482e382b7443c6bd36f6cd35083409 : FAILURE in 1h 32m 21s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9b1e9f0303ee49bb82a816e93e864b6a : SUCCESS in 1h 26m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/262980dcc2ea410eb7b63476167e26b5 : SUCCESS in 1h 46m 03s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/3c89aa3582904712bc01d95333c65942 : SUCCESS in 1h 35m 53s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/83864ec0d209457aa67ede4dff579b0a : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e803432cb635accf6d44246ed9249a4c2a40f323","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-01 16:56:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/87/853787/3/check/DellEMC_Unity_FC_os-brick/3bd279f/DellEMC_Unity_FC_os-brick/255 : SUCCESS in 1h 28m 49s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/853787/3/check/DellEMC_Unity_iSCSI_os-brick/fd9ea91/DellEMC_Unity_iSCSI_os-brick/269 : FAILURE in 2h 45m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1b49c1f78826647ab4e418a36639d29796fafa79","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-02-01 17:01:52.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34f4d193580dee0ffd0965b433c1e10d3bf35af8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 18:23:52.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e5e15e38516648afac3654b1c73c5a4a : POST_FAILURE in 1h 37m 33s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"c6c85f334c135223d4cb58b4b335b329af73691f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-02-02 06:48:41.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"100d57c51452afcdb99214186e8477d966f9437f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-02 06:54:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/ce88cb0820b54ebda09b95092320811b : POST_FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d77dd6f39fa25b952fc8d6b7a2af291c0753cd6a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-02 07:13:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/853787 : FAILURE in 23m 36s","accounts_in_message":[],"_revision_number":4},{"id":"03ae8c4062ba03c7c94c826c488b4d8aca409156","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-02 08:43:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/64d590a/ : SUCCESS in 1h 41m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6db7f74/ : SUCCESS in 1h 52m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/52a3242/ : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":4},{"id":"cbfe316e86f486979344bf27d48b37138f6da9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 09:06:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6dde218128247119a507b3fab7159ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f81588f89d5429b82a32fdbcd8df28d : TIMED_OUT in 2h 04m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339ccb98d8d64332a71ace30596e5bab : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbb09b07807a46c4909d6fdadbf2690c : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84a56d9bfefe46ada57ee1795b802344 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42621ab94b5f4e55a326ee3f989ea714 : SUCCESS in 6m 52s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/65980b371f094282b709f90b6ee4112c : SUCCESS in 4m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/71eeeb9e59ce4079a683e12c71b035e9 : SUCCESS in 1h 50m 25s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e1a4e1ebd8514c1ba93cad8e6bde334c : SUCCESS in 1h 23m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ce5c1232a1264da8a8478048c3fbf533 : SUCCESS in 2h 07m 20s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cb117518f148495aac1068680d405443 : SUCCESS in 1h 44m 47s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f62d7dae5ce54d219c374371b2bde3a5 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edae0bd7feb6eff1435593ae868dc7e3b5644b3d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-02 11:08:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_Unity_FC_os-brick/60cc7d2/DellEMC_Unity_FC_os-brick/257 : SUCCESS in 1h 13m 35s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_Unity_iSCSI_os-brick/0db17e6/DellEMC_Unity_iSCSI_os-brick/271 : FAILURE in 2h 25m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"894df331b7412d4223e6bbd1114c4a64166404d4","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-02-02 13:14:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c69582a1e9bc30b7add09ca59c9c33d4c7d87af","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-02 13:21:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3cb33cf9d6d5455a803b29e98391ea49 : POST_FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":4},{"id":"230e38055bbf82acf3e15363d79621116056a2fc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-02 15:04:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e6961a0/ : SUCCESS in 1h 48m 40s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f3d5d79/ : SUCCESS in 1h 41m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/853787/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7588944/ : SUCCESS in 1h 36m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d607c7d279731e3daaf6579ba7fc7f0e421b6323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 15:20:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55d59cafe3504a31aea357077eec9f35\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cb3070084144fa28ca52ff5f442aa8d : SUCCESS in 1h 57m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5444224aaf744578ce13da492f8f947 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691d059eee6f48668ac5186f49561ee4 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1217c78bfb045ae920a078994633fd5 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9eee2be1b741e0b911c9505b861d97 : SUCCESS in 7m 06s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8a3be6efc8ba40cb8b6d2309cf6aeb1b : SUCCESS in 4m 13s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/0986d4af3ed643aeae79aba0a3b1451e : SUCCESS in 1h 49m 45s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/c887378bd07243aeaa447c73ff9a74ff : SUCCESS in 1h 07m 58s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bd3614e166b244ff97c90a822fb56522 : SUCCESS in 1h 47m 53s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/12d2887653324db1bfcc6d241e83f9d1 : SUCCESS in 1h 27m 43s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/eae0cebf9c1b4e258239e2f6c322722e : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59707d868bbf12db7bf57d142507ded9ee31859d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-02 16:20:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/853787/4/upstream-check/cinder-tempest-plugin-ontap-fcp/a7f438e/ : FAILURE in 57m 58s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/853787/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/e5d430c/ : SUCCESS in 43m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/853787/4/upstream-check/cinder-tempest-plugin-ontap-nfs/96845ce/ : SUCCESS in 42m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/853787/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/208cf89/ : SUCCESS in 43m 36s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"11aba080a82be4ded3e96fc557fbf66a30e30aa0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-02 16:54:31.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_PowerStore_FC-ob/e00a167/DellEMC_PowerStore_FC-ob/202 : FAILURE in 1h 33m 22s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_PowerStore_iSCSI-ob/03be441/DellEMC_PowerStore_iSCSI-ob/123 : SUCCESS in 1h 29m 01s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_PowerStore_NFS-ob/5e28e2a/DellEMC_PowerStore_NFS-ob/95 : SUCCESS in 1h 18m 14s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_PowerStore_NVMeTCP-ob/109a1e3/DellEMC_PowerStore_NVMeTCP-ob/75 : SUCCESS in 1h 25m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"692d6c121c7616e8f420b1c38749f01ed39ffd4e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-02 20:09:11.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0e8bf4a4655d4e62b0f1c0dd348bd7e1 : POST_FAILURE in 1h 26m 23s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/99238c9df5bb42ed84bce49ce4019f6a : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d18ff7dd57d0f22ed6ddfac289bca745909522c6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-02 23:44:24.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_Unity_FC_os-brick/f7becfe/DellEMC_Unity_FC_os-brick/258 : SUCCESS in 1h 19m 35s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/87/853787/4/check/DellEMC_Unity_iSCSI_os-brick/74185c9/DellEMC_Unity_iSCSI_os-brick/272 : FAILURE in 2h 22m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"19af8d7e215f3a01f3c637f5217c838d873c803c","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-03-13 16:33:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c92c7a37829228d1a2299ca16b2e6d970facf6a","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-16 10:01:08.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9873179f4222f0d860acbd094b143ab75577475b","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-03-30 11:50:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"767c9536d385e4a548bbcafc86a51225fa0ca89d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-30 16:39:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-87-853787-4/FC/ : SUCCESS in 2h 27m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-87-853787-4/iSCSI/ : SUCCESS in 1h 54m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"be8d3c934348f3720ae8881edc41f8e04ee11387","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-04-03 12:35:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2315abbf865b52df14b4bdc3f09df340a0a7311","author":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"date":"2023-04-14 15:51:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6227c8c1948539a3e0e1c2586430b0c7371b332e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-08 22:55:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"927817796d40d3fffb60cd3481ad7d7dd1e6186c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-11-09 21:00:10.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/87/853787/4/fc : SUCCESS in 0h 41m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/87/853787/4/iscsi : SUCCESS in 0h 55m 42s","accounts_in_message":[],"_revision_number":4},{"id":"c21217dab799674a19c0098c6dd33da1b10562e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-18 15:12:25.000000000","message":"Abandoned\n\nNo activity in over two years on this and the cinder patch.  Feel free to restore if you are still interested in working on this.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a40f15da59809a58acbb3778f9a67b716f8d6fe4","revisions":{"be47bbd06b23dbcc3d551087cc8dc90a6032a107":{"kind":"REWORK","_number":1,"created":"2022-08-22 16:41:15.000000000","uploader":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"ref":"refs/changes/87/853787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/853787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/853787/1"}}},"commit":{"parents":[{"commit":"2a071914f53cf382e296578b2227c8bc1ab199ca","subject":"Bump mypy version to 0.960","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2a071914f53cf382e296578b2227c8bc1ab199ca"}]}],"author":{"name":"Jean-Sebastien BEVILACQUA","email":"realitix@gmail.com","date":"2021-08-11 09:43:25.000000000","tz":120},"committer":{"name":"Jean-Sébastien Bevilacqua","email":"realitix@gmail.com","date":"2022-08-22 16:41:00.000000000","tz":120},"subject":"Add Lustre support in remote fs","message":"Add Lustre support in remote fs\n\nChange-Id: I08440acb4582e29e4c6398be524e790957c9c6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/be47bbd06b23dbcc3d551087cc8dc90a6032a107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/be47bbd06b23dbcc3d551087cc8dc90a6032a107"}]},"branch":"refs/heads/master"},"1ee4a3c5dfe555707435c50d0c6ff507b6552fd4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-21 14:38:03.000000000","uploader":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"ref":"refs/changes/87/853787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/853787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/853787/2"}}},"commit":{"parents":[{"commit":"c3f14b467f4196ff166132e6f3f069bbf1dd9994","subject":"Merge \"nit: correct spelling of Rescanning in debug log\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c3f14b467f4196ff166132e6f3f069bbf1dd9994"}]}],"author":{"name":"Jean-Sebastien BEVILACQUA","email":"realitix@gmail.com","date":"2021-08-11 09:43:25.000000000","tz":120},"committer":{"name":"Jean-Sébastien Bevilacqua","email":"realitix@gmail.com","date":"2022-12-21 14:37:46.000000000","tz":60},"subject":"Add Lustre support in remote fs","message":"Add Lustre support in remote fs\n\nChange-Id: I08440acb4582e29e4c6398be524e790957c9c6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1ee4a3c5dfe555707435c50d0c6ff507b6552fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1ee4a3c5dfe555707435c50d0c6ff507b6552fd4"}]},"branch":"refs/heads/master"},"c1fddacc1fb2e2f6739ad2e56eba110698540862":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-01 14:11:12.000000000","uploader":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"ref":"refs/changes/87/853787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/853787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/853787/3"}}},"commit":{"parents":[{"commit":"655fcc41b33d3f6afc8f85005868d0111077bdb5","subject":"Merge \"Handle FileNotFoundError on get_system_uuid()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/655fcc41b33d3f6afc8f85005868d0111077bdb5"}]}],"author":{"name":"Jean-Sebastien BEVILACQUA","email":"realitix@gmail.com","date":"2021-08-11 09:43:25.000000000","tz":120},"committer":{"name":"Jean-Sébastien Bevilacqua","email":"realitix@gmail.com","date":"2023-02-01 14:11:02.000000000","tz":60},"subject":"Add Lustre support in remote fs","message":"Add Lustre support in remote fs\n\nChange-Id: I08440acb4582e29e4c6398be524e790957c9c6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c1fddacc1fb2e2f6739ad2e56eba110698540862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c1fddacc1fb2e2f6739ad2e56eba110698540862"}]},"branch":"refs/heads/master"},"a40f15da59809a58acbb3778f9a67b716f8d6fe4":{"kind":"NO_CHANGE","_number":4,"created":"2023-02-02 06:48:41.000000000","uploader":{"_account_id":10848,"name":"JS","email":"realitix@gmail.com","username":"realitix"},"ref":"refs/changes/87/853787/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/853787/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/853787/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/853787/4"}}},"commit":{"parents":[{"commit":"655fcc41b33d3f6afc8f85005868d0111077bdb5","subject":"Merge \"Handle FileNotFoundError on get_system_uuid()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/655fcc41b33d3f6afc8f85005868d0111077bdb5"}]}],"author":{"name":"Jean-Sebastien BEVILACQUA","email":"realitix@gmail.com","date":"2021-08-11 09:43:25.000000000","tz":120},"committer":{"name":"Jean-Sébastien Bevilacqua","email":"realitix@gmail.com","date":"2023-02-02 06:48:33.000000000","tz":60},"subject":"Add Lustre support in remote fs","message":"Add Lustre support in remote fs\n\nChange-Id: I08440acb4582e29e4c6398be524e790957c9c6d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a40f15da59809a58acbb3778f9a67b716f8d6fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a40f15da59809a58acbb3778f9a67b716f8d6fe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
