)]}'
{"id":"openstack%2Fos-brick~853789","triplet_id":"openstack%2Fos-brick~master~I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6","project":"openstack/os-brick","branch":"master","topic":"mypy","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-19 16:19:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6","subject":"mypy: improve executor coverage","status":"MERGED","created":"2022-08-22 19:25:18.000000000","updated":"2026-08-19 16:21:04.000000000","submitted":"2026-08-19 16:19:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853789-mypy","meta_rev_id":"f3d2c896ae1435de876eee99017d2eef9f109c0c","_number":853789,"virtual_id_number":853789,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 16:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-08-19 13:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-08-19 14:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-08-19 14:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 19:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 19:36:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-22 20:00:00.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-22 21:01:53.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-08-22 23:43:38.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 23:44:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-23 14:00:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-23 16:59:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-23 20:45:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-23 23:34:29.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-25 14:03:22.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-02-10 15:58:35.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-10 16:02:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-10 19:08:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-14 05:20:31.000000000","updated_by":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"reviewer":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"state":"CC"},{"updated":"2023-04-10 19:30:40.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-12 10:27:38.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-12 12:30:25.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-13 19:36:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-17 22:00:11.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2026-08-03 15:58:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-19 13:26:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-08-19 14:17:16.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"49aabc2a314f9a8aab55fd90bcd6920246f88439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-22 19:25:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8b140a3fb8e304451d0f59fea3965e9e5b2b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:25:50.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 849021,6","accounts_in_message":[],"_revision_number":1},{"id":"a97512e92441751a5d2eb47944fe992069ec138a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-22 19:36:36.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"3793cf98f0a6008ee3ee16245afdfbd4c5fe707a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-22 20:00:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661196328552 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"57ad896fcb64ab84326ff3c1b9882a4d137fb7cb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-22 21:01:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/853789/1/iscsi/ : SUCCESS in 1h 30m 23s\n- os-brick-smb http://cloudbase-ci.com/os-brick/853789/1/smb/ : SUCCESS in 1h 27m 34s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/853789/1/ceph/ : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":1},{"id":"da38c4b6e16735bbac31646f1879a4629eddb2c2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-22 23:43:38.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/853789/1/check/devstack-plugin-ceph-dsvm-focal-py3/fdf05d5/ : TIMED_OUT in 3h 08m 15s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/853789/1/check/tempest-dsvm-full-focal-py3/f622256/ : TIMED_OUT in 3h 09m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9502a179f451d5fbcd0e2797d5bc348b9ab68864","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-22 23:44:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/89/853789/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/30ce3db/ : SUCCESS in 1h 29m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/89/853789/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fbb2235/ : SUCCESS in 1h 39m 13s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/89/853789/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/3f4b081/ : SUCCESS in 1h 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5adf93866537d2b6da38a2e86bdbebab8bd44a3a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-23 14:00:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/1/check/DellEMC_Unity_FC_os-brick/428819b/DellEMC_Unity_FC_os-brick/212 : FAILURE in 17m 41s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/1/check/DellEMC_Unity_iSCSI_os-brick/3c74731/DellEMC_Unity_iSCSI_os-brick/227 : SUCCESS in 1h 48m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"51c2f9ed1d9ccf55d31b782aba1f46b1675fd29f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-23 15:16:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"be6e55af8b28ac2284fce926de388b2877d4ff41","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-23 15:16:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"25fcfcdfa7d646b9c817382010277c1e53bf336d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-23 15:18:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"1a0f7bcaa312019eb3f71bc6277d95f2b417e269","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-23 15:18:59.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"6394a780e2ce5e6e3cb4df27473f118674340eb6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-23 15:21:18.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"f19ff710b2539f209738aa50e904c96acbcf121f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-23 16:53:07.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661267815641 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d1a5e81eb60b4ef38e58ef20986769e8995c759d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 16:59:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82aaee8707e34e06abe7bda61167f809\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a50429f9ef1c494cb4c1af4749481721 : SUCCESS in 1h 40m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb9e2d1959e4a1cb7e69836fdd74aeb : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3728b0828edc405a90af8aa93f034dfc : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd2ef1901c4b41a08d12c82892082326 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ab2d2a3ec4c483b824794e7f795596f : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1723b5f89b454cb0aec908d5f884aff1 : SUCCESS in 8m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a876885833514aa291896d0deec4f6d8 : SUCCESS in 6m 00s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/286d55ffa7bc41ef82e54ab41934333a : SUCCESS in 1h 20m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/57f31164e079484cb9bba65829a3f3c5 : SUCCESS in 1h 37m 36s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/11fd08bc3d294fb5814cabaea0f52d05 : SUCCESS in 1h 30m 14s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b817b394387b4de38ab86de84eb6e271 : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e62b82efacdbc912cb2505567f4ee2fdd8849fc7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-23 16:59:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_PowerFlex_os-brick/8561020/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8169b8b88cfb4bb18ebc575d24a5c00ebe09d527","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-23 20:29:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/89/853789/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/dc4ecbb/ : SUCCESS in 1h 46m 51s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/89/853789/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c3b3960/ : FAILURE in 30m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/89/853789/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/aec3bab/ : SUCCESS in 1h 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"3e51851476a38e9530ed57d229e77737dd9da1af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-23 20:45:01.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/23_08_2022_20_24_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"0df0c543bb634bab89b166e8427db20ddbc7a3b0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-23 20:47:37.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/853789/2/check/devstack-plugin-ceph-dsvm-focal-py3/81e4830/ : TIMED_OUT in 3h 16m 08s\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/os-brick/89/853789/2/check/tempest-dsvm-full-focal-py3/4c6322c/ : SUCCESS in 2h 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"02ae670da3ea1feb72a652d0a96aefb103821c2d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-23 23:34:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_PowerStore_FC-ob/f41ec02/DellEMC_PowerStore_FC-ob/172 : FAILURE in 1h 20m 43s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_PowerStore_NFS-ob/2bab3fa/DellEMC_PowerStore_NFS-ob/68 : SUCCESS in 1h 30m 51s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_PowerStore_NVMeTCP-ob/8a7c068/DellEMC_PowerStore_NVMeTCP-ob/49 : FAILURE in 1h 51m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"dda6167901e344903170fc4041930b76521d16be","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-24 12:17:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_Unity_FC_os-brick/082fa68/DellEMC_Unity_FC_os-brick/221 : FAILURE in 14m 31s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/2/check/DellEMC_Unity_iSCSI_os-brick/d78442a/DellEMC_Unity_iSCSI_os-brick/236 : SUCCESS in 1h 52m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"883af36fea1192c583a3c26660cdcc47432c12a5","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-25 14:03:22.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"f1cab7ed18ed7ba676c1a748a6f90f6fdfcba949","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-25 14:03:35.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"74a64f508ae3227bd58df8fd7fde23823a4cb5ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-31 14:12:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8fe82e0a2190d4e1901d2b6221bd97a402d6481b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-08-31 15:20:49.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/853789/3/iscsi/ : SUCCESS in 53m 36s\n- os-brick-smb http://cloudbase-ci.com/os-brick/853789/3/smb/ : SUCCESS in 1h 03m 19s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/853789/3/ceph/ : FAILURE in 1h 08m 12s","accounts_in_message":[],"_revision_number":3},{"id":"058667c0e78ce6f389f3e7dc54dff99c58c54eca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:16:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc656d0b974847e693eaf886a69a281b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91597ee2c55a492481355a28038b9d89 : SUCCESS in 1h 59m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5058bfe08ae34da8980c9ef470ced2db : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d18776a1b8b14009ad1e99369bd5a125 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70d6efac9dd241c4acd7e5c9349994db : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c5cbace87644dfbea1c99ac9927ec7 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fea3e60d7f4bfd9feebe7f42da8cf5 : SUCCESS in 10m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b8b13c697c494789b3ddba48f889071e : SUCCESS in 7m 59s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/0c10efc8df5c47559d2dfdfa8b51fa94 : SUCCESS in 1h 37m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d63e4c9f20a40cf9706febdaacd4a8e : SUCCESS in 1h 59m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/08bbd54e6407482e9459e2bad07a5134 : SUCCESS in 1h 43m 59s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/0de0db370a8c4a13b77e4f64fa93bf96 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b71d839cf0980456a584b44ab668cea7ae5f0b87","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-31 16:17:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_PowerFlex_os-brick/5c8b836/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"682979c48400db9c81be0d860a6ba4d6fef5aec9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-31 16:24:15.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/31_08_2022_16_16_48\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"b0ad967f06423dc1ebf5b03ba65ef134bc2dbe53","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-31 22:41:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/89/853789/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bedef25/ : SUCCESS in 1h 42m 39s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/89/853789/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/37ecf79/ : SUCCESS in 1h 35m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/89/853789/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/446daeb/ : SUCCESS in 1h 52m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7df4fdc2698359b3005b088afc9f460884f6a829","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-01 20:46:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_PowerStore_FC-ob/63ca30a/DellEMC_PowerStore_FC-ob/185 : SUCCESS in 1h 19m 36s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_PowerStore_iSCSI-ob/6494a51/DellEMC_PowerStore_iSCSI-ob/106 : ABORTED in 3h 00m 06s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_PowerStore_NFS-ob/e64599d/DellEMC_PowerStore_NFS-ob/80 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_PowerStore_NVMeTCP-ob/0772d3f/DellEMC_PowerStore_NVMeTCP-ob/61 : FAILURE in 1h 38m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"b2283bb51de6cb85ff30ab23cf4fbe089f8382e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-02 03:11:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_Unity_FC_os-brick/9db1acd/DellEMC_Unity_FC_os-brick/231 : SUCCESS in 1h 22m 25s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/3/check/DellEMC_Unity_iSCSI_os-brick/1da4829/DellEMC_Unity_iSCSI_os-brick/245 : FAILURE in 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b52a785f277827a1c113df5c619314c6eafc05d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-02 07:48:00.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/530","accounts_in_message":[],"_revision_number":3},{"id":"227b1a3652b825c05341c3f44f3ab9db40129ba9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-14 14:41:23.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d187a25918de391ae28590eb323dac9e69134c2a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-09-14 15:36:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/853789/4/iscsi/ : SUCCESS in 52m 01s\n- os-brick-smb http://cloudbase-ci.com/os-brick/853789/4/smb/ : SUCCESS in 53m 02s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/853789/4/ceph/ : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":4},{"id":"facc8e4d3916939bb0801630d95619a7eb4daee5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-14 16:32:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/89/853789/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cd65687/ : SUCCESS in 1h 42m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/89/853789/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7ab95eb/ : SUCCESS in 1h 50m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/89/853789/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9efd5c6/ : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":4},{"id":"88bf6bc351faabb864999e721af3990c36914249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 16:33:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba3ed2d9a9c400abb085df47492e5db\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83f87319d2245978e0a21479ad03590 : SUCCESS in 1h 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c5b3ae5f7f40de90eb2b13b6b6d9a3 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceb903615b0e4f13ad78f93c7b7f7377 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/807be52ea38e4f24a4b14399a045fee4 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea725d5fac140519e89c7c04a84e330 : SUCCESS in 5m 29s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d2378b031c3448bd84d8a702db1c20cd : SUCCESS in 3m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/705787a46b4a4674b8c46f01c1aea85c : SUCCESS in 1h 22m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4fafc587704b42ceb9900fb886e5c5c8 : SUCCESS in 1h 44m 51s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/22079d00d0664003aa9e1b26ae0d9395 : SUCCESS in 1h 30m 05s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/534291f6b43e4b928896bb64626c9f3b : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a0b1aeee84ef48fa990c32d09fbf5c048cd854d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-14 16:33:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_PowerFlex_os-brick/4f27112/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c0c01e102e1678cd7f0fa7556938738ac6d23d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-09-14 16:41:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_09_2022_16_33_20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"6bf2d17b7ad3477cf1639d76bd0e4aa925afc839","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-14 17:18:36.000000000","message":"Patch Set 4:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/623","accounts_in_message":[],"_revision_number":4},{"id":"0b9dd2f2aabfdba8bf85165135eddfd0f26d21aa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-14 22:07:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_PowerStore_FC-ob/e79357e/DellEMC_PowerStore_FC-ob/192 : SUCCESS in 1h 22m 59s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_PowerStore_iSCSI-ob/893eae0/DellEMC_PowerStore_iSCSI-ob/113 : FAILURE in 1h 57m 10s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_PowerStore_NFS-ob/4701162/DellEMC_PowerStore_NFS-ob/88 : SUCCESS in 1h 37m 25s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_PowerStore_NVMeTCP-ob/43c39fb/DellEMC_PowerStore_NVMeTCP-ob/68 : SUCCESS in 1h 31m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"a8f0f787061ab4268b04c5a72da04b8084447874","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-15 04:16:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_Unity_FC_os-brick/0810cda/DellEMC_Unity_FC_os-brick/241 : FAILURE in 1h 16m 53s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/4/check/DellEMC_Unity_iSCSI_os-brick/bbc9e9c/DellEMC_Unity_iSCSI_os-brick/255 : FAILURE in 2h 06m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1fb7a06d638ffb68ac68c9db56bb43717db162a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-02-10 15:38:07.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d82dd403141812c106c46edffcf28c2b49dc0701","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-10 15:58:35.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/ff571923ab5946fa8d83f0ca1988125e : POST_FAILURE in 19m 53s","accounts_in_message":[],"_revision_number":5},{"id":"6344c16880696f2309c2c18f8e6fd433acbaf633","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-10 16:02:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/853789 : FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":5},{"id":"ba0b2985895d8f9f9bbf1933470477667b57c90d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-10 18:09:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f4f6e72/ : SUCCESS in 1h 38m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0a902ac/ : FAILURE in 2h 12m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e3ca3e4/ : SUCCESS in 1h 38m 39s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/f5f7df7/ : SUCCESS in 1h 55m 09s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/4ce0f8b/ : FAILURE in 2h 30m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/89/853789/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/a3d8db6/ : SUCCESS in 1h 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c07442670d20bcf039e3ec169ccd82c86ce06e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 18:10:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3e978e63454b46803eacf06f370338\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bb7d40eb7de41d1b31099170c11b948 : SUCCESS in 1h 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5befc7288b47da80786d2d1668eab9 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a3bfefc4cc24947a0009f35dade471a : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4595358486114f4cab2b2af0e5a12239 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2b9b74d1854cc4a0ea5e8a4de41890 : SUCCESS in 7m 34s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9c56679f6af24d06b202e317f08ce819 : SUCCESS in 4m 28s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/6f51ab24fd7f4efa9e575df70f2c2718 : SUCCESS in 1h 08m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/478031d5cefc4e3ca14e7fb5dbfa51f5 : SUCCESS in 2h 07m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6fac5186590c4f8c82b99943dc286ec3 : SUCCESS in 1h 23m 27s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/863c7c7646df4e5db4045214995384d0 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe2f1b88201ff298b6599f1adc5b6b0fdee13d1e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-10 19:08:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/5/upstream-check/cinder-tempest-plugin-ontap-fcp/c607d0d/ : SUCCESS in 57m 41s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/5daefe8/ : SUCCESS in 47m 23s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/5/upstream-check/cinder-tempest-plugin-ontap-nfs/af06729/ : SUCCESS in 45m 15s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0f07f69/ : SUCCESS in 42m 40s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"308187dc64160da449890189420f86cb16db2e06","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-10 22:13:10.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/5/check/DellEMC_Unity_FC_os-brick/94e973e/DellEMC_Unity_FC_os-brick/267 : SUCCESS in 1h 26m 20s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/5/check/DellEMC_Unity_iSCSI_os-brick/20f18db/DellEMC_Unity_iSCSI_os-brick/281 : SUCCESS in 1h 38m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"58cf6a3cafec795aee972d6c379fcac643c0a401","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-11 04:52:55.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b435e763fac347d4b4406b0e7e21549c : FAILURE in 2h 11m 51s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"7af459e870056189ff79152b548c6f33b7e23b4c","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-02-14 05:20:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5248a71ac5194bdd25f9d365ec2c70c1a727338d","author":{"_account_id":34148,"name":"Aboubacar Diare","display_name":"abdi","email":"diare@hpe.com","username":"abdi"},"date":"2023-02-14 05:23:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab1c0bfe34a0b8267df954b2d735fc70d570f52b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-10 15:10:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"64daefcedd054747e7725a7be16d8ad9256f5c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 17:06:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2c7f118037f4b1c85f047912cff9901\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa19f13bb8b349f2a1c2829ee549cabc : SUCCESS in 1h 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b8802f5b7940a5be48c6f62bea2e06 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/748cbc27ea3c4d2081203e8567806fcc : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f713f669657470da50f5799390390af : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa0f9bea31243b9bb869104342b773f : SUCCESS in 6m 51s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9a0e6cef0016404dade9ee0c96a4b091 : SUCCESS in 3m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e3384828b8dc4abe96514c2c024fee76 : SUCCESS in 1h 23m 02s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df97820d8d2c460dab7cba697e3afd84 : SUCCESS in 1h 51m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/32283a9fdeee4ef39a38461f4074b334 : SUCCESS in 1h 31m 48s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f0ecafb30daa476f830eb62e60451191 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4bc78e3d38e1498704ec9b557b29b12f368018c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-10 19:07:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/6/upstream-check/cinder-tempest-plugin-ontap-fcp/be01ba2/ : SUCCESS in 1h 59m 59s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/f7df93e/ : SUCCESS in 40m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/6/upstream-check/cinder-tempest-plugin-ontap-nfs/18441e5/ : SUCCESS in 43m 28s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1cf9661/ : SUCCESS in 41m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"8072286910c8cedb61a3511ea2b5006055d87360","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-10 19:30:40.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-853789-6/FC/ : SUCCESS in 2h 59m 13s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-89-853789-6/iSCSI/ : SUCCESS in 2h 45m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"0f491494f748b2711485127d2344b587d43647d5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-10 20:56:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/89/853789/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/1da5092/ : SUCCESS in 1h 57m 02s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/89/853789/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/1f7a855/ : FAILURE in 1h 54m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/89/853789/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/21ffc9a/ : SUCCESS in 1h 42m 26s","accounts_in_message":[],"_revision_number":6},{"id":"68de3d3bc8a8c9feae5c4ba909ab6afbf37d3cfa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-10 22:00:38.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/os-brick/89/853789/6/check/devstack-plugin-ceph-dsvm-focal-py3/c55eaf5/ : TIMED_OUT in 3h 13m 22s\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/os-brick/89/853789/6/check/tempest-dsvm-full-focal-py3/392742a/ : SUCCESS in 2h 23m 52s","accounts_in_message":[],"_revision_number":6},{"id":"e4f4e039948109398adafc00f334fd5f82025e5b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-10 23:35:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/8954531fc8754245ae7ff3a0fb31ee16 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":6},{"id":"0cb056a159934dde90bd66535494060059640583","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-12 10:27:38.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/6/20230412121131/NEC-FC : SUCCESS in 3h 26m 44s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/6/20230412121131/NEC-ISCSI : SUCCESS in 7h 16m 03s","accounts_in_message":[],"_revision_number":6},{"id":"74d79521ec0ce1bfaf7bfad56b620ea942959a80","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-12 12:30:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/6/20230412181801/NEC-V-ISCSI : SUCCESS in 2h 55m 29s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/6/20230412181801/NEC-V-FC : SUCCESS in 3h 12m 21s","accounts_in_message":[],"_revision_number":6},{"id":"2f2cb7a4135e8f346e770f8d8ae557672f41f6df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-13 19:31:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b6a961b538d4029c56d1232da0a125f01448cabf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 19:36:00.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5290bf49a3d846c984da45b9e4dd50cf : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"b869a4953ca5366f0ce4cd17c6eca1c460055bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 21:19:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/510ba997cea9491eadb18fff8491d9b9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00ecd4fbcb184b67a885d33c42bd433f : SUCCESS in 1h 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5b85e4fdfe43d9a5e2b8f963f2f1c8 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d273b09ddd48dc9e41b32fe92c1666 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5938d67959f499b9c599987f6b9c010 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16f12880ef543958caad7a5d7a3ce2b : SUCCESS in 6m 55s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/298f3bf29bf64a6f88fc955b0a1286c0 : SUCCESS in 3m 52s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/763b00cee83e4afbac3f5e06b94b6c33 : SUCCESS in 1h 40m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3506898480c14decb59b95d0470a778d : SUCCESS in 1h 46m 23s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1f41b8f5ce7242669cc8fb2399bf90d5 : SUCCESS in 1h 07m 15s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/637e20ca4d1f4b6ba91c3f43e3a56086 : SUCCESS in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7436aec3a11aad2296ad7cf3fad92690209fd37d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-13 21:45:59.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/89/853789/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2923cd6/ : SUCCESS in 2h 13m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/89/853789/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/a722ebd/ : FAILURE in 2h 02m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/89/853789/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c61dc7c/ : SUCCESS in 1h 50m 01s","accounts_in_message":[],"_revision_number":7},{"id":"4e26c601b0aa6eb47a0949ee0abf3242b9cb9ccf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-13 22:06:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/7/upstream-check/cinder-tempest-plugin-ontap-fcp/5740e23/ : RETRY_LIMIT in 7m 39s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/4f14095/ : SUCCESS in 46m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/7/upstream-check/cinder-tempest-plugin-ontap-nfs/95c40b3/ : SUCCESS in 46m 17s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/89/853789/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8a1077d/ : FAILURE in 44m 18s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"fc4a202e45aee1c5901f4b167dd493069c544b8a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-14 03:27:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/89/853789/7/check/DellEMC_Unity_FC_os-brick/822868d/DellEMC_Unity_FC_os-brick/31 : FAILURE in 1h 53m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/89/853789/7/check/DellEMC_Unity_iSCSI_os-brick/91fe185/DellEMC_Unity_iSCSI_os-brick/31 : FAILURE in 1h 46m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4c6e49248ea88690a34c5f07d6675ac2e45f556c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-14 13:35:02.000000000","message":"Patch Set 7:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1586/","accounts_in_message":[],"_revision_number":7},{"id":"bea70025aef03cc524c2f8bd0a3c3ed32c63d64c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-15 05:50:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/7/20230715115516/NEC-V-ISCSI : SUCCESS in 2h 33m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/7/20230715115516/NEC-V-FC : SUCCESS in 2h 55m 36s","accounts_in_message":[],"_revision_number":7},{"id":"f0736d009a2e19c2de9400b49502275470795729","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-15 16:51:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/7/20230715172539/NEC-FC : SUCCESS in 3h 47m 45s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/89/853789/7/20230715172539/NEC-ISCSI : SUCCESS in 8h 26m 14s","accounts_in_message":[],"_revision_number":7},{"id":"04156ca50e66ff51c2ed1a0ae4bd013e02b1f78c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-17 22:00:11.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/853789/7/fc : SUCCESS in 0h 54m 16s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/89/853789/7/iscsi : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":7},{"id":"3b68ff3f4657a9c1b25d26331fea0fba2dcec6b2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-03 14:41:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"e59889737e995844b5f03be924113d99b39222d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-08-03 15:01:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"cc6c22e102756a5e4fa20159ca6a6eb24eb5ee2d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 15:15:12.000000000","message":"Patch Set 8:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4827/.","accounts_in_message":[],"_revision_number":8},{"id":"07a6887f812bff636c0c1abaf3f49c127c808c00","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-08-03 15:16:24.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/5003/.","accounts_in_message":[],"_revision_number":8},{"id":"084e26a9a8d07499dce720669900efddb605582b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 15:19:22.000000000","message":"Patch Set 8:\n\nPure Storage CI logs (all jobs): https://github.com/Everpure-Zuul-CI/os-brick-CI/tree/gh-pages/dbf4ab90ac98464d88d41f215f99d43b/","accounts_in_message":[],"_revision_number":8},{"id":"270fbca1c13c07c7a370b2c548e8d6844055e228","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-08-03 15:48:25.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dbf4ab90ac98464d88d41f215f99d43b\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b16677419cff444891355b63be902e8b : SUCCESS in 42m 30s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/49b302e507c84259a926fc42e21a8b77 : SUCCESS in 32m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6b7f2ce7504540fda7f6b3a0a8ca012b : SUCCESS in 46m 26s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/de03ce09a5b9499a85bc7fa21803d061 : SUCCESS in 30m 36s\n- pure-devstack-cinder-tempest-nvme-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/22223d6b833b47e5ad3f96e902d692e4 : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e6362090359029c1b2db0995a56d24499e6ce0e0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 15:58:30.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6 .","accounts_in_message":[],"_revision_number":8},{"id":"fe1a22d237b8d5b5fd72fe8a42d899bd46ff3fb6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-08-03 16:06:44.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/0c047d1ff5e042cc96fdaadf2ac2f5f3 : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":8},{"id":"fa56fb9a5c38523923628a2b6dcd17df43bac77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 16:53:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd2fbde7648f4d839e375180b71e57ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0d0814a137f4217af7c825d37272f3c : SUCCESS in 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86da7abedc5649448b075aead10d3068 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a64702b59de04462bd8a062fc678123c : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4dd839e3aff4246b6a9c2d40d3244f4 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b3b67b3fe72446b691f52db541b1a0a6 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c54fe91cd84249858baafc74a02d30 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d321b8f19d554b6fbbc67ce721433b47 : SUCCESS in 6m 04s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c265dc193813470eb57bcf9f2dddf0b7 : SUCCESS in 3m 07s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/1b39c0b583fa427f82e43ea2bf1a98d8 : SUCCESS in 1h 45m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4fd9e35987842c8a0e8c08913f45a3b : SUCCESS in 1h 46m 36s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/dd82317787d94a70bf30b29e79f798f9 : SUCCESS in 1h 07m 49s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4d2b50d4e78645eb8cd130b494e4c60b : SUCCESS in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4cf9a0bfb798b7ec6eca6c65709f7d5c3c4f7c96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-03 17:04:06.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_08_2026_16_57_13\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"05b9b28cb0fe720e2e6b87ccbab8570dac8bf3ef","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-08-08 17:01:39.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/853789/8/20260808162137/NEC-ISCSI : FAILURE in 9h 24m 59s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/89/853789/8/20260808162137/NEC-FC : FAILURE in 9h 39m 59s","accounts_in_message":[],"_revision_number":8},{"id":"a15c9fce0b348d9cf6ef81e8a3ae5608a3032ab3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-08-19 13:26:34.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a481a51e357cd9334220bfa5e3a1290da2b9817","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-08-19 14:17:16.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"35daedc2f52d3ef37aac5782f0e74506081922a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 14:17:55.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f4ad911e08529c81b80f2f9cb555e2ab7526074a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:19:15.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3334dd0e04a4eedac20a4b848f4801f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22f3faebb2004b86bbcf7bea3b17b293 : SUCCESS in 1h 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53115153288a4e54971b39e3fa7ac55c : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c717cdd6ff5a4e10b9d5517f87e8701c : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84a1190b7c6541708ca305811bb82f26 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab540ebd843249a39aa9a40e919ba9b0 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff0fa771c384d5db692baea9f7d269b : SUCCESS in 6m 50s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f540ec43f36046d4aba1e1641680b39f : SUCCESS in 1h 46m 01s","accounts_in_message":[],"_revision_number":8},{"id":"98925fb1f2dd9ba091acfbefc62c5fba35599eb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:19:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"f3d2c896ae1435de876eee99017d2eef9f109c0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:21:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f16bb783c4345f386d253c975c317ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a467bb0cbbe4410ad42a3c9053a2e05 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"30f7e1d4ef2a0524daad82997398dd0337833ac3","revisions":{"24bf24806973293b636f870afad230098ec2f682":{"kind":"REWORK","_number":1,"created":"2022-08-22 19:25:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/1"}}},"commit":{"parents":[{"commit":"7ec15d594ab6102b1d6453522094e73fa43bddb9","subject":"mypy: lvm.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ec15d594ab6102b1d6453522094e73fa43bddb9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:24:08.000000000","tz":-240},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/24bf24806973293b636f870afad230098ec2f682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/24bf24806973293b636f870afad230098ec2f682"}]},"branch":"refs/heads/master"},"f004ac01e4cb2bd55015740a13668ba204095b28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-23 15:16:47.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/2"}}},"commit":{"parents":[{"commit":"4c64879c39a172aca38156b25077b340b0ebf06e","subject":"mypy: lvm.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4c64879c39a172aca38156b25077b340b0ebf06e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-23 15:16:47.000000000","tz":0},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f004ac01e4cb2bd55015740a13668ba204095b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f004ac01e4cb2bd55015740a13668ba204095b28"}]},"branch":"refs/heads/master","description":"Rebase"},"d44fd92758225802e983a8948a39bc8a69649edd":{"kind":"REWORK","_number":3,"created":"2022-08-31 14:12:17.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/3"}}},"commit":{"parents":[{"commit":"1f928e301519fb8a684d549c06bd0b641edcc252","subject":"Merge \"Add mypy-report directory to .gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1f928e301519fb8a684d549c06bd0b641edcc252"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-31 14:10:39.000000000","tz":-240},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d44fd92758225802e983a8948a39bc8a69649edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d44fd92758225802e983a8948a39bc8a69649edd"}]},"branch":"refs/heads/master"},"a7dc8af1f11ce19b847f2bcfc4e6172c7b27ccfd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-09-14 14:41:23.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/4"}}},"commit":{"parents":[{"commit":"4a12f1b4dad4bbdd30389e8c0e5731766aeeca29","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4a12f1b4dad4bbdd30389e8c0e5731766aeeca29"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-14 14:41:23.000000000","tz":0},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7dc8af1f11ce19b847f2bcfc4e6172c7b27ccfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7dc8af1f11ce19b847f2bcfc4e6172c7b27ccfd"}]},"branch":"refs/heads/master","description":"Rebase"},"daa74c9bef6c04be061969f0de9e05ddc773f438":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-10 15:38:07.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/5"}}},"commit":{"parents":[{"commit":"65f922b16706adf75c4bffe2790ee2a5f1dea412","subject":"Merge \"Update minimum requirements in os-brick\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/65f922b16706adf75c4bffe2790ee2a5f1dea412"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-02-10 15:38:07.000000000","tz":0},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/daa74c9bef6c04be061969f0de9e05ddc773f438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/daa74c9bef6c04be061969f0de9e05ddc773f438"}]},"branch":"refs/heads/master","description":"Rebase"},"455b21411cc692c84531b140d3413d2d857690f8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-10 15:10:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/6"}}},"commit":{"parents":[{"commit":"9496bb2e477e2b519271b021c1bb6aab58df0749","subject":"LVM: Fix supports_full_pool_create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9496bb2e477e2b519271b021c1bb6aab58df0749"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-04-10 15:10:22.000000000","tz":0},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/455b21411cc692c84531b140d3413d2d857690f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/455b21411cc692c84531b140d3413d2d857690f8"}]},"branch":"refs/heads/master","description":"Rebase"},"a38db7b6d473ffd5411c7d34435b3e62889548db":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-07-13 19:31:04.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/7"}}},"commit":{"parents":[{"commit":"6d70ace1bbc8cd94280ac7efc9d33066ff7bbf83","subject":"Merge \"linuxrbd: Remove rados_connect_timeout parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6d70ace1bbc8cd94280ac7efc9d33066ff7bbf83"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-07-13 19:31:04.000000000","tz":0},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a38db7b6d473ffd5411c7d34435b3e62889548db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a38db7b6d473ffd5411c7d34435b3e62889548db"}]},"branch":"refs/heads/master","description":"Rebase"},"30f7e1d4ef2a0524daad82997398dd0337833ac3":{"kind":"REWORK","_number":8,"created":"2026-08-03 15:01:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/89/853789/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/89/853789/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/89/853789/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/89/853789/8"}}},"commit":{"parents":[{"commit":"379b7a5fe27ee22148fe48b68c454059485b8ec9","subject":"Update python tested versions for hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-08-22 19:23:28.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-08-03 15:01:25.000000000","tz":-240},"subject":"mypy: improve executor coverage","message":"mypy: improve executor coverage\n\nChange-Id: I33bcd9c53eaa6a0c33fa90c08d2fb442cfb375b6\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30f7e1d4ef2a0524daad82997398dd0337833ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30f7e1d4ef2a0524daad82997398dd0337833ac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
