)]}'
{"id":"openstack%2Fos-brick~857044","triplet_id":"openstack%2Fos-brick~master~I18b55da0d1f142818e7ea14f6eebcc0f0f0cd23f","project":"openstack/os-brick","branch":"master","topic":"bug/1989280","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2023-02-08 20:03:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18b55da0d1f142818e7ea14f6eebcc0f0f0cd23f","subject":"Fix wrong assertion methods","status":"MERGED","created":"2022-09-12 11:57:12.000000000","updated":"2023-02-09 03:53:40.000000000","submitted":"2023-02-08 20:03:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857044-bug/1989280","meta_rev_id":"f8412d8d704c16cc532e3413b5c44cc5df1d216f","_number":857044,"virtual_id_number":857044,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 20:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-02-05 22:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 07:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-07 07:15:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 12:51:56.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-09-12 13:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 13:48:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-12 13:56:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-09-12 14:29:29.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-12 16:53:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-12 17:30:55.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-12 19:38:41.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-14 15:13:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-14 14:53:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-05 22:37:46.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-06 01:32:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-06 01:40:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-06 05:05:04.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-07 07:15:03.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-02-08 16:09:32.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"}],"messages":[{"id":"b0beffb73649238b6bb249078662df2ee0de1c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-09-12 11:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da415af208d486aace2da86ad7f743613b640122","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-09-12 12:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/857044/1/iscsi/ : SUCCESS in 54m 23s\n- os-brick-smb http://cloudbase-ci.com/os-brick/857044/1/smb/ : SUCCESS in 48m 56s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/857044/1/ceph/ : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a20f3a8739b21c00d2eabfc571ecb7b00083152e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 13:48:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9396947e5f0b45bab370f6e3795c44fc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50083aca51f94355ac677f33aced62f4 : SUCCESS in 1h 50m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec3817546d74a2e9b263ef1984327e7 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e839155ff334891a12fff1a87aced24 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77468a09ad374f8ab00f33d822dd9049 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9413467583d488dbce69ccfeb7a7fd1 : SUCCESS in 7m 17s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9c13376824e14c6da6e2fdebaf3520aa : SUCCESS in 5m 34s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/35b144ad13ae40e794feac3d929c252f : SUCCESS in 1h 24m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2414fa7c4a174616b75edd94b4edc9ab : SUCCESS in 1h 38m 17s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1f763618baa345d4811792f218bbb571 : SUCCESS in 1h 29m 19s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/fea13bd72f214f2c87923498708041cc : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"088f6b9bef9492dcc516491f566c38138242c491","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-12 13:48:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_PowerFlex_os-brick/e0b331e/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"be9a286b1872a2f61bccfea46d410adac4282341","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-09-12 13:56:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/12_09_2022_13_48_33\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a234bdae54f6c82f5c2249c4a4aa9756288d632","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-12 14:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/857044/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/df53526/ : SUCCESS in 1h 30m 55s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/857044/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/762f820/ : SUCCESS in 1h 45m 14s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/44/857044/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/32545f1/ : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"66e916796ea0e447e1ade193d26a656fc901b4fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-12 16:53:08.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_Unity_FC_os-brick/879ab6c/DellEMC_Unity_FC_os-brick/239 : SUCCESS in 1h 22m 00s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_Unity_iSCSI_os-brick/00c90c6/DellEMC_Unity_iSCSI_os-brick/253 : SUCCESS in 1h 57m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b6db7fa79fe3f4e6f14bb377f54b8d24e016effb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-12 17:30:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_PowerStore_FC-ob/e3b01dd/DellEMC_PowerStore_FC-ob/191 : SUCCESS in 1h 24m 46s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_PowerStore_iSCSI-ob/bff92a3/DellEMC_PowerStore_iSCSI-ob/112 : SUCCESS in 1h 59m 26s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_PowerStore_NFS-ob/f807722/DellEMC_PowerStore_NFS-ob/87 : SUCCESS in 1h 37m 23s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/44/857044/1/check/DellEMC_PowerStore_NVMeTCP-ob/77d7e72/DellEMC_PowerStore_NVMeTCP-ob/67 : FAILURE in 1h 47m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6aead5a75295edd92519ae874f5c9fa404af035d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-12 19:38:41.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/597","accounts_in_message":[],"_revision_number":1},{"id":"01fd8d18cba0f380bbf23b40dfa4459a1c16dca8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-14 15:13:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5d4b6dc434193eea6b3c34e475e102339da1791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-11-14 13:01:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4cd8904954c2cca78caa6b6ca40545ff52cc59d3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-14 13:03:27.000000000","message":"Patch Set 2:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/816/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"4a22c16c0ccb8fb77e72d3c6bb1a5f7b1f7c8739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 14:23:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/091f4180d3434dfc9a9175bdb6c957a5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f30abacfdd141dda3e48128d5b5c071 : SUCCESS in 1h 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a73077e53445208368ec2a05e7ba2d : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06b24d013f13432bbf5b05a1019d3741 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df013e30e3e246fc9e7146f40756c361 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7896ec06af4150955539e96eeef0e1 : SUCCESS in 6m 34s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/015dfc8a292d44e7b94505311af175f2 : SUCCESS in 3m 27s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/c5ae5d6b62614e3191283813ce7c8ac8 : SUCCESS in 1h 08m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2671c4d2b43249999c5930ede39c45f1 : SUCCESS in 1h 16m 53s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/082b3810d2f94fdabfe7578c2ad898b7 : SUCCESS in 1h 20m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f1fb8a798a0a465cb9d0a31ca04b21df : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d9774429d0cb3d533c7b75e3093bd7b01402f89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-14 14:31:32.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_11_2022_14_23_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"b15d640eb3320bed0acc15aee5cb5afc6b8e10f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-14 14:53:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/857044 : FAILURE in 1h 50m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c5a0af467d8bf3a15e6bdfd3ee18c18525a9e8dd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-14 15:04:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/857044/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/31cd380/ : SUCCESS in 1h 32m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/857044/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/caf6478/ : SUCCESS in 1h 32m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/44/857044/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c2286a0/ : FAILURE in 2h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"34be1cb728cd3bf4e52328096fc6cca0ff30c83a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-14 16:29:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/44/857044/2/check/DellEMC_Unity_FC_os-brick/bff42b5/DellEMC_Unity_FC_os-brick/249 : SUCCESS in 1h 03m 49s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/44/857044/2/check/DellEMC_Unity_iSCSI_os-brick/7a3f59d/DellEMC_Unity_iSCSI_os-brick/263 : FAILURE in 2h 05m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d35f92336d1907cb621e9a9d9122448961b9cb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2023-02-05 22:33:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"632b601f073be3adf868bab9799919f6cab49d92","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-05 22:37:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/29da018003b447f6af6c222d62d6bd00 : POST_FAILURE in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"4698f574ade7da1223e59fd7b4a73bcdd81186c4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-05 22:58:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/857044 : FAILURE in 23m 42s","accounts_in_message":[],"_revision_number":3},{"id":"942c73ba26bdb2dc9d009520100f7f18d989a5e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-06 00:15:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/44/857044/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/921e9ce/ : SUCCESS in 1h 39m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/44/857044/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e23631b/ : SUCCESS in 1h 32m 53s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/44/857044/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/dfe781a/ : SUCCESS in 1h 30m 29s","accounts_in_message":[],"_revision_number":3},{"id":"10a20cfa5dceffb16ea011c20d022508204315dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 00:27:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebab0aa036a2465dabaaaa179ccdcd4f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05d46f2783b441968e78f17f79adc794 : SUCCESS in 1h 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55dc46c884f0451c91beeef042ca5ecc : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8022e2e4824e4e4d97d17f93f7581327 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2e64129363046fdafd723c1b01319df : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c81345ffc149daae80d5e21fd89ffe : SUCCESS in 5m 21s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/14df1ce36f7843448c32b90a57b3cc10 : SUCCESS in 3m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/f7013819bdc741abb2165373d52dc7cf : SUCCESS in 1h 19m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d505ff4496334921a0cd383b0bc65a59 : SUCCESS in 1h 21m 44s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c3e9732fdea34392a14a1c0bfd50bc09 : SUCCESS in 1h 47m 02s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a457138ecb4d467887b1e4f66376cc07 : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7371e2981206b18c98dd44690c4c92830d441d3c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 01:07:04.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_Unity_FC_os-brick/a664683/DellEMC_Unity_FC_os-brick/259 : SUCCESS in 1h 43m 28s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_Unity_iSCSI_os-brick/907b0b2/DellEMC_Unity_iSCSI_os-brick/273 : SUCCESS in 2h 33m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3adcf1b1c669406c3c93fa5c95e957892ecc8974","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-06 01:32:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-fcp/e611ebd/ : SUCCESS in 59m 00s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/3e167f6/ : SUCCESS in 44m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-nfs/24de042/ : SUCCESS in 42m 54s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4a490a8/ : FAILURE in 43m 34s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"525f7da86ffff3d144d819694e8a794734c72591","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-06 01:40:21.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e7e4ac30e21c4397920ec51d4f351211 : POST_FAILURE in 1h 39m 41s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/af81c5b571bc46cb82dad17e39cff51f : POST_FAILURE in 2h 11m 23s","accounts_in_message":[],"_revision_number":3},{"id":"28b1c120f5069557c6afdcc800cedc6078c50d5c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-06 02:01:15.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_FC-ob/4515de2/DellEMC_PowerStore_FC-ob/203 : SUCCESS in 1h 25m 35s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_iSCSI-ob/f3130f5/DellEMC_PowerStore_iSCSI-ob/124 : SUCCESS in 1h 32m 58s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NFS-ob/17616e5/DellEMC_PowerStore_NFS-ob/96 : SUCCESS in 1h 21m 49s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NVMeTCP-ob/14a0f2b/DellEMC_PowerStore_NVMeTCP-ob/76 : SUCCESS in 1h 30m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"df14818ade84fc52c208fa0f5a4a6cf121ebca57","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-06 05:05:04.000000000","message":"Patch Set 3:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/44/857044/3/fc : SUCCESS in 2h 31m 18s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/44/857044/3/iscsi : SUCCESS in 3h 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"855d00957b8b322164894777684e12f5a4f3ee7c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-07 07:15:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6906d0ef9353711fc17295ec14cca16bde0f229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 07:15:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5712a2b711d3a87fed03b5a79df22c15999a13a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 09:22:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d854981ef4ba4fb0a5d106488dc9e65e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1867d7c0056e4c80972878c23687f2ff : FAILURE in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd4c825508a49878c68fbc829f9f0f8 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/936fef77c9f44c04b6d9d11a726f9896 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bc66a08c6aa453cb0d7caaf157815dc : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e45cd3e822447e89d60f82c5aaa4d29 : SUCCESS in 6m 11s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e97bac1823be4bb4bc53784cca8d11b1 : SUCCESS in 1h 55m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d90d52439c5aefd8cf5f911ffcb41fcfa82e10df","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2023-02-07 19:57:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"abdab21bb77a08640f9f3fd2a3b8421dcf1d5a29","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-07 20:03:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c9fc7ccc9e084569be9335549913167b : POST_FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"03c01f863bffe2535ab0caec682c8e80bec3d565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 22:14:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdb66b5dc54435093acb2803a7d3438\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cd1ef1d00d84ded908916adc38915e9 : SUCCESS in 1h 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af19e11ff1e64e5faa239d86babd99fe : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064fa35745aa4609becade9f6ffc3b6c : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0acac875a9204c02bc5716689019583f : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465fa06b503b459cbff9e93c8334e551 : SUCCESS in 5m 44s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0a9f08bf1ddf48f1ab727b474742a839 : SUCCESS in 8m 14s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a4cafbe71f1a4594934bd527bd268bf1 : FAILURE in 1h 31m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fec255bd210447bca7ced3bf9c18c984 : SUCCESS in 2h 14m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c7b31ce6624f4903955f87020f529c39 : SUCCESS in 1h 50m 22s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/3399050e8380418e846b0616ed3c5dff : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7391cca0bb98db26be0cde521b1260a471599707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 22:15:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa20d985691c68cc305ea2cb96e475efb091446d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 01:22:21.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6dd3d4349cb42b4a274af98fdd761fb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddf1746336974938a3575968845e2108 : TIMED_OUT in 2h 04m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dad6543a6fc469294e3295faa86d562 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b8741dcfe8b4a80bcc9496fd6e1bc8a : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e648ac2d013a405b9f0baa82a2d23ebe : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4f571b534c4b898755d58824c15fc1 : SUCCESS in 6m 11s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f7f70990edde4f1ba27f43e2a82e668d : SUCCESS in 2h 50m 06s","accounts_in_message":[],"_revision_number":3},{"id":"45756c15cc98662df24ae4f0210239bf991c9431","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-08 02:02:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_FC-ob/62de4be/DellEMC_PowerStore_FC-ob/207 : SUCCESS in 1h 32m 07s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_iSCSI-ob/62bbb95/DellEMC_PowerStore_iSCSI-ob/128 : FAILURE in 1h 28m 22s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NFS-ob/d7ea93b/DellEMC_PowerStore_NFS-ob/100 : SUCCESS in 1h 14m 14s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NVMeTCP-ob/39d0a76/DellEMC_PowerStore_NVMeTCP-ob/80 : SUCCESS in 1h 25m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"ef11f29f536f686284ebf21f46fa579f915169a4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-08 03:24:32.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/aa7c14a94d96485984fb0935ffc20a0d : FAILURE in 1h 50m 43s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"79735f9a28657ee69d67a218ec3edd49810c5f7a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-08 07:51:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ebae61c45e5ace6f6c7600ecc584e90d1f8efd3","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-08 07:57:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/4a6d6d2100e446d590f4953b387f5c19 : POST_FAILURE in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"753356af26cddf46a807e5ee340425c313feb926","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-08 09:53:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-fcp/483b7e5/ : SUCCESS in 1h 53m 15s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/0b8a25d/ : SUCCESS in 55m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-nfs/6d16231/ : SUCCESS in 55m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/857044/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d292f79/ : FAILURE in 55m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c5cc8ace2253436dfe27d3bc3b52ac0a9069a861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 09:57:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24fc5b12b1524e76a887fc657601512b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2ed7006bc74476ba8bb189ea5cbd09a : TIMED_OUT in 2h 04m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b2f29df60e4f2480235e4ee30f4cb3 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d9d7b3bd9394c3e8bf038e84f50f905 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26c700198f8943a0810a458e6104ea13 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561c72802f5a4201ab3a01eb05b9b069 : SUCCESS in 5m 26s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b0107012d86646a88b30b17f8569cfd5 : SUCCESS in 5m 58s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/50bba5630c0849d79dd95882b8616845 : SUCCESS in 1h 08m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e07925825d26430e9e96c9100fbe5100 : SUCCESS in 1h 45m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cf27b1df60694f8bab06d41e28a0d3e6 : SUCCESS in 1h 37m 15s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/7b1b83ad2d90484dab7ccc0d79b52d08 : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f12073f59e1a474557348bf940a492315609eb00","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 15:34:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_Unity_FC_os-brick/4ed0e22/DellEMC_Unity_FC_os-brick/264 : SUCCESS in 1h 16m 41s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_Unity_iSCSI_os-brick/bc7a426/DellEMC_Unity_iSCSI_os-brick/278 : ABORTED in 3h 00m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"7283582a7c1fb5c2178ed852889e03591dbcf4eb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-02-08 16:09:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97cdbdc79271ca8e4982e6531c92349ddf2dcdd6","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-08 16:23:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c8e28d14849741fe99608fce09fed8d9 : POST_FAILURE in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7972645f320504b9680c5ebc16636265224f1be8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 18:12:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4194b48b4a49968042511bae0a63e5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/749cb158f8cd463585c6efacc1901e67 : SUCCESS in 1h 36m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4cc0abc3804cdb8eb0efe8c97f89c5 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782cea8aae6e4eeeb3d83a092af726ac : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4c0989487794fe88c7bcbbcd62eb22c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d20c97b1076420e98f6e06db496db2c : SUCCESS in 7m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b59957d7ef0f4d639ee822617446d791 : SUCCESS in 4m 34s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2be91e1c5a1942dabdc50e989bcba87c : SUCCESS in 1h 44m 52s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2ec05f83f03d485d9a354cea9a4387dc : SUCCESS in 1h 37m 31s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/fcca8d8520bd40b286ccf46f1896e0cc : SUCCESS in 1h 55m 19s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a991c6b942724ae8ba77b081441d8106 : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12d8d91c6c914d2e5fbcbb78bd63210807f45dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 18:13:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a943d96dac4215bdf1833cd559ec7a3eeec0bdb3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-08 19:54:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_FC-ob/05cdc73/DellEMC_PowerStore_FC-ob/208 : FAILURE in 1h 35m 31s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_iSCSI-ob/a9404ac/DellEMC_PowerStore_iSCSI-ob/129 : FAILURE in 1h 27m 26s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NFS-ob/8624a0c/DellEMC_PowerStore_NFS-ob/101 : SUCCESS in 1h 14m 56s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/44/857044/3/check/DellEMC_PowerStore_NVMeTCP-ob/8df38ab/DellEMC_PowerStore_NVMeTCP-ob/81 : SUCCESS in 1h 27m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"7b6dc779587a68bd64b049268cc6764a5fbfc8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 20:03:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23a2d9430de429fa31aa17a23eecb4c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a06b3f99f944e1aab8a5ec24e9f499b : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc421192255c4d3fb3b992710abaf000 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fee993746b164e1b8d437c4ac27f5504 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38c6424f74a743e78fe04e3c239526e8 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feef06fcb21c4e569d73e910050d6233 : SUCCESS in 5m 43s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6c1dc9be28ef4c8cb85714977744f196 : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a335c384ae62d7eaeabdb16339e359c2130df386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 20:03:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b1648c4a361b3b7428df75c86a9264cef668812c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 20:04:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415359c30f0841c381f3247d3534e40c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf1f638af5b4cee898a6a4856e4a9f9 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3},{"id":"d72842a76e53a92967de96a6d1382cb95f3fc76d","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-08 21:05:04.000000000","message":"Patch Set 3:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/44/857044/3/abktql/fc : SUCCESS in 2h 01m 11s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/44/857044/3/abktql/iscsi : SUCCESS in 2h 12m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f8412d8d704c16cc532e3413b5c44cc5df1d216f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-09 03:53:40.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/a4fc920bf37f40deadec25a2991790d6 : FAILURE in 1h 32m 57s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d8fe939d7fb547e29466c705be7975d6 : SUCCESS in 3h 16m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7b1426e31ecd4a5e177036248df1cf8733e1e14","revisions":{"850e67f50a8cfe794f5463a8efe17b022596c14f":{"kind":"REWORK","_number":1,"created":"2022-09-12 11:57:12.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/857044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/44/857044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/44/857044/1"}}},"commit":{"parents":[{"commit":"4a12f1b4dad4bbdd30389e8c0e5731766aeeca29","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4a12f1b4dad4bbdd30389e8c0e5731766aeeca29"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-12 11:54:06.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-12 11:55:08.000000000","tz":540},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027called_once_with\u0027 with \u0027assert_called_once_with\u0027.\n\nChange-Id: I18b55da0d1f142818e7ea14f6eebcc0f0f0cd23f\nCloses-Bug: 1989280\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/850e67f50a8cfe794f5463a8efe17b022596c14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/850e67f50a8cfe794f5463a8efe17b022596c14f"}]},"branch":"refs/heads/master"},"dbec249645c4963d4080dd1944745efa7a7ce602":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-14 13:01:56.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/857044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/44/857044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/44/857044/2"}}},"commit":{"parents":[{"commit":"0639af4fc8d7f3abdbfe6326b778b219652af864","subject":"Update metadata in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0639af4fc8d7f3abdbfe6326b778b219652af864"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-12 11:54:06.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-11-14 13:01:56.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027called_once_with\u0027 with \u0027assert_called_once_with\u0027.\n\nChange-Id: I18b55da0d1f142818e7ea14f6eebcc0f0f0cd23f\nCloses-Bug: 1989280\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/dbec249645c4963d4080dd1944745efa7a7ce602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/dbec249645c4963d4080dd1944745efa7a7ce602"}]},"branch":"refs/heads/master","description":"Rebase"},"e7b1426e31ecd4a5e177036248df1cf8733e1e14":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-05 22:33:35.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/44/857044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/44/857044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/44/857044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/44/857044/3"}}},"commit":{"parents":[{"commit":"655fcc41b33d3f6afc8f85005868d0111077bdb5","subject":"Merge \"Handle FileNotFoundError on get_system_uuid()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/655fcc41b33d3f6afc8f85005868d0111077bdb5"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-09-12 11:54:06.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2023-02-05 22:33:35.000000000","tz":0},"subject":"Fix wrong assertion methods","message":"Fix wrong assertion methods\n\nReplace \u0027called_once_with\u0027 with \u0027assert_called_once_with\u0027.\n\nChange-Id: I18b55da0d1f142818e7ea14f6eebcc0f0f0cd23f\nCloses-Bug: 1989280\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e7b1426e31ecd4a5e177036248df1cf8733e1e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e7b1426e31ecd4a5e177036248df1cf8733e1e14"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
