)]}'
{"id":"openstack%2Fos-brick~862183","triplet_id":"openstack%2Fos-brick~master~Ib1e203c8fb4a94671513dee17edfe4c5ce1fee34","project":"openstack/os-brick","branch":"master","topic":"migrate-to-jammy","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-16 16:11:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib1e203c8fb4a94671513dee17edfe4c5ce1fee34","subject":"WIP: Test Ubuntu 22.04 (Jammy) migration","status":"ABANDONED","created":"2022-10-20 18:32:58.000000000","updated":"2023-02-16 16:11:26.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ae5e629620214365910694016444a119f9b4ef8","_number":862183,"virtual_id_number":862183,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-16 14:43:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 19:00:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 19:26:24.000000000","updated_by":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"reviewer":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"state":"CC"},{"updated":"2022-10-20 19:33:50.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-20 19:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 20:26:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-20 20:45:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-21 06:57:53.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-11-16 12:43:22.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2022-11-16 14:52:28.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"b16e6c35710ac7103af6324a46c2f142d3ffc5e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-20 18:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"015242e586680946420f6120b3c6d546653f7b7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-20 19:00:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/862183 : FAILURE in 21m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e5a9d9d27c42fb7e891fa1093062e6493d2c5633","tag":"autogenerated:cloudbase-ci","author":{"_account_id":27710,"name":"Cloudbase OSBrick Windows CI","email":"windows_osbrick_ci@cloudbasesolutions.com","username":"windows-osbrick-ci"},"date":"2022-10-20 19:26:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase OSBrick Windows CI, add a review comment with run-Cloudbase OSBrick Windows CI\n\n- os-brick-iscsi http://cloudbase-ci.com/os-brick/862183/1/iscsi/ : SUCCESS in 51m 06s\n- os-brick-smb http://cloudbase-ci.com/os-brick/862183/1/smb/ : SUCCESS in 50m 31s\n- os-brick-ceph http://cloudbase-ci.com/os-brick/862183/1/ceph/ : FAILURE in 53m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6ffa3151c8cf9fedbe8c19c5d059ff202afe88f6","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-20 19:33:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-os-brick https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/cd97dcddc505416594f6439996b3e016 : FAILURE in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d5882d22ad00f81a707226d8a4bec28afaf54be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 19:54:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c04e3298180c42de9b539b7ba616a2ff\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cef96e9d4844e89a0db51e9ed038810 : SUCCESS in 1h 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b87f687ff9408293a096d928f5dc5d : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105ed4cf343240ee9661f8e62bee5c28 : FAILURE in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac362fc89a45471da92fdc85ba5be274 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e9f1ca4de347daa95a4bb8cecbac18 : SUCCESS in 5m 03s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7d1f94b91ff647e7a07dc3836886f7b6 : SUCCESS in 6m 51s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/eeaaa471d1b24eec847b42a3c060b8b5 : POST_FAILURE in 7m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e524b2c09d5042a9b828adfcae8930aa : SUCCESS in 1h 11m 43s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/9aa8fa63bfc949ada86725b7ac3f9fdb : SUCCESS in 1h 07m 49s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/087040f7fba4416ea034be311bcf9878 : SUCCESS in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23a287b76a3e9c38f683639848def9c1c0b6a0a4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-20 20:26:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_FC_os-brick/9da419c/DellEMC_Unity_FC_os-brick/246 : SUCCESS in 1h 10m 11s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_iSCSI_os-brick/f547dc0/DellEMC_Unity_iSCSI_os-brick/260 : SUCCESS in 1h 53m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"aaf43f2a96c563712ff679c91cb3d28ec93d006a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-20 20:45:55.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5a9c50218610405ab7c7704866f18a3f : POST_FAILURE in 2h 09m 56s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c9cb442581854be1b517ece8134dd7ec : POST_FAILURE in 1h 07m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6e2b37319ecaeb64e34555e71498e6a92d4331b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-21 06:57:53.000000000","message":"Patch Set 1:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/768/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"d4c88557805b563f1b9f43ee81e7c1a5fcdf6047","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-21 07:35:47.000000000","message":"Patch Set 1:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/769/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"48ad693c49dab5840d29b4b338b104e95d585756","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-16 12:43:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c8a49a054ed349b56f71547fba2d8f077397c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 14:43:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a4a1e62f0c464d8299c13597f8645e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bf72a2bc2984f43a37cba5e77731567 : SUCCESS in 1h 52m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2e3d5533bb407c9b56412c294a0d61 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7635158c98794aca98019c88b6096edf : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c3cd02cec314c57bd09286384d7f21b : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bfb3b6cf10a4de4ae9c3b6b56c368ce : SUCCESS in 7m 12s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/382186673359479a96b99dff1841b5e0 : SUCCESS in 7m 27s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/134980de25af4710a27fabc653a0b0f8 : FAILURE in 29m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/47de4d4187c849df90350eba815b6e34 : SUCCESS in 1h 39m 30s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/792414b99b374c6bb39eed8efa063763 : SUCCESS in 1h 35m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4fd51451a4584d3a815d782f3fc701bd : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"042ddff74352c0c89bc9574529200a4d475078b3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-16 14:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_11_2022_14_43_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"d219cbf7ffe477b6fa6ece404b7fa957cc882ca9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-16 17:25:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_FC_os-brick/337ce2b/DellEMC_Unity_FC_os-brick/252 : SUCCESS in 1h 06m 52s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_iSCSI_os-brick/9cfae31/DellEMC_Unity_iSCSI_os-brick/266 : SUCCESS in 2h 06m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd427fa33fe4cc67130a19e0dc7fde75b5897a55","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-17 21:18:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b0d61c5a1fbbdf55093bc94e6a7176a6440f887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 23:32:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a230995ded448569ebeb6f1f4c17c74\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77a7ccaa369845d997d52ba2c11c04e6 : SUCCESS in 1h 44m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e2315a010e4465986ba52d26a12a60 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9694a641f2a74c71a0286d4dc6ef64c1 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baff4cb99afb4744893ecccf73f58404 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681e8a088f3e472a84d0a74f9579401f : SUCCESS in 5m 58s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d5edeb6dacb340d7bfeac7f2e244a42d : SUCCESS in 3m 49s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/804ce5a10e534b6a9e7d56050eb252de : FAILURE in 34m 32s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/671f9cb48fab4302ad47e7af41802216 : SUCCESS in 2h 06m 27s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/58654e715e1747959dc36508f6e4d4f9 : SUCCESS in 1h 41m 21s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b42808015bfc49e480f1573a3523cf21 : SUCCESS in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85c8c466fd9d8d05905e1a64fcfaa0724535c930","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-18 04:02:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_FC_os-brick/62d3f87/DellEMC_Unity_FC_os-brick/253 : SUCCESS in 58m 57s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/83/862183/1/check/DellEMC_Unity_iSCSI_os-brick/1f0711f/DellEMC_Unity_iSCSI_os-brick/267 : SUCCESS in 2h 16m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2ae5e629620214365910694016444a119f9b4ef8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-16 16:11:26.000000000","message":"Abandoned\n\nI guess this isn\u0027t needed any more?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72e8d50e9b547a1a8df770fb1defc7cca860b7ac","revisions":{"72e8d50e9b547a1a8df770fb1defc7cca860b7ac":{"kind":"REWORK","_number":1,"created":"2022-10-20 18:32:58.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/83/862183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/83/862183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/862183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/862183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/83/862183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/83/862183/1"}}},"commit":{"parents":[{"commit":"4a12f1b4dad4bbdd30389e8c0e5731766aeeca29","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4a12f1b4dad4bbdd30389e8c0e5731766aeeca29"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-10-20 18:32:23.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-10-20 18:32:23.000000000","tz":-240},"subject":"WIP: Test Ubuntu 22.04 (Jammy) migration","message":"WIP: Test Ubuntu 22.04 (Jammy) migration\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/861110\nChange-Id: Ib1e203c8fb4a94671513dee17edfe4c5ce1fee34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/72e8d50e9b547a1a8df770fb1defc7cca860b7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/72e8d50e9b547a1a8df770fb1defc7cca860b7ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
