)]}'
{"id":"openstack%2Fos-brick~868447","triplet_id":"openstack%2Fos-brick~stable%2Fzed~Ida6359ddebc59091f2f2dfc951950538793d0109","project":"openstack/os-brick","branch":"stable/zed","topic":"tox-4-postponed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-04-20 06:18:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida6359ddebc59091f2f2dfc951950538793d0109","subject":"[stable-only] Pin tox \u003c4","status":"MERGED","created":"2022-12-22 23:11:54.000000000","updated":"2023-04-20 06:19:55.000000000","submitted":"2023-04-20 06:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"868447-tox-4-postponed","meta_rev_id":"e744095cbe83f173f294e09a79e9438d3253df85","_number":868447,"virtual_id_number":868447,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 06:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-02-18 15:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 04:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 04:51:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-23 00:48:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-23 01:13:26.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-23 01:47:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-12-23 03:50:48.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-01-04 11:10:29.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-02-18 15:25:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4efda3ec067e32a1cc0b7e30d24b4e0dbeda288d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-22 23:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b44a387f5f1c49084044d5ad7b121de3b8aa04b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 00:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50927f81db034170ae9611bc7764a657\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c91d4d313b7849ffa38e574fba0be3f5 : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75a04ce17618468ab03427d595673887 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0b0f4b2ec274f4f8c5133a3968a74d6 : SUCCESS in 1h 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd9126ea5504810af503a21add5ddfa : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5fe21784f5947078fa9e173215ccc70 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7b706d04f624b2db25211812d4dee20 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5842ab725fd64c3aa7cbe2e49372875f : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef95c3b7f39410192697c8060562980 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a8ea41925d741a3afff628f239d5fe3 : SUCCESS in 3m 42s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/852fb5f787ca4e12af333dc5fd34bb54 : SUCCESS in 3m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/46457329ab18441782b60858321ce57b : SUCCESS in 57m 26s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e19af3102d1f4613b6a93a446ed1f3b9 : SUCCESS in 1h 17m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f29e4192762544acb59770898d9e66ae : SUCCESS in 1h 16m 47s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c717865b4ffe4cabaffe95ac9595af1b : SUCCESS in 1h 31m 51s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5dc2380cc8cc4dd4a4343580137f6b47 : SUCCESS in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4439b233929a5f2c70a70f900f79a64a2ac64160","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-23 01:13:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/868447 : FAILURE in 1h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"32b465990a50ba58e64fc5a4212dc4066f9104cc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-23 01:47:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/63d7cd4/ : SUCCESS in 1h 38m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fda15e0/ : SUCCESS in 1h 48m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/6a50c03/ : SUCCESS in 1h 46m 32s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/5c02afa/ : SUCCESS in 1h 47m 04s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/355881a/ : FAILURE in 1h 18m 46s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/47/868447/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/4272066/ : FAILURE in 1h 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9995891a41a425a8f34b27d650ef222f8d5a9b3e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-23 03:50:48.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/908","accounts_in_message":[],"_revision_number":1},{"id":"93b907cac6cb5e8e17f1af9e2a24c32ca2950e11","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-04 11:10:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ab143af39248dd2f65ee19faf21797b5049154d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-04 23:29:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"914d4015236552caaa202c3a22c4b423e1f85f4d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-05 14:46:57.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0e158f1314708c2fb136750f3c1a306e8c97048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-05 16:36:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"958f04fb98ce66a6cb146817e3df039072c64105","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-05 16:44:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56d4a9724d99b2a5d36e14874d6bbc219e0d369e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-05 17:36:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/868447 : FAILURE in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"989bd2f4c8532969a18ec3ed8b101edf3b35117c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 18:12:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/602f4ed1c04e4f3fba428ffe372a5d6d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0e9cfc79d464f949fdba57c363bbd87 : SUCCESS in 2m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75f5aa415baf4dbb9f6c392acf15d364 : SUCCESS in 2m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6220b1eda8f42f4a17f23a023b8c910 : SUCCESS in 1h 31m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc3f53977da4e5e8b657304c468ba33 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26f3716e8f9d4130931b30e387942546 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ee6ca667cf345c9aaa22b63b1f2ca61 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c36dcbdc91a44ef90e1e22e0e8d21f7 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e364d76afe07455b997884f84402a8ed : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bb7e496ea0c494ab875f5ba65e4cc27 : SUCCESS in 5m 00s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c5f13dcf5dbb4b1aa52e974668e67c1a : SUCCESS in 3m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/ce1eb719005c4397b5a46182d3255c25 : SUCCESS in 1h 11m 18s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a421b3415cc847c8a9397ef353d92c4c : SUCCESS in 1h 05m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/145caffea1d64294ab296668237d2e26 : SUCCESS in 1h 33m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c7ffa87663bf4fad89b48ecb86eb2e5d : SUCCESS in 1h 25m 41s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5e52658609744eeba0bca7a2466b30bf : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56465edef9c74a7523fa4c8c11aae19fef95c71c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-05 21:03:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/868447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a6fc1ab/ : SUCCESS in 1h 36m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/868447/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/30a5c29/ : FAILURE in 1h 58m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/868447/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/559b974/ : FAILURE in 2h 18m 24s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/47/868447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/d32ce9f/ : TIMED_OUT in 3h 06m 20s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a370eeb352aeb0eacba5e3a1985fb82c2794e3a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-02-18 15:25:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e288d6a6c2a1f211ecc4b8f53ddd46e9fa3a6052","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-04-20 04:51:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1c19baf8f55a059920cb1b33b914a50abedd06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 04:52:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ff71d53ca64f254da686a69fa7d443807c9b045","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-04-20 04:52:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bc182beafdcdb1255ae813b619a968745d33f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:18:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4202592815ef4a619f0ef1ad82fd6fbe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e671e31d896846458f280f1311cb8e61 : SUCCESS in 1m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ebd972b536c042ea9013aa0b1092e08f : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f2f0afb1bbc45faac44e2bcf416b891 : SUCCESS in 1h 01m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4d9881e0254130ae64b279136801fb : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a90e29f461e94de98231ba55ac69713b : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2467b1c50042467c96d6309f73a8c96a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f153a643a2640b586ec0bc1b91fb581 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98cd48c5923e4ba291fa252ee649e889 : SUCCESS in 4m 31s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/503dbbd63a3447118148d039c0d2535c : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":2},{"id":"78b07aa9f8966d062c2d8bea213c2cbcf19dacb3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e744095cbe83f173f294e09a79e9438d3253df85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/846df4171b1446fc9c0b7f8c37708332\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a934eb3e79a4495b82b5def1faddbdf5 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ec54c1bce48476ba881cb0a35788012 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7dfaf4243ae4a75971462ad35765acfedca6a89","revisions":{"420f5e1fd0db387385fb6c5f013c597c914bf51d":{"kind":"REWORK","_number":1,"created":"2022-12-22 23:11:54.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/47/868447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/47/868447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/47/868447/1"}}},"commit":{"parents":[{"commit":"ef4982cf64791f40973121bfde3ba5d25be8d169","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ef4982cf64791f40973121bfde3ba5d25be8d169"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-22 23:11:07.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-22 23:11:07.000000000","tz":-300},"subject":"[stable-only] Pin tox \u003c4","message":"[stable-only] Pin tox \u003c4\n\nContinue to use tox 3 until we decide how to do the transition\nto tox 4 in the stable branches.\n\nChange-Id: Ida6359ddebc59091f2f2dfc951950538793d0109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/420f5e1fd0db387385fb6c5f013c597c914bf51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/420f5e1fd0db387385fb6c5f013c597c914bf51d"}]},"branch":"refs/heads/stable/zed"},"a7dfaf4243ae4a75971462ad35765acfedca6a89":{"kind":"REWORK","_number":2,"created":"2023-01-05 16:36:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/47/868447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/47/868447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/47/868447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/47/868447/2"}}},"commit":{"parents":[{"commit":"ef4982cf64791f40973121bfde3ba5d25be8d169","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ef4982cf64791f40973121bfde3ba5d25be8d169"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-12-22 23:11:07.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-05 16:29:51.000000000","tz":-300},"subject":"[stable-only] Pin tox \u003c4","message":"[stable-only] Pin tox \u003c4\n\nContinue to use tox 3 in the stable branches.\n\nTwo changes:\n- .zuul.yaml: set the ensure_tox_version to use \u003c4, which will\n  tell zuul to install tox\u003c4 if it\u0027s not present\n- tox.ini: set requires\u003dtox\u003c4 so that if tox has been installed\n  already, our tox-based jobs will use tox 3 to run the tests\n\nChange-Id: Ida6359ddebc59091f2f2dfc951950538793d0109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7dfaf4243ae4a75971462ad35765acfedca6a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a7dfaf4243ae4a75971462ad35765acfedca6a89"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
