)]}'
{"id":"openstack%2Fos-brick~876282","triplet_id":"openstack%2Fos-brick~master~Ie3eaf0d16e27e35ec30635c82b38c634db9c4541","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-03-09 23:07:14.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-03-09 16:40:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ie3eaf0d16e27e35ec30635c82b38c634db9c4541","subject":"Add Python 3.10 to setup.cfg metadata","status":"MERGED","created":"2023-03-09 16:22:25.000000000","updated":"2023-03-10 22:25:25.000000000","submitted":"2023-03-09 23:07:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876282","meta_rev_id":"9ffc0298ce23f9a6bd31b05a7df5c617db8699de","_number":876282,"virtual_id_number":876282,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 23:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-03-09 16:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-09 16:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-09 16:40:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 16:36:13.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-03-09 16:39:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-03-09 16:39:36.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-09 16:47:26.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-09 18:25:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 22:23:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-10 00:10:27.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-10 01:21:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-10 01:46:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-10 05:17:03.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-10 05:33:01.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-10 22:25:25.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"}],"messages":[{"id":"a086accc0b9e788592360d9ebd07a751bebbe6ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-03-09 16:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6df806705ae3fc8f63016423e8dddfc545218fb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-09 16:36:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d2cbafd1f268b718d2e33d713e6f97b525cca50","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-09 16:39:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3eefadb255ce36e3d5944b5831c4bdaf3453932d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-09 16:39:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3a3c4e7b80c69abef08500f9a29955e64468141f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-09 16:40:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4701e112be94c57a9cca9144cfe2ec56d56030d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 16:47:26.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/56/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"e2d2e0ffac3643aea2c12d52a45241f13e0dad45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:25:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb45ceb98b7b468c806a4bd72f7116cd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7cbab6f299d74db3994d84b2c7fee04d : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18a4a344f66340e18153837c60963689 : FAILURE in 1h 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a6b100d83c401284a6328659166934 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/317c919658d34be283b7d1c05947bdc1 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb63f7e47df4446f9bf317216b59d802 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd8f13399c4406cb1cc7717efe0079d : SUCCESS in 6m 27s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c6a754957f0240139941213d55d5247e : SUCCESS in 4m 58s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/8daa8ee4f3754681b5b0b31ef2d2891b : SUCCESS in 1h 12m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b6b4d0941b6542cd8844709cf1b2e1e1 : SUCCESS in 2h 00m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/61f776f9ea864ca298d2427c3ca1c9ea : SUCCESS in 1h 11m 05s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4cbe8613f0eb448cabbaa85e1341663f : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2db643ccb146392a063ec0f9db93fb27e0ba7fd0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-03-09 19:34:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86cc5992bfc56b10347f70b2ebfeb41de08150f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 19:34:04.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/56","accounts_in_message":[],"_revision_number":1},{"id":"136ce2de67de2b4af3b413cd52861907812672a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 19:34:07.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/62/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"08aa447a8ce07eefb1f7c1a506848557ac8fd973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:25:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b93ae306bfd45bd8481e0d0c581c07d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c885507a8c3d4cd6816e38de562f99b1 : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0e8bbef17a842a09a52ca4054ae9c32 : SUCCESS in 1h 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a36d52c1dd940a79fee1e2cecc2db2c : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6195824832e4002967aea468b0f4f89 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/432b1ef4f95c43b8a6ba47511545f755 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cafed57a314827a7163c8b741f83ca : SUCCESS in 5m 51s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/285fc7315b8b44faa357020467fe9fe5 : SUCCESS in 4m 21s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e9808c76eec74d94acd6a45c6bf3b283 : SUCCESS in 1h 15m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/273c43920a7b428eb74b8ba1f0110d85 : SUCCESS in 1h 41m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/77efe320cf4d4fec8165ee6fbdf9e42b : SUCCESS in 1h 25m 09s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c44e755708cc40ccbfe40bac3dc4a88d : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"998aff1365f5c16d05c46c53b8c4c82d6dbd280b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78d43489d875f2bdd49f8043ebbb11bcc451117d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 21:33:38.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/62","accounts_in_message":[],"_revision_number":1},{"id":"53d5ff0987b99bc141eefda9fb2b25e7047d8826","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-09 22:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/876282/1/upstream-check/cinder-tempest-plugin-ontap-fcp/d5b0f1c/ : SUCCESS in 57m 18s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/876282/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/207e4b9/ : SUCCESS in 44m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/876282/1/upstream-check/cinder-tempest-plugin-ontap-nfs/1017ac9/ : SUCCESS in 43m 35s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/82/876282/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/25de17e/ : FAILURE in 42m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b9fea6b726815a0d2715fceece7c44430165b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 23:07:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ff4e8b3d884c3388e822844ac11f91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3f7222cb5a3447c092201592c001a04c : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82fb4a31e5cf45bdb666876947790edb : SUCCESS in 1h 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6dbbd1b8264c73941fa6f075b55eb0 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9494618f0d474deb8753f2aac3964da8 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd7905ff7bf5400885f37c880385cb2a : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce25f47c34de4492a4f169427c34626e : SUCCESS in 5m 11s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/055dff2eaf95418ea66cb4987a74a180 : SUCCESS in 1h 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5c92ef69456720c26b8d3a6a71b33a231afbf765","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 23:07:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e774da39b68a0f6a04adb9e5111fb0f32579d920","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 23:07:22.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/65/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"183d9a6c4b4bfad93543fb32d709c27743831941","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 23:07:23.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/65 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"46e28c4c261be3fc34ef6eb1e80c05684584912a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 23:08:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59f3ebdc6ed04f558d9d3619882c1e72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4243cf647e747aca1f1d991485b6d89 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1},{"id":"46a651b42f00f09a7014b9ae1c6029b11f687bcc","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-10 00:10:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/bd277396f2d049b2943a3bdb56138b4d : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b609237bac2dc675a21896cf92443935605e04e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-10 01:21:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_Unity_FC_os-brick/ab2889e/DellEMC_Unity_FC_os-brick/278 : SUCCESS in 1h 30m 46s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_Unity_iSCSI_os-brick/ccbd089/DellEMC_Unity_iSCSI_os-brick/292 : FAILURE in 1h 44m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb1758d8069724cc611ba9b1c4a0585a5b2f3d17","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-10 01:46:07.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/93b0ab6ace0e45659a3a11c4020e11d4 : SUCCESS in 2h 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"629783028585992597ee768c803d5f26f79a5d05","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-10 05:17:03.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_PowerStore_FC-ob/8ec2779/DellEMC_PowerStore_FC-ob/214 : SUCCESS in 1h 28m 06s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_PowerStore_iSCSI-ob/8b7319c/DellEMC_PowerStore_iSCSI-ob/136 : SUCCESS in 1h 35m 51s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_PowerStore_NFS-ob/83f73bf/DellEMC_PowerStore_NFS-ob/108 : SUCCESS in 1h 16m 13s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_PowerStore_NVMeTCP-ob/7a9a0d9/DellEMC_PowerStore_NVMeTCP-ob/88 : SUCCESS in 1h 31m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f3f2393c21affb90dfbc1ac278c175c49d1c0976","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-10 05:33:01.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_SC_FC-os-brick/7de7b6e/DellEMC_SC_FC-os-brick/176 : SUCCESS in 1h 03m 32s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/82/876282/1/check/DellEMC_SC_iSCSI-os-brick/2c9b26a/DellEMC_SC_iSCSI-os-brick/171 : FAILURE in 1h 35m 18s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ffc0298ce23f9a6bd31b05a7df5c617db8699de","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-10 22:25:25.000000000","message":"Patch Set 1:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/876282/1/fc : FAILURE in 0h 38m 03s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/82/876282/1/iscsi : FAILURE in 0h 48m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e15edf6c17449899ec8401c37482f7cb5de207d3","revisions":{"e15edf6c17449899ec8401c37482f7cb5de207d3":{"kind":"REWORK","_number":1,"created":"2023-03-09 16:22:25.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/82/876282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/82/876282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/876282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/876282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/82/876282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/82/876282/1"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-03-09 16:20:37.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-03-09 16:21:49.000000000","tz":-300},"subject":"Add Python 3.10 to setup.cfg metadata","message":"Add Python 3.10 to setup.cfg metadata\n\nWe support 3.10 as of Antelope.\n\nChange-Id: Ie3eaf0d16e27e35ec30635c82b38c634db9c4541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e15edf6c17449899ec8401c37482f7cb5de207d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e15edf6c17449899ec8401c37482f7cb5de207d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
