)]}'
{"id":"openstack%2Fos-brick~876284","triplet_id":"openstack%2Fos-brick~master~I2399b0b2af8254cd5697b44dcfcec553c2845bec","project":"openstack/os-brick","branch":"master","topic":"bug/2004630-patch7","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-21 14:44:01.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-01-19 14:19:36.000000000","reason":"\u003cGERRIT_ACCOUNT_35075\u003e replied on the change","reason_account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2023-09-26 11:46:12.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-04-20 12:11:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-04-20 12:11:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-10-30 17:30:47.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"35063":{"account":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-02-21 16:47:55.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-09-28 10:14:55.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I2399b0b2af8254cd5697b44dcfcec553c2845bec","subject":"Dell PowerFlex: Unnecessary login happen","status":"MERGED","created":"2023-03-09 18:39:48.000000000","updated":"2024-02-21 16:54:59.000000000","submitted":"2024-02-21 16:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876284-bug/2004630-patch7","meta_rev_id":"bf9dcbe0ec9f9f2c6e36233bb425184133d13ba5","_number":876284,"virtual_id_number":876284,"owner":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 16:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2024-02-21 14:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-01-19 14:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2023-10-30 17:30:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2023-10-08 03:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-09 01:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2024-02-21 14:38:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-02-21 14:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 18:56:38.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-03-09 19:32:03.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2023-03-09 20:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 20:53:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-09 21:40:55.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-10 03:48:57.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-10 03:57:38.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-10 04:23:19.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-10 08:12:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-10 10:29:04.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-10 22:25:27.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-03-13 17:21:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-13 18:04:09.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 15:19:47.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-03-22 08:06:36.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-03-22 08:06:36.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-03-22 08:06:36.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-24 11:29:00.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-30 11:33:34.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"CC"},{"updated":"2023-03-30 21:37:06.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-19 15:01:00.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-04-19 15:08:11.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-04-20 12:11:30.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-05-24 06:30:09.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-05-24 15:31:59.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-07-05 03:47:54.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-07-05 11:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-21 14:54:46.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2023-07-26 17:21:53.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-26 20:12:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-27 02:40:04.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2023-07-27 04:59:37.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-09-20 07:40:35.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"CC"},{"updated":"2023-10-08 03:48:10.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-10-09 01:39:51.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-10-29 14:15:45.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"6609b8560c0278e06e859d1f4e05d1a82a1f390b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-03-09 18:39:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c710ef66f357231728b08cd9ce8017a3af3b265b","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-03-09 19:32:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bb60f5bd098f7d4665acb08a110436eddd4e905","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-03-09 19:48:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4538924b04d397b2c64d4b7d4074f79f9f5d86f","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-03-09 19:49:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c7eb4c8b34ca9b045d946095dc23558524bed34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:28:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eaed0f37b2b4a9f822e177d6655ec27\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/893654d2cbce42ac91436c69392061ef : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12fffbb4bcb40688049cfbe44e3c871 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce67fae4f3d94d6dbb33195c23980b1e : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7abd761a842745f19dd08b9d051c82fe : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de8f3d841714f3aa47ba946c87965ac : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d4d86fd15a848c69ce6a433fd7c0103 : SUCCESS in 4m 11s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d62e73b035cc494c8d6a21f3ea1e06ec : SUCCESS in 4m 03s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/272e52deb00d401e8f3609773c4cb343 : SUCCESS in 1h 20m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02fe63b684ce4f64b572574f04db5ea6 : SUCCESS in 1h 45m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/18ae342e503f498e92d04da5104cb609 : SUCCESS in 1h 18m 12s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c8b814bb6d4b41a1a61c27f83c10eab1 : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56d2af31f037102780dccd2162356fdab885bd51","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-09 20:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/84/876284/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/545ee2d/ : SUCCESS in 1h 49m 10s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/84/876284/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7adff45/ : FAILURE in 2h 11m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/84/876284/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/cd73eb3/ : SUCCESS in 1h 42m 06s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/acc82f5/ : SUCCESS in 1h 45m 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/cc5fe80/ : SUCCESS in 1h 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"760038c6452b2b6567a0c792e6e80965d5ac45db","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-09 21:40:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/1/upstream-check/cinder-tempest-plugin-ontap-fcp/166a00e/ : SUCCESS in 57m 46s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/83cd99a/ : SUCCESS in 1h 09m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/1/upstream-check/cinder-tempest-plugin-ontap-nfs/ee3fb45/ : SUCCESS in 1h 11m 27s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8153f46/ : SUCCESS in 42m 48s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"64142b753f4c0543132fded8f6edccd61159c398","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-10 03:48:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_PowerStore_FC-ob/7bab14c/DellEMC_PowerStore_FC-ob/213 : SUCCESS in 1h 29m 47s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_PowerStore_iSCSI-ob/2ca05d4/DellEMC_PowerStore_iSCSI-ob/135 : SUCCESS in 1h 26m 51s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_PowerStore_NFS-ob/6374804/DellEMC_PowerStore_NFS-ob/107 : SUCCESS in 1h 19m 26s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_PowerStore_NVMeTCP-ob/815b729/DellEMC_PowerStore_NVMeTCP-ob/87 : SUCCESS in 1h 32m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2fce63e2bb7a8b79175ccee9a20ee2bf64112ad","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-10 03:57:38.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_SC_FC-os-brick/3d7f82b/DellEMC_SC_FC-os-brick/175 : SUCCESS in 1h 06m 39s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_SC_iSCSI-os-brick/49efad7/DellEMC_SC_iSCSI-os-brick/170 : SUCCESS in 1h 33m 26s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"f07a3c01a900ef5d28edf7f5136d940d9dd8caf2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-10 04:23:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9cca242c362d495aaaa81825784419d6 : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"71814a5d2cd91f18475b537eedbca114105c98e5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-10 08:12:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/aad90297dcb4449f9f5a056e8b68eef3 : SUCCESS in 2h 59m 13s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/73d0c87138fd4527ba47b28abf799236 : SUCCESS in 2h 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5142924a03a79c018cbc72bce324742b3fef81b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-10 10:29:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_Unity_FC_os-brick/c1f03bc/DellEMC_Unity_FC_os-brick/279 : SUCCESS in 1h 38m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/1/check/DellEMC_Unity_iSCSI_os-brick/1083f0f/DellEMC_Unity_iSCSI_os-brick/293 : FAILURE in 1h 26m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ef26d08e70e99b440343304d9f39e291a12469d","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-10 22:25:27.000000000","message":"Patch Set 1:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/84/876284/1/fc : FAILURE in 0h 41m 34s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/84/876284/1/iscsi : FAILURE in 1h 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"94aea74e47a8b6b70ab4401cc803126dbc13278e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-03-13 17:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d2700fac851f185ca4f0c44075418de3dd0c96c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-13 17:21:59.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/88/ (1/2)","accounts_in_message":[],"_revision_number":2},{"id":"ce135159d6ffc495cab9fdfe7cc1e1b3d2625501","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-03-13 17:23:11.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ef51cdbd4b7aac1c037d71b99b1e75e52207d01","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-13 17:47:44.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"0ae0af105b41649a4cb5aefb5408176f8cf9b5cb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-13 18:04:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick finger://openstack-ci-fc/4c4b24bdb5d540f9af00c77efc0e30c1 : RETRY_LIMIT in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"33699e0d36935c809d2bcbcdebea2169398db059","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-03-13 18:07:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02e71552dae10f255bfc343fb3f7d3ac2cdaa067","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-13 18:35:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/4603c584faac47199860e525805f46b9 : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a4ff95e5697c03cf4cea60e6e9f3884991155eec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-13 18:55:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/2/check/DellEMC_Unity_FC_os-brick/4e53b52/DellEMC_Unity_FC_os-brick/281 : SUCCESS in 1h 33m 01s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/2/check/DellEMC_Unity_iSCSI_os-brick/6db1885/DellEMC_Unity_iSCSI_os-brick/295 : FAILURE in 1h 25m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"eea92e4368632db29aa5034c43b1fb163710b29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 18:57:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b7f133187e94c6f8d47de39c2aac221\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bad09e8e2fbb4636a4087f08c24d8369 : SUCCESS in 1h 07m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63db914cc4574c228c27de29ce341ccc : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7144e31ffa3b44689866ccfb7370191a : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f23d4be88844dcf9cd2d3ce6700694e : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de98f5eccabc48eba6b3f48510b4f22a : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f8d4c6c14ab4c289e31152e05efa1d8 : SUCCESS in 4m 23s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7c9b689c914d4a34b885a756f56c8a7b : SUCCESS in 3m 59s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2422925a150543558aa1a15a6fb80c02 : SUCCESS in 1h 29m 28s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d2cc5c9fdcab40c9ba2bf06e97b9b7c9 : FAILURE in 1h 23m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/899960dc3f7043559fcb852e72a41bf6 : SUCCESS in 1h 11m 54s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d664d46d34e64f76a913755f6538f2ce : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1de9521729c27c566823f22400cbc1b15a79d98","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-13 19:29:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/84/876284/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/066f18c/ : SUCCESS in 1h 53m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/84/876284/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f3a9efd/ : FAILURE in 2h 05m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/84/876284/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/0027ce9/ : SUCCESS in 1h 31m 28s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/402e084/ : SUCCESS in 1h 40m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/186829c/ : SUCCESS in 2h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7efab86f043d892df7f891ca4e154e44aaedca5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-13 23:53:59.000000000","message":"Patch Set 2:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/88/ : ABORTED\n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/1169/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f36bf63f052564f2f046bffa08eb67a4870d047c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-03-14 13:53:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4fa724277b1f7684eabdb3345ca75b8406c6d1da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-14 13:53:54.000000000","message":"Patch Set 3:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/94/ (1/2)","accounts_in_message":[],"_revision_number":3},{"id":"97ab1fd01636e1f515f04d04a043176bafba5d62","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-03-14 13:54:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbf552eaf5f7733bde922bce5b884f96cb38f016","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-14 14:15:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-os-brick finger://openstack-ci-fc/528d44a5db89444abf55785b7989de00 : RETRY_LIMIT in 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0712b6ba6d07455a6cb38e1dd6333553eaba6a28","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-14 15:05:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/2f10d8dd290c4523b43b60d7db175b0b : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c170d0e0d4374e535daf548639c89e793ace8772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 15:36:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3eb90308c664292a3c515140f881653\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4328381f82a749839dadf05a69f614a5 : SUCCESS in 1h 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3752d27e9864cfd94835d524e41a89d : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eca78deb4dd4dab8a2ad102d456847e : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f69ddf6464ee4c608783826c4c7a6d01 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6f1cf2152b459799675d1e140f728b : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d2fd3c24034b48a2261c910d043c35 : SUCCESS in 4m 09s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/1fb956a301204c9aad788bdf30791222 : SUCCESS in 3m 52s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/af2cdcb8f6df4ac1beb96e4b8597c56e : SUCCESS in 1h 23m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/221645f2ce944bcf99ccac3a0d0ae44f : SUCCESS in 1h 38m 24s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/06b45dd824b4472d92c39ffb73ad6ab8 : SUCCESS in 1h 23m 52s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f11a368880484f7cbdecdd1ba29bd400 : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"78051301dbaeeaf1b92c50a25a07828bc81d86c1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-14 16:29:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/84/876284/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/beb0d87/ : SUCCESS in 1h 42m 02s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/84/876284/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/54abbbb/ : FAILURE in 2h 32m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/84/876284/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/fddc36f/ : SUCCESS in 1h 44m 45s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/f6c3b4e/ : SUCCESS in 1h 59m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/96e0052/ : SUCCESS in 1h 59m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f5715af78621e0ba74caff3002809161fa0f424","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-14 16:38:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/3/upstream-check/cinder-tempest-plugin-ontap-fcp/f2fad9b/ : SUCCESS in 1h 01m 24s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/408879d/ : SUCCESS in 44m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/3/upstream-check/cinder-tempest-plugin-ontap-nfs/6ce5e72/ : SUCCESS in 45m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4b18de3/ : FAILURE in 42m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1a2a6b6b4509c3ec87de1736fe69feb6f4796336","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-14 17:08:49.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_PowerStore_FC-ob/c58ecfb/DellEMC_PowerStore_FC-ob/216 : FAILURE in 1h 28m 59s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_PowerStore_iSCSI-ob/fd86e52/DellEMC_PowerStore_iSCSI-ob/138 : SUCCESS in 1h 32m 29s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_PowerStore_NFS-ob/c3999b8/DellEMC_PowerStore_NFS-ob/110 : SUCCESS in 1h 18m 27s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_PowerStore_NVMeTCP-ob/a44c936/DellEMC_PowerStore_NVMeTCP-ob/90 : SUCCESS in 1h 30m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"564af56424e4d6585e4a8fdaef5ba4cd8b9f9695","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-14 17:22:55.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_SC_FC-os-brick/be9f251/DellEMC_SC_FC-os-brick/178 : SUCCESS in 1h 04m 30s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_SC_iSCSI-os-brick/c5e1081/DellEMC_SC_iSCSI-os-brick/173 : SUCCESS in 1h 46m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"3bea8a8e68a5db1cd14435e58a4addca190e09f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-14 18:00:21.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_Unity_FC_os-brick/af10e6f/DellEMC_Unity_FC_os-brick/282 : FAILURE in 1h 58m 26s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/3/check/DellEMC_Unity_iSCSI_os-brick/457a117/DellEMC_Unity_iSCSI_os-brick/296 : FAILURE in 2h 15m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"05b483c234ef55bbdd816b755bb2e53a666b946e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-14 18:33:12.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0a4baca8d87b43b0ae076d13501103a8 : POST_FAILURE in 3h 34m 11s\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f574f56522f34fb192ed2cced82679d3 : POST_FAILURE in 2h 27m 19s","accounts_in_message":[],"_revision_number":3},{"id":"123b2284ac7b599754895214a0685801943c1a1c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-15 01:46:15.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1175","accounts_in_message":[],"_revision_number":3},{"id":"ed7b0ae35de0644f37e23712ab1e04fe44a51422","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-17 15:19:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/3/20230317212744/NEC-V-ISCSI : SUCCESS in 2h 36m 25s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/3/20230317212744/NEC-V-FC : SUCCESS in 2h 52m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7eef0cba00128af38ecae8e79a550fa0bdde48cb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-24 11:29:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/3/20230324135621/NEC-ISCSI : SUCCESS in 3h 20m 46s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/3/20230324135621/NEC-FC : SUCCESS in 6h 32m 36s","accounts_in_message":[],"_revision_number":3},{"id":"87eebe318b6fb654e663efdc6dae4e0174095c92","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-03-30 11:33:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b05a4d6d68550f9799c2e988dc0f13e2ed01569f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-30 21:37:06.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-3/FC/ : SUCCESS in 2h 6m 56s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-3/iSCSI/ : SUCCESS in 2h 3m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"05e469e72496e9f808b1dab4c76bb0d55c06b4cd","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-04-19 15:01:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58665bb204982bc128dd6bc8e7857cf56607801f","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-04-19 15:03:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2934099bbd3c96155ef5dc2eb3ede172d3c52339","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-19 15:08:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"74b22f2c186f14e65c265b969e55915e2803fe2c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-04-20 12:11:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cc7f192bb14762142375b762372ff78fffbe66d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-19 16:45:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d2b6249ed016feeed48f7fcd89f221cbc579456d","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-19 16:49:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d5c99e0f0468cc94a44113434d784e9d4846e6d0","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-19 16:49:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"327c416d2c1917f0e83563c085138d06da472c6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-19 16:51:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fd992e6d4c8f7477f45831c482049b33ac2ccb7b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-19 17:36:50.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/3527d25431cb4630ab74ef2fa25e5a50 : POST_FAILURE in 44m 56s","accounts_in_message":[],"_revision_number":5},{"id":"21cdf670ed5dd1f6033f188d49d807c4c40cac58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:34:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c731fd6ef1434a8653f2328a8ba7fb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5eed6fdad07c434e9f9c7819cf4c6b7d : SUCCESS in 1h 26m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca12c0bcacb144bba255e2a8c55074c9 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c348db2979f8403ea99c8590f1bbe8df : FAILURE in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9492886d0c824502adc4d0ed74d479c3 : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d272f5c0ad45aea56442159a2c1099 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb12c3eef1ac4be3a024c183fed9b631 : SUCCESS in 4m 33s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c2db8429e20d498d96c80f68189eb115 : FAILURE in 3m 38s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2225b2c5858a41848b768f6ad683e780 : SUCCESS in 1h 29m 34s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0dc82541a47f4c1ca46459bb0146c33f : SUCCESS in 1h 41m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/bb5699a985b84b93978b272c2bb25304 : SUCCESS in 1h 21m 32s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/57c43109e6c447a798b14eef704c6f1f : FAILURE in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70a10968daebaf9d684e75aff0d1acd5563889a3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-19 18:59:59.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/5/check/DellEMC_Unity_FC_os-brick/a53ba88/DellEMC_Unity_FC_os-brick/13 : SUCCESS in 1h 50m 43s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/5/check/DellEMC_Unity_iSCSI_os-brick/ff29ae3/DellEMC_Unity_iSCSI_os-brick/13 : SUCCESS in 2h 08m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"d2bf15191f35a1c6c00bbd312c352e1dc3ebee29","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-19 20:16:10.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"98a880867cb688ef16a4f499abce5b07bfbdb20f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-19 22:05:50.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/5/20230520040327/NEC-V-ISCSI : SUCCESS in 2h 22m 13s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/5/20230520040327/NEC-V-FC : SUCCESS in 3h 02m 20s","accounts_in_message":[],"_revision_number":5},{"id":"f2b60b44e24aaa1bd3838923ec48fef041ea7dcd","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-20 07:39:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/5/20230520090742/NEC-ISCSI : SUCCESS in 3h 20m 04s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/5/20230520090742/NEC-FC : SUCCESS in 7h 32m 09s","accounts_in_message":[],"_revision_number":5},{"id":"6df84f75140ecbf5eda8ee3e2a59bf5757366dc7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-21 17:14:35.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/79c5303fe94f491d8b58d173b2e88b78 : RETRY_LIMIT in 1m 00s\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/os-brick/84/876284/5/check/tempest-dsvm-full-focal-py3/20d90a4/ : FAILURE in 2h 04m 09s","accounts_in_message":[],"_revision_number":5},{"id":"a4f65b1dbdfd2e33ac94a82565434b1da9d2c8e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-22 16:15:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a243eaf20895e6aa1ee5c0e8caa7c4634095253a","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-22 16:17:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5f75242f0341f2075d8465517093b924f64cc742","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-22 16:23:15.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"39886462ab0a57f3f44c817b49840fe5eeaf06fe","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-22 17:02:09.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/115a8e7966644c40b7bfab9d1d2c865e : POST_FAILURE in 46m 18s","accounts_in_message":[],"_revision_number":6},{"id":"27361626be163c7eea8c5a53cd4161d2e76cd237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 17:57:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e86297af1464cf6bc0d327e7b8b417c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13caedc5f1754736b741268aa36d650e : SUCCESS in 1h 07m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048388269ba1429b893d6d4fbda9a9db : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff0abfedacb845f9a195b91cde26a91e : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7fd21df0d1c498584185bd8b6e8d679 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddae67809a83455688a54206978e6b23 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6ec54afaee41f8bc2644e594aca6cb : SUCCESS in 4m 24s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9772e76001344f0387896db8bd68f551 : SUCCESS in 4m 54s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e834db26ff784c598153baaec36360e5 : SUCCESS in 1h 40m 37s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96b3363877f640a9a691c87887374dbb : SUCCESS in 1h 34m 19s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ba33961a78794f6f96a13726ea2aa9a0 : SUCCESS in 1h 05m 17s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/78d4a94637c04765b3dbc7a02a993ed2 : FAILURE in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cdca67687c560ad2ab42762c6716d8a701ec89ca","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-22 18:18:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-6/FC/ : SUCCESS in 2h 2m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-6/iSCSI/ : SUCCESS in 1h 48m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"50d83c12477d9dedc4397888d40be6a09135103c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-22 18:27:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_Unity_FC_os-brick/d461d6e/DellEMC_Unity_FC_os-brick/14 : FAILURE in 2h 06m 13s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_Unity_iSCSI_os-brick/f296289/DellEMC_Unity_iSCSI_os-brick/14 : FAILURE in 2h 12m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"9bdf1d8facf595bf415a5ac1fa9a643557d1c460","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-22 18:51:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/6/20230523011513/NEC-V-ISCSI : SUCCESS in 2h 36m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/6/20230523011513/NEC-V-FC : SUCCESS in 2h 36m 20s","accounts_in_message":[],"_revision_number":6},{"id":"f9bf54165944a77c5a5f2484a78a32324aebb31e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-22 19:08:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/6/upstream-check/cinder-tempest-plugin-ontap-fcp/b196f4a/ : RETRY_LIMIT in 8m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/f590e1d/ : SUCCESS in 1h 08m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/6/upstream-check/cinder-tempest-plugin-ontap-nfs/2e89000/ : SUCCESS in 47m 21s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8903fec/ : FAILURE in 1h 07m 46s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"e8ea090809819366849d52159f4ae148a086b034","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-22 19:36:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/6/20230523011512/NEC-FC : SUCCESS in 3h 21m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/6/20230523011512/NEC-ISCSI : SUCCESS in 3h 21m 28s","accounts_in_message":[],"_revision_number":6},{"id":"829639c9298ba368f8f9ab24cd0568dd11c3a257","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-22 19:57:20.000000000","message":"Patch Set 6:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1362/","accounts_in_message":[],"_revision_number":6},{"id":"9951bcfe61fe7f387eb234e4e84d8ce7409e88e1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-23 00:04:11.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_PowerStore_FC_os-brick/295c0ba/DellEMC_PowerStore_FC_os-brick/6 : SUCCESS in 1h 16m 01s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_PowerStore_iSCSI_os-brick/8a5560d/DellEMC_PowerStore_iSCSI_os-brick/6 : SUCCESS in 1h 04m 06s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_PowerStore_NFS_os-brick/33f52de/DellEMC_PowerStore_NFS_os-brick/6 : SUCCESS in 47m 16s\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/6/check/DellEMC_PowerStore_NVMeTCP_os-brick/87fe85e/DellEMC_PowerStore_NVMeTCP_os-brick/6 : SUCCESS in 1h 00m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"d3622bb4c3c5b1257edc257b6fef806435e75a97","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-24 06:30:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b2570739626e31ba912f9e9cfcd6c25876944f7","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-24 15:31:59.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ca45d8158dfe8e2b2677ff8f837546fa6c7bc9cc","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-05-30 13:50:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5bd7f9c1b1ad09e1eb0d8649be8dfc527be42021","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-06-12 09:14:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"57f1e38476cf01162423d68ff1f139be9d8acda7","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-07-05 03:47:54.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"adac4da022f6c86b07ce2dda783bdc1fb4a1b57c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-05 11:00:08.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/6/fc : SUCCESS in 0h 42m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/6/iscsi : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":6},{"id":"4227c3e9daf34fade141311b21781ce4a98a2e15","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-07-21 14:54:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"722cb5259c53145c8b90633a2e77851439be04f8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-21 15:23:55.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f98e36eb01c4731fbec9a06ee430e9d447fc532a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-07-26 17:16:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ed0d69e646f71d40626eea19cfba697a9e297e5c","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-07-26 17:18:55.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7ffc5a66bd5a464a118855adeabe0bfb4ebb3efb","author":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"date":"2023-07-26 17:19:14.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b34f1da14d88d7427621c6b4a2e6c850a60b7184","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-26 17:21:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6d287bda163e476481117d7b0dba9e10 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"800f73e7f755bd08b1303e91396a55b8578fdaff","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-26 19:37:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/56f832f/ : SUCCESS in 1h 57m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/84/876284/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/fedb98a/ : SUCCESS in 2h 01m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/4e4ce92/ : SUCCESS in 2h 18m 40s","accounts_in_message":[],"_revision_number":7},{"id":"136bdbf78a85681aa3617826028483d6ac706a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 19:51:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14688db64234152ba1c9a6ab2769eaf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1c3df40a8f14b9b8ab5768fe1049f64 : SUCCESS in 1h 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d43424bae74267bea05b5b85081ffc : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be30dc4b5d194abfa4b262d886e9caea : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/201f2c53766f4ae2863c824a4d0ffb9a : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa66783973944db386622df805b206a7 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88f0b0bb3cb841bc82e5bcb152b38b4b : SUCCESS in 4m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9793faf54e0c445d83cb11ffc07aaeb9 : SUCCESS in 3m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7123d1a59eba46758340c858010e5058 : FAILURE in 2h 33m 27s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a7581d4b16c24ad2a2cb6cf694f3600a : SUCCESS in 1h 40m 31s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c203709f02be4a37829a8f7ccac11b77 : SUCCESS in 1h 40m 32s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/00ef4ced5df048c3aed7ffc37478cf6d : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cfd90a01fc1384c832807dd81c35b229c463416d","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-26 20:00:08.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/7/fc : SUCCESS in 0h 40m 44s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/7/iscsi : SUCCESS in 0h 47m 13s","accounts_in_message":[],"_revision_number":7},{"id":"90e30bb5a88929dc93a94a26437d93b4bc462534","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-26 20:12:46.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerFlex_os-brick/063a9ae/DellEMC_PowerFlex_os-brick/18 : FAILURE in 1h 48m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"52c56e91422cd811cbb2eeae0dda0960210186e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-26 20:52:59.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_Unity_FC_os-brick/e595b51/DellEMC_Unity_FC_os-brick/37 : FAILURE in 1h 57m 47s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_Unity_iSCSI_os-brick/220ad2e/DellEMC_Unity_iSCSI_os-brick/37 : FAILURE in 2h 19m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"d4e1112a1438c14c3071c7ab1aef86c70f58ee57","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-26 21:12:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/7/upstream-check/cinder-tempest-plugin-ontap-fcp/8e8a5f9/ : SUCCESS in 1h 17m 47s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/8faee50/ : SUCCESS in 43m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/7/upstream-check/cinder-tempest-plugin-ontap-nfs/5b6ba38/ : SUCCESS in 1h 18m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ba94e89/ : SUCCESS in 1h 19m 44s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"a780e0e13f5b1d35cd3bdb5647105c5834d8e7ac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-26 22:08:59.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerStore_FC_os-brick/8227e47/DellEMC_PowerStore_FC_os-brick/25 : FAILURE in 1h 17m 55s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerStore_iSCSI_os-brick/88269d7/DellEMC_PowerStore_iSCSI_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerStore_NFS_os-brick/389f2fc/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerStore_NVMeTCP_os-brick/9648511/DellEMC_PowerStore_NVMeTCP_os-brick/25 : FAILURE in 1h 14m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"25535960cbab93243db9366ca7829de39d7000de","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-27 01:55:40.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/7/20230727043121/NEC-V-ISCSI : SUCCESS in 2h 39m 33s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/7/20230727043121/NEC-V-FC : SUCCESS in 6h 24m 16s","accounts_in_message":[],"_revision_number":7},{"id":"5e3bcdd257153a7185e0a34d7ba49ee11ebd75a3","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-07-27 02:40:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cf66e2358b9624e9c0088c96ec9b8392637a8578","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 04:25:38.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerFlex_os-brick/b57c2be/DellEMC_PowerFlex_os-brick/20 : FAILURE in 1h 45m 19s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"8ab02aa60aa647f3049f3938dd68413c50c6ff7b","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-27 04:59:37.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/7/check/tyds_test/5eaedbf : FAILURE in 4h 34m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c96db92ff38cce341b25fe7bd4a192009f07efbd","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-07-27 06:18:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d860acb9f0b10afc9527ab7bd6fda2d6a79c6566","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 08:22:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerFlex_os-brick/c5e0445/DellEMC_PowerFlex_os-brick/21 : FAILURE in 2h 04m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"8ec5e210a658af5d292fe1726ebf24232cd9a437","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-27 11:54:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/7/20230727163851/NEC-FC : SUCCESS in 3h 59m 24s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/7/20230727163851/NEC-ISCSI : SUCCESS in 4h 15m 22s","accounts_in_message":[],"_revision_number":7},{"id":"88ce0ce81ebba069747864d9a24cce6c5733a842","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-08-24 03:13:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c5da879bbce36a209e2eb9546c8190526fc6879d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-24 06:31:14.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/7/check/DellEMC_PowerFlex_os-brick/b34e5f7/DellEMC_PowerFlex_os-brick/35 : SUCCESS in 1h 45m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"e3aa57ade21e6f238801ed7076bdfb931a531522","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-11 09:50:44.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"11de43c0bfaa688f8322a5749c94254e539aff23","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-15 14:16:41.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0dc1e1b099e8c2cf4123ecdd828f2872fbea2c33","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-09-15 14:33:12.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2f2baac5a887a4e5b854cf0a1243e2fa57e3307e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-15 16:02:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"052ef82c58213d84b1ea8df6e9ae024fca0bd0c4","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 07:40:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a170948d1964224b6ee0feacbe1f406055594200","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-20 07:42:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"908ce971ba9bb46877adb17093e41813d179eb45","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-20 10:56:23.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"28b91e7922257620e0cc3322361fc6eaf2dd7cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 11:44:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"511cdab939a3c5255bade3d3844a413abeda1f7b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-20 11:45:04.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8c7e2935aa8a4b0da207f6642409a281 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"cca9ab02c824c68348a2f5e868038bf44b228e99","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 11:52:56.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"972be69cf71b8b0bb9f4237872f8daf9318cef21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 11:54:29.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"09834ce7e10d9b030c67fefcad2bc53b92c641b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 11:55:00.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"6417880040c1211fc8f68122f5b3935950192407","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-20 11:56:16.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/56576e76cd144d2a9ea5242a4bd13cbb : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"16091ec3abe81cd181136d58c53093ea18368fe9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-20 12:27:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/79ca384f51764c7bad2d0d0415ba25e9 : RETRY_LIMIT in 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/39db24f37c9f43c69cf2ae5c3759d7b5 : RETRY_LIMIT in 41s","accounts_in_message":[],"_revision_number":10},{"id":"d4279587afbe1956a53587eb04a81b1a1c699221","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-20 13:27:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1871/","accounts_in_message":[],"_revision_number":10},{"id":"0327345536ed098a5ac3738f642ee7825c6bc06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 13:58:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e805c8625442fab2db3d7447f05727\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98fa8b5a55814ff7978d68ec200d6075 : SUCCESS in 1h 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb91c99a53b4ca2a2453dedf65bedb0 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10dbd081472a40f1b5ee68fc838658b8 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b5b470cb623482e979ae8aacc4fe838 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/132e6dc9a51d416da323a1a47d652e58 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d0ef7665eea490aaa36302d915ee1b4 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1353249ac4a447789335b92d5c602b04 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9647f889b384caead8695a5ac5d911b : SUCCESS in 3m 59s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/e6072a171efd40b7aa6a447680a02c05 : SUCCESS in 4m 19s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7c0b04c4a50b4e04ac44f72e1af2809f : SUCCESS in 1h 56m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c229869796eb4e5d98b02ded3bd18977 : SUCCESS in 1h 51m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7a72a223ac7f477fb47976d2ae36b76e : SUCCESS in 56m 00s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/ebba7e548edb4ab5b94df72828e2971b : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"25ba11321187b5c2b7025f285810561d4d8886ff","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-20 14:27:18.000000000","message":"Patch Set 10:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/10/check/tyds_test/761864f : FAILURE in 9m 31s","accounts_in_message":[],"_revision_number":10},{"id":"064fd22d14aff8e1f6f1b16fda250cf126ce9a0a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-20 14:48:32.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2fcdc949552d9f270d4717f7c0c50972675d3c6e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-20 15:28:29.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/10/upstream-check/cinder-tempest-plugin-ontap-fcp/20a2fc5/ : RETRY_LIMIT in 7m 54s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/83885b5/ : SUCCESS in 42m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/10/upstream-check/cinder-tempest-plugin-ontap-nfs/8ca27ba/ : SUCCESS in 46m 06s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/5c10869/ : SUCCESS in 1h 28m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"165d726ef2703e5545a299b79faf1b3cf61865c3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 15:59:14.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/10/check/DellEMC_Unity_FC_os-brick/bf7ec83/DellEMC_Unity_FC_os-brick/106 : FAILURE in 1h 32m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/10/check/DellEMC_Unity_iSCSI_os-brick/7c1da18/DellEMC_Unity_iSCSI_os-brick/106 : FAILURE in 1h 38m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"aa3694fc2a583d814d98b8792c3c203431eb211b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 16:11:11.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"3b7e4592776e16ee6f353bd2e6f53c8c95b044e7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-20 16:12:31.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3341d89365fe421ca6963d9011c2bdea : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"3b4a4d765b365d6f12d7789954f3d814b2558ba6","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-20 16:20:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6b1537c186e96a5a2a1456751da331172ce53f65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-20 18:04:03.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/11/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/81209cd/ : SUCCESS in 1h 51m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/e0290d8ad69d45bdac55f21f2ab3e4c0 : RETRY_LIMIT in 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/a1c870c662b44686be40c570785e92f9 : RETRY_LIMIT in 42s","accounts_in_message":[],"_revision_number":11},{"id":"2e9ee2ffa0645d1981cdd340308c09ea37d9b841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 18:21:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e09f140d66435bb32073a49c792a36\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d1d2744c1c84aa8b24363941530f200 : SUCCESS in 1h 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e7a86d94de499abd9f19f605151e7e : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea22c870457743f3aee97f5377ab9f21 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a0a840ba0e54760959853497b99fcae : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4c538cac2ec4dfeb432f158f31a57aa : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65706b2abcc84f65ada57c791ff203e8 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb7e0bcfef24eecb3926e4fbf4719f0 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0eae115aec641dbac241ec7d88cc3fd : SUCCESS in 5m 27s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/214c63e296ec4695ae2dd5743205d653 : SUCCESS in 4m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a31aabf33106435f8f50a4fd14324438 : SUCCESS in 2h 02m 15s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d814aaa2963b4d519e172c436de5318d : SUCCESS in 1h 20m 19s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0bd970ad46a648c2828dc72bd725b7a8 : SUCCESS in 1h 26m 48s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c2ed6066cc2e4b6cbc96e12ec5d6353e : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"233978397fc2b09ba7e521aeeb069327f3611825","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-20 19:41:58.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/11/upstream-check/cinder-tempest-plugin-ontap-fcp/c18c3c1/ : RETRY_LIMIT in 7m 58s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/d135c4f/ : SUCCESS in 40m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/11/upstream-check/cinder-tempest-plugin-ontap-nfs/d87ef8e/ : SUCCESS in 45m 19s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/11/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/43662ea/ : SUCCESS in 1h 19m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"dc938e06cc8ed8d8ac1df8d636193f205659612a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 20:34:26.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_PowerFlex_os-brick/4385290/DellEMC_PowerFlex_os-brick/70 : SUCCESS in 1h 06m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"ebb8dba1f94b09e4d11655c02d210ab92cc70570","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-20 20:35:11.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-9/FC/ : UNSTABLE in 3m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-9/iSCSI/ : FAILURE in 7h 54m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"ca849c77ccb08f8fbd1a4a30e7c0fb5dc14f8418","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 21:07:13.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_Unity_FC_os-brick/c33a98b/DellEMC_Unity_FC_os-brick/109 : FAILURE in 1h 32m 46s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_Unity_iSCSI_os-brick/f5d6225/DellEMC_Unity_iSCSI_os-brick/109 : FAILURE in 1h 42m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"6a145193e45dff6fe8c54ab2765cbb7fd75fad9f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-21 00:54:48.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_PowerStore_FC_os-brick/ce9c4b8/DellEMC_PowerStore_FC_os-brick/61 : FAILURE in 1h 12m 28s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_PowerStore_iSCSI_os-brick/8e5a0cf/DellEMC_PowerStore_iSCSI_os-brick/54 : FAILURE in 1h 11m 37s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_PowerStore_NFS_os-brick/3aebad0/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/11/check/DellEMC_PowerStore_NVMeTCP_os-brick/3b64757/DellEMC_PowerStore_NVMeTCP_os-brick/65 : FAILURE in 1h 10m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"9ccc1f8ad00a8a19249d14c769f0b5ed8977010e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-21 01:16:14.000000000","message":"Patch Set 11:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/11/check/tyds_test/3d8ae2a : FAILURE in 9m 19s","accounts_in_message":[],"_revision_number":11},{"id":"6d27e46d2c8125a1b625d2a3f644c94933dadbbd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-09-21 07:36:48.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/11/20230921132831/NEC-V-ISCSI : SUCCESS in 2h 54m 31s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/11/20230921132831/NEC-V-FC : SUCCESS in 3h 08m 14s","accounts_in_message":[],"_revision_number":11},{"id":"f45d824341a0b0804739892c8defe305a9dd5605","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-09-22 03:33:10.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/11/20230922042429/NEC-ISCSI : SUCCESS in 3h 43m 21s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/11/20230922042429/NEC-FC : SUCCESS in 8h 08m 38s","accounts_in_message":[],"_revision_number":11},{"id":"8ed2b46f1ac56c81a657c452ef8f4eaf4cacd545","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-26 08:01:59.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"83c49846d56271068dc54aa5896b96b3554733a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-26 09:33:47.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"de4279689d7bcafe66da35a3b59b217eec571165","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-26 09:34:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"33dac05183e9f6abb66e40425b7518219a633148","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-26 09:34:58.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5c6667107bf04418a034aa12e1d0e40d : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"2f8c361eda270bb6d4293bc14bcd9a2687a70937","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-26 10:50:49.000000000","message":"Patch Set 12:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1918/","accounts_in_message":[],"_revision_number":12},{"id":"6a5e48429ff28de6f78ec48d05bac56580bf9e12","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-26 11:27:36.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/fd63698/ : SUCCESS in 1h 51m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/84/876284/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/ab69b61/ : SUCCESS in 1h 50m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/12/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5726d50/ : SUCCESS in 1h 52m 42s","accounts_in_message":[],"_revision_number":12},{"id":"66a1dd881f8cf4cc2cd800e62f13cdea352e6b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 11:44:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1beb473a274d48a68d8c1d17525945\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a383cd4c07c46eaa0e70e5db9891d6b : SUCCESS in 1h 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6d91c8e0c14ddfb9650b2877bda834 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52a32d5a93784f229c0f284b112f7194 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60cdbce140b94749b155b7a2ee9d953b : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a43de41615e49eaa9a82027b7755ab5 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51cc8a45c0f54563955b672a1d3be255 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6520e06124a341e8ad0b88762b71e061 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3d8039298f74fbc94db562ae66c1dfc : SUCCESS in 4m 01s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c61833b5a80f41a892cfffe3855b10e7 : SUCCESS in 3m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/d979117bca4a46f38778b0ac96bfb64f : SUCCESS in 1h 49m 10s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0624442e5e444cffa6123dd784645a9c : FAILURE in 2h 04m 42s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/86d853ae30a94e7eaa9c973b101e6095 : SUCCESS in 1h 36m 05s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/46257255d3124776b027829ea806b8dd : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7b56c8fbf5815f6ec9a8fd1d4777cfe5626a0f7c","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-26 11:46:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1ffa2e1c627ef034f2ab5661016d378bc5a12f3a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-26 11:47:23.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/da859520cdbc4af9ad6fcb11373d660c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"db2c55d4410ebfb04ad59e87933cc82d165dd935","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-26 11:52:37.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-12/FC/ : FAILURE in 2h 18m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-12/iSCSI/ : FAILURE in 1h 28m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"27f2eecd7373fee9dd3c8e740d0a6ac8759c3022","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-26 12:00:07.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/12/fc : SUCCESS in 0h 42m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/12/iscsi : SUCCESS in 0h 48m 34s","accounts_in_message":[],"_revision_number":12},{"id":"93d5ab3474154d544b70f84913b1baeff11a14a9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-26 12:25:30.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerFlex_os-brick/1501b9e/DellEMC_PowerFlex_os-brick/72 : FAILURE in 2h 51m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"97fa54759811ddd4cb21238c1646c6148756ce7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 13:53:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147bf16a55dd499ea4e169d369bb5c91\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2e27053ea684b2c82ef0a8f01572ffc : SUCCESS in 1h 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5440a764f104aa88859ad3850061401 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b20bf257cb04f49b1cc2a684f679218 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d55cebb20f084070bb5bb1ec31fd98a7 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e31008d5c044c4ba14fe29bf8f40162 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1b88477f9bb4924812c40f41bfa94d8 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6305ff3d17b40a4bcee05dfdcc7eeaf : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8decd846f952452eb2090f21b916e0f5 : SUCCESS in 4m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c1c74bf93adf48db9e10e34bb629eb8e : SUCCESS in 4m 56s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/43b0776872fc479784979f27e1d0131c : SUCCESS in 2h 01m 48s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e307b989d88d4aba9d7d8cae16ddaa04 : SUCCESS in 1h 54m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6d165797b3924fba8fc9cfbe86070efb : SUCCESS in 1h 30m 36s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/122dc6a8ef87483a93f0ce85bed5855b : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5e8796087dee0c452cb786dc0528808b3e2e8064","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-26 14:40:51.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/12/upstream-check/cinder-tempest-plugin-ontap-fcp/a93464c/ : RETRY_LIMIT in 8m 05s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/a146d2c/ : SUCCESS in 42m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/12/upstream-check/cinder-tempest-plugin-ontap-nfs/db263ca/ : SUCCESS in 45m 34s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d3e8f6d/ : SUCCESS in 43m 22s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"dc73908986c43299fcf13bd455664f61c598efb5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-26 15:03:48.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerStore_FC_os-brick/d86831f/DellEMC_PowerStore_FC_os-brick/62 : FAILURE in 1h 09m 44s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerStore_iSCSI_os-brick/05a8c3b/DellEMC_PowerStore_iSCSI_os-brick/55 : FAILURE in 1h 09m 54s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerStore_NFS_os-brick/031fee1/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerStore_NVMeTCP_os-brick/5f4d907/DellEMC_PowerStore_NVMeTCP_os-brick/66 : FAILURE in 1h 05m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"39f9f60c66036ad3603d1f32feec7fb8ec6dfcab","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-26 16:44:54.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerFlex_os-brick/86e67de/DellEMC_PowerFlex_os-brick/73 : FAILURE in 2h 51m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"eadb0fcae173d52ec1617f8be82b1c39de961c2e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-26 20:55:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/703/","accounts_in_message":[],"_revision_number":12},{"id":"a7d62916c30e0d6e9d154a811cb69ca2a0e646d1","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-26 21:20:51.000000000","message":"Patch Set 12:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/12/check/tyds_test/0e7337e : SUCCESS in 3h 20m 55s","accounts_in_message":[],"_revision_number":12},{"id":"a747d753cda5c4d2e346559d023a21585ada02e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-26 22:23:15.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_Unity_FC_os-brick/9b89d85/DellEMC_Unity_FC_os-brick/112 : FAILURE in 1h 43m 03s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_Unity_iSCSI_os-brick/3ef6211/DellEMC_Unity_iSCSI_os-brick/112 : FAILURE in 2h 55m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"e4a91220ca541d2f34149c3c7ac04b206e9ed856","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-09-27 04:50:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"98a601ce893ddefab500cc58b34c68fb764f3dd6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-27 05:55:06.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/12/check/DellEMC_PowerFlex_os-brick/bba0da2/DellEMC_PowerFlex_os-brick/74 : ABORTED in 1h 04m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"e79fa009fc00097503d4f81d454f5e25fad10814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-27 07:23:07.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"4c2d7f3f1aeffa353c56882f29431869946012a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-27 07:24:28.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1f6988b03a27467b8b36d42645526342 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"fd921bcf86a4f7695051887d3db30691c6322c10","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-27 08:51:02.000000000","message":"Patch Set 13:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1924/","accounts_in_message":[],"_revision_number":13},{"id":"9f37ba1178a168e5432036b3fb7bf6ae15c00924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 09:01:03.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f6f726e096746f59607335e496d3706\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b005978d4a014682827b4130fc3e2118 : SUCCESS in 1h 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32f27c99ac94321b813aa8d7aea002b : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdca77c2110b41d39b35d07add3a63cc : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe7cb5b66bc4480aad82d316aebde49 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/541a1cdd49ee4643b8e840596228ec35 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/269e3bcfbe144213a96297e24d77c5c0 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebf8f738ccd4cf8a97cd09ea5b699a6 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53728d545bd74ca1b4d0cb54e333bf65 : SUCCESS in 4m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4df5fbc4c926457699ce57d7bdf8bdd3 : SUCCESS in 3m 39s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a49ee229bfae406587fd9fd2402f2ea0 : FAILURE in 1h 32m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a8e73d6dc50f4fdd901d18c59aef9a2d : FAILURE in 1h 36m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a6466ee22da74065b05c65dbc7803325 : SUCCESS in 59m 40s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/388e1c0928eb437e9610abbbfc0db717 : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"dffd4456994c619413bcf73b1af9200f29790f7e","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-27 09:04:41.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5b6c5b323c869ecda7a5d67f79874af728a91a2f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-27 09:06:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a473f3e9cd0a43e1ac3d735674dbb18c : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"da7bd56272b7d3cf39ad98c7e6935abd2411b07c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-27 09:12:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/94d0a20/ : SUCCESS in 1h 42m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/1f9a32d/ : SUCCESS in 1h 47m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e76446e/ : SUCCESS in 1h 40m 08s","accounts_in_message":[],"_revision_number":13},{"id":"8426bbac12af9aa85b5d6facd91c72fafc05e548","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-27 09:28:07.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-13/FC/ : FAILURE in 2h 4m 40s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-13/iSCSI/ : FAILURE in 1h 42m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"90188cc98e187ccaef6dc4ec339e4f32ab267545","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-27 10:00:06.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/13/fc : SUCCESS in 0h 29m 35s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/13/iscsi : SUCCESS in 0h 43m 53s","accounts_in_message":[],"_revision_number":13},{"id":"a989abee185030c13ad9c8fbddc582c99231f7f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 11:11:47.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0672a3115804889ba58bd34b2986ea5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14151953693d4913a5ac826fa7820a3d : SUCCESS in 1h 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080f707575944f888f0d8ae28f90dce3 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/181978faad1341e09f0dfa7ff6e187c3 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01269854ad6e47f7a4def7a1b488c2ae : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299b394aa16a47c28f5e97ba9f43a994 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b6c33f021c849d4bacdb2d3d46ba937 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2721218c92c441a491d845cd65a055f9 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1ec29c34da14510a6ed722d48b1b1f4 : SUCCESS in 6m 23s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c6ff1ddb6df0409f927d0da5e8bc2ce5 : SUCCESS in 4m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/37692680ea70435fa775beaa4f1b8c98 : FAILURE in 2h 01m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a7fb8dd0aff4d70b419b26835d38056 : FAILURE in 1h 48m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/247644dad2504ab69de0fe0896b4ca94 : SUCCESS in 1h 12m 45s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/3ac708ece2b949c3840991927d13cfd7 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"e9d4537e237dfefb9e49f40aa7e6ee1a1f08023b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-27 11:13:34.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"03b682dd965adc74c80f4d3b257bfe70ac22a289","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-27 11:14:52.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b60797d7257a44b4af93678c55a982c2 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"243e73091eb7ca09175cc09703f650d4030302b6","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-27 12:01:43.000000000","message":"Patch Set 13:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/13/check/tyds_test/ae783eb : SUCCESS in 3h 39m 03s","accounts_in_message":[],"_revision_number":13},{"id":"ae7cbfcad41fb19d6c2cabfc98eb6ddcd3b731e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 13:16:45.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0478d0c03814dcebcb2a43d3dfa25b3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/434130599da949cea5ec67a0b58445b6 : SUCCESS in 1h 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a38f4f5ab0e45b5b061c2f331fd9ad5 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22ff3c65db264acaa51400ed7309e385 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aa56fe729634f8c93b7c9e7f205ac3b : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddb007b91342441cb322f554a1265e9a : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23532a871d014a5c8271578a2200e403 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1998a36ef6094cbdbcad044ee08a734e : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01f3f2dbfdb14a988f42297afff405c5 : SUCCESS in 4m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/691a2ab4637d4dce80c2b51adabae84f : SUCCESS in 4m 46s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/d1d7c633a3474a89ac3fbd18c6a40182 : FAILURE in 1h 57m 47s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5d3ba9100c7b4978b61ffc5915253e90 : FAILURE in 1h 45m 00s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d6561087c2294099882a85268cd953c9 : SUCCESS in 1h 41m 36s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/eacdfca46dca4aa7bcdbd837479c7936 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b8539ce845588e9b8c8226f97916df15e604b283","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-27 13:28:14.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/b95a0ffd871d4218a95d6b28c6404965 : RETRY_LIMIT in 43s","accounts_in_message":[],"_revision_number":13},{"id":"fee3c046f36169b9653b14b051f00726b7338c89","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-27 14:42:00.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"aa25d30130478c0998b77f170161dedf79bf2053","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-27 21:13:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e2177d849ef4845505ca5b6a78acba02e4556bae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-27 21:14:31.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/69f427691c254314a1311d0abb5149bc : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"1158c5f91f7355e43e02b1afc4329051163e64eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 22:35:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1325abce35904c92bdfd3cf1020548e9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce91f65db08f45b3ba1f8a0a1242a7c8 : SUCCESS in 1h 05m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212482a0409e48ce92bf3a050a2b7880 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c0ac0f74214d9492047066bd8af47f : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02adf13295ee48e88cf9cc486adcf0da : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d4f61488b744f0ab6a0767b0aa0692 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a385143c7b848f6895f8e8bce2e6c20 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07c697cc6aa4f0fa96bfb445ebad881 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38e6fab39d249dcbd640736a0ce40f2 : SUCCESS in 4m 18s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/047ad85a201e4fa49c49dc09402ee131 : SUCCESS in 3m 49s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7282d2b0450e4e2fb7ca86aa122cc37b : SUCCESS in 1h 20m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e7245796815b4214b1c1c2f4bc07653c : SUCCESS in 1h 12m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5e26da0a45184413aea9bdcdb60de6c3 : SUCCESS in 1h 03m 35s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/ed3822d221344a2eb7403bd6d645fa03 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a2eb5fae4fb823f18bc13ab311a7aef38c9bb051","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-27 23:51:33.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_FC_os-brick/0221782/DellEMC_PowerStore_FC_os-brick/63 : FAILURE in 1h 16m 02s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_iSCSI_os-brick/0bed098/DellEMC_PowerStore_iSCSI_os-brick/56 : FAILURE in 1h 06m 42s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_NFS_os-brick/53ebc8a/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_NVMeTCP_os-brick/619cbc5/DellEMC_PowerStore_NVMeTCP_os-brick/67 : FAILURE in 1h 12m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"ad49269daea209dfea9500a1345a27a835939c14","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-27 23:57:43.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-fcp/14d7b71/ : RETRY_LIMIT in 8m 17s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-iscsi/b45180e/ : SUCCESS in 1h 21m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-nfs/ea8fe70/ : SUCCESS in 45m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/fe4b583/ : SUCCESS in 1h 20m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"e58115d03ff22b9632217c907b9a84dc64974150","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 01:26:55.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerFlex_os-brick/26086d7/DellEMC_PowerFlex_os-brick/75 : FAILURE in 2h 51m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"86b41c2af6cdcc0a07dffb62829bc173f1533e51","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-28 03:39:07.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"03a8d90ce700881178785dac9e7ee6c826d59770","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-28 04:54:34.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"9faa7178f7c1c67510e5ec5d988b51c50cdb9c52","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-09-28 05:08:48.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"99fae8a1bf68504c051e4b544de27373dc40e6a6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-28 05:59:27.000000000","message":"Patch Set 13: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"167218c616c9bf5c9a416474fa25f9b694f38423","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 06:30:51.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerFlex_os-brick/a3a65aa/DellEMC_PowerFlex_os-brick/76 : FAILURE in 2h 51m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"5bb63fd09e456c43129f82e1f9acf6a41695c325","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-28 10:14:55.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"40256d025df1c5ab34de0f041ce645d571d98e66","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-28 17:28:20.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c5ffd85fbceb7a03ef9ae928fbda8297a3ddcb00","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-09-29 00:03:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/13/20230929055603/NEC-V-ISCSI : SUCCESS in 2h 49m 50s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/13/20230929055603/NEC-V-FC : SUCCESS in 3h 07m 10s","accounts_in_message":[],"_revision_number":13},{"id":"f80d073e6d66c74ada724482af896c2227d9f9d1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-09-29 11:15:53.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/13/20230929093319/NEC-FC : SUCCESS in 3h 27m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/13/20230929093319/NEC-ISCSI : SUCCESS in 10h 42m 30s","accounts_in_message":[],"_revision_number":13},{"id":"0931148462996272e4574b6b6bc3b07b50fe9b5d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-10-02 15:14:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0868320d5b6cc116d71bffd28a8c45f561a2ca64","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-02 16:44:18.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerFlex_os-brick/c9a6f9d/DellEMC_PowerFlex_os-brick/80 : SUCCESS in 1h 03m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"661a6efa56557b7948df72afc6e5de3b7336c27e","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-10-03 11:52:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"faae28ad6dd4de3d52574fe4e6f2dddfb4b31220","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-10-03 11:53:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"96a99f9fefdbf448f1d3da62cb0675efef3565a2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-03 11:53:23.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b813a0e53cea44b2adc36fdc474c2803 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"cef7237c4ed9a6555f91d9afe1532ba30ee064cf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-03 13:45:47.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/5ee205f/ : SUCCESS in 1h 50m 09s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/defc8ca/ : SUCCESS in 1h 47m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/13/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/a26b63d/ : SUCCESS in 1h 51m 48s","accounts_in_message":[],"_revision_number":13},{"id":"6b192e90be89bb211fc7f782afc955e9c2439fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 13:48:15.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4a7cf7b36e46a88823a177bbc2eb48\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a245dd26a034fcf848e47becc137ea7 : SUCCESS in 1h 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a48e823ec04c49b3f0f086e84643b0 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55b3788ab408428aa3a5097a29f3ac26 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b6707928c324706afb4cb8a96b94ba6 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/640f35ac772447488dbf26f7ae2bd638 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36a46e4476754bf9b40ee6ac886dac5e : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e71b98dca94428fa5adaa3cde228291 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/236bbfe3dbdb427bb17413c4832d339d : SUCCESS in 4m 12s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/1a2eec50d0114227ae26c9f673f4a266 : SUCCESS in 5m 04s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/d09d0461340f414883edf757af6c3505 : SUCCESS in 1h 50m 29s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d28c1bff91b84224993c7731079440b0 : SUCCESS in 1h 26m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/33914e8ca71f4604b25343c36d7b9985 : SUCCESS in 1h 19m 57s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/77907b1696b2448d986bb412ca8a287d : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"89e5629b963ef273c546c2b04e85c542d7206e52","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-03 15:04:33.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-fcp/66612d7/ : SUCCESS in 1h 03m 18s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-iscsi/105aadc/ : SUCCESS in 41m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-nfs/0eea4cc/ : SUCCESS in 1h 15m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/13/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/21e8e0a/ : SUCCESS in 41m 42s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"2455b9ea8f3a4f88106b74c7dacdd927d05f7b51","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-03 15:24:42.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_FC_os-brick/011ebc9/DellEMC_PowerStore_FC_os-brick/65 : FAILURE in 1h 11m 33s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_iSCSI_os-brick/4db1d9b/DellEMC_PowerStore_iSCSI_os-brick/58 : FAILURE in 1h 36m 13s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_NFS_os-brick/94d25b9/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerStore_NVMeTCP_os-brick/5a1f684/DellEMC_PowerStore_NVMeTCP_os-brick/69 : FAILURE in 1h 21m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"e8eca61a878ae43ac2786d86fd7522936a3f155f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-03 20:20:08.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/13/check/DellEMC_PowerFlex_os-brick/18b64db/DellEMC_PowerFlex_os-brick/81 : FAILURE in 1h 13m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"a16631682bdf5ee4c7084e662393f4497bcfe81d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-10-07 02:16:26.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"4347c418a203b2011232c98dd9e2223cec55102d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-07 02:17:37.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/19d7f3b104654e29bb75bfb3edc39221 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"2ee661d3c669b2ee5d104e0a279a8339d91a666d","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-10-07 02:20:54.000000000","message":"Patch Set 14:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/84/876284/14/check/tyds_test/0afc110 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":14},{"id":"d98cfa112bf511651d77fb74289641b0031ba3f3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-07 03:19:03.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerFlex_os-brick/6ea2d30/DellEMC_PowerFlex_os-brick/82 : SUCCESS in 1h 02m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"fabdd48be33d137a5a4050f5fe372d86bb0ef522","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-10-07 03:30:58.000000000","message":"Patch Set 14:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1962/","accounts_in_message":[],"_revision_number":14},{"id":"d5b0cb2d666b9f7b78b8a24ccb0a4c683390dffe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-07 04:00:25.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/84/876284/14/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/ee9efe5/ : FAILURE in 1h 42m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/84/876284/14/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/dc66c6b/ : SUCCESS in 1h 40m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/84/876284/14/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/38ae237/ : SUCCESS in 1h 39m 49s","accounts_in_message":[],"_revision_number":14},{"id":"daeb4f3a3a8d9508500d876df2761dbc94bc1212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 04:24:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726ea51001cf465dac443c3aea368b08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7db9c70ce793442f817946b27dab6380 : SUCCESS in 1h 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad2b9f1e3ed40c782810043d8b1d091 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da2bd60429f6449cbbc326687d0c20e7 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f50730d0436400082f1e2ef4b029afc : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b152ec89ef84c669ac5ae364274e00e : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5160a928f5f14ceb8561ecf5b8ebc973 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26bb00b757c42e280a57b9eb11042d1 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dccd080373b4b5bbeb969caa6cfbe8d : SUCCESS in 5m 25s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/95bf8274cde849609d7b87471e75cf12 : SUCCESS in 4m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/8ddca0f898484515b3711fa06c85ec1c : SUCCESS in 2h 00m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d82e0d6871014ba0a57e6c49ddaba3c6 : SUCCESS in 1h 46m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ecb2922aab7649e7a6bd69c5662b150c : SUCCESS in 1h 32m 42s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b87ef4f57b6242c9830f0b9b526a3eac : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dde4507bb9c3263cca7c3889bba3e8ae7ab3bcfb","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-10-07 05:00:06.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/14/fc : SUCCESS in 0h 31m 05s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/84/876284/14/iscsi : SUCCESS in 0h 42m 46s","accounts_in_message":[],"_revision_number":14},{"id":"3f1518219571f33e6672dbaf4c0a17b2157a0f2b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-10-07 05:28:01.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/14/20231007111627/NEC-V-ISCSI : SUCCESS in 2h 54m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/14/20231007111627/NEC-V-FC : SUCCESS in 3h 11m 30s","accounts_in_message":[],"_revision_number":14},{"id":"3c3dcf5be4a447f24050edfc01ab6c11e78ce3f3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-07 05:31:20.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerFlex_os-brick/5aa7691/DellEMC_PowerFlex_os-brick/83 : SUCCESS in 1h 06m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"0d166235e4f9c7190bdc5b70add784f8bd9cb486","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-07 05:39:58.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerStore_FC_os-brick/3607241/DellEMC_PowerStore_FC_os-brick/66 : FAILURE in 1h 14m 49s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerStore_iSCSI_os-brick/8e76e83/DellEMC_PowerStore_iSCSI_os-brick/59 : FAILURE in 1h 03m 19s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerStore_NFS_os-brick/be1170d/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/84/876284/14/check/DellEMC_PowerStore_NVMeTCP_os-brick/51c406c/DellEMC_PowerStore_NVMeTCP_os-brick/70 : FAILURE in 1h 05m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"07950eca631d86b5eade8ce43d435ff9cf19de65","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-07 05:49:04.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-14/FC/ : SUCCESS in 3h 32m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-84-876284-14/iSCSI/ : SUCCESS in 1h 33m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"45aad611720a2265089242a1bf99639a5562190c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-07 05:58:16.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/14/upstream-check/cinder-tempest-plugin-ontap-fcp/a1019b6/ : SUCCESS in 1h 01m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/14/upstream-check/cinder-tempest-plugin-ontap-iscsi/49293cd/ : SUCCESS in 1h 25m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/14/upstream-check/cinder-tempest-plugin-ontap-nfs/67ae057/ : SUCCESS in 46m 33s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/84/876284/14/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/53b566a/ : SUCCESS in 1h 26m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"1611af9480de26e26347ab9418a29f0ee58741c8","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-10-08 03:48:10.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"3791383f5eff603ae5e4ac16c2684c6f102d0b4f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-10-08 20:06:20.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/14/20231008181734/NEC-ISCSI : SUCCESS in 8h 16m 20s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/84/876284/14/20231008181734/NEC-FC : SUCCESS in 10h 48m 43s","accounts_in_message":[],"_revision_number":14},{"id":"62e413c8b7cb3d37b1437f141f85b9ed2d66172b","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-10-09 01:39:51.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"f3d6694f1a379b50bfeb084e135cb7ac8f693ea5","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2023-10-17 06:59:06.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5b5ae8aeddae3194f4d3f561a744355e43a1354b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-10-29 14:15:45.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_10_2023_14_09_10\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"16df5554c6f46f9fa22a204d4262fe82d7020dcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-10-30 16:10:45.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_10_2023_16_03_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"5ec822df4388dbff87ce96b8f0a73ad915688582","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-10-30 17:30:47.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9b101bd5e9849e24c11ea4388e5cde7f55f4aa86","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-19 14:19:36.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"eff5e0f8b0a39f0f7cd080f5388850aaf931fc09","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-21 14:38:44.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7f7a45a43c317c547d0abdc6312f87de36fcb94c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-21 14:44:01.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"64ffab7c4d81ae9ee10f8683ae137f69753772d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 14:44:34.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"ebde558b4e3b728fb9342672bdfa8bad8370beb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:47:55.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edacb8e66bce493a8348d33e7fcce036\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e03b658f831247d193b2cd65bbef9082 : SUCCESS in 1h 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bc391c6ed642d4813d5c6d5bfe5d72 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11aa0bd3bd64f5f86f3c9dadb2d23f3 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18b37ac4df344466bafffd59b3bfc7f2 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf6adf3bea648b98214c92e3138be2a : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42b5c7a80cdb405081fde81daec15817 : SUCCESS in 3m 53s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2889fa45bb5244d0bfbd92a74efda251 : SUCCESS in 1h 57m 16s","accounts_in_message":[],"_revision_number":14},{"id":"894a458aac70e20e30dacfc1078634782fe8026b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:47:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"bf9dcbe0ec9f9f2c6e36233bb425184133d13ba5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 16:54:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc89024123944d6492099a34a6163dfd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae7c6cedeea4f58bc7603ebed0964ab : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea616e377fb479dbcbe88cf4df10878 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"14b024c505e68617a50af655d04aaa86b34d15c0","revisions":{"402e64bb6a6a322ceb4fcdf42cb6d3417f69574b":{"kind":"REWORK","_number":1,"created":"2023-03-09 18:39:48.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/1"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:37:11.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nFixes issue in os-brick that causes token to never be cached and\ncauseing unnecessary logins. Now it tests to see if token exists\nand caches it for future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/402e64bb6a6a322ceb4fcdf42cb6d3417f69574b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/402e64bb6a6a322ceb4fcdf42cb6d3417f69574b"}]},"branch":"refs/heads/master"},"7bc6b5fb85d3b75883bf67e8d9210a3596d80dde":{"kind":"REWORK","_number":2,"created":"2023-03-13 17:21:54.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/2"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-13 17:19:04.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7bc6b5fb85d3b75883bf67e8d9210a3596d80dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7bc6b5fb85d3b75883bf67e8d9210a3596d80dde"}]},"branch":"refs/heads/master"},"3cd2f8c107a33d2097dc4762ff5dd38511638c52":{"kind":"REWORK","_number":3,"created":"2023-03-14 13:53:48.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/3"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-14 13:53:18.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3cd2f8c107a33d2097dc4762ff5dd38511638c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3cd2f8c107a33d2097dc4762ff5dd38511638c52"}]},"branch":"refs/heads/master"},"4368f1e813de8b4542eba4cfb488bda13f1fbdb1":{"kind":"REWORK","_number":4,"created":"2023-05-19 16:45:23.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/4"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-05-19 16:32:28.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4368f1e813de8b4542eba4cfb488bda13f1fbdb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4368f1e813de8b4542eba4cfb488bda13f1fbdb1"}]},"branch":"refs/heads/master"},"4f767da9c5f7773128abf0c05de1787070cdb154":{"kind":"REWORK","_number":5,"created":"2023-05-19 16:51:10.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/5"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-05-19 16:50:34.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f767da9c5f7773128abf0c05de1787070cdb154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4f767da9c5f7773128abf0c05de1787070cdb154"}]},"branch":"refs/heads/master"},"bb2ee29a5a971fe7c6e3fa05743dc69b0db7b5d4":{"kind":"REWORK","_number":6,"created":"2023-05-22 16:15:11.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/6"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-05-22 16:14:41.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bb2ee29a5a971fe7c6e3fa05743dc69b0db7b5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bb2ee29a5a971fe7c6e3fa05743dc69b0db7b5d4"}]},"branch":"refs/heads/master"},"64b1262a40649ab6b90a59e5372350aee5d3759e":{"kind":"REWORK","_number":7,"created":"2023-07-26 17:16:53.000000000","uploader":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"ref":"refs/changes/84/876284/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/7"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"tony-saad","email":"tony.saad@dell.com","date":"2023-07-26 17:16:23.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the Dell PowerFlex (scaleio) connector for os-brick\nand enables the token to get cached and reusable. It avoids\nunnecessary logins. If the token already exists, then it get cached\nfor future use.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/64b1262a40649ab6b90a59e5372350aee5d3759e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/64b1262a40649ab6b90a59e5372350aee5d3759e"}]},"branch":"refs/heads/master"},"1d0fc5c5e1b651bb0a24d30638d83619c34e6244":{"kind":"REWORK","_number":8,"created":"2023-09-20 11:44:02.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/8"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-20 11:43:59.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the unnecessary login issue introduced\nin Change-ID Iab54c515fe7be252df52b1a0503a251779805759\nin the Dell PowerFlex (scaleio) connector for os-brick.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d0fc5c5e1b651bb0a24d30638d83619c34e6244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1d0fc5c5e1b651bb0a24d30638d83619c34e6244"}]},"branch":"refs/heads/master"},"0ad438e62025534b40c036770e07811ebc2ed2a2":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-09-20 11:54:29.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/9"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-20 11:54:29.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the unnecessary login issue introduced\nin Change-ID II6f01a178616b74ed9a86876ca46e7e46eb360518\nin the Dell PowerFlex (scaleio) connector for os-brick.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ad438e62025534b40c036770e07811ebc2ed2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ad438e62025534b40c036770e07811ebc2ed2a2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3225edc574fb44c4ef57af5a9fdcc9102ab67bb7":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-09-20 11:55:00.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/10"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-20 11:55:00.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the unnecessary login issue introduced\nin Change-ID I6f01a178616b74ed9a86876ca46e7e46eb360518\nin the Dell PowerFlex (scaleio) connector for os-brick.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3225edc574fb44c4ef57af5a9fdcc9102ab67bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3225edc574fb44c4ef57af5a9fdcc9102ab67bb7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"96ee4d1e425ad55f11b79a19bed7ab70b1cc3763":{"kind":"REWORK","_number":11,"created":"2023-09-20 16:11:11.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/11"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-20 16:10:35.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThis patch fixes the unnecessary login issue introduced\nin Change-ID Iab54c515fe7be252df52b1a0503a251779805759\nin the Dell PowerFlex (scaleio) connector for os-brick.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/96ee4d1e425ad55f11b79a19bed7ab70b1cc3763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/96ee4d1e425ad55f11b79a19bed7ab70b1cc3763"}]},"branch":"refs/heads/master"},"9efa035366f61ae6286f30530f5d114f6be86d05":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-09-26 09:33:47.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/12"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-26 09:33:47.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThe Dell PowerFlex (scaleio) connector maintains a token cache\nfor PowerFlex OS.\nThe cache was overwritten with None by misktake\nin Change-ID I6f01a178616b74ed9a86876ca46e7e46eb360518.\nThis patch fixes the broken cache to avoid unnecessary login.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9efa035366f61ae6286f30530f5d114f6be86d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9efa035366f61ae6286f30530f5d114f6be86d05"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3c2cdc12b417fcc81e72d10b056241960f1ba61e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-09-27 07:23:07.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/13"}}},"commit":{"parents":[{"commit":"07b0e4a94be8bef55377af17b5c708e92af73557","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/07b0e4a94be8bef55377af17b5c708e92af73557"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-09-27 07:23:07.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThe Dell PowerFlex (scaleio) connector maintains a token cache\nfor PowerFlex OS.\nThe cache was overwritten with None by misktake\nin Change-ID I6f01a178616b74ed9a86876ca46e7e46eb360518.\nThis patch fixes the broken cache to avoid unnecessary login.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3c2cdc12b417fcc81e72d10b056241960f1ba61e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3c2cdc12b417fcc81e72d10b056241960f1ba61e"}]},"branch":"refs/heads/master","description":"Rebase"},"14b024c505e68617a50af655d04aaa86b34d15c0":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-10-07 02:16:26.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/84/876284/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/876284/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/876284/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/876284/14"}}},"commit":{"parents":[{"commit":"fb23a40a65f1a6e06b4580ca309469f9ae2cced6","subject":"Merge \"Revert \"Dell PowerFlex password appears in plain text when creating a volume\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fb23a40a65f1a6e06b4580ca309469f9ae2cced6"}]}],"author":{"name":"Tony Saad","email":"tony.saad@dell.com","date":"2023-03-09 18:32:28.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2023-10-07 02:16:26.000000000","tz":0},"subject":"Dell PowerFlex: Unnecessary login happen","message":"Dell PowerFlex: Unnecessary login happen\n\nThe Dell PowerFlex (scaleio) connector maintains a token cache\nfor PowerFlex OS.\nThe cache was overwritten with None by misktake\nin Change-ID I6f01a178616b74ed9a86876ca46e7e46eb360518.\nThis patch fixes the broken cache to avoid unnecessary login.\n\nCloses-Bug: #2004630\nChange-Id: I2399b0b2af8254cd5697b44dcfcec553c2845bec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/14b024c505e68617a50af655d04aaa86b34d15c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/14b024c505e68617a50af655d04aaa86b34d15c0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
