)]}'
{"id":"openstack%2Fos-brick~876992","triplet_id":"openstack%2Fos-brick~master~Icd68f845d83df2d4875e61a7e92630e73e4564c4","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-03-17 16:12:56.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-04-21 16:39:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd68f845d83df2d4875e61a7e92630e73e4564c4","subject":"Set packages in setup.cfg","status":"MERGED","created":"2023-03-09 16:16:29.000000000","updated":"2023-04-21 16:40:11.000000000","submitted":"2023-04-21 16:39:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876992","meta_rev_id":"856c8a15d7e36ea858ca7ba3e955c072a858f298","_number":876992,"virtual_id_number":876992,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 16:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-03-17 16:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-03-16 10:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-21 14:42:51.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-21 14:42:51.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 16:20:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 16:32:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-09 18:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 19:34:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-09 21:35:48.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-09 21:55:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-10 00:44:09.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-10 00:51:42.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-10 22:25:33.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-03-14 09:32:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 10:18:51.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-03-15 22:00:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2023-03-16 10:19:45.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-03-17 16:12:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-04-21 14:42:51.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"}],"messages":[{"id":"aa4434bcc8ca4ba15abda6805450810c342a8ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-03-09 16:16:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1389a2199cddf23ca9b456dcb4afc699322c2841","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-09 16:20:10.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"8e1d41f7ae4232212cc29c025306306c05445604","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 16:32:06.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/54/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"5de9fd3fc4b773dd309183bcc1c2b947dfbf5966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:07:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffaeec74db2d45cc8d0cf6f92b3a12ef\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e06c033555a247a69be94ca32f8321e4 : SUCCESS in 1m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92cba1f4ac4f4537a40e4a351517389b : SUCCESS in 1h 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d853581d8803410094d0e6dde1515725 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/510679c7600045cf883c2d10f3f35f4d : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4f3267ca0a346a3beb167af36451b48 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce3bca9d5624a5a84339fcfe47d158f : SUCCESS in 4m 57s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3e32ea701feb443f91830fe5b345f34c : SUCCESS in 4m 22s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9f48d83668b744f0ba4cac7ef2383f2c : SUCCESS in 1h 24m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7a91218d891a4c2a8a8e6fa45b6feffb : SUCCESS in 1h 48m 45s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/06863799f74049e99208d7827a33ee32 : SUCCESS in 1h 06m 28s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/507288f1ce7e40f0a01492301adcf1e4 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1be1a70133747fbf8ca0b8243b73e2d8a3b0c2eb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-09 19:34:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp cinder-tempest-plugin-ontap-fcp : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/876992/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/ec6dc80/ : SUCCESS in 1h 25m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/876992/1/upstream-check/cinder-tempest-plugin-ontap-nfs/675aeb2/ : SUCCESS in 1h 22m 14s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/876992/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/242bdcf/ : FAILURE in 42m 42s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e3ee7aaac11ad9e10d35706db9755cf0d5579c43","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 21:33:35.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/54","accounts_in_message":[],"_revision_number":1},{"id":"2d7067e7314550dd9629045569def18435359f19","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-09 21:35:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/5aa7be5a107f485a9cfca93a214955ce : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"53445e8195b35704ab842abd58ef500390ad5bf4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-09 21:55:02.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_Unity_FC_os-brick/83898cb/DellEMC_Unity_FC_os-brick/277 : SUCCESS in 1h 53m 35s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_Unity_iSCSI_os-brick/5bec9a1/DellEMC_Unity_iSCSI_os-brick/291 : SUCCESS in 1h 22m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6261f95489e50350fd3063b50c78be3b3b808282","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-10 00:44:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_PowerStore_FC-ob/a9f9142/DellEMC_PowerStore_FC-ob/212 : FAILURE in 1h 33m 37s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_PowerStore_iSCSI-ob/3f26dca/DellEMC_PowerStore_iSCSI-ob/134 : SUCCESS in 1h 28m 53s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_PowerStore_NFS-ob/dde2468/DellEMC_PowerStore_NFS-ob/106 : SUCCESS in 1h 13m 25s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_PowerStore_NVMeTCP-ob/0c081d2/DellEMC_PowerStore_NVMeTCP-ob/86 : SUCCESS in 1h 31m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"e9ad28b0b9658cd97fb504a650adb8dd8ae09257","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-10 00:51:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC-os-brick https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_SC_FC-os-brick/522b2ad/DellEMC_SC_FC-os-brick/174 : SUCCESS in 1h 04m 17s\n- DellEMC_SC_iSCSI-os-brick https://elab-os-logsrv.delllabs.net/92/876992/1/check/DellEMC_SC_iSCSI-os-brick/5ab4c20/DellEMC_SC_iSCSI-os-brick/169 : SUCCESS in 1h 35m 10s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"64baa0c898ad96ffc3e855e620befaa2cf80d758","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-10 22:25:33.000000000","message":"Patch Set 1:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/92/876992/1/fc : FAILURE in 0h 52m 14s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/92/876992/1/iscsi : FAILURE in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2030c92bf414b1d1dbe7f5dd04eab6d71435c5f7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-14 09:32:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/876992/1/20230314150843/NEC-FC : SUCCESS in 3h 05m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/876992/1/20230314150843/NEC-ISCSI : SUCCESS in 3h 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"088d104502ca8e2a9913849a194744c852c9cd52","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-15 10:18:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/876992/1/20230315131217/NEC-V-ISCSI : SUCCESS in 2h 42m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/876992/1/20230315131217/NEC-V-FC : SUCCESS in 6h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"803bbb85a22f163c3ee6e028a2ea97a9a35f66dd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-15 22:00:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4089d189582d1d32b52b03541713bc66ef191d4f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-16 10:19:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72584d5f6b9681388a3e3effacd6fe5e699269ec","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-03-16 12:58:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"931db9929c83e32d72d501afe897a9e6b1d91b0d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-17 16:12:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b410cbcdd5f64dc12f6cf87c31ce9611d0b21f0","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-21 14:42:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09f100e8379361fcab95aa8c03872c99c16ddc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 14:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"227114994f9501a290ebdc4e4ea048f9b6c66525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:39:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1925ccf92edc4632a2fc0dccc28ce1b5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e79a5bd831f24318b98fa9e45116670b : SUCCESS in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3996ac7869a40aea51c230395df7d13 : SUCCESS in 1h 08m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d315cf84148943beb9b0870bd301e9b9 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d74283cda534da3b3ad2b45d69d0b8b : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c1447e65e19444b801ab112cda7da0b : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13951a3f4ad440299c050de04cc1f87 : SUCCESS in 5m 41s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb77025840104291a80eecd0daecec82 : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c725bbcdbc55b775955e072ee9c35f8cf9461fff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:39:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"856c8a15d7e36ea858ca7ba3e955c072a858f298","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 16:40:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0e3e75915049049d22c5507b2a8d89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d3e0207d3644719b02c2b71d8c0664 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f184c8a8091201fb4e2a6024391f1e4594711b98","revisions":{"f184c8a8091201fb4e2a6024391f1e4594711b98":{"kind":"REWORK","_number":1,"created":"2023-03-09 16:16:29.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/92/876992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/92/876992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/876992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/876992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/876992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/92/876992/1"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-03-09 16:02:13.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-03-09 16:02:13.000000000","tz":-300},"subject":"Set packages in setup.cfg","message":"Set packages in setup.cfg\n\nWithout this, \"tox -e py38\" fails with an error from\nsetuptools:\n    error: Multiple top-level packages discovered in\n    a flat-layout: [\u0027etc\u0027, \u0027os_brick\u0027, \u0027releasenotes\u0027].\n\nChange-Id: Icd68f845d83df2d4875e61a7e92630e73e4564c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f184c8a8091201fb4e2a6024391f1e4594711b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f184c8a8091201fb4e2a6024391f1e4594711b98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
