)]}'
{"id":"openstack%2Fos-brick~882840","triplet_id":"openstack%2Fos-brick~master~Ia74ecfba03ba23de9d30eb33706245a7f85e1d66","project":"openstack/os-brick","branch":"master","topic":"bug/2004555","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-05-10 20:03:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia74ecfba03ba23de9d30eb33706245a7f85e1d66","subject":"Support force disconnect for FC","status":"MERGED","created":"2023-05-10 14:39:38.000000000","updated":"2023-05-11 09:44:08.000000000","submitted":"2023-05-10 20:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"882840-bug/2004555","meta_rev_id":"c6ea3f76b8370951c8d231dfa38c571dbf58a5d3","_number":882840,"virtual_id_number":882840,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 20:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-05-10 15:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-10 18:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-10 18:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-05-10 15:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 15:24:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-05-10 16:18:02.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-05-10 16:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 17:50:06.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-10 17:54:58.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-05-10 18:22:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-10 18:26:23.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-05-10 19:59:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-11 09:44:08.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e95e515330ec420e08f84cf5288beb7a548c710c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-05-10 14:39:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8438eaa5c7035218e890f71377af36bb85a6561","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-10 15:24:03.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56cd84b25dd22a4d06800a841b47e724c944d7e1","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-10 16:18:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-os-brick https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/43f40b9c2f654b31ad58c22840057db4 : POST_FAILURE in 49m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8776dcfd8baa73df44bf1bef66f988f0f88c42d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 16:50:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c71fb36d96124485a41cba4c5ba5ce00\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bbce6f4a8f14b3aa25e14bbbcc58d26 : SUCCESS in 1h 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232df23f933b45edbe997bed76762079 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77eda42e8208470787d890ac821d1dab : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76e476fb1323418696f4f088876f4310 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a7e65475bd4b26bc174e1d963745d2 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8b192ad491a40b1bb584d4178dec922 : SUCCESS in 4m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0f9e583ed0e4476992a589fb6a4252f0 : SUCCESS in 4m 45s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/4e4361e7483a41b8ad42002be497c185 : TIMED_OUT in 2h 04m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c8c350a23b9f41cb827bde0f255b7fec : SUCCESS in 2h 02m 52s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cddd4207db1f45b5a8bf7b62371b3390 : SUCCESS in 1h 33m 21s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/36a6c891a9f8407daf7ee6c6743323b3 : FAILURE in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb41fefcd4587d4b450d62b65faa1cf75fd5daf1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-10 17:50:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-882840-1/FC/ : SUCCESS in 2h 27m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-882840-1/iSCSI/ : SUCCESS in 2h 57m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"35ac02737f11fc3deb39ab8fe34538861fab0702","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-10 17:54:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_PowerStore_FC_os-brick/09aef38/DellEMC_PowerStore_FC_os-brick/4 : SUCCESS in 1h 04m 08s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_PowerStore_iSCSI_os-brick/09b5b5d/DellEMC_PowerStore_iSCSI_os-brick/4 : SUCCESS in 1h 03m 00s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_PowerStore_NFS_os-brick/bcf4e26/DellEMC_PowerStore_NFS_os-brick/4 : SUCCESS in 48m 47s\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/3cf3b2c/DellEMC_PowerStore_NVMeTCP_os-brick/4 : SUCCESS in 1h 04m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"cf7b103428aa0d1f0d6ba4197aa97a7a1a8d15f2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-10 18:22:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/882840/1/upstream-check/cinder-tempest-plugin-ontap-fcp/de65d88/ : RETRY_LIMIT in 8m 52s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/882840/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f2f1fac/ : SUCCESS in 1h 12m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/882840/1/upstream-check/cinder-tempest-plugin-ontap-nfs/c64ccf5/ : SUCCESS in 1h 19m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/882840/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/19f5807/ : SUCCESS in 1h 11m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c1580556aa155efd70d4bd2b3144a8152637d32e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-05-10 18:26:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"176acf1bf7d96c8fe4520c1776600f594a6d4dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 18:26:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d88b664e9a48639c7dfe016100508a7844fd15d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-10 19:59:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_Unity_FC_os-brick/6dfe1ae/DellEMC_Unity_FC_os-brick/8 : SUCCESS in 1h 37m 36s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/40/882840/1/check/DellEMC_Unity_iSCSI_os-brick/edf8f90/DellEMC_Unity_iSCSI_os-brick/8 : SUCCESS in 2h 07m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec65f2f1a2740dfba60c5000710a84b118bbedca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5cc5940f67d425da387d98b400de9c0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2ffd630f829492fa0509f6739ed98cc : SUCCESS in 1h 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883755b8d449483ab61c261f09589b85 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e24d22b405542cc92caf8ca8df19bea : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1ff5a48f65f47ca95ca13eb5d56da4a : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0050a7ce7a454dbb036194e5ed0ff4 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe3410fa41bd499999eca3730fabcc4b : SUCCESS in 4m 47s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/039a7db58a164dcdbcec12e49f29cc8f : SUCCESS in 1h 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c7a46be960bc32c755012ba66a52210110416eff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:03:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3efb81d4c4a3188aec6e2caf3025cd6e291ca9f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:04:25.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97578ebcdfeb448aa506fb8b2fcd61a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00152eef3fad413c8db7d8bc329871ec : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea5d5bdbe78c4e9c926e26283769ba33 : FAILURE in 39s","accounts_in_message":[],"_revision_number":1},{"id":"c6ea3f76b8370951c8d231dfa38c571dbf58a5d3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-11 09:44:08.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/os-brick/40/882840/1/check/devstack-plugin-ceph-dsvm-focal-py3/deed4ec/ : FAILURE in 50m 40s\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/os-brick/40/882840/1/check/tempest-dsvm-full-focal-py3/fbe7f05/ : FAILURE in 1h 43m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"570df49db9de3030e658619138588b836c007f8c","revisions":{"570df49db9de3030e658619138588b836c007f8c":{"kind":"REWORK","_number":1,"created":"2023-05-10 14:39:38.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/40/882840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/40/882840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/882840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/882840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/40/882840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/40/882840/1"}}},"commit":{"parents":[{"commit":"a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a0bc09ba50aae1a3f6c3601a4b90c5cb1a84b565"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-03-01 12:08:16.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-04-24 09:39:37.000000000","tz":120},"subject":"Support force disconnect for FC","message":"Support force disconnect for FC\n\nThis patch adds support for the force and ignore_errors on the\ndisconnect_volume of the FC connector like we have in the iSCSI\nconnector.\n\nRelated-Bug: #2004555\nChange-Id: Ia74ecfba03ba23de9d30eb33706245a7f85e1d66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/570df49db9de3030e658619138588b836c007f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/570df49db9de3030e658619138588b836c007f8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
