)]}'
{"id":"openstack%2Fos-brick~887576","triplet_id":"openstack%2Fos-brick~master~I6030d433dcf45f496bfee79aa00b2af9b51fd420","project":"openstack/os-brick","branch":"master","topic":"storpool-driver-fixes","attention_set":{},"removed_from_attention_set":{"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-01-22 13:10:12.000000000","reason":"\u003cGERRIT_ACCOUNT_31779\u003e replied on the change","reason_account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}},"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2024-07-05 16:55:39.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-07-05 14:50:32.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-07-05 16:55:39.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-05 16:55:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6030d433dcf45f496bfee79aa00b2af9b51fd420","subject":"StorPool: Raise on spopenstack, SP_OURID issues","status":"MERGED","created":"2023-07-04 08:31:25.000000000","updated":"2024-07-05 16:56:40.000000000","submitted":"2024-07-05 16:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887576-storpool-driver-fixes","meta_rev_id":"7cccdf7b268bdadd476005696d44d21a9f7151b5","_number":887576,"virtual_id_number":887576,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 16:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"recommended":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"all":[{"value":1,"date":"2024-07-05 05:52:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2024-07-04 03:29:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":2,"date":"2024-07-05 14:50:32.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-07-04 04:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-04 03:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":1,"date":"2024-07-05 14:50:32.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"}],"CC":[{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-04 11:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-04 12:26:38.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-04 12:32:50.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-04 12:36:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-04 14:59:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-04 17:48:55.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-06 06:16:41.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-06 21:47:16.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-10 10:22:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-10 11:00:12.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-10 11:00:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-13 18:58:32.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-07-13 19:03:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-29 07:33:43.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2023-09-07 16:47:47.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-09-15 07:56:20.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2023-09-20 15:10:09.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-11-10 04:17:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-01-16 14:31:47.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-01-22 13:10:12.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2024-07-04 03:30:36.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2024-07-04 04:00:03.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"REVIEWER"},{"updated":"2024-07-05 05:52:11.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"}],"messages":[{"id":"950825c70736e2cd43c9aabdf7322bd35be0bef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-07-04 08:31:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea2400907dff7c195c29ed03b29744ce6fb01f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 11:01:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d999eec6b434f69b96de37a0021814a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50d854c8fd7c4d49bf1d05f77040cbe1 : SUCCESS in 1h 31m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2024eee8438d4ecda3f9472479a287fa : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03509d85e0db4a788d5e02d562b381e0 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3bca89bb2cf465091db8c9fa3999d9d : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8390a0850344e1b9b6041b68408c54 : SUCCESS in 6m 58s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9caaa5702d93403e9d4be15667b354f0 : SUCCESS in 4m 12s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/adf90bcc6e6b4c749693363b57ad6876 : SUCCESS in 2h 22m 20s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aab28561af834c77a2a9ef3137137c34 : SUCCESS in 2h 17m 55s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/fb6532aa5a27454ea45654b7a5d63383 : SUCCESS in 1h 49m 06s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/6f6689d8e31947febab6b781df3bf333 : FAILURE in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34025e48e8f463fb5ffb4fdeabda4156485d5a77","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-04 12:26:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_PowerStore_FC_os-brick/e755897/DellEMC_PowerStore_FC_os-brick/18 : FAILURE in 1h 25m 00s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_PowerStore_iSCSI_os-brick/a72fea4/DellEMC_PowerStore_iSCSI_os-brick/18 : FAILURE in 1h 08m 11s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_PowerStore_NFS_os-brick/d2c4b89/DellEMC_PowerStore_NFS_os-brick/18 : FAILURE in 55m 40s\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/2c41a98/DellEMC_PowerStore_NVMeTCP_os-brick/18 : FAILURE in 1h 08m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"d441c01a9bd6f05052a419d489332e29fcf17f9d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-04 12:32:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_PowerFlex_os-brick/1afd929/DellEMC_PowerFlex_os-brick/7 : FAILURE in 1h 41m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a9cd0eb6bc7253b1b7d57681b7b718cfb0e3f58f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-04 12:36:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/1/upstream-check/cinder-tempest-plugin-ontap-fcp/f20bcb2/ : RETRY_LIMIT in 10m 57s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/6c8a64e/ : RETRY_LIMIT in 9m 32s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/1/upstream-check/cinder-tempest-plugin-ontap-nfs/64e5310/ : SUCCESS in 1h 28m 44s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/fa3aafb/ : SUCCESS in 1h 33m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2162bfe065c0c83fcfdbdd0381b84e62f8b67020","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-04 14:59:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_Unity_FC_os-brick/100a9e4/DellEMC_Unity_FC_os-brick/26 : FAILURE in 1h 56m 25s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/1/check/DellEMC_Unity_iSCSI_os-brick/387c5ee/DellEMC_Unity_iSCSI_os-brick/26 : FAILURE in 2h 10m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ced60b32c05b89a99938be1e4f7eb48a2ccac363","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-04 17:48:55.000000000","message":"Patch Set 1:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1541/","accounts_in_message":[],"_revision_number":1},{"id":"3d8e667f132f2a4a90a020b48d2909d578ed137f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-06 06:16:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/1/20230706121936/NEC-V-ISCSI : SUCCESS in 2h 49m 57s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/1/20230706121936/NEC-V-FC : SUCCESS in 2h 57m 02s","accounts_in_message":[],"_revision_number":1},{"id":"63f25e6f803e57a4abf77b8c9d101dc7b792d19b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-06 21:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/1/20230706224742/NEC-ISCSI : SUCCESS in 3h 45m 08s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/1/20230706224742/NEC-FC : SUCCESS in 7h 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"03d1c4ce87e43b8420d4f26e10b3b0d1ba6199fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-07-10 07:57:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"464deb1e10eefa136a3d7b079e10d884adc7ba01","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-10 10:22:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/3638e8d/ : SUCCESS in 2h 22m 00s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/9c26b0a/ : SUCCESS in 1h 56m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/1a1757b/ : SUCCESS in 2h 02m 58s","accounts_in_message":[],"_revision_number":2},{"id":"efbe1fe0b9b39040e878329ca9fe7bf6524ee254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 10:31:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/089188d1499d465e928ad7ab93f466a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5732dcea75149f4a35b13d76fc88083 : SUCCESS in 1h 36m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f932c5a8c814a069d5f81824830ede2 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc0072bb0438419da666eabc4ed672c0 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06a24f452b0f402994d4d3cfa12bc653 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f4531f450b4c27b8b34b25c054718a : SUCCESS in 5m 08s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/30e6e25ee3eb4030a096a9300fe9dada : SUCCESS in 5m 02s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a1e6f82f5f1846138b05e2f3ed374f21 : SUCCESS in 2h 19m 19s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/135b5314db394559bbc0145edd35cdd0 : FAILURE in 2h 26m 30s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/146d5ee5caeb498aa85c09e3d02a7f1e : SUCCESS in 1h 27m 33s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/ab6810ee2c7d4ee2b578c266f7802078 : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4bfdd96cf83dabf873f5b326ce18632776fdf45","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-10 11:00:12.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/2/fc : SUCCESS in 0h 45m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/2/iscsi : SUCCESS in 0h 43m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d4839b372de25bb2d339edf09964cd0e240191bd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-10 11:00:29.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-2/FC/ : SUCCESS in 2h 34m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-2/iSCSI/ : SUCCESS in 2h 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"716d0706bb14948c7e29823d013a2383c5914184","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-10 11:33:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/2/check/DellEMC_PowerFlex_os-brick/a045d74/DellEMC_PowerFlex_os-brick/10 : FAILURE in 1h 52m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3aa89fe3630fb205716322be35cb39738f903d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-10 12:16:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/2/check/DellEMC_Unity_FC_os-brick/9f26432/DellEMC_Unity_FC_os-brick/28 : FAILURE in 1h 52m 46s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/2/check/DellEMC_Unity_iSCSI_os-brick/3e41791/DellEMC_Unity_iSCSI_os-brick/28 : FAILURE in 2h 14m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"09d402faf938b961f116a112a29eec3d1b5a93a9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-10 13:27:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/2/20230710194250/NEC-V-ISCSI : SUCCESS in 2h 36m 32s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/2/20230710194250/NEC-V-FC : SUCCESS in 2h 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"df0703d64c7a059279a73a25002f9c2d5ffb4c73","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-10 13:55:48.000000000","message":"Patch Set 2:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1562/","accounts_in_message":[],"_revision_number":2},{"id":"35b663b8a3ada8660bc28b5b519ed895cf7d576b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-11 10:31:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/2/20230711151123/NEC-FC : SUCCESS in 3h 50m 32s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/2/20230711151123/NEC-ISCSI : SUCCESS in 4h 20m 03s","accounts_in_message":[],"_revision_number":2},{"id":"38386ce8211b987e6142c2e8a3ae14aac0f64613","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-07-13 18:58:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc7720634eb8cc575bc6fc50c692ac2db433247d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-13 19:03:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ac72fb8fbca24b1fa2726faca03af38d : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"23d6ae86f8796736e2b992fda8be2b970795b40d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 21:13:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcdd53ac2b904060a145f0f39928b02a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e0703000ddd4ebf94d0407f21bda011 : TIMED_OUT in 2h 04m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae91211a48d4b67a2dd46f12931e8e1 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10f54193dd984229b5267d6f6910bf2c : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38161310aded41668fa44305dc6fd4af : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16534b2e630f4612a4d4098fe9ed0c97 : SUCCESS in 10m 19s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c8f836f5c4514a328fb588e5243e6fb3 : SUCCESS in 3m 29s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/80e26db025854732a442fe34278c80b3 : SUCCESS in 1h 58m 01s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c390fd87fa884a4c9af0a9e610f46525 : FAILURE in 1h 41m 57s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f9ba24f907ac409aa3f0f61fb508f6a3 : SUCCESS in 1h 24m 13s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d8fc78f1f20b46b6a67668b1b374e599 : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9338fed07ea18593d996bb1de2bde8c3e3fe7fc2","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-07-18 11:24:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"348aadc761eb4a70e6461a4ae4c6d6e08cc2ce13","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-18 14:35:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fba521a2c09e48bd9f1ca33d4734b67e : FAILURE in 1h 14m 53s","accounts_in_message":[],"_revision_number":2},{"id":"e390f220e4d3c06f6c51ad6d1b67ba2444db1dab","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-04 14:39:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b30cd46fb6479c817d69a9a81c08e78a0f501b0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-04 17:49:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/be4d72eff4f0496a8cd5dda279d6922e : FAILURE in 1h 12m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9036e21312c756571608656d0769bc6b05f6bab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-10 17:51:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"33519875729929971bdd92ccf36d58cd689be6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 19:34:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38cae1fc02254823aa345a7d0b59d7eb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b84cdfae1dbb488b8174e6b2adf57b46 : FAILURE in 1h 07m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d53e3f0f6b4fb6863b4f72687310e4 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e73b5a67214898a60a329c8ebb6339 : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab81e613ce914262b97247f819d23707 : FAILURE in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f4ad11711f41db8c6e09ada9fa0624 : SUCCESS in 5m 34s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/50e2d7faa45f4200a5575ff2a4e9acef : FAILURE in 4m 08s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/adee7036ab604751afa30f9cd71bd14f : SUCCESS in 1h 41m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61510e8171d4452785a6f399c46cf1e7 : SUCCESS in 1h 39m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d342bf4077824dc08b2903d2e4ef00b5 : SUCCESS in 1h 16m 37s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4f749380a8f44fbe897b3f9458bd67e1 : SUCCESS in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c8e3459f1ee23b9026ac89dc69b9328cbb8ec6d3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-10 20:21:09.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-3/FC/ : SUCCESS in 2h 14m 22s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-3/iSCSI/ : SUCCESS in 1h 44m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"69f39e79bfc31882813fc52d065c4e7c2fecb419","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-10 21:27:14.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/4f252e1/ : FAILURE in 1h 58m 21s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/066258a/ : SUCCESS in 1h 45m 46s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/453df68/ : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"2062152806eac89b2fccfcd881d3819c36c7d6c8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-10 22:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/3/fc : SUCCESS in 0h 27m 10s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/3/iscsi : SUCCESS in 0h 41m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7b0e620eff38d53169ec5d37f7cdad4665b24990","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-10 22:05:16.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/3/check/DellEMC_PowerFlex_os-brick/6da6f57/DellEMC_PowerFlex_os-brick/28 : FAILURE in 1h 44m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"9f07daed122e5e54aaf5638d11b6e4f087f714e9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-10 22:27:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/506/","accounts_in_message":[],"_revision_number":3},{"id":"7b69743ecb602127164f51f775482b58a191ad42","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-11 00:34:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/84081f2ed7b84c5599aafba2cd9902b2 : FAILURE in 2h 27m 12s","accounts_in_message":[],"_revision_number":3},{"id":"5ec345a038fdc5081311888d9f70dfd6013dcd47","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-11 01:50:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/3/check/DellEMC_Unity_FC_os-brick/9622d7a/DellEMC_Unity_FC_os-brick/44 : FAILURE in 1h 54m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/3/check/DellEMC_Unity_iSCSI_os-brick/c1e50ac/DellEMC_Unity_iSCSI_os-brick/44 : FAILURE in 2h 32m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"761f6d628c4e53577a8272db2ab5ecb93ec15dd4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-12 02:02:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/3/20230812080622/NEC-V-ISCSI : SUCCESS in 2h 52m 38s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/3/20230812080622/NEC-V-FC : SUCCESS in 2h 55m 37s","accounts_in_message":[],"_revision_number":3},{"id":"9719a6fcc82cee2c8d9c52b3ed98ebf7aed1d6b9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-12 23:53:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/3/20230813003352/NEC-ISCSI : SUCCESS in 3h 54m 19s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/3/20230813003352/NEC-FC : SUCCESS in 8h 19m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3a702faeb6952ae6be12e97be9c698b71c3e8bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-17 11:13:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2ca9cf2e8e6803bad6eb5f95a418c57d9f13b484","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 12:09:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_Unity_FC_os-brick/13b1681/DellEMC_Unity_FC_os-brick/48 : FAILURE in 28m 15s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_Unity_iSCSI_os-brick/94708a1/DellEMC_Unity_iSCSI_os-brick/48 : FAILURE in 14m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8615ac33546bf46ec643360b12332982f70060cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:58:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dabf80f192a405d9acd72dd7558e827\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fd1b2c0a57d463cb15ac37155c90564 : SUCCESS in 1h 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3724b4d06c46b8ada96071f53ec846 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/162f952008d0439381fc7bd744163113 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09ef3f7850ed4d31ae444e6b75f3ea88 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf7213a4d0a40038e214cde5bdbc180 : SUCCESS in 7m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f3306493b07a47658a119d3024884834 : SUCCESS in 4m 24s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/455f29e4a51e4ea0bf7706f3f7206279 : SUCCESS in 2h 01m 03s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e75a1f352ced442c9e5b12f62700f517 : SUCCESS in 1h 50m 43s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1d87b2c9cc72449c87e5470f49e5154b : FAILURE in 1h 03m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c2ac47cedff9440f92e154d7ae5cf4ac : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8e608f6f1cf498c275a6f79ea76d666fc32689e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-17 14:14:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_PowerStore_FC_os-brick/7fde14a/DellEMC_PowerStore_FC_os-brick/30 : FAILURE in 15m 53s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_PowerStore_iSCSI_os-brick/b77415e/DellEMC_PowerStore_iSCSI_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_PowerStore_NFS_os-brick/41e47a7/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_PowerStore_NVMeTCP_os-brick/fb75c25/DellEMC_PowerStore_NVMeTCP_os-brick/30 : FAILURE in 13m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"b521d2bf7a3918572f528ababd98387c292d7024","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-17 15:36:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/4/upstream-check/cinder-tempest-plugin-ontap-fcp/861762e/ : SUCCESS in 1h 09m 34s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/746f797/ : SUCCESS in 1h 33m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/4/upstream-check/cinder-tempest-plugin-ontap-nfs/16a7f45/ : SUCCESS in 1h 36m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/aef18a4/ : FAILURE in 1h 32m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a16d91b1d7213a9a4460d29056059066ebec6b6b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 15:44:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_Unity_FC_os-brick/4e59d69/DellEMC_Unity_FC_os-brick/52 : FAILURE in 28m 07s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/4/check/DellEMC_Unity_iSCSI_os-brick/2002585/DellEMC_Unity_iSCSI_os-brick/52 : FAILURE in 14m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d647f14c05069bc7b86114255e050d8fde58a402","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-17 15:46:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b5a4b88/ : SUCCESS in 2h 04m 03s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/e256cd2/ : SUCCESS in 2h 00m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/07d9ec8/ : SUCCESS in 1h 53m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9f1796ea278e044f3556200fc4a70b8692f1df2f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-17 20:33:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9caea00da6fc4e4a83395f7be7ccb11b : FAILURE in 2h 55m 21s","accounts_in_message":[],"_revision_number":4},{"id":"df1f52ebfc9ca90aac671bca136513d1dc46896f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-18 23:30:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/4/20230819052349/NEC-V-ISCSI : SUCCESS in 3h 01m 35s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/4/20230819052349/NEC-V-FC : SUCCESS in 3h 06m 19s","accounts_in_message":[],"_revision_number":4},{"id":"ca461cce246c3db3f2cfd32171321a2f98d07b04","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-19 20:36:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/4/20230820013734/NEC-FC : SUCCESS in 3h 48m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/4/20230820013734/NEC-ISCSI : SUCCESS in 3h 58m 24s","accounts_in_message":[],"_revision_number":4},{"id":"15223ed6a0916ffec86f3cf73cd13506bb4a771f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-24 17:39:40.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e7d8b97086c9e5e05ce90bd734a8eb58fce85fa2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-24 19:02:47.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/5/check/DellEMC_Unity_FC_os-brick/83f4c50/DellEMC_Unity_FC_os-brick/76 : FAILURE in 34m 48s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/5/check/DellEMC_Unity_iSCSI_os-brick/cf10a42/DellEMC_Unity_iSCSI_os-brick/76 : FAILURE in 14m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"a312cecb4beb9d8c59e0fb326fe2b32a70169b05","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-24 19:09:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c252c277191044bcb81b601586d6a56a : SUCCESS in 1h 27m 35s","accounts_in_message":[],"_revision_number":5},{"id":"411ff0dc1945c4de11115a84b37b5ee8db474edc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-24 19:39:30.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/52dc826/ : SUCCESS in 1h 52m 57s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/c84c674/ : SUCCESS in 1h 55m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5ee168e/ : SUCCESS in 1h 49m 36s","accounts_in_message":[],"_revision_number":5},{"id":"6d630fa9e04668ed2dd03d901b26fdb7add48fa1","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-24 20:00:07.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/5/fc : SUCCESS in 0h 31m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/5/iscsi : SUCCESS in 0h 48m 53s","accounts_in_message":[],"_revision_number":5},{"id":"e70c6c58ef766b11576b2692ee6f886201328975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 20:08:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c9609a295b04ab7897401b54e2bb222\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74f54dc924f542b7885b431b63c9a1f3 : SUCCESS in 1h 08m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4fc27e0a3f4030b4ae33ebb407966d : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/070e5570ad4248e9a666da23d8417a79 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da8b81050914498b8cfbb941b07fc32e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa061f231c0d4a4aac362c2e978c7d3e : SUCCESS in 5m 17s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/15ca90cfd32949a3943bf201fe6c3fb5 : SUCCESS in 3m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a67bee10cec24a429b0b3fd8a592709d : SUCCESS in 1h 50m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/874519f46616454d9c715ab1461a2ddc : FAILURE in 2h 20m 44s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/81430ae57c184835b86f1010b0359a1a : SUCCESS in 1h 03m 13s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a565d041eb354efea34612bf1eee60db : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7f34a5572eb9d8f9f8c0fdf8aae19872937e7710","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-24 20:20:32.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-5/FC/ : SUCCESS in 2h 25m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-5/iSCSI/ : SUCCESS in 1h 50m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"296a64af2b261defd1a100ae6cc129682f66f274","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-24 22:53:46.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/5/check/DellEMC_PowerFlex_os-brick/1755daf/DellEMC_PowerFlex_os-brick/42 : FAILURE in 1h 46m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"afc9a47ab6f5d237e2778f00c65998b8d53cfb73","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-25 00:37:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1763/","accounts_in_message":[],"_revision_number":5},{"id":"fbd8f19c9d7ea39b97f3c681da89864f838290a0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-25 17:44:17.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/5/20230825234159/NEC-V-ISCSI : SUCCESS in 2h 59m 35s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/5/20230825234159/NEC-V-FC : SUCCESS in 3h 02m 15s","accounts_in_message":[],"_revision_number":5},{"id":"138e1d23bb238ccac1aea82b689836c810785c09","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 12:58:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/5/20230828133842/NEC-FC : SUCCESS in 3h 56m 11s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/5/20230828133842/NEC-ISCSI : SUCCESS in 8h 19m 17s","accounts_in_message":[],"_revision_number":5},{"id":"e7759583d2d7634bd431c7a4dad84c1dd74e3f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-28 14:30:16.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1467b178fa97b4e1ecd04e89a137b6abe80730f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 14:33:25.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_Unity_FC_os-brick/8fc6f30/DellEMC_Unity_FC_os-brick/83 : FAILURE in 2m 28s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_Unity_iSCSI_os-brick/1ccca07/DellEMC_Unity_iSCSI_os-brick/83 : FAILURE in 1m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"8e24f64ee6af75076309789f37f76252a1354a8a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-28 15:50:18.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ae1f356088ee4e74b420c0fc0fb5ebfc : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":6},{"id":"2c81bff0a6bd3d87114209fdc597916bf961acfa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-28 16:30:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/0c616fe/ : SUCCESS in 1h 47m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/98deeb7/ : SUCCESS in 1h 53m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/dcf9146/ : SUCCESS in 1h 52m 40s","accounts_in_message":[],"_revision_number":6},{"id":"2506c2c4566df57cdbc0ba9c791928426430da62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:34:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a94f9118d94252ae9aa839c0116d36\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16653ccb4623490ba43ef24ae895ecb4 : SUCCESS in 1h 39m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab0d874afc045cabd7a2c570b60e935 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f743c788c3424be897c05558500aa86f : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be5d63df7e1846b2998efbf34bdc9da7 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b767d43acd9441c95a04639ec4ae881 : SUCCESS in 6m 36s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8fac1f0eea2848ffa8f43eeb6343c443 : SUCCESS in 4m 14s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/8e0078072ca947ff897b1b534e66eba3 : SUCCESS in 1h 58m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6002918a8b2541d784b01fed8e7f1dd2 : SUCCESS in 1h 52m 40s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f272100812b6441a839491d959044443 : SUCCESS in 1h 27m 26s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/45ee5ba904ac4e22aa8e31a67e59d432 : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"32aa72972a88f5d521c4b3755f18395472d8c578","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 16:36:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_Unity_FC_os-brick/3ea0967/DellEMC_Unity_FC_os-brick/84 : FAILURE in 2m 28s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_Unity_iSCSI_os-brick/a543bda/DellEMC_Unity_iSCSI_os-brick/84 : FAILURE in 1m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"412386d21358827709efb49debf6a8d6e83f5b0c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-28 17:32:32.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-6/FC/ : SUCCESS in 2h 33m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-6/iSCSI/ : SUCCESS in 2h 48m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"5ffb51d086fffc7516c1023b793c9b638353e26f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-28 17:46:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/6/upstream-check/cinder-tempest-plugin-ontap-fcp/f137b7e/ : SUCCESS in 1h 04m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/0146c76/ : SUCCESS in 47m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/6/upstream-check/cinder-tempest-plugin-ontap-nfs/69c6a0a/ : SUCCESS in 1h 12m 00s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/6141773/ : FAILURE in 44m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"90fd51ea212c4228454e3b567c437b5bf97e5958","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-28 18:25:42.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_PowerStore_FC_os-brick/b1c42b9/DellEMC_PowerStore_FC_os-brick/47 : FAILURE in 1h 18m 41s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_PowerStore_iSCSI_os-brick/1c42cc5/DellEMC_PowerStore_iSCSI_os-brick/36 : FAILURE in 1h 15m 19s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_PowerStore_NFS_os-brick/5f69466/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_PowerStore_NVMeTCP_os-brick/9e8d84b/DellEMC_PowerStore_NVMeTCP_os-brick/47 : FAILURE in 1h 25m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"9cf49e53ca8da125c914813042fd84f9637000d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 18:34:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/6/check/DellEMC_PowerFlex_os-brick/0a3b7aa/DellEMC_PowerFlex_os-brick/49 : FAILURE in 2h 00m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"0f9ed82b82f7278d57c7fc063976966f930513a5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-29 00:03:15.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/6/20230829055645/NEC-V-FC : SUCCESS in 2h 57m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/6/20230829055645/NEC-V-ISCSI : SUCCESS in 3h 06m 26s","accounts_in_message":[],"_revision_number":6},{"id":"23a9ec348891ae45ca561e20653cc704098129ff","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-29 01:00:08.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/6/fc : SUCCESS in 0h 32m 43s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/6/iscsi : SUCCESS in 0h 49m 23s","accounts_in_message":[],"_revision_number":6},{"id":"ab03d24f798cddb5d3d8b23b25dd20c9ef022ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-29 07:23:18.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"25bc342c7d508e87ba6a5069b469241445264a62","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 07:28:32.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/7/check/DellEMC_Unity_FC_os-brick/572a52d/DellEMC_Unity_FC_os-brick/86 : FAILURE in 2m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/7/check/DellEMC_Unity_iSCSI_os-brick/bdc05c8/DellEMC_Unity_iSCSI_os-brick/86 : FAILURE in 1m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8273851a68a9c3b1f0ff146755ffc056fd2d833f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-08-29 07:31:11.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"056472fed08b109177ae7bad31519e1ebca82e11","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 07:36:25.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_Unity_FC_os-brick/3ecd5b8/DellEMC_Unity_FC_os-brick/88 : FAILURE in 2m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_Unity_iSCSI_os-brick/c05d57d/DellEMC_Unity_iSCSI_os-brick/88 : FAILURE in 1m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"f9f4dc45a55befee96b581eec01d7d34e9c9e312","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-29 08:46:10.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/de6187b74292400c8585ac5452049729 : SUCCESS in 1h 12m 50s","accounts_in_message":[],"_revision_number":8},{"id":"0b04c36d3f9db4c4cfa9c6003f3f1cab9c6894ab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-29 09:29:17.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/cd004fc/ : SUCCESS in 1h 50m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/24e38f3/ : SUCCESS in 1h 54m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/b05e4cd/ : SUCCESS in 1h 50m 51s","accounts_in_message":[],"_revision_number":8},{"id":"a6cf78578f4870a9178cd2974d2e65c9548aa3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:33:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d86c98cafc5744bebfe99c58dd2ed9b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66785f5fcebc4b5da1908c52224f6b2d : SUCCESS in 1h 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321f9d931e3a418f9a01770a0ea9d395 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/950b9d14c7b74cc6abbc42c1e2b95546 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a49a3b20ff5845ceb52121c41f6a2743 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88660ba5567946a195380ee9ff451ac4 : SUCCESS in 6m 28s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/36ff8588b57441e388e39570139abb3c : SUCCESS in 3m 47s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/f79fc1ec51964a22a9852ba718c3b472 : FAILURE in 1h 46m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/912ba3e9739d401bbf7caac2bf8abe0f : SUCCESS in 1h 55m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0dc32739ae4d4351b057e842022ca2e3 : SUCCESS in 1h 26m 01s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/390c57627806422a9292bcb366408fcb : SUCCESS in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ade61a630d39900bfa00a6bcd9c623b7d1eed62b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-29 11:06:42.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_PowerFlex_os-brick/dcd7d7f/DellEMC_PowerFlex_os-brick/53 : FAILURE in 1h 58m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"eff795481ac2199716f3cfa03416fa500bcee7d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 11:11:20.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_Unity_FC_os-brick/ca3a24d/DellEMC_Unity_FC_os-brick/90 : FAILURE in 36m 20s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_Unity_iSCSI_os-brick/1d93409/DellEMC_Unity_iSCSI_os-brick/90 : FAILURE in 10m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"3bb5c138147fb448044969d39b8d01e348340ba1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-29 11:17:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/8/upstream-check/cinder-tempest-plugin-ontap-fcp/db949f2/ : FAILURE in 1h 21m 34s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/46d1723/ : SUCCESS in 46m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/8/upstream-check/cinder-tempest-plugin-ontap-nfs/4a8fba7/ : SUCCESS in 1h 41m 57s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f21614a/ : SUCCESS in 1h 39m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"0557e552129d9177bcb9d267ed5883021a178a52","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-29 12:09:19.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_PowerStore_FC_os-brick/bf3a7d5/DellEMC_PowerStore_FC_os-brick/49 : FAILURE in 1h 28m 43s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_PowerStore_iSCSI_os-brick/3e3d716/DellEMC_PowerStore_iSCSI_os-brick/38 : FAILURE in 1h 11m 52s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_PowerStore_NFS_os-brick/32258ba/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/76/887576/8/check/DellEMC_PowerStore_NVMeTCP_os-brick/03033d5/DellEMC_PowerStore_NVMeTCP_os-brick/49 : FAILURE in 1h 20m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"1a93e363db767fee282f1ff92a8b3703cc374270","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-29 13:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/8/fc : SUCCESS in 0h 44m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/8/iscsi : SUCCESS in 0h 56m 06s","accounts_in_message":[],"_revision_number":8},{"id":"23492e144d5bc93e91e4b400f3a7c85cf2da414a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-29 14:52:24.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-8/FC/ : FAILURE in 6h 43m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-8/iSCSI/ : UNSTABLE in 10m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"5d61f395ddca2627349ed993da1f4c366115c547","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-29 15:55:19.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/8/20230829213605/NEC-V-FC : SUCCESS in 3h 15m 09s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/8/20230829213605/NEC-V-ISCSI : SUCCESS in 3h 19m 10s","accounts_in_message":[],"_revision_number":8},{"id":"ed5a44d626f5404ab65bbcade0c80a901a009006","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-30 19:08:48.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/8/20230831002234/NEC-ISCSI : SUCCESS in 3h 46m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/8/20230831002234/NEC-FC : SUCCESS in 3h 46m 10s","accounts_in_message":[],"_revision_number":8},{"id":"b098004ac709194c0aa39c1bbe84723daee7ea78","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-04 10:23:27.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/085dfcfe83064d8487ab8ee5608cf275 : FAILURE in 1h 40m 09s","accounts_in_message":[],"_revision_number":8},{"id":"7f49e1dbd8e81c97e66ac600622a2285c6d293c2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-04 12:28:36.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/f27d3af5de5e4ae68a83b6cb8a38dd0f : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":8},{"id":"465c58e10773bf792b8b95eb9c8b310b51477adc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-05 17:13:50.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-8/FC/ : SUCCESS in 7h 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-8/iSCSI/ : SUCCESS in 7h 46m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"1e4e3d685735ebbdc40bd9a6ad73f21e9c66334b","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-07 16:47:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/76/887576/5/check/tyds_test/8e149dc : SUCCESS in 4h 28m 55s","accounts_in_message":[],"_revision_number":5},{"id":"d5ab2bc95d891bde2b5967beab39c0ec24004d77","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-07 22:01:55.000000000","message":"Patch Set 6:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/76/887576/6/check/tyds_test/dfab5a5 : SUCCESS in 4h 53m 58s","accounts_in_message":[],"_revision_number":6},{"id":"b0132f3c1e3bd449d324c79724ff331f51767a5c","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-15 07:56:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5fc53a66e30e26c06100fcd673fe267e08796b3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-09-20 15:05:33.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e329f7ee394fd01990a1acbbb5a2c0df0d1e4d3b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-20 15:06:50.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a7b346a087d548bdad531b93dabbf48a : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"6102c431f1ed0f166e5404dfa36952214241983f","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-09-20 15:06:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"84915829f174c9585b4569fcfc4fd093ad6cefe1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-09-20 15:10:09.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f95c1451b953f0defc37aefc32187c31a870f0d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-20 16:28:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1872/","accounts_in_message":[],"_revision_number":9},{"id":"9ce48eaf16564f07b040375058127c00ab0bece9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-20 17:10:16.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/76/887576/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/e9b7bac/ : SUCCESS in 1h 56m 05s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/76/887576/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/a0d46d8/ : FAILURE in 2h 00m 39s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/76/887576/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ea85b34/ : SUCCESS in 2h 02m 10s","accounts_in_message":[],"_revision_number":9},{"id":"684611390701c8e9d58399944d701b8f6bcd6293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:12:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b71f16ec724b69b3c2ce67c23a1ce4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e3de8c4589f47d68aa931f6e9f33e6d : SUCCESS in 1h 07m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba206028c8ba49888343842322bbf9cc : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a88736e4c0d4cd4bbc5e5140f37731d : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c15a1e00c46435390be6ab00b73ae62 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00b657bc63964083b5913414bbfb6592 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65f4ea31a88e4a0b95a847dc1203a7a7 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1e62cb9ea842bdac0f071551c213de : SUCCESS in 6m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7c412e4eebe74026806b641388ca15e2 : SUCCESS in 4m 45s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/89bd1074c461491e9efe3535568c0807 : SUCCESS in 2h 04m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d592a7d9a24549e488948cea1981d828 : SUCCESS in 1h 56m 16s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/820f75d0df3c43a3a07b29507fc90d48 : SUCCESS in 56m 20s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/96627db8e46847f1a5a2e4810c475338 : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e50c953856ae7c78f59cd54094bbb2abaa61461a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-20 17:59:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/666/","accounts_in_message":[],"_revision_number":9},{"id":"79e84677b79d56e0f2e605b244077c91d5886564","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-20 18:56:44.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/9/upstream-check/cinder-tempest-plugin-ontap-fcp/3619f14/ : RETRY_LIMIT in 8m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/74e67b1/ : SUCCESS in 1h 34m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/9/upstream-check/cinder-tempest-plugin-ontap-nfs/bed7fbe/ : SUCCESS in 1h 36m 22s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f3342cc/ : SUCCESS in 1h 36m 38s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"36f6618b42b36dc23fc9986e13f34f3174b6e121","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 19:24:47.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_Unity_FC_os-brick/c874c21/DellEMC_Unity_FC_os-brick/108 : FAILURE in 1h 26m 35s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_Unity_iSCSI_os-brick/26bdf62/DellEMC_Unity_iSCSI_os-brick/108 : FAILURE in 1h 41m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"3baab47da6f2fea1381f4be3d869ce6a736220d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 19:27:43.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_PowerFlex_os-brick/ad04f56/DellEMC_PowerFlex_os-brick/69 : SUCCESS in 1h 08m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"8aad5777ba0bf0f8215bd7fe44e5ed566a46e8e6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-20 23:42:20.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_PowerStore_FC_os-brick/d55b060/DellEMC_PowerStore_FC_os-brick/60 : FAILURE in 1h 15m 01s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_PowerStore_iSCSI_os-brick/fe28289/DellEMC_PowerStore_iSCSI_os-brick/53 : FAILURE in 1h 08m 52s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_PowerStore_NFS_os-brick/3aa7b16/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/76/887576/9/check/DellEMC_PowerStore_NVMeTCP_os-brick/05ecab9/DellEMC_PowerStore_NVMeTCP_os-brick/64 : FAILURE in 1h 05m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"6e329136eb5fe72f654401ed4ea6656717d4df4c","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-21 01:04:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/76/887576/9/check/tyds_test/51b7390 : SUCCESS in 3h 30m 32s","accounts_in_message":[],"_revision_number":9},{"id":"cb86f4120ec1f13e4b6b11bb8e401198689e2b93","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-09-21 04:28:28.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/9/20230921071215/NEC-V-ISCSI : SUCCESS in 2h 43m 59s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/9/20230921071215/NEC-V-FC : SUCCESS in 6h 16m 09s","accounts_in_message":[],"_revision_number":9},{"id":"8f65fced891d004fa4e4fcfa28faa9b454dd937c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-21 06:00:07.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/9/fc : SUCCESS in 0h 59m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/9/iscsi : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":9},{"id":"db5d21a6b0b3e19124aae7c646914ec6f3766826","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 19:24:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/9/20230922003315/NEC-ISCSI : SUCCESS in 3h 44m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/887576/9/20230922003315/NEC-FC : SUCCESS in 3h 51m 07s","accounts_in_message":[],"_revision_number":9},{"id":"58ad58b0b20984e2eb17e00c20279e657efbbb3f","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-09-29 11:56:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3bd67bdb7ad0d99a6c6c8d926db6598dfccdc5d9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-10 04:17:27.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c7c5ecca1618f1e590d922521a2a46690d0eeaee","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-11-22 15:12:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"45f2c9a2239e450e26f193462d88cc6df8660da6","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-12-01 07:39:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"edb02b5c6a902fa9ccd70ec6abd69b49698b9155","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-01 08:54:04.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6b0fc0f1ccc24473a144aff686e46407 : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":9},{"id":"3e25c244ca04dc9f3764d270d4e0fc2e58313d88","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2023-12-06 09:20:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e906d1f344ffbe1a8478ed3bb497a9ed9465a1d9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-12-06 15:13:14.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cc1f3837430c2f812a6253689c08617d34c521a3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-01-10 15:12:58.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a4c97929255f04f8c40d37f56d1af2a0c1f25b95","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-16 14:31:47.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8d12861daf7cb0fdef59ecf852595d25eda7c99f","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-01-22 13:10:12.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e6b3cbec9b0407c6a7da9ba3704121b5e49ec5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-04 03:29:49.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"984ff493270acbab2a34fbcec8831ef2368e461f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-04 03:30:36.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"457cee4ea53bee492505728b00aba439a5579e69","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2024-07-04 04:00:03.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f6ed10e4d404fb3b0f1c4adb9f9842dd876fd58d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-04 04:52:38.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ad38745019504c80979a0975ca13a5ef : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":10},{"id":"1bc2a3965aeb819b55ab86ff38ebea503f82e4e5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-04 05:01:56.000000000","message":"Patch Set 10:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/679/.","accounts_in_message":[],"_revision_number":10},{"id":"6b49c481ecf0f3ae687205059c71cebcd83a9d45","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-04 05:33:51.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_Unity_FC_os-brick/6f8a545/DellEMC_Unity_FC_os-brick/184 : SUCCESS in 1h 30m 04s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_Unity_iSCSI_os-brick/21d53a0/DellEMC_Unity_iSCSI_os-brick/181 : SUCCESS in 2h 03m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"e0f926a57828dab061c6bcb6faa0d6e6f2443ae0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-04 05:39:11.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerFlex_os-brick/5dae12e/DellEMC_PowerFlex_os-brick/103 : SUCCESS in 2h 09m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"3e227c4d51673040c9c8199e7d5633dc13239924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 05:49:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/986fe5c027df48ab817a589dcc5415df\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f154434173a42c8883459b4bc5e1696 : SUCCESS in 1h 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239025d7dbbb4047bf9b4ab433c0c21f : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55473db12a44bb890426dc1c65a1d3b : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0db4208820e486d99fdc9608e73e945 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0eb4a13b9dda44b5b690f3b26e4d5f63 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55dabe7402042c881ce48a835aec3c6 : SUCCESS in 5m 30s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a7fb1867dd2c471ca8d9dafaf8a5fb7e : SUCCESS in 4m 20s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2d75a674e7884e059dfb77b573aadc06 : FAILURE in 2h 19m 24s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01a531ce3b274c5a964bdea05707d74f : SUCCESS in 1h 39m 50s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d1120e2b99754b799acd073decc36784 : SUCCESS in 1h 27m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a582d12eb9a14e2890fb8695bdaa5dff : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1b31008fadac910c14e45ff372ff6cac8b050436","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-04 06:09:37.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-10/FC/ : SUCCESS in 2h 39m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-887576-10/iSCSI/ : SUCCESS in 1h 44m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"d36cf82380b21e3ca8eea7c35b49235cce527251","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-04 07:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/10/fc : SUCCESS in 0h 49m 15s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/76/887576/10/iscsi : SUCCESS in 0h 53m 10s","accounts_in_message":[],"_revision_number":10},{"id":"a7a05045312b03e9c9d93f9b849018737a227428","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-04 07:21:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/14b9837/ : SUCCESS in 58m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/10/upstream-check/cinder-tempest-plugin-ontap-nfs/f671dbf/ : SUCCESS in 53m 43s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/887576/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7590261/ : SUCCESS in 1h 31m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"a6fc9e5476cb06b815163f6e3f77b672e609783f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-04 09:58:23.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerStore_FC-ob/3322c4f/DellEMC_PowerStore_FC-ob/236 : FAILURE in 10s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerStore_iSCSI-ob/5133812/DellEMC_PowerStore_iSCSI-ob/156 : SUCCESS in 2h 03m 30s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerStore_NFS-ob/7822de2/DellEMC_PowerStore_NFS-ob/131 : FAILURE in 1m 35s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerStore_NVMeTCP-ob/cab40f0/DellEMC_PowerStore_NVMeTCP-ob/110 : SUCCESS in 2h 15m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"8908c7cc7784bd719eefb2f1bd6bcbf1c0977fec","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-04 10:42:20.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_PowerFlex_os-brick/a64a1af/DellEMC_PowerFlex_os-brick/104 : FAILURE in 1h 43m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"deaeca23f111be44696087b0bce9a0ce5bbf8cdd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-04 11:18:01.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_Unity_FC_os-brick/82fb3d4/DellEMC_Unity_FC_os-brick/185 : SUCCESS in 1h 24m 19s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/76/887576/10/check/DellEMC_Unity_iSCSI_os-brick/dcdbeed/DellEMC_Unity_iSCSI_os-brick/182 : SUCCESS in 1h 53m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"f395c6929e635ad4c54e9a0b98ef690fa521c566","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2024-07-05 05:52:11.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d4834887882b0e94f12cc3cf888a28b285e5e903","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-05 14:50:32.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"57b2a2b07dca5a9416f4d995da64e8d39175bc0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 14:51:05.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"810ab1f97c31799155501df2b75f85907a09407f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 16:55:39.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38afd22552fe4fc68d517efb93d3cb4b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/387bce829cd5492889994cffc52650ee : SUCCESS in 1h 48m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf1ab808cbd4923909403c87d1cc854 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c823d228f3e84afaa5e40426077ccadf : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea40788bd13a4a99b2f76f6c8f641cea : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45da8ecaa3ef4e0b9b7311c82a97322b : SUCCESS in 6m 04s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0a6795404a8419ba1f45734838973a7 : SUCCESS in 2h 02m 58s","accounts_in_message":[],"_revision_number":10},{"id":"d6085060b574da6044d1ba275216eaaf6c16522d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 16:55:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"7cccdf7b268bdadd476005696d44d21a9f7151b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 16:56:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acaa9ecba6f84911ac9d40c7dd704b11\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba84d7965a04099bfb36f15b4f18a90 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b290711cfb23603627e9e28c733fceb5184589b5","revisions":{"101fec2c2d4d89a88561c23700f49f7d38d7ec3c":{"kind":"REWORK","_number":1,"created":"2023-07-04 08:31:25.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/1"}}},"commit":{"parents":[{"commit":"1c48f00b15d4fee1a76035c4dc195ad506af8544","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1c48f00b15d4fee1a76035c4dc195ad506af8544"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-07-04 06:46:11.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/101fec2c2d4d89a88561c23700f49f7d38d7ec3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/101fec2c2d4d89a88561c23700f49f7d38d7ec3c"}]},"branch":"refs/heads/master"},"8d7bdc0013da26c06f971d510d55fe745dfa87d6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-10 07:57:36.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/2"}}},"commit":{"parents":[{"commit":"f0f987383ecd02b5396f939c1e87bd5ff134d64c","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f0f987383ecd02b5396f939c1e87bd5ff134d64c"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-07-10 07:57:30.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8d7bdc0013da26c06f971d510d55fe745dfa87d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8d7bdc0013da26c06f971d510d55fe745dfa87d6"}]},"branch":"refs/heads/master"},"432ed4668edf3e1e3ea05111d6cf551b78cc1e50":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-10 17:51:27.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/3"}}},"commit":{"parents":[{"commit":"54d45be30378d32b55bf5b1840fd1ab93f8a66eb","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/54d45be30378d32b55bf5b1840fd1ab93f8a66eb"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-10 17:51:17.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/432ed4668edf3e1e3ea05111d6cf551b78cc1e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/432ed4668edf3e1e3ea05111d6cf551b78cc1e50"}]},"branch":"refs/heads/master"},"d7caf1daeaa2f4fcf6ce3799e9cb0281f700ec2b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-17 11:13:07.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/4"}}},"commit":{"parents":[{"commit":"ce8972e40c1967f01f3d258634c0022bc4b0911c","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ce8972e40c1967f01f3d258634c0022bc4b0911c"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-17 11:12:48.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d7caf1daeaa2f4fcf6ce3799e9cb0281f700ec2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d7caf1daeaa2f4fcf6ce3799e9cb0281f700ec2b"}]},"branch":"refs/heads/master"},"4e5b1ff2e11c22bf13f4f259069b3b5920730b93":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-08-24 17:39:40.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/5"}}},"commit":{"parents":[{"commit":"9796697a95172b9c83b1e549f476b9ac4506ef69","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9796697a95172b9c83b1e549f476b9ac4506ef69"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-24 17:39:33.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4e5b1ff2e11c22bf13f4f259069b3b5920730b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4e5b1ff2e11c22bf13f4f259069b3b5920730b93"}]},"branch":"refs/heads/master"},"f445e83af2b152ec7decd2d3ef515f458f5d65de":{"kind":"NO_CHANGE","_number":6,"created":"2023-08-28 14:30:16.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/6"}}},"commit":{"parents":[{"commit":"2bad4d7bad0b31ce8691aca79d13f213e2af2eaa","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2bad4d7bad0b31ce8691aca79d13f213e2af2eaa"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-28 14:30:16.000000000","tz":0},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f445e83af2b152ec7decd2d3ef515f458f5d65de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f445e83af2b152ec7decd2d3ef515f458f5d65de"}]},"branch":"refs/heads/master","description":"Rebase"},"78c570773c00dfb5e80a44fd3945af4dfa63979c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-29 07:23:18.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/7"}}},"commit":{"parents":[{"commit":"96aa98afc9f979fc0b4f10468e382aa494299f6e","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/96aa98afc9f979fc0b4f10468e382aa494299f6e"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-29 07:22:13.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/78c570773c00dfb5e80a44fd3945af4dfa63979c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/78c570773c00dfb5e80a44fd3945af4dfa63979c"}]},"branch":"refs/heads/master"},"3abb5932a693933ec231b99d49eefae9161adfbd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-29 07:31:11.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/8"}}},"commit":{"parents":[{"commit":"9f7e659ee9bf41151df94d591e40aefa1ebb9409","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9f7e659ee9bf41151df94d591e40aefa1ebb9409"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-08-29 07:30:34.000000000","tz":180},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3abb5932a693933ec231b99d49eefae9161adfbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3abb5932a693933ec231b99d49eefae9161adfbd"}]},"branch":"refs/heads/master"},"f2574ec01b3b0144dcceefd41208b573844b8bbb":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-09-20 15:05:33.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/76/887576/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/9"}}},"commit":{"parents":[{"commit":"9f7e659ee9bf41151df94d591e40aefa1ebb9409","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9f7e659ee9bf41151df94d591e40aefa1ebb9409"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-09-20 15:05:33.000000000","tz":0},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nThe module \u0027spopenstack\u0027 is required for this connector, but is\ninstalled as needed. Detecting early on if it is missing makes debugging\nfaster.\n\nThe same is true for the \"SP_OURID\" configuration item in\n`_attach.config()`.\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2574ec01b3b0144dcceefd41208b573844b8bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2574ec01b3b0144dcceefd41208b573844b8bbb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b290711cfb23603627e9e28c733fceb5184589b5":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-07-04 03:29:49.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/76/887576/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/76/887576/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/76/887576/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/76/887576/10"}}},"commit":{"parents":[{"commit":"7c2c80c9a54e73bf56a0644e02f0918d29749749","subject":"storpool.py: Use StorPool\u0027s API for Attach/Detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7c2c80c9a54e73bf56a0644e02f0918d29749749"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2023-06-02 14:43:45.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-04 03:23:03.000000000","tz":330},"subject":"StorPool: Raise on spopenstack, SP_OURID issues","message":"StorPool: Raise on spopenstack, SP_OURID issues\n\nThe module \u0027spopenstack\u0027 is required for this connector, but is\ninstalled as needed. Detecting early on if it is missing makes debugging\nfaster.\n\nThe same is true for the \"SP_OURID\" configuration item in\n`_attach.config()`.\n\nChange-Id: I6030d433dcf45f496bfee79aa00b2af9b51fd420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b290711cfb23603627e9e28c733fceb5184589b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b290711cfb23603627e9e28c733fceb5184589b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
