)]}'
{"id":"openstack%2Fos-brick~889093","triplet_id":"openstack%2Fos-brick~master~Ie49b70e69ed07589a9ca155524685d95c460374f","project":"openstack/os-brick","branch":"master","topic":"remove-egg-info","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-11-14 19:38:46.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-11-10 15:30:23.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"Ie49b70e69ed07589a9ca155524685d95c460374f","subject":"Remove egg_info from setup.cfg","status":"MERGED","created":"2023-07-20 12:58:59.000000000","updated":"2023-11-14 19:39:47.000000000","submitted":"2023-11-14 19:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889093-remove-egg-info","meta_rev_id":"961075eabdf1f254e581cccdd44cc63099042ab7","_number":889093,"virtual_id_number":889093,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 19:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"recommended":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2023-11-09 06:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-11-10 15:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-24 12:29:39.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2023-07-25 05:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2023-11-09 09:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-11-10 15:30:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 14:22:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 15:49:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 15:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 16:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-20 16:15:36.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-07-20 16:22:18.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-07-20 16:42:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-20 17:31:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-20 17:48:13.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-07-20 18:37:02.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-07-20 19:49:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-21 18:33:02.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-21 18:43:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-24 12:29:39.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-07-25 05:45:10.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-11-09 06:50:00.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-11-09 09:32:13.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"2814d240bc6b2726a4b48f379de8b517d403a8eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-20 12:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f43630a9334373fcd05d2c69161e75f118c1bae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-20 14:22:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0c279edc465e43729683eaf228e9536e : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1c9040f9ffaa938a7313830d801864bf58d4cbf7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 15:49:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/889093/1/check/DellEMC_PowerFlex_os-brick/731c163/DellEMC_PowerFlex_os-brick/13 : SUCCESS in 1h 35m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"93829fc7304bb0bbd80e9abd93dd8c5e831b8096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 15:50:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca1ce60ccb544c42801aad50d5354263\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0de5b8fae4e44fbdb087ad0b4b15723d : SUCCESS in 2m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30d1208c5c904564bdfe51cbd9c1d449 : FAILURE in 2h 01m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f4f081f76548af9f85c85a363e0ea2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2260e3fda6034cd09b1b7b84c208c0eb : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2441f572a1774b5b99a19c08e2584edb : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f9512b5dac4007bc1cfcfbe16aa92c : FAILURE in 5m 13s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d8d43b1a8d754581a6c5ddc3caf8ff75 : SUCCESS in 4m 27s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/3991be15fad84ea3bd3e5ee5f8df1db9 : SUCCESS in 2h 23m 48s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3924e301f4ca4458bee465e726e1f97d : FAILURE in 2h 47m 10s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e4af207d444245d28172a12aa6af6932 : SUCCESS in 2h 09m 29s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/bfceefab96014869b2688c1ea14209d9 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62b4b438a0945fca1763861627fab248a684b29f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-20 16:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/889093/1/fc : SUCCESS in 0h 33m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/889093/1/iscsi : SUCCESS in 0h 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"77466364c35ebe11b48fec28896692e438490f8a","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-20 16:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/93/889093/1/check/tyds_test/0c51583 : SUCCESS in 3h 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"af27956c4b3adc3fdc1a52f572cdb706c97fb384","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-20 16:22:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/1/20230720215902/NEC-V-ISCSI : SUCCESS in 2h 36m 28s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/1/20230720215902/NEC-V-FC : SUCCESS in 3h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b10bd461383f25cfd99378581a72ea3019710926","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-20 16:42:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/889093/1/check/DellEMC_Unity_FC_os-brick/02d2a70/DellEMC_Unity_FC_os-brick/32 : FAILURE in 2h 14m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/889093/1/check/DellEMC_Unity_iSCSI_os-brick/03f1dcc/DellEMC_Unity_iSCSI_os-brick/32 : FAILURE in 1h 58m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c98fc20d9a195ac1977e6729cf4071d4f5e6ca8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-20 17:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-889093-1/FC/ : SUCCESS in 2h 44m 14s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-889093-1/iSCSI/ : SUCCESS in 2h 17m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"996dbe6f8e7d7b294dedb0fbe40fe0d55ab78757","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-20 17:48:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1626/","accounts_in_message":[],"_revision_number":1},{"id":"45c7704fbb06ac3d426d297ee2eeaee0a1f17fca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-20 17:52:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"595de1b9515aa436370a670e57d07ad9af0a3722","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-20 19:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/553cd194a8594317893a4a523c5a5bc1 : SUCCESS in 1h 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"081542deb47e2c5117791befa3f7153ff813b9f5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 19:49:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/1/20230721004151/NEC-FC : SUCCESS in 3h 39m 52s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/1/20230721004151/NEC-ISCSI : SUCCESS in 4h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4817f903b824195ad2000e5df1f68fe83911d0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 20:21:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45990bcacb4445e7b7ca7841c23ddaa7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/904fe3924f804f97bf55fe1a8b046ac0 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7464f2f4030742a9bf8a43504b0fd05f : SUCCESS in 1h 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da779bfafe3e4c4f9b7b77de63819ec0 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0c61fd80d79436d9c0a946826b67a73 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b04a6ea791b4793ab3bb6623ee5d779 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b820f2ed4bca4bd59b8ee920983a0eb7 : FAILURE in 4m 48s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c99124df509f4f79a69d5c9f5958a46e : SUCCESS in 4m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7d19b328af2c49b1adb4802588e6b349 : SUCCESS in 2h 19m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fc607330af7848c2ae3b9442719beda4 : SUCCESS in 2h 25m 16s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ec1abd211920494395fd0162346e69e2 : SUCCESS in 1h 45m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4f819fe48c4c40d385c1cc277866f207 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa5708a1e938522c21f60ab12e428e5eec0894e0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-21 12:33:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3a4b1b97fe4a8d13bc9f1dfc76d74b923d2a720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-21 12:33:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"495ea6ee7d451285bc068f85ee9347fd69c2b041","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-21 14:10:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b1b34ea05d5141089a12037e204454a2 : SUCCESS in 1h 35m 42s","accounts_in_message":[],"_revision_number":2},{"id":"81a53a451e1d9634896ec8ffc5ec592beca6c466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 14:38:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89c7b638085c494380a88fe0e5b3f6d7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6107bb9b97348cba1803f9b66594966 : SUCCESS in 1m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32bded4da39e483d9c5ce41dae07c068 : SUCCESS in 1h 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0981427814504061a15962db3e6c4f59 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfedad7b4de84300bc92d7a8d54f32ec : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85272f1a105648a99bcaac2b21c783a5 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad4fd4b821e4c20be40f610b455995c : SUCCESS in 5m 01s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/48cdbc70d7874c78870ed4362aef3609 : SUCCESS in 5m 05s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/53a02b48e7df4ccaaaa00b741da5b1ff : SUCCESS in 2h 03m 38s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/741faa91d199440eb314c53b115cdab7 : FAILURE in 1h 55m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b84a04700e154897b1780860e652fcd1 : SUCCESS in 1h 32m 44s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/6a015e8107d540ec9ab0e12010272496 : SUCCESS in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"527c62f79e7424e91546b37a4b590c85ced572af","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-21 15:11:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"963fe5c3b7a10f8c13c0989876028449f6b47192","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-21 16:27:03.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-889093-2/FC/ : SUCCESS in 3h 40m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-889093-2/iSCSI/ : SUCCESS in 3h 5m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0b2a7326a0c853650e98ad0fb16989fbfb652f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 17:10:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0782763fea84b3ba6eafb781992d3d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c9cebfb045974412bfead7df0866162b : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3117655e7dd94ebd822c2a835eb76ca2 : SUCCESS in 1h 32m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205a7a63854e4be3947cb47570d56f6a : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c33a490a8d246cca70e86b4449057b6 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f80e773179f84e40ae7f479976d0fadd : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9990353e564c45d3a1ee53ff85939eed : SUCCESS in 6m 36s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/44bd6b359d1b48a49da73190d0d2395a : SUCCESS in 4m 03s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/3a0ee1c672314284aa6bf73da54edac1 : SUCCESS in 1h 57m 18s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55ba336fdb4e45f3b5e9d89f72b10331 : SUCCESS in 1h 57m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7176cbb1e1064261a6631dc4f8479318 : SUCCESS in 1h 38m 35s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4c38289363894e9897549f41d96b8021 : SUCCESS in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c246b7d556cb0eb90c4fa25d60472e98c3964655","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-21 17:25:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_PowerFlex_os-brick/7d80bce/DellEMC_PowerFlex_os-brick/16 : FAILURE in 2h 08m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7c73b713c09e2b865f3439b82592760d6dce54d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-21 17:30:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fb0651db42a34f3aa0e137ed74ddb23d : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":2},{"id":"beaa39e2a3cfe5101c60b9dac296ee96c17c6914","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-21 18:10:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_Unity_FC_os-brick/294302a/DellEMC_Unity_FC_os-brick/35 : FAILURE in 2h 01m 03s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_Unity_iSCSI_os-brick/519e489/DellEMC_Unity_iSCSI_os-brick/35 : FAILURE in 1h 55m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"bda8471a8ffe1e3a68bdfd2b70f17a041d35c285","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-21 18:11:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1631/","accounts_in_message":[],"_revision_number":2},{"id":"c415ec8593449a39fec10e7f6db5816514a327ce","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-21 18:33:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_PowerStore_FC_os-brick/c16bda8/DellEMC_PowerStore_FC_os-brick/23 : FAILURE in 1h 22m 17s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_PowerStore_iSCSI_os-brick/5632410/DellEMC_PowerStore_iSCSI_os-brick/23 : FAILURE in 30m 21s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_PowerStore_NFS_os-brick/e67e3bf/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/93/889093/2/check/DellEMC_PowerStore_NVMeTCP_os-brick/0fc85ea/DellEMC_PowerStore_NVMeTCP_os-brick/23 : FAILURE in 1h 14m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"3967bb85b631f5828c2ac16f7cc2d861dcdd3f05","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-21 18:43:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-fcp/0605937/ : SUCCESS in 1h 02m 34s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/d5f069e/ : SUCCESS in 45m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-nfs/7fc3ada/ : SUCCESS in 46m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1bfa374/ : FAILURE in 1h 31m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f92c0be6b7f3be640f44f558f9e87cd4cefc2a04","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-22 00:00:30.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/2/20230722014641/NEC-V-FC : SUCCESS in 6h 44m 33s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/2/20230722014641/NEC-V-ISCSI : FAILURE in 7h 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5c48a5395ffbfef475dfcb88433e57c5480fffab","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-07-22 00:45:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/93/889093/2/check/tyds_test/df2050a : SUCCESS in 7h 00m 34s","accounts_in_message":[],"_revision_number":2},{"id":"29d64c7b1e637b1699947a3b5151430f120271ef","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-22 05:17:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/2/20230722095957/NEC-FC : SUCCESS in 4h 17m 35s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/889093/2/20230722095957/NEC-ISCSI : SUCCESS in 4h 17m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a61207c2463dcb0b1af0348df9e6b54d3e080c73","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-07-24 12:29:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4479821292b4bcddc400cb5e50f1f5883abee710","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-25 04:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/889093/2/fc : SUCCESS in 0h 39m 52s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/889093/2/iscsi : SUCCESS in 0h 46m 30s","accounts_in_message":[],"_revision_number":2},{"id":"87faf1a8de4d7f59162f62c276e174b16dd3425a","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-07-25 05:45:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a8e36cf94e84d1c6b487451b13646d6a5fa0bb01","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-09 06:50:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6ab58197b7f646eaabad6abf1b0d7eaed73a9ad","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-11-09 09:32:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ecfa76c64bffacd7efd2c5659c03cc3a56ecaef","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-10 15:30:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"40f64fb4c5232e10e9f1a1794c07ac515ae3b3fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 15:30:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80fd388331e1fa6e5e534a74475ce26523542671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:36:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db5d94d253a941258fd76b07c827015a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/325afa69bcc3433ea849ff7aba0f2053 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97d654b99c494a568877e6c0cf0e5b3e : SUCCESS in 1h 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b812ba7a2dd4892a55f76c244552f10 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d5f3a7cc6845ed9813fe592dcf3d8b : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a312a885c7d24bdf8073ac8c159cc830 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde2fc93fb9c415e896d234ba1d310eb : SUCCESS in 5m 54s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5681a60b955d4eb9ba5a73c6fcbb7f26 : FAILURE in 2h 00m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b21a728b880c679e97ace7a99d01f115e9d66e26","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-14 15:22:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81f9cbd28733ec62dc2a96ad971c2450bac5a0de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-14 15:41:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ed4abb462dc44d63a831acd3f29a50b1 : RETRY_LIMIT in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5d28680e6cd60f45c72cbe2286ac1d1c03ad9d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:29:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54de6d18a77249cfadfd4e7eb6d30aea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b67b317f30a42c787acb0fe1f993694 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f10ae1f88d3f4518948665e24add5c62 : SUCCESS in 1h 46m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b2be2496d64f20b04aa2f1700274ce : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f042b98f2d8d489dab11e0ca6b9ed906 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0167093598de4b60bd227878103ebdf9 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f15ffdf942b437c9f1a084022d3fc00 : SUCCESS in 5m 44s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/dd080f52186b480b9609874a1f838928 : SUCCESS in 4m 10s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2014b9dda3b74b8a9c3917539742e538 : SUCCESS in 2h 05m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bbdf844e6fe04458ae2a41efc99e36b7 : SUCCESS in 1h 49m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d516b4caf49c4f779cacfe21b9e625b9 : SUCCESS in 1h 26m 29s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f362ad0788b4446cb2c0488fec35505b : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf6c668653a6cbeb4cd1a4c474918ec6dbfd2b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 17:30:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09c600617074d17c9cff034ff96604499b82eb5a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-14 18:40:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-fcp/987ca91/ : SUCCESS in 1h 10m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/59cee7a/ : RETRY_LIMIT in 8m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-nfs/9b778d5/ : RETRY_LIMIT in 7m 50s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/889093/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ea9f1bf/ : RETRY_LIMIT in 7m 53s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8512936459474a5f8e4a21fa1f0379956a745311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:38:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/207cd32ecc5247c9971c114cfb3e7e4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dffe52d80ccf49e98e4d4d58e14fa33d : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc7aea20bfd64a4a86cd2d42ae332b6d : SUCCESS in 1h 24m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327ee6b059294147bc7da7f2f617945f : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9db30578dbf4639b2db222d7f3af618 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd72c6e1dcb64583891a30560a81dd23 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf21a87097745408094c639c93f0848 : SUCCESS in 5m 49s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c23ba51169444dc3b7aff34ae187fd09 : SUCCESS in 1h 59m 39s","accounts_in_message":[],"_revision_number":2},{"id":"df63ced1d0723a36a9cdba556241ab387aab2529","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:38:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"961075eabdf1f254e581cccdd44cc63099042ab7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:39:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50cbd2f00b004be8b7002fcff3565c68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f3dd27cc7c45c7848fd0e85a704825 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d38bcebe69d02a78a474f5411647e5ce8c155bcf","revisions":{"512b3e264839c1c5f8f0fd78923a3b3951e66814":{"kind":"REWORK","_number":1,"created":"2023-07-20 12:58:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/93/889093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/889093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/889093/1"}}},"commit":{"parents":[{"commit":"6d70ace1bbc8cd94280ac7efc9d33066ff7bbf83","subject":"Merge \"linuxrbd: Remove rados_connect_timeout parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6d70ace1bbc8cd94280ac7efc9d33066ff7bbf83"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 12:54:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 12:54:07.000000000","tz":-240},"subject":"Remove egg_info from setup.cfg","message":"Remove egg_info from setup.cfg\n\nThis was removed from cinder by change If7db350ea55a that removed\npy27 support, but apparently we missed it in os-brick.\n\nChange-Id: Ie49b70e69ed07589a9ca155524685d95c460374f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/512b3e264839c1c5f8f0fd78923a3b3951e66814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/512b3e264839c1c5f8f0fd78923a3b3951e66814"}]},"branch":"refs/heads/master"},"d38bcebe69d02a78a474f5411647e5ce8c155bcf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-21 12:33:32.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/93/889093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/889093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/889093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/889093/2"}}},"commit":{"parents":[{"commit":"c187a33b7b15b1c17d4d007fbf1b62b76a1c575c","subject":"[docs] add autodoc_mock_imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c187a33b7b15b1c17d4d007fbf1b62b76a1c575c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-20 12:54:07.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-21 12:33:32.000000000","tz":0},"subject":"Remove egg_info from setup.cfg","message":"Remove egg_info from setup.cfg\n\nThis was removed from cinder by change If7db350ea55a that removed\npy27 support, but apparently we missed it in os-brick.\n\nChange-Id: Ie49b70e69ed07589a9ca155524685d95c460374f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d38bcebe69d02a78a474f5411647e5ce8c155bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d38bcebe69d02a78a474f5411647e5ce8c155bcf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
