)]}'
{"id":"openstack%2Fos-brick~890159","triplet_id":"openstack%2Fos-brick~master~I5fc02efc5e9657821a1335f1c1ac5fe036e9329a","project":"openstack/os-brick","branch":"master","topic":"bug/1609753","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-08-25 16:32:11.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-08-25 16:32:11.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-08-24 12:11:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9535\u003e replied on the change","reason_account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}}},"hashtags":[],"change_id":"I5fc02efc5e9657821a1335f1c1ac5fe036e9329a","subject":"Check paths before extending device","status":"MERGED","created":"2023-07-31 19:37:17.000000000","updated":"2023-09-07 05:17:00.000000000","submitted":"2023-08-25 16:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890159-bug/1609753","meta_rev_id":"ff200d8184ebde49bfb7546d3571e568fbf643c8","_number":890159,"virtual_id_number":890159,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 16:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-24 12:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-25 08:53:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 21:18:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-31 21:21:52.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-31 21:29:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-31 21:44:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-31 21:58:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 22:25:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-07-31 23:18:33.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-31 23:37:52.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-01 04:28:54.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-08-01 05:04:48.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-01 06:22:17.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-08-01 07:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-08-01 12:31:26.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"CC"},{"updated":"2023-08-23 10:55:42.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2023-08-23 16:00:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-08-24 14:30:27.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"CC"},{"updated":"2023-08-25 12:07:08.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2023-09-07 05:17:00.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"}],"messages":[{"id":"dfd6a68916ddff76d9415652582593389222b581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-07-31 19:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"045951cf80372f48f985042fc9005e146b72540d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-31 21:18:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bcadbc05e9684b6a933d684e203d9552 : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7b92a681ebd089ec2d88cc21603202a1cf3f1dab","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-31 21:21:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy pure-devstack-cinder-tempest-iscsi-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/59/890159/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e9fb8ad/ : SUCCESS in 1h 43m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3836266c5f2a8344bd451ab7538f0a4a3c7a1174","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-31 21:29:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerFlex_os-brick/378b5e9/DellEMC_PowerFlex_os-brick/24 : FAILURE in 1h 43m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e835fae62fde1b20f659721581609a3a381dbc81","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-31 21:44:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-1/FC/ : FAILURE in 2h 6m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-1/iSCSI/ : FAILURE in 1h 19m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"26db2d9ec9c2a665271ed3c1d6e93dfe07d03710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 21:58:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/159b13cba92e48ad8c4cab6c6bd449b5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf4b1a6c21534f33b6fd5196668b3529 : SUCCESS in 1h 04m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cc9a91f0c24f73a750bad1bd3d57fa : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fdf1ae4cb1340ee95f547dabf4f65f2 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/596ca753b62f466fa198b0728d58c1dc : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8490ab9aa646db9c12f9f9d0b3b788 : SUCCESS in 6m 09s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/cb7f0f58f32f4427a83a748db78be444 : SUCCESS in 5m 19s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/35a81dd268a04aec821a85ee221f347e : FAILURE in 2h 09m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1e41536f32134bfa8d014e7b43526088 : SUCCESS in 2h 15m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/94cbfead655c48069c67737175d30bb1 : SUCCESS in 50m 28s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/aaf425d6005745fa93f8a8164b46432b : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c15e7d6f54741bb1edf4d6bb48f265e7bbe6e527","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-31 22:25:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_Unity_FC_os-brick/a534735/DellEMC_Unity_FC_os-brick/41 : FAILURE in 2h 11m 01s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_Unity_iSCSI_os-brick/b8a6288/DellEMC_Unity_iSCSI_os-brick/41 : FAILURE in 1h 59m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2bfa7580d854fa2b4f9e6baebd91d64459cd9a48","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-31 23:18:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerStore_FC_os-brick/c860113/DellEMC_PowerStore_FC_os-brick/27 : FAILURE in 1h 19m 36s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerStore_iSCSI_os-brick/2243af1/DellEMC_PowerStore_iSCSI_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerStore_NFS_os-brick/90340af/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/4748b17/DellEMC_PowerStore_NVMeTCP_os-brick/27 : FAILURE in 1h 14m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"78205fad1461ca2373042634c23203dab339b3c0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-31 23:37:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/1/upstream-check/cinder-tempest-plugin-ontap-fcp/e637313/ : SUCCESS in 1h 04m 12s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/739a4a7/ : SUCCESS in 1h 31m 37s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/1/upstream-check/cinder-tempest-plugin-ontap-nfs/4df97ee/ : SUCCESS in 1h 37m 06s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4f12eb5/ : FAILURE in 1h 32m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ccd0d54410b90a10e1134fe03cef15f6bf6aec33","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-31 23:42:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/1/check/DellEMC_PowerFlex_os-brick/e8db584/DellEMC_PowerFlex_os-brick/25 : FAILURE in 1h 43m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"fff8671c5322013df10f3f6ec31fa869bae38c78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-01 04:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1667/","accounts_in_message":[],"_revision_number":1},{"id":"9e0647d423e14c4de5e65b523ac94566ac81d242","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-01 05:04:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/1/20230801100558/NEC-FC : SUCCESS in 3h 56m 39s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/1/20230801100558/NEC-ISCSI : SUCCESS in 3h 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d890d47efc39e6bd82910521242465047b0c6aa8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-01 06:22:17.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/1/20230801093606/NEC-V-FC : SUCCESS in 2h 53m 57s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/1/20230801093606/NEC-V-ISCSI : FAILURE in 5h 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7872439fa97fb44651199762c13955ed73b577c4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-01 07:00:08.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/1/fc : SUCCESS in 0h 29m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/1/iscsi : SUCCESS in 0h 44m 07s","accounts_in_message":[],"_revision_number":1},{"id":"356eb6eed2ce0ccaf837259930d1f3db944c28f9","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-08-01 12:31:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07a1e872da162d72274764e880188bb24ecf9c35","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-01 15:38:08.000000000","message":"Patch Set 1:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/435/","accounts_in_message":[],"_revision_number":1},{"id":"b35ecc63e07f529e6e1369c769f2c1c7cf565ca3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-01 17:44:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1670/","accounts_in_message":[],"_revision_number":1},{"id":"8ce71667e20eaa759a315b1bb5d42b432a6c7d95","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-02 01:28:51.000000000","message":"Patch Set 1:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/441/","accounts_in_message":[],"_revision_number":1},{"id":"169ff668a0f2758b48473adee14a5b1e16625121","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-08-02 13:59:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/446/","accounts_in_message":[],"_revision_number":1},{"id":"26ab9aaaffdcf6e88d129df1e5a4f83eaebbf4b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-22 09:13:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c7a53ee9d873446909197b60ea46161b199fde2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-22 09:39:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"808f258028ec8bc95ac51c5d20c0029f758bb5ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-22 10:08:28.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_Unity_FC_os-brick/6cb54ee/DellEMC_Unity_FC_os-brick/58 : FAILURE in 28m 13s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_Unity_iSCSI_os-brick/bb86192/DellEMC_Unity_iSCSI_os-brick/58 : FAILURE in 14m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4dcf28ad3acd8976efddf9d8bf56952c494aaeaa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-22 11:40:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/59/890159/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/233a7ad/ : SUCCESS in 1h 44m 44s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/59/890159/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/ed1ff59/ : SUCCESS in 1h 58m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/59/890159/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/edcc199/ : SUCCESS in 1h 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b10f902cde56c4e7ac46b13998826465ead894b4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-22 11:46:22.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-2/FC/ : SUCCESS in 2h 32m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-2/iSCSI/ : SUCCESS in 2h 31m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"362e7226e310299703e042b1ebe217747d8343b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:11:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b51a39d212f4b63be57df6bec76a849\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9366550df2664fe5b26859b98af7ca9f : SUCCESS in 1h 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c7d8004e8d4037ae1521ad094ddb73 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbe0b6e819544ee9996c9694f542e4a4 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e61f5f2c6d144be1aaed059757f12947 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9abfb9c35a4ea48ee30bc77b7503d9 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10cda40be2ac4058ab8d2a73565f6afd : SUCCESS in 4m 08s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ee47bbe2cfcf45e3aef9bd10e417588b : SUCCESS in 5m 09s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/9d023d7105f74b259e5eb8e1f617d21c : SUCCESS in 2h 30m 17s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/73ddd2a93b394bb980ffaae69584b20d : SUCCESS in 1h 46m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1e9392af7f354e93a4451813f17c0182 : SUCCESS in 2h 00m 09s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/63a5220fc7ce41a0a09e4bbc9e36d644 : SUCCESS in 1h 15m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/dd76f049c07447c1b8f63a37bde7797b : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f83e321e801db59a80c3fe625b651fcfa642ef95","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-22 12:27:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_PowerStore_FC_os-brick/4e935d1/DellEMC_PowerStore_FC_os-brick/34 : FAILURE in 15m 37s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_PowerStore_iSCSI_os-brick/8beb493/DellEMC_PowerStore_iSCSI_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_PowerStore_NFS_os-brick/241b521/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_PowerStore_NVMeTCP_os-brick/6c8967e/DellEMC_PowerStore_NVMeTCP_os-brick/34 : FAILURE in 13m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"06336265764541ea4b06ec7f02b94ee606638172","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-22 12:40:17.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_Unity_FC_os-brick/5303194/DellEMC_Unity_FC_os-brick/59 : FAILURE in 28m 51s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/3/check/DellEMC_Unity_iSCSI_os-brick/1bb8eab/DellEMC_Unity_iSCSI_os-brick/59 : FAILURE in 14m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c5f2a8b8c53aca6b48778b3afce7e4b2d71f8897","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-22 13:35:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/3/upstream-check/cinder-tempest-plugin-ontap-fcp/47faef4/ : FAILURE in 1h 07m 28s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/1c20c8d/ : SUCCESS in 45m 40s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/3/upstream-check/cinder-tempest-plugin-ontap-nfs/ba6bef9/ : SUCCESS in 1h 22m 08s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4082540/ : FAILURE in 1h 23m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"17b3fce0909782502d138751efceef83fa438fac","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-22 16:27:39.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-3/FC/ : FAILURE in 6h 47m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-3/iSCSI/ : SUCCESS in 1h 58m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"932f84efa6f2843bfbb123450de772591d2996de","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-22 18:45:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8d0b0b65f1034aa1abb60ec4ee403c60 : FAILURE in 1h 51m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ddab501cdfe5f5f9596531d8e44ef597e9463678","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-23 02:12:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/3/20230823080400/NEC-V-FC : SUCCESS in 2h 55m 43s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/3/20230823080400/NEC-V-ISCSI : SUCCESS in 3h 08m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d9d60ec9d1dea5a2d4afd58e74f86b7603138f3a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-08-23 10:55:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d56ea55bf976cdd0185601f834a6e269c3c3a35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-23 11:27:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"df0c2a6a46098f49a25793da1ebf33471ade286c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-23 11:27:47.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e7bafaf66ebcdee96e976e6ad6f11fea65aed392","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-08-23 11:29:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2badcd0ef98eef472f475d0acaf3969fc9999f4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-23 12:47:10.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_FC_os-brick/0230edf/DellEMC_Unity_FC_os-brick/62 : FAILURE in 28m 17s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_iSCSI_os-brick/b278b39/DellEMC_Unity_iSCSI_os-brick/62 : FAILURE in 14m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"aacbe7f5500d767aa78d8d23ba56daf73a0bda45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 13:39:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/167be049850c4d33ac3f0f2c4a78d60b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5df6be662dad413f8ce55568ec25b9f5 : SUCCESS in 1h 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029731ace8a64b8aa821039b70706ae0 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ce878103d4459290f2a5eccaf44df6 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b716551fb6d14461bd7fda27c2fa9c12 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd5a4abbfccc4bdab25a357d2c685a5d : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4d6432139d4a69b9ffd9df74dea50b : SUCCESS in 4m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3fd6f5e97df1417890725d5e91010e52 : SUCCESS in 4m 29s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/ab28242fdfa94417935bb5b85d3606a3 : SUCCESS in 2h 01m 17s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e9da89559c054627bbe82b7eb624eabf : SUCCESS in 2h 06m 36s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a5e3a8111b340d3ba0c30f1cb9457f0 : FAILURE in 1h 29m 09s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/df903f4360534f2793d8927e32e0f4fd : SUCCESS in 1h 19m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/beb01dca51bd4593a2557580d697ca62 : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68a54bff4d6feac9d7c064a8e6401cefc40757f4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-23 15:18:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/59/890159/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/843111f/ : FAILURE in 2h 17m 59s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/59/890159/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/734f05d/ : SUCCESS in 2h 32m 40s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/59/890159/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/0110161/ : SUCCESS in 2h 30m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ab633efcdf49d2123f28510bef84463ebddcc5bd","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-23 15:39:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88f9470d6aeb238b4734c2544d4f256f4a422b81","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-23 16:00:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1d8d7151eac0df4a8061dbc5c208e6e74bf12bd2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-23 17:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/4/fc : SUCCESS in 0h 40m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/4/iscsi : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3a2a7654576a2e66e60f4c561534dc1744d823e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 17:55:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6174c6e522496bb5331963fcf3ddd6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c32d466e01534b1687192f88f2355c2a : SUCCESS in 1h 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2011ed2300418aab08de8694ef2ba8 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c92fbf9926b0413e9da6a76d0b8bb3b1 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcc44faf78b243ee84027a02d32a3649 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1f33768b7141f680d4158cae6b99db : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b164fe63b1854e259e3aceaddb06b075 : SUCCESS in 4m 25s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/507b48cefc6b407ab827f9530eb1b194 : SUCCESS in 3m 54s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/870b2b80ab1c485ca813522c9c19aea0 : SUCCESS in 2h 15m 51s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/da217e8e8c7b4eb8aaa3da20f8af18db : SUCCESS in 2h 00m 01s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32dcd5f9a91749a897f168fedcfa907f : SUCCESS in 1h 51m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/bb002ec7497f489d959af1a9db3bfae2 : SUCCESS in 1h 24m 04s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/3b463868c30b40f286da8416aad71631 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4405981e055e6722c6f744aab21559033c47f0d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 17:56:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e422169235842205f72eb75a978c2935384334f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-23 19:17:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_FC_os-brick/6fa66df/DellEMC_Unity_FC_os-brick/66 : FAILURE in 48m 34s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_iSCSI_os-brick/4c79465/DellEMC_Unity_iSCSI_os-brick/66 : FAILURE in 14m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf2f267f67884623cfc619b4bb4f05d6a5ce31e8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-23 19:34:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-fcp/94aa1a8/ : FAILURE in 1h 37m 16s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/3828642/ : SUCCESS in 46m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-nfs/6979dd8/ : SUCCESS in 50m 30s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/33c053c/ : SUCCESS in 43m 10s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"55e207238f94b2e7d79d7a943699cff41c7cd437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 19:36:42.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a51723e8c5844afaea8eb648093f66e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d1af79616754650a9abf39fb903fde7 : SUCCESS in 1h 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2ea9f0710e445abe2ea79cd51e28a8 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1a5f7a61abf421f8dcbb954c707d1aa : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc3743a06659452da3fff3778f78a743 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93a65af3da040d581734dea54404783 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b86fc290fc994f76ab67112ed20c3006 : SUCCESS in 5m 08s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9a3aced4e7f8438497ea276bc55427cb : FAILURE in 1h 20m 30s","accounts_in_message":[],"_revision_number":4},{"id":"836231c9d06b2e9ed16cf90421f29317ffa0e4ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-23 19:42:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c509d0f8195e664f4e2caf431b8b0635897d01cf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-23 20:21:10.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerFlex_os-brick/da73b09/DellEMC_PowerFlex_os-brick/31 : SUCCESS in 1h 38m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"79148fc1139d9cbeb1b3dad88d615eacaec4535a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-23 20:29:07.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-4/FC/ : FAILURE in 8h 48m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-4/iSCSI/ : SUCCESS in 3h 58m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"9a3b0c00eabb69e71fad5ef4e612e31321ec5d21","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-23 21:16:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/df6a0d63b6fb4742a435374072b9ef5a : FAILURE in 1h 23m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1fd552dda12bdc40dfbe048a4ba3bc6868d030d2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-23 21:21:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_FC_os-brick/37e6c04/DellEMC_PowerStore_FC_os-brick/36 : FAILURE in 1h 28m 35s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_iSCSI_os-brick/2dc9a3a/DellEMC_PowerStore_iSCSI_os-brick/25 : FAILURE in 1h 10m 27s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_NFS_os-brick/61965c8/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_NVMeTCP_os-brick/4b61a8c/DellEMC_PowerStore_NVMeTCP_os-brick/36 : FAILURE in 1h 13m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"ff0a30eedf02215f21c916de089680af5d65d44e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 21:54:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f251166842a44dcf9fbed55b8bf080f4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e42e044fd234f9aaca2bca150872fb9 : SUCCESS in 1h 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d16a26e273df47d1bce8bbb150374d1d : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38446101730045609c3f214393243540 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b10ec2294c274af39161c7188746f24e : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156a678ba119489a856a1bb039f17555 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23cc337dd2e446eca517e6ecbb22746c : SUCCESS in 4m 07s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/5f4696269eb44836a8b57a1936a38818 : SUCCESS in 4m 09s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/5b622d926a5f4c38b74f00d898bc6acf : SUCCESS in 1h 52m 56s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/98b28d3e4687495a8816c398e67cecc1 : SUCCESS in 1h 54m 43s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01f96d373c4b4a75a6e1a2593369abf8 : FAILURE in 2h 02m 56s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8c08e0db74b44995aa2a4a8f5958f5e0 : SUCCESS in 1h 28m 55s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/0336fdaad4124656b0cb8dd03955d271 : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ec8574b3b8830bac7693543f889859334a76e053","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-23 22:27:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"13b592bac5e0e893115c72de2df8f2fbba073c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 00:25:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4260856de63644588c1be53f9eeaf6de\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cfd839f13be4dd4ab10cf411861b3d9 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70985f3717c4ebfae6e3d81b8fd1f71 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a6e048dc784431c85f28056bc3613ec : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/059e66705ac247f8a25e9d2edb1366e4 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b92a786e0dc4f989ae58d1243766965 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03354afb400c47219718b5653b2d680a : SUCCESS in 3m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9380b5d5e0b34ca78c46b4e81394b0f5 : SUCCESS in 3m 46s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/696e4ea3c20e48daa9f709aed9aa1000 : SUCCESS in 1h 49m 08s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/c96151839a634172a38533c43c165056 : SUCCESS in 1h 57m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d8d2f5870bd4c69983db8c253087a7e : FAILURE in 1h 49m 56s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4561fef65798409b8fb060acdcabb7a0 : SUCCESS in 1h 10m 24s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/9fdd7fe810f34b8baad79224efde9f0e : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"43d03560dd14042ba6618cabdcc88e14cd8199b0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-24 01:59:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f92eadbbf696d132a8ed653345835352072c9047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 04:27:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a86ce8eebfaa4625a700468f47e47399\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a441decb852422fb906e1923423081a : SUCCESS in 1h 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6831e2dcc7d648dda0aedcca775d0f67 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ce6a30f63a4aabb82f4e8a8af853cd : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1a3c9c9bca64ba7a705e0773927ee8b : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3001c3cb04a24dc58908a5f351b47fc9 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb15dfc931d8465e9a3b5f3b20d66619 : SUCCESS in 4m 22s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b5ead4d9ebd54aea97da8b1ce38c084f : SUCCESS in 5m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/9b07fae5fa8c4c2f86287d6cd2033afc : SUCCESS in 2h 17m 02s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/5af68572c1884e23a3b586a4ae236419 : SUCCESS in 1h 56m 35s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6ccec56bcd504a808847b327db85de5d : SUCCESS in 2h 00m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e9badc70f6904687bb1bc1754d1ba7c0 : SUCCESS in 1h 23m 01s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1542e305d38847e29969a6230e5d9562 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef4805c935458238841a6564426c7a4271ab79e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 04:28:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d674092ed048b8c63219c10d7cc467325bdcf909","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-24 05:09:50.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_FC_os-brick/2f57539/DellEMC_Unity_FC_os-brick/69 : FAILURE in 41m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_Unity_iSCSI_os-brick/06a5892/DellEMC_Unity_iSCSI_os-brick/69 : FAILURE in 15m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c1cf1a1e1e7db4d471a5cefa868599740a612b4c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-24 05:52:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_FC_os-brick/db579c2/DellEMC_PowerStore_FC_os-brick/39 : FAILURE in 1h 24m 47s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_iSCSI_os-brick/559177f/DellEMC_PowerStore_iSCSI_os-brick/28 : FAILURE in 1h 12m 10s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_NFS_os-brick/e21df6c/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerStore_NVMeTCP_os-brick/965825a/DellEMC_PowerStore_NVMeTCP_os-brick/39 : FAILURE in 1h 20m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"7b67d9a97966f1f54ffff5af11a936d02cc74818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 06:40:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fe598f4ba6d42038403c658b8d751de\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0069319e7204371a10958e1ab66392c : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccb96234b1a4ba6b51521cc07a9673d : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed584db2eb194a4896819521e079ca9b : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a48ea5d1bce4eb78585120fbdc929b2 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2e654a236d4dfb93d958422f711d05 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f72fe7b93e24dbf84fa44805dbf0bef : SUCCESS in 4m 03s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cba978bce79b46d78d66247d56b325bc : FAILURE in 2h 07m 35s","accounts_in_message":[],"_revision_number":4},{"id":"bdc4612664b0a0cb1b6916111f098b3320411c90","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-24 07:38:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7d5e0b57b413466a86ee10839e17a60f : FAILURE in 1h 25m 40s","accounts_in_message":[],"_revision_number":4},{"id":"a4efeb76b6e68247ba3478a6b9fc6d23b5f1e536","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-24 07:39:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-fcp/baaa1ac/ : SUCCESS in 1h 09m 58s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/5dfeca0/ : FAILURE in 56m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-nfs/e1dd8f2/ : SUCCESS in 1h 14m 52s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/19ac219/ : FAILURE in 53m 53s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e39c5a8a424cb50e58f8b71117d21925df3baa10","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-24 08:03:08.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/4/check/DellEMC_PowerFlex_os-brick/e4a194d/DellEMC_PowerFlex_os-brick/36 : SUCCESS in 1h 31m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"14dc46cf565f78de11c79578664d581be1170fba","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-24 10:04:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/4/20230824122152/NEC-V-ISCSI : SUCCESS in 2h 59m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/4/20230824122152/NEC-V-FC : SUCCESS in 6h 42m 39s","accounts_in_message":[],"_revision_number":4},{"id":"76372e8896a347281e7fc418900b21b40652ea89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-24 11:40:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"40cc2eed95ba8980a4d90692f6e3bcc58fef619c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-08-24 12:11:12.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"11460cf1c008226a0937958fa46cd014db5a8b87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-24 12:16:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_FC_os-brick/b8d6831/DellEMC_Unity_FC_os-brick/71 : FAILURE in 35m 43s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_iSCSI_os-brick/8f5c6e9/DellEMC_Unity_iSCSI_os-brick/71 : FAILURE in 14m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"bbf6804adde5db416e28303a80e78a740c838fa1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-24 13:45:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b269bf0/ : SUCCESS in 1h 59m 06s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/92772b4/ : SUCCESS in 1h 54m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/c509e19/ : FAILURE in 1h 56m 49s","accounts_in_message":[],"_revision_number":5},{"id":"e60fd57b2ed6c6aab75f9cb048de00c824ca39ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 13:59:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8402111fe6d4b1aa7c7336d5260955c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13572672ea1a45e986c6eab09dc16ac0 : SUCCESS in 1h 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1349045516fe4ce6bd2c1f1963ce8e83 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89931266db8a40a1bdfe4e15e150eda5 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df6d27ec6a1d4ff480fd9efa77d1994d : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3275686c8aa44574966ec880c5e45a08 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed3a114b57e9420a81f8914ee55bcb63 : SUCCESS in 5m 26s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/88aedc60014246d7bbd7569e79c397d0 : SUCCESS in 3m 40s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/f00badcc4c0e4b449846561da5dcfc99 : FAILURE in 35m 11s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2416f01e1006489d8ce495ca352ff76e : SUCCESS in 2h 10m 11s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d984e5c1be1432cb93844e8fd6a66cc : SUCCESS in 1h 39m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7ab141f8b88d4b0aaca5663570563d9d : SUCCESS in 1h 03m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/182c48ae192c4f669231c523938f120e : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c0b1c356879e320bf18eb18a27449541741246a7","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2023-08-24 14:30:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d943060d97eeed5e55e79b2170b973e59a5afe6b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-24 14:52:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerFlex_os-brick/4b191e3/DellEMC_PowerFlex_os-brick/38 : FAILURE in 1h 48m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"01c4156bad83a0e8b9b9f90794f29535d5d8407c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-08-24 15:00:06.000000000","message":"Patch Set 5:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/5/fc : SUCCESS in 0h 44m 53s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/59/890159/5/iscsi : SUCCESS in 0h 51m 48s","accounts_in_message":[],"_revision_number":5},{"id":"c4a0f732c12af24c762a0f4546c2567acb6ad516","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-24 16:05:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b18ebb571ff34156baf5eccfbf2020a8 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":5},{"id":"c9b431bbb59403ef0690d36756767ee62f8fe7f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-24 16:29:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/f9cc0ef/ : SUCCESS in 1h 48m 24s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/10db0e2/ : FAILURE in 1h 56m 32s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/59/890159/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/d19cc28/ : SUCCESS in 1h 55m 33s","accounts_in_message":[],"_revision_number":5},{"id":"191e56adc0a7ffffeb5d9f9bf747fa089535bf0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:45:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/029d47922388427187fa2b38dfdd67b6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9475e677ac89456ba76dc25164c04c48 : SUCCESS in 1h 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d707f763fcaa4aa4bfa35e39654bb554 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/239d8a990ce04117829404d7569bfc1a : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f24eb56d3d61400caf49289145a0c608 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d74ba31312432f8825a607a89a0bd8 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61caf37aea174e3fad6920b7bd1422a0 : SUCCESS in 4m 21s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/421b4c8186c54ce1b48cd0e6d689ac56 : SUCCESS in 4m 07s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/f489992fc9754c54a730037120e2be34 : SUCCESS in 2h 07m 12s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/d756c395279b45009365ea84f59ed7eb : SUCCESS in 2h 09m 19s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d3e6bcb4b8b479c9f5296b373d94a0a : FAILURE in 1h 51m 40s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c5e3879d234c4d8ea59db76fd0288081 : SUCCESS in 1h 19m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/2c7237eed0854c07a4bccc34043dc7d2 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"df4cf3c8411d9528a37d3f0413117b869c16f411","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-24 17:36:36.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-5/FC/ : SUCCESS in 5h 37m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-890159-5/iSCSI/ : FAILURE in 5h 55m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"900e1150057c21842f613ba7dd7405e6ed0df652","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-08-24 18:01:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"335cc0752ae9a75cc176ab035413bcc460d97479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 20:14:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc56d7f30e5b463f9a66451bf16483c0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7a08c692ece419f9253e7419f9d8aec : SUCCESS in 1h 32m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2005f64c401a4db7af806ccead80df94 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e0df8d82e34f01b4183679c44537db : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fcf84bb8376427ebfeafa068567cb7d : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734f895469fb429ba0510865e851da90 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f49b4bbaa8d47ba8ab6e8296e39d7c1 : SUCCESS in 4m 08s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b16a0b62abba4a4c9d050565da9cd460 : SUCCESS in 4m 24s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/6c2fe73731c94a42b08933501634ee1e : SUCCESS in 1h 34m 48s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/489d68cf28fc47dc86617f69f44dde23 : SUCCESS in 2h 07m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bdd7cc8f15564dc286792037a79cb9f2 : SUCCESS in 1h 46m 34s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b9455d547553401ba35c78044a308589 : SUCCESS in 1h 25m 47s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/9562fb9edb88408cb48f9a930525c187 : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c9acb0d15c041f8c40e06c84c3484f9f6bbd899e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-24 20:26:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8a1d1953bc34402f84a260ec011fe6e8 : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":5},{"id":"81d11c81cc2a529236a69a97005f7294c6e85d2e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-24 20:54:53.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_FC_os-brick/d9bc147/DellEMC_Unity_FC_os-brick/78 : FAILURE in 40m 06s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_iSCSI_os-brick/a7b52f8/DellEMC_Unity_iSCSI_os-brick/78 : FAILURE in 14m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"eb7d8c778037b0de5d99af287c66b4603073475c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-24 21:59:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-fcp/56649a5/ : FAILURE in 1h 30m 31s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/58b70f1/ : SUCCESS in 1h 43m 51s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-nfs/886cddf/ : SUCCESS in 1h 40m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/41dd3ce/ : FAILURE in 1h 36m 19s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"bd2de11f3965bf2eba6066a46ca029ad2a4e5e9d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-24 22:16:19.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_FC_os-brick/6c226b6/DellEMC_PowerStore_FC_os-brick/43 : FAILURE in 1h 21m 26s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_iSCSI_os-brick/652741a/DellEMC_PowerStore_iSCSI_os-brick/32 : FAILURE in 1h 09m 36s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_NFS_os-brick/0a5ef19/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_NVMeTCP_os-brick/6256d57/DellEMC_PowerStore_NVMeTCP_os-brick/43 : FAILURE in 1h 11m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"538a0594fa7292e8cb945a25098bdbc21d24a774","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-25 02:02:13.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerFlex_os-brick/ad64e8a/DellEMC_PowerFlex_os-brick/44 : FAILURE in 1h 34m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a01af8fe5a88b8768bb3a6b263c6518dda7d1973","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-25 04:22:52.000000000","message":"Patch Set 5:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/5/20230825072612/NEC-V-ISCSI : FAILURE in 5h 31m 09s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/59/890159/5/20230825072612/NEC-V-FC : SUCCESS in 5h 56m 36s","accounts_in_message":[],"_revision_number":5},{"id":"7984dc983b5407d7700f4cc5d10f1b35795921de","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-08-25 08:53:10.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfbc3a9c617cdcd9b95e6b6842bcf4a38b76c191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 08:53:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ecff204cd0d18b05af34e57f9d3659806d7eed6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 10:34:54.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/baf1e7f5d684456291fccdd0e6186fe6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6be01b47ce7342e78b6ffce331206d1b : SUCCESS in 1h 36m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629f72ff07aa40b5aa3545cebe558fa8 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b917dff6d775467387ba0eac3e5bd7af : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baf84ceec7544ef7899e86711eed8a4b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d17f1a66a9e44f38a64a00c265b60bd : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5728aa80a6ea4042be75984d3cf30fdb : POST_FAILURE in 4m 03s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0302ebe684c948cf8b9a2cf590196baf : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":5},{"id":"94a08520942a39a1b40f06863bf4515b19405ffb","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-08-25 12:07:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"759953e11e60c3e6af21389c386b2db9a1df3978","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-25 13:23:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/274c53b7c31f415f8f7641b7b479a68c : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":5},{"id":"456588765bb6539ed6614275a91e6efc8b0e01a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:38:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d966dbce57477eb4b2ff847ea297a9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9414e4e6ac74177933110eeaacbf3d2 : SUCCESS in 1h 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e5e79eecb54d72a7540ce17ab62952 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c720eb6ef024e4cbf4c8b10fb5937d8 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7956e96318d646c0bbdaf252387dc712 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8a173c18a8487bbd2bcf4b1db12f8a : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34c9d1aae4494661b9cf7b7e768edf6f : SUCCESS in 4m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/5d6fa144e3704ad58a0ab929780ce00c : SUCCESS in 4m 25s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/00de05a4e5034b50aba5ecb9d9418ee9 : SUCCESS in 2h 24m 48s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/1945d97e1d63474daf2703ba9e0c9856 : SUCCESS in 2h 12m 32s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/701d7d3b6e024dcfa5313cf79593318f : SUCCESS in 1h 53m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e00c57fbb29a43fe8158e5533d1d1c57 : SUCCESS in 1h 26m 31s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/7b2d32c6eccb4fb89797c272c35f584a : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6708b314feaf4c3983c0ae3296d5bf89caae940f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:38:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7506ba6f68d698f0aa92bcdbca20d820c0d8fb5a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-25 14:41:09.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_FC_os-brick/9f9830a/DellEMC_Unity_FC_os-brick/81 : FAILURE in 2m 29s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_Unity_iSCSI_os-brick/99cc796/DellEMC_Unity_iSCSI_os-brick/81 : FAILURE in 1m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"64aa0ce85d950b43ce527d8a07e63668344fafbe","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-25 15:56:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_FC_os-brick/ca93d7e/DellEMC_PowerStore_FC_os-brick/46 : FAILURE in 1h 18m 12s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_iSCSI_os-brick/ef8d736/DellEMC_PowerStore_iSCSI_os-brick/35 : FAILURE in 1h 10m 50s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_NFS_os-brick/e1a7db1/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerStore_NVMeTCP_os-brick/f87b917/DellEMC_PowerStore_NVMeTCP_os-brick/46 : FAILURE in 1h 16m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"058cf2376e27ec1a8ea0b14a16198c41f0b04e52","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-25 16:06:43.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-fcp/2016d82/ : FAILURE in 1h 22m 06s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/de3a4fb/ : SUCCESS in 1h 26m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-nfs/7ced1df/ : SUCCESS in 1h 26m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/890159/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7487181/ : SUCCESS in 1h 24m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"5a9045ecb257d8229e7a93e54f269a441ed5fcd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:32:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1873de49934eafa8f7add6f53b320d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77ff5cda7c504145957c9d56971e89de : SUCCESS in 1h 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb2176de4be48999f3943aeb1c0be4e : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd8cb070d944d0d81431ec4866ae56c : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f74da94f7fbc43059c69dbcd85da817f : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1ccc3100894b53ba9cc398b38c244c : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/836ace88164d4a16be011891d68bbebe : SUCCESS in 4m 23s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de9efda0a6734b7b950699a9252dd7e1 : SUCCESS in 1h 48m 31s","accounts_in_message":[],"_revision_number":5},{"id":"bb02208dc521d21495203a625806cf7d7b4d4992","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:32:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"01704bfec1c9f614557f8c4558aac1d9802f52e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 16:33:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a9b5a65278b4266b3098e1d05022f41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6530b0bf99cf437c9bab113dfea76a2d : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bda3cabc6594ff39e9ce4fb942669e7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5},{"id":"5deec6da787627865d845d39d19e7d708b1bba83","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-25 16:40:28.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/59/890159/5/check/DellEMC_PowerFlex_os-brick/7cc14ba/DellEMC_PowerFlex_os-brick/47 : FAILURE in 2h 01m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"ff200d8184ebde49bfb7546d3571e568fbf643c8","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-07 05:17:00.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/59/890159/5/check/tyds_test/a06b52e : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1435059da5c3b66e83ec9108cd85d17471a8d747","revisions":{"8f1eb75c7844dce92d78985cf93b72dc1ae9eb44":{"kind":"REWORK","_number":1,"created":"2023-07-31 19:37:17.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/59/890159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/890159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/890159/1"}}},"commit":{"parents":[{"commit":"75714711f2631dbba19750c55a7b049bb07da379","subject":"Fix multipath resize map with friendly names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75714711f2631dbba19750c55a7b049bb07da379"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 09:14:59.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 19:35:30.000000000","tz":0},"subject":"WIP: Check paths before extending device","message":"WIP: Check paths before extending device\n\nCheck if all paths are up before extending the device.\n\nChange-Id: I5fc02efc5e9657821a1335f1c1ac5fe036e9329a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8f1eb75c7844dce92d78985cf93b72dc1ae9eb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8f1eb75c7844dce92d78985cf93b72dc1ae9eb44"}]},"branch":"refs/heads/master"},"f508bcf0f08e5f5115c13630af95667a607ad2d0":{"kind":"REWORK","_number":2,"created":"2023-08-22 09:13:54.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/59/890159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/890159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/890159/2"}}},"commit":{"parents":[{"commit":"75714711f2631dbba19750c55a7b049bb07da379","subject":"Fix multipath resize map with friendly names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75714711f2631dbba19750c55a7b049bb07da379"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 09:14:59.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-08-22 09:13:42.000000000","tz":0},"subject":"Check paths before extending device","message":"Check paths before extending device\n\nIn a multipath enabled deployment, when we try to extend a volume\nand some paths are down, we fail to extend the multipath device and\nleave the environment in an inconsistent state. See LP Bug #2032177\nfor more details.\n\nTo handle this, we check if all the paths are up before trying to\nextend the device and fail fast if any path is down. This ensures\nwe don\u0027t partially extend some paths and leave the other to the\noriginal size leading to inconsistent state in the environment.\n\nCloses-Bug: 2032177\n\nChange-Id: I5fc02efc5e9657821a1335f1c1ac5fe036e9329a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f508bcf0f08e5f5115c13630af95667a607ad2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f508bcf0f08e5f5115c13630af95667a607ad2d0"}]},"branch":"refs/heads/master"},"c5e97671968ce24b77da717d5078ce41cf39a112":{"kind":"REWORK","_number":3,"created":"2023-08-22 09:39:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/59/890159/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/890159/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/890159/3"}}},"commit":{"parents":[{"commit":"75714711f2631dbba19750c55a7b049bb07da379","subject":"Fix multipath resize map with friendly names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75714711f2631dbba19750c55a7b049bb07da379"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 09:14:59.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-08-22 09:39:32.000000000","tz":0},"subject":"Check paths before extending device","message":"Check paths before extending device\n\nIn a multipath enabled deployment, when we try to extend a volume\nand some paths are down, we fail to extend the multipath device and\nleave the environment in an inconsistent state. See LP Bug #2032177\nfor more details.\n\nTo handle this, we check if all the paths are up before trying to\nextend the device and fail fast if any path is down. This ensures\nwe don\u0027t partially extend some paths and leave the other to the\noriginal size leading to inconsistent state in the environment.\n\nCloses-Bug: 2032177\n\nChange-Id: I5fc02efc5e9657821a1335f1c1ac5fe036e9329a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c5e97671968ce24b77da717d5078ce41cf39a112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c5e97671968ce24b77da717d5078ce41cf39a112"}]},"branch":"refs/heads/master"},"5b4ab0027e56997585c1a1ba7e19f7ddea722fa6":{"kind":"REWORK","_number":4,"created":"2023-08-23 11:27:31.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/59/890159/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/890159/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/890159/4"}}},"commit":{"parents":[{"commit":"75714711f2631dbba19750c55a7b049bb07da379","subject":"Fix multipath resize map with friendly names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/75714711f2631dbba19750c55a7b049bb07da379"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 09:14:59.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-08-23 11:27:03.000000000","tz":0},"subject":"Check paths before extending device","message":"Check paths before extending device\n\nIn a multipath enabled deployment, when we try to extend a volume\nand some paths are down, we fail to extend the multipath device and\nleave the environment in an inconsistent state. See LP Bug #2032177\nfor more details.\n\nTo handle this, we check if all the paths are up before trying to\nextend the device and fail fast if any path is down. This ensures\nwe don\u0027t partially extend some paths and leave the other to the\noriginal size leading to inconsistent state in the environment.\n\nCloses-Bug: 2032177\n\nChange-Id: I5fc02efc5e9657821a1335f1c1ac5fe036e9329a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5b4ab0027e56997585c1a1ba7e19f7ddea722fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5b4ab0027e56997585c1a1ba7e19f7ddea722fa6"}]},"branch":"refs/heads/master"},"1435059da5c3b66e83ec9108cd85d17471a8d747":{"kind":"REWORK","_number":5,"created":"2023-08-24 11:40:59.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/59/890159/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/59/890159/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/59/890159/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/59/890159/5"}}},"commit":{"parents":[{"commit":"eb76f8283ed24df856dffbd5f0ba9b3be257720d","subject":"Merge \"NVMe-oF: Create /etc/nvme/hostid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb76f8283ed24df856dffbd5f0ba9b3be257720d"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-07-31 09:14:59.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2023-08-24 11:40:32.000000000","tz":0},"subject":"Check paths before extending device","message":"Check paths before extending device\n\nIn a multipath enabled deployment, when we try to extend a volume\nand some paths are down, we fail to extend the multipath device and\nleave the environment in an inconsistent state. See LP Bug #2032177\nfor more details.\n\nTo handle this, we check if all the paths are up before trying to\nextend the device and fail fast if any path is down. This ensures\nwe don\u0027t partially extend some paths and leave the other to the\noriginal size leading to inconsistent state in the environment.\n\nCloses-Bug: 2032177\n\nChange-Id: I5fc02efc5e9657821a1335f1c1ac5fe036e9329a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1435059da5c3b66e83ec9108cd85d17471a8d747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1435059da5c3b66e83ec9108cd85d17471a8d747"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
