)]}'
{"id":"openstack%2Fos-brick~892426","triplet_id":"openstack%2Fos-brick~master~I99b36cb6c906872efa4c7f943001511af917c3be","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-11-29 15:15:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-12-19 13:11:13.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I99b36cb6c906872efa4c7f943001511af917c3be","subject":"tox: Don\u0027t share envdir between pep8 and fast8","status":"MERGED","created":"2023-08-23 12:29:19.000000000","updated":"2023-12-19 19:23:35.000000000","submitted":"2023-12-19 17:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892426","meta_rev_id":"f723500b32866da86b7ba1429b20ba412bfb4bac","_number":892426,"virtual_id_number":892426,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-19 17:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2023-11-29 15:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2023-11-09 21:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-30 01:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2023-11-17 14:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2023-11-29 15:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 14:11:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-08-23 14:32:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 19:40:02.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-08-23 22:12:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-23 22:32:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-24 13:17:34.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-08-24 20:54:52.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-27 08:40:22.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-06 01:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-09-06 22:02:12.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-11-09 21:10:34.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-11-17 14:33:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-11-28 13:13:20.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-11-30 01:52:13.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"b9a1303cadaf874344080f28173087c454c6fb6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-08-23 12:29:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b744bbbd9817246d4ad19bb42ed7ec809b21b09","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-23 14:11:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_FC_os-brick/5fb5058/DellEMC_Unity_FC_os-brick/64 : FAILURE in 28m 25s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_iSCSI_os-brick/f0757e1/DellEMC_Unity_iSCSI_os-brick/64 : FAILURE in 14m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1cfa8e8f93009af6759acadbd4a4982910c5b1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 14:32:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef1aabddd3724819b28f38393ba40d0b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a401d35abfc84f52a3a4ec2f29a6338a : SUCCESS in 1h 08m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9380d1f24ec94d9bba755d1cb0451293 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd6554c053245c18e249b5d57b43f9d : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5bc7b42678b4f3588ac6962a5bd8afc : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d33d0823b9459b9fec48ed294deb22 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c104d1a24747479c086ca1f027806b : SUCCESS in 3m 54s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/73edccb4d1a94bf2a06947c0e622985c : SUCCESS in 4m 40s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2473280f0ae5456ebe62f7d73bf440ef : SUCCESS in 1h 58m 11s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e043e89bac442f4b1ff63b0647fbd39 : FAILURE in 1h 30m 55s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/218a4d4028d84f198269a1460610f510 : SUCCESS in 59m 57s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/2fa2370a458a40dea91650b5bb07db98 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8302bff640b2a01a2e1dd92bd721e0e9784dac58","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-23 19:40:02.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-892426-1/FC/ : FAILURE in 7h 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-892426-1/iSCSI/ : SUCCESS in 2h 31m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b746d5eea45eb7290f853294f2c5a4fb925e5a6d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-23 22:12:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerFlex_os-brick/118359e/DellEMC_PowerFlex_os-brick/32 : FAILURE in 1h 51m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1b96ade956c8b24fc139861947e2ce98ca0cc2b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-23 22:32:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/88d59f4defcd4936905eac14bc7f4992 : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7ede75ff9b61452489d7aeaf957042c33550b3ca","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-08-24 13:17:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/892426/1/20230824190438/NEC-V-FC : SUCCESS in 3h 05m 57s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/892426/1/20230824190438/NEC-V-ISCSI : SUCCESS in 3h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0daed8fcd6efd1e15d4442263913cad270fd4fe6","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-08-24 17:08:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed6302f42ab6b83cbb036973a639175fe0da72a1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-24 18:48:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/63e04a2146a4484287d8cb1c94ec5d81 : FAILURE in 1h 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ffa5435495be5149a38c19b26d211bdf30b8550a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 19:23:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34fcbaff3f242bebeec90d09a1f8eb8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e1203dd730a45d58608b27e0a6631e5 : SUCCESS in 1h 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121482412b784aa688eb5ccb9beec9cf : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1955e457bc0e45b0b6e10945709ad8b9 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92e8320ab9d24bce973729156ca0ad34 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535762f9b16040d688368f29ec16c5d0 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a81b43e30c3b442eb11552a40aca52d2 : SUCCESS in 5m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7277434b95314b8fa9d48df01fa399c5 : SUCCESS in 3m 47s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/635fcf84d9ef44598c1765f21a94bc25 : SUCCESS in 2h 07m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68c61495b27a458b8fb4520cef1acb3a : SUCCESS in 2h 00m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/65f5920770f94e9e8b561789cbe5c5cb : SUCCESS in 1h 27m 03s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/eaac8fc22bfd4098a42ad9b8e8660b12 : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec7951e330ae949578aa22f2daac7906677a6fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-24 20:11:07.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_FC_os-brick/019e45d/DellEMC_Unity_FC_os-brick/77 : FAILURE in 47m 13s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_iSCSI_os-brick/66bbf63/DellEMC_Unity_iSCSI_os-brick/77 : FAILURE in 14m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d87e77ee63bbdc16c64356255f4ff110a82e5074","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-24 20:54:52.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_FC_os-brick/4d75fb4/DellEMC_PowerStore_FC_os-brick/42 : FAILURE in 1h 30m 58s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_iSCSI_os-brick/4305045/DellEMC_PowerStore_iSCSI_os-brick/31 : FAILURE in 1h 12m 54s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NFS_os-brick/9c1dd6f/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/2b495eb/DellEMC_PowerStore_NVMeTCP_os-brick/42 : FAILURE in 1h 10m 34s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f5e2c62d02597eb5c113f18d3b3b2925394fa135","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-25 00:27:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerFlex_os-brick/609c50f/DellEMC_PowerFlex_os-brick/43 : SUCCESS in 1h 33m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"422a6e069d7f99f2c770a7e4ad828a7152d53e84","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-08-27 08:40:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/26/892426/1/20230827131250/NEC-ISCSI : SUCCESS in 3h 42m 40s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/26/892426/1/20230827131250/NEC-FC : SUCCESS in 4h 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"eb63f05487894963a60d9c1c73e33645a89fc57a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-05 23:02:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-892426-1/FC/ : SUCCESS in 10h 20m 31s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-26-892426-1/iSCSI/ : SUCCESS in 9h 19m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"8e1acc4a0dfff9e0e7cbb87ef66bd49cab5d5551","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-06 01:00:07.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/892426/1/fc : SUCCESS in 0h 32m 00s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/26/892426/1/iscsi : SUCCESS in 0h 45m 35s","accounts_in_message":[],"_revision_number":1},{"id":"341350908fe371963daa1d386a4efe6886148532","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-06 22:02:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/26/892426/1/check/tyds_test/d6c717a : SUCCESS in 3h 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d3492066858fe2f76769fa2b51a9fed8140e41e4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-09 21:10:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69979e182db5bd868c1e0ed9e95fb312f5900b30","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-17 14:33:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36ee63fcd19c39f339c54cbce51685a141f55b64","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-11-29 15:15:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08b36d8c7ea8dd82006f4ee5602fb662d236414a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:15:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d39e2b0ed19415470441266e5f145d81285ca6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 17:10:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee8b1189e2c44a1fa716a0a4df31f1dd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cd504e66f5646f3a40d849eb56372e2 : SUCCESS in 1h 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f5a50ebd9147c2bd9e317953123bb9 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/914619996c9347cbbd455c85392dd3d3 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c641944b18e4294974cee5692009046 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e59edab7ba4223bc7b407148133327 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a32ee7355924a96ba86d1c9e07c4f8d : SUCCESS in 4m 28s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/af9eedcfabc640749291b2e404a2053c : FAILURE in 1h 53m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ee3eff3d92d5cd36c9f60c4ba80d812208966d65","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-29 22:31:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50d28fc8bfda0afbd7a646f200c7b7b15c91a89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5007b3006d0044f5aa0c11a973780d26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d78152005134c88836cfa52c0b277fb : SUCCESS in 1h 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4782416d9b494d8811ea89a97c372b : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4179f88c158c4de0baa925e764fabcc1 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17c5520af85049d99ee8a91c7cd510ed : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1735c45608f244a9b766408c811dbf0b : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2be6dd122ed4dcbb8ae750ab2f0891e : SUCCESS in 4m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9a1d175d844c488a82ad1fd4cb968451 : SUCCESS in 5m 03s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/27d4860c265a490b859e9f6390ba50bb : SUCCESS in 1h 40m 12s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7eda05c531824e6b8b556284e99936c7 : SUCCESS in 1h 28m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/73332363f6ee4c7eb5cdde26bf5e2af5 : SUCCESS in 1h 58m 24s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/346e5d6e1c63423c8ccf60aad500e208 : SUCCESS in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ce3405b0dfdec2cb37dcf7bef28322471d35f56","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-30 00:36:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerFlex_os-brick/86daeb2/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"50473628e46ad394e74b0861c72d602bf74c827f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:36:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d7de902917ccdbb79bec16d4847c85524098a68","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-30 01:42:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/027f7b90358642a0ab0aa355b2698e61 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d00504702564cc81745848051bf6615056142f6d","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-11-30 01:52:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a2dd7fd47ad5051182bca3656feb8a778975016","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-11-30 02:06:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_FC_os-brick/0ee6d20/DellEMC_PowerStore_FC_os-brick/71 : SUCCESS in 1h 30m 01s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_iSCSI_os-brick/6645a89/DellEMC_PowerStore_iSCSI_os-brick/64 : SUCCESS in 1h 24m 15s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NFS_os-brick/01c74ea/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/c1e1030/DellEMC_PowerStore_NVMeTCP_os-brick/75 : SUCCESS in 1h 28m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"c231d5b2a0684b5148f8261637feb96fd053d74e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 02:09:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0c41039a2af42df872faad1a0246b22\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e861b809dfb4e85b5710237a3556461 : SUCCESS in 1h 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba529dc97f59416b9054748d60791d09 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/138d210572bf46cf9adbc11acacca78f : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1db65ab3ff04ca1aafcdad012daf917 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2f968ab831492193297f185c31b342 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58dd796de7847b08436fe666492b02a : SUCCESS in 5m 16s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b70e0bfcb20d41d0874063817df1d514 : FAILURE in 1h 31m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6fa9d86816cd45f777082de55efe3a8466b9d7d8","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-11-30 14:20:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cc67d79c83f5fa5337d6679d9143aabc4a3aff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 16:51:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a6bd3f6d6c54e4b9066ff4fc239df34\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06967415082b4f3eafb73abe2309c379 : SUCCESS in 1h 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f2443b915244039aef439069c68375 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5c516eb9f444a08aad8d6a5818f009 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9e3e507eea8420799d18cd6130151ab : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f2fba751074919a778024396e3bb43 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbb79eb61e5546fe8e3ed46eed4fc4fe : SUCCESS in 4m 22s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/71488c1106004e859e712fecbdb09ab7 : SUCCESS in 3m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ae0bf8195b9c4e0694a30a265d50d953 : SUCCESS in 2h 25m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fcf2b0210930471ba85c17c0142e6495 : FAILURE in 2h 08m 03s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/776778560a7d48ab8b07eee466ba5213 : SUCCESS in 1h 23m 25s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/51c557e1051d4213bf7c4b3d57267158 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2369f324dce528da15ec9962e2454bc02118e568","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-30 17:38:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/521a2ffb78f146df99ed2a65368d78a7 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d815c37ad827ed0f6412b16cf85237f92425c4bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-01 09:38:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_FC_os-brick/1f6185a/DellEMC_Unity_FC_os-brick/120 : SUCCESS in 2h 06m 48s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_iSCSI_os-brick/1f27361/DellEMC_Unity_iSCSI_os-brick/121 : FAILURE in 2h 40m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"03ab495c9b239768ce7ca104613d27ff6bd88b0a","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-12-18 15:33:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a2d9065534bab6ba2aea11a5b55eb9fc256e966","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-18 17:57:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7a52b891b6e44cf288730b68de4d8903 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0df930b9f37d4539d1215280bdc7b518a4aeef94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 18:02:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe5e9ecb6d947d58b20d667d106e713\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3a656a6627347a0ad598f828495d417 : SUCCESS in 1h 31m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c88ce5946c40ef8be8c7aa7f2742ea : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09baa67a06ec426bb84029b7107a6a36 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7d4dbb9a16541ebb1939703e0c5ff9d : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6610d264de4fe88cc00ee8b61935c4 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec879de9810948c08ace64bb41b02d48 : SUCCESS in 4m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/418b555fff8c40c19b3556c93e0f6cc7 : SUCCESS in 4m 21s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/90c1e0a508e5487e8b2f8c64760a84e9 : SUCCESS in 1h 49m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8821804929c64544b11d3ae064d77b78 : SUCCESS in 2h 20m 39s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/90f0852cbd2e4a268e9c248a0271702d : SUCCESS in 1h 38m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/9a7f6ac053eb44f39480911db7d578b7 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90190ecfa54da9d9c425a2c0df9ad5bdd2001172","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-18 18:02:44.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerFlex_os-brick/cb3383c/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"18ec291af8fe07dc3f63a22e4d746dc717b9b509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 18:03:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77da392fa045a89c61dc0b057c50edd20fd013d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-18 20:00:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_FC_os-brick/7b01fd9/DellEMC_Unity_FC_os-brick/130 : FAILURE in 1h 14m 42s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_iSCSI_os-brick/aef1006/DellEMC_Unity_iSCSI_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4aedae992f772a5aa494003807bd53a748381b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 20:19:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70eaa5ed6b87413b9c418be754c7aa9e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7007ccfa2e049b1aef15ee3464f1516 : SUCCESS in 1h 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/104e9d2f38cc482b9657e5c6e7dfd047 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/056e0db0b84a4f14adb4e93d8042c738 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec6e563c5a024d7a9376c57ac3e32059 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cce33b18086484fb0c8d48fb92dfb5a : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28b26af1aeb94188a7d9a174f0469382 : SUCCESS in 4m 16s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f0c9b40148c4ca98a217133303a8801 : FAILURE in 2h 08m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8497d92b54441a4eefb6a2dfd8cc55169eeec5d0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-18 20:30:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_FC_os-brick/f8ab4a5/DellEMC_PowerStore_FC_os-brick/74 : SUCCESS in 1h 31m 35s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_iSCSI_os-brick/a8f5752/DellEMC_PowerStore_iSCSI_os-brick/67 : SUCCESS in 1h 26m 06s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NFS_os-brick/6edd885/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/d687d52/DellEMC_PowerStore_NVMeTCP_os-brick/78 : SUCCESS in 1h 34m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"58e8319b01222dc5c39f0a9d6a868c9c545e9c6f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-12-19 13:11:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c893c7cbb4178a287c2f06281a1bab06e77eea53","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-19 14:25:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/010cf58749cd4027904aabe1060acaaf : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7aaf0c61d7fcae793f6ff127630587bb68b3a816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad0c06eea9b34098983f009d37fc1ae7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d779998c5fd4b8aaf15402d5fc3526d : SUCCESS in 1h 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcadaceaf9a4b58b57ccfd4ef9a61aa : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c37db6328bf428aaefd5cd9d7b7f423 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9138a6c025174758bd631bcc2f3d31a8 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d83bc01fe3c40ceb5c09a79bc483e1b : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e92abeab1b4a1da1aa64b4e259a6bd : SUCCESS in 4m 11s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f703e53276d84a0cb0a122140263cb4c : SUCCESS in 4m 13s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/c87ee64e258e41c5aaf3c8b11a0eecf9 : SUCCESS in 2h 16m 22s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d6aa4daa275407aa96a589572c0514d : SUCCESS in 1h 54m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b19beb8e9866403f9250805e6f32679a : SUCCESS in 1h 14m 29s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1bcab15668c64a0c981562b9d2c9cb5c : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0474e06ad858c5bf2cdc9945b6417e46a852acc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-19 15:34:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerFlex_os-brick/8b6bef6/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"614775e8a8fd2cc2edce8ba7092d46a067c32805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:34:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f1ffd6a1c8d30e795a297b7d707aa55db565366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 17:33:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6cb204df844c55b76b304e022097d4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbd1e9d6968746bf86a660f9b4d046c7 : SUCCESS in 1h 26m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c68552abe824efd9425c3a9787fa823 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/159e7784fe2249fb89f79942fae40fa0 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a411ed2b1c644835a4f6679cdfa37325 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87c11db5c134e9b869d5ad77164aed2 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca7be9954a2a4b3581b3d1827e7b20c9 : SUCCESS in 4m 26s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cb5c2480cdd8489e8aebdebf41908381 : SUCCESS in 1h 54m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5c798cd7afa514ddfe2710fc1bc93a881043cb9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 17:33:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d461b26038370faed5238bd3e12c855e1ac3a784","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 17:35:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7afaf6792f4379a3c02f5d886a37ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d555164e4778439c8bff520e8e3fc816 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52dad13901bf4d81b83bad542ab6d8c8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1},{"id":"e2714995e5fbd3162767aef428875f084377c373","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-19 18:36:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_FC_os-brick/51b82fc/DellEMC_PowerStore_FC_os-brick/76 : SUCCESS in 1h 26m 54s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_iSCSI_os-brick/733dda2/DellEMC_PowerStore_iSCSI_os-brick/69 : SUCCESS in 1h 29m 07s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NFS_os-brick/6db220b/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/b5221fb/DellEMC_PowerStore_NVMeTCP_os-brick/80 : SUCCESS in 1h 35m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f723500b32866da86b7ba1429b20ba412bfb4bac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-19 19:23:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_FC_os-brick/7737273/DellEMC_Unity_FC_os-brick/132 : FAILURE in 1h 15m 16s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/26/892426/1/check/DellEMC_Unity_iSCSI_os-brick/84d16e0/DellEMC_Unity_iSCSI_os-brick/131 : FAILURE in 1h 56m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04a23c59a7e0c21b59e3d5957589a4e38148bba7","revisions":{"04a23c59a7e0c21b59e3d5957589a4e38148bba7":{"kind":"REWORK","_number":1,"created":"2023-08-23 12:29:19.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/26/892426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/26/892426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/26/892426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/26/892426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/26/892426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/26/892426/1"}}},"commit":{"parents":[{"commit":"bfb55b7b58b4a89a690c9c72d7c0ec45b3b66493","subject":"Merge \"nvmeof: Call findmnt with -v\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/bfb55b7b58b4a89a690c9c72d7c0ec45b3b66493"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-23 12:27:24.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-08-23 12:27:24.000000000","tz":-240},"subject":"tox: Don\u0027t share envdir between pep8 and fast8","message":"tox: Don\u0027t share envdir between pep8 and fast8\n\nThis no longer works, so just drop it.\n\nfast8: recreate env because env type changed from\n{\u0027name\u0027: \u0027pep8\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027} to\n{\u0027name\u0027: \u0027fast8\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027}\n\nTox stated here in 2022 that this functionality is\nnot intended to be supported:\nhttps://github.com/tox-dev/tox/issues/425#issuecomment-1011944293\n\nChange-Id: I99b36cb6c906872efa4c7f943001511af917c3be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/04a23c59a7e0c21b59e3d5957589a4e38148bba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/04a23c59a7e0c21b59e3d5957589a4e38148bba7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
