)]}'
{"id":"openstack%2Fos-brick~896536","triplet_id":"openstack%2Fos-brick~stable%2F2023.2~I928c97ea977f6d0a0b654f15c80c00523c141406","project":"openstack/os-brick","branch":"stable/2023.2","topic":"bug/2013749-stable/2023.2","attention_set":{},"removed_from_attention_set":{"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2023-10-04 16:43:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I928c97ea977f6d0a0b654f15c80c00523c141406","subject":"Dell Powerflex: Add new VOLUME_ALREADY_MAPPED_ERROR","status":"MERGED","created":"2023-09-27 14:27:45.000000000","updated":"2023-10-04 16:44:58.000000000","submitted":"2023-10-04 16:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896536-bug/2013749-stable/2023.2","cherry_pick_of_change":885772,"cherry_pick_of_patch_set":3,"meta_rev_id":"28c177bb4b96aee30d041f449730959666ed28eb","_number":896536,"virtual_id_number":896536,"owner":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 16:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-04 14:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2023-09-27 19:17:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-04 14:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 14:27:45.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":35063,"name":"Tony Saad","email":"tony.saad@dell.com","username":"tony-saad"},"state":"CC"},{"updated":"2023-09-27 14:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 15:50:07.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-09-27 15:58:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-27 19:17:58.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-10-04 14:41:22.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"dde781cb1b1670d60cc3af3379badcb8b3a331b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-27 14:27:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"16213e1da2cef0434149fdffb4667ce89ba9731c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 14:39:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10bf7d1864cf4472a14f0ebb5d060491\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0913f9e5704ecaa0e81113fbc56945 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc6f25e268f74e65bdf686058cb1d455 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98c108b2bfc54d77984928fa35213770 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55a43f9cd2f449ab862c2856641ab95 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/630dc2c91e9e434ab31215a7338cad98 : SUCCESS in 4m 15s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b147d93456bb48ee84dfa117c2935664 : SUCCESS in 5m 20s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a803a1f7e23643e99891b2ac2e4a4860 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40af27ca8737fe30d03aaac2974f3cc0a50c226e","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-27 15:50:07.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/36/896536/1/check/tyds_test/fe601c6 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cc345c8d681aa1912f4e252f0d6be4c5cde63ca5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-27 15:58:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/896536/1/upstream-check/cinder-tempest-plugin-ontap-fcp/5d61897/ : RETRY_LIMIT in 8m 31s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/896536/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f2cb2e1/ : SUCCESS in 43m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/896536/1/upstream-check/cinder-tempest-plugin-ontap-nfs/acaf0c4/ : SUCCESS in 1h 18m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"94732ec6d7717fd87543bfd8aa444459a0fb7f44","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-27 19:17:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25dcbc27c2eb5ac0595fdc978dab6a471e2a089e","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-10-04 14:41:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afdc0a85ed41289a7e08d4036be3904ffc778cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 14:41:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65fcb3a47d76bf0afb8192ac40eddb10b50ca5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:43:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794fd56f4e0c4d548c507191dc368a72\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1c77070a8f1405b8e8d7b9f505145bb : SUCCESS in 1h 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad3ba248b0b425da925250b1aef4c70 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c8296b19f9c4160ac2d66765f4eca6d : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5150e04ac4d74404b272a01f7cd4323f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c8be00cce548c891000d55db6ced5c : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/364f3e52d1224b838998a99bffa09874 : SUCCESS in 4m 59s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2cf327cc28664f289154f3337f836d8e : SUCCESS in 1h 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f7222bd35f479d168804f0fe5699080234f86f6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:43:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"28c177bb4b96aee30d041f449730959666ed28eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:44:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35989510bd6846ef8f46127242e01a4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553bbabf0d06477f8c4e42af56a25fc3 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c90825287da4fedba402888ef8c4f58 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8e7dfbb47aeb846e8ca417e79eb56fd908e0b50","revisions":{"c8e7dfbb47aeb846e8ca417e79eb56fd908e0b50":{"kind":"REWORK","_number":1,"created":"2023-09-27 14:27:45.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/36/896536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/36/896536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/36/896536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/36/896536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/36/896536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/36/896536/1"}}},"commit":{"parents":[{"commit":"b2f1508fae58e00697ade2d5d3f5a779a56f24c5","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b2f1508fae58e00697ade2d5d3f5a779a56f24c5"}]}],"author":{"name":"tony-saad","email":"tony.saad@dell.com","date":"2023-06-09 17:17:41.000000000","tz":0},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-09-27 14:27:45.000000000","tz":0},"subject":"Dell Powerflex: Add new VOLUME_ALREADY_MAPPED_ERROR","message":"Dell Powerflex: Add new VOLUME_ALREADY_MAPPED_ERROR\n\nDell Powerflex 4.x changed the error code of VOLUME_ALREADY_MAPPED_ERROR\nto 4037. This patch adds that error code.\n\nCloses-Bug: #2013749\nChange-Id: I928c97ea977f6d0a0b654f15c80c00523c141406\n(cherry picked from commit 42d9a892805225c13b87d5e68b6000fe01b7ae99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c8e7dfbb47aeb846e8ca417e79eb56fd908e0b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c8e7dfbb47aeb846e8ca417e79eb56fd908e0b50"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
