)]}'
{"id":"openstack%2Fos-brick~900913","triplet_id":"openstack%2Fos-brick~master~I97a441700304765e6db33f3827d5146fed09a919","project":"openstack/os-brick","branch":"master","topic":"deprecate-windows","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-21 15:22:18.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-02-21 19:17:10.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-02-21 15:23:47.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-21 19:17:10.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-02-21 19:17:10.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-02-21 19:17:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97a441700304765e6db33f3827d5146fed09a919","subject":"Deprecate Windows OS support","status":"MERGED","created":"2023-11-14 15:49:24.000000000","updated":"2024-02-23 10:43:08.000000000","submitted":"2024-02-21 19:17:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"900913-deprecate-windows","meta_rev_id":"a750861e06d3e7c2775ff18cc3961597a1e3b06a","_number":900913,"virtual_id_number":900913,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 19:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-21 15:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2024-02-21 15:23:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-21 15:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 16:08:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-14 17:41:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-14 18:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 19:37:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-14 20:50:24.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-15 01:54:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-11-15 14:17:27.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2023-11-16 22:58:19.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-11-17 14:38:03.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2023-11-18 02:33:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-14 15:33:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-02-16 15:16:45.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-02-21 14:24:53.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-02-21 14:28:29.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-02-21 14:29:04.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-21 15:10:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-02-21 21:23:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-21 21:39:22.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-02-23 10:43:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"92f23224cf24e302e10882053db0509c14fbe969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 15:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57480417e6dd404e1680975ac8714053349d062b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-11-14 16:08:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/45b650c361d844d3938fd5c1d47f3955 : RETRY_LIMIT in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"35cf37f555ba5bc0ee3b0407479509c81334fb3d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-14 17:41:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/13/900913/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/37ec995/ : SUCCESS in 1h 51m 14s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/13/900913/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/2402373/ : SUCCESS in 1h 48m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/13/900913/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5b54567/ : SUCCESS in 1h 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"235994aa161e3d995d0948fc676ea5bc9bfbea2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 18:06:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c311257bf44088b4870895374fb3bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cdaec6fa6c2479b880b0cc68d741f7b : SUCCESS in 1h 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582e1bffbd6d474c92e582a0846778f4 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e93f34dfea9c4cd09f65a15de452cfaf : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/702dc8ed27364e4ebf268b4815a713e4 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32e72ae6d6647d4ab2e13ce7baaba52 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69dd6f6d49e84f848200cced4cec542c : SUCCESS in 4m 11s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ee9efaf9cce444c8bd3867cf4d00f2dd : SUCCESS in 5m 03s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7bf9a2e2de1948bab5b54fb5baf31d2a : SUCCESS in 2h 16m 05s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/223243e7f92e44238f4120bdaf2783b1 : SUCCESS in 1h 58m 11s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/dd622e305f29403dae5163c80e50fa25 : SUCCESS in 1h 24m 10s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/a7513497f1854bd3850ee1540ca12740 : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"107d168c99c7c4482e4c52d37261b0ad386c3bf3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-11-14 19:37:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/1/upstream-check/cinder-tempest-plugin-ontap-fcp/7ef7f85/ : SUCCESS in 1h 30m 22s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/a42feae/ : RETRY_LIMIT in 7m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/1/upstream-check/cinder-tempest-plugin-ontap-nfs/2f84a17/ : RETRY_LIMIT in 8m 02s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1cd2caf/ : RETRY_LIMIT in 7m 51s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"9f0ff8ea72fdc0023f9b3637291aa9e2cf688536","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-14 20:50:24.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-900913-1/FC/ : UNSTABLE in 2m 37s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-900913-1/iSCSI/ : FAILURE in 3h 56m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0d3c1e802e68a62aafa2f800bb0c15b68b73811","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-15 01:54:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/96/.","accounts_in_message":[],"_revision_number":1},{"id":"4310bf38a58cbc02943ba748637e352cc6400097","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-11-15 14:17:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37ce7bb14e527f63366048803380e1a347e6dca5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-11-15 16:35:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/99/.","accounts_in_message":[],"_revision_number":1},{"id":"782c823dec46d6dc67fc5d122eb8e91c4bc45000","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-11-16 22:58:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/13/900913/1/20231117011109/NEC-V-ISCSI : SUCCESS in 3h 16m 42s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/13/900913/1/20231117011109/NEC-V-FC : SUCCESS in 6h 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5e0e8638dfa79d92e9b070c3daf438522f3467b4","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-17 14:38:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c850a9fa79ee6a059f0393dc8b8fc03fdee93b1b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-11-18 02:33:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/13/900913/1/20231118071416/NEC-FC : SUCCESS in 4h 07m 50s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/13/900913/1/20231118071416/NEC-ISCSI : SUCCESS in 4h 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ecc430cf7c560661984b2eb5ee2e20c5094f8e17","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-16 15:16:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6077c86be2a0356ddde047b5469a1d6b2a442bb7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-16 15:21:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f6e8d1ba8deb03b3e0fa181b36a16bd3fabf10","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:54:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10dbbb19d2f2f8afd55c04845d67009d823d0dc9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:55:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60b5cbdd1ee8388de45ae025c1f90809aa63b1d2","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-21 14:28:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d53c442788caded33057dc33e7cce1e72f0a556","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-21 14:29:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2ea5466a7b71d596c10f2daf7c33e370df577b65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 14:59:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79b2595cb748bddef78dff49b369283ae567ffc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 15:00:18.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"410fbe9e87f3212a41d3f70aae3ea8c856d1250c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 15:05:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce48ed465dd9c5d1d830ddf2c697841ee3aa3980","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-21 15:08:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5ebad2023afaa42c04d4b4e6a522646019b036f3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-21 15:10:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25407372264e4e07dbf98912d989eb8b5ca08ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 15:15:27.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cde7cd09a329c5f3b9000543ae10e5b4c38dd578","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 15:15:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"acb76e86fbaa2dac84a26a440ea21252de59471a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-21 15:22:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fed84fc1cdbbd8f1c718ca1823b0be7d31d0dc8c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-02-21 15:23:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b6e8d48800b1402393ae2f9f9e88e9dfe765e7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:31:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03adb101a7554c25ac8631817e7069ae\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76fd63c3ad6d4718aea4903e76231495 : SUCCESS in 1h 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3271b53f43c246559435c326a47a4839 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eecb0fae611143d9830b86922444337f : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8feb0ea5670412a8ac57e883ebd3fed : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32668a0441bb4fdba97c0cc6e979c0eb : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61fec65d7d454349a8be2f650a0f9b14 : SUCCESS in 3m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/771810ba090947468e7db0e589ab2cfc : SUCCESS in 3m 39s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/890e2eb165324ed18e87e1d6731e54aa : SUCCESS in 2h 15m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e45a5bde3e34e709afc755c0893af35 : SUCCESS in 1h 51m 13s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/432e985f5ce54e46854a2317f5f57c3f : SUCCESS in 1h 12m 06s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/285fca7d53e24cbfa6845b8a7d732ff1 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10d872e67343ad53b467d518446dc82f283465e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 17:32:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e52357894ab15958bfc0b315d89761c35a20cf38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:17:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49a49417edd341f59176ba3701bb3bca\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf6fdcc67e91475898a372a9540e0b30 : SUCCESS in 1h 39m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c332f6c2098e4a6b901683de849ba8af : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b0b2a92b6f145aaaff87c2fb86db86c : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76dc42b1bb41473c9f847180caf66b79 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41bff7acbcb4e2493aeb67879e3ca03 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff92d16bde94b598fd9bef19dc62a18 : SUCCESS in 3m 47s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e157e38d5a104e418a2365e1f4404aa6 : SUCCESS in 1h 43m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b155f0531a2b5068e6bc9622529332f8b07c0861","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:17:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ffb6ecc4f55b94e63c54231584c08e6b29e6d860","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 19:18:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e2138cfe5ca4326af20ece747d85321\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ca0bfa48e24ac38d6b0bff8eefa8f0 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85be196fedcc421b94ee92ad46c716ba : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3},{"id":"69e24d711b7eccab57f11f3631aa2b6f07302fe0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-21 19:54:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/763/.","accounts_in_message":[],"_revision_number":3},{"id":"bd877f0fe5bbc63022d1d16a979944585703878c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-21 19:58:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/3/upstream-check/cinder-tempest-plugin-ontap-fcp/dea1df5/ : RETRY_LIMIT in 8m 21s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/72471f9/ : SUCCESS in 2h 19m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/3/upstream-check/cinder-tempest-plugin-ontap-nfs/fa6b82e/ : SUCCESS in 2h 19m 18s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/900913/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/2ff60c1/ : SUCCESS in 44m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"311071ed35292ce2c7fd6307e3dc01a94e718bf1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-21 20:28:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3cec21bb72a240cd81a716992ddf951a : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":3},{"id":"35fcf693944a6f5e76a475a6ae738bb6d78b4b71","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-21 21:23:57.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_PowerFlex_os-brick/dd8524d/DellEMC_PowerFlex_os-brick/109 : SUCCESS in 1h 41m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d5fb1fcaa24972d4bbf07a5f64e2a1bcc338aeba","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-21 21:39:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_PowerStore_FC-ob/b30699b/DellEMC_PowerStore_FC-ob/216 : FAILURE in 1h 38m 48s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_PowerStore_iSCSI-ob/4e29876/DellEMC_PowerStore_iSCSI-ob/136 : SUCCESS in 1h 33m 25s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_PowerStore_NFS-ob/ecccb0c/DellEMC_PowerStore_NFS-ob/111 : SUCCESS in 1h 02m 14s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_PowerStore_NVMeTCP-ob/aa54542/DellEMC_PowerStore_NVMeTCP-ob/91 : FAILURE in 1h 44m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"6711e348169ff86a85df7cc9469c7e68d1831d51","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-21 23:01:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/752/.","accounts_in_message":[],"_revision_number":3},{"id":"a750861e06d3e7c2775ff18cc3961597a1e3b06a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-23 10:43:08.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_Unity_FC_os-brick/bc3e8c0/DellEMC_Unity_FC_os-brick/148 : SUCCESS in 1h 12m 16s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/13/900913/3/check/DellEMC_Unity_iSCSI_os-brick/6cea478/DellEMC_Unity_iSCSI_os-brick/147 : SUCCESS in 1h 31m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"278fdf8b3ad9be4c87146d086b7c4da1ce76c7ee","revisions":{"cd57168700d6a9c47b9cbf86436d4a1384446ab4":{"kind":"REWORK","_number":1,"created":"2023-11-14 15:49:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/900913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/13/900913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/13/900913/1"}}},"commit":{"parents":[{"commit":"fb23a40a65f1a6e06b4580ca309469f9ae2cced6","subject":"Merge \"Revert \"Dell PowerFlex password appears in plain text when creating a volume\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fb23a40a65f1a6e06b4580ca309469f9ae2cced6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 15:48:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 15:49:21.000000000","tz":540},"subject":"Deprecate Windows OS support","message":"Deprecate Windows OS support\n\nUnfortunately the Winstackers project has been retired because of no\nmaintainers[1][2], and OpenStack on Windows host is no longer\nmaintained.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-November/031044.html\n[2] https://review.opendev.org/c/openstack/governance/+/886880\n\nChange-Id: I97a441700304765e6db33f3827d5146fed09a919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cd57168700d6a9c47b9cbf86436d4a1384446ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cd57168700d6a9c47b9cbf86436d4a1384446ab4"}]},"branch":"refs/heads/master"},"26eef68d1fdfa43129c1d575962e89dc7e6728c7":{"kind":"REWORK","_number":2,"created":"2024-02-21 15:00:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/900913/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/13/900913/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/13/900913/2"}}},"commit":{"parents":[{"commit":"fb23a40a65f1a6e06b4580ca309469f9ae2cced6","subject":"Merge \"Revert \"Dell PowerFlex password appears in plain text when creating a volume\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fb23a40a65f1a6e06b4580ca309469f9ae2cced6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 15:48:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 15:00:18.000000000","tz":0},"subject":"Deprecate Windows OS support","message":"Deprecate Windows OS support\n\nUnfortunately the Winstackers project has been retired because of no\nmaintainers[1][2], and OpenStack on Windows host is no longer\nmaintained.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-November/031044.html\n[2] https://review.opendev.org/c/openstack/governance/+/886880\n\nChange-Id: I97a441700304765e6db33f3827d5146fed09a919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/26eef68d1fdfa43129c1d575962e89dc7e6728c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/26eef68d1fdfa43129c1d575962e89dc7e6728c7"}]},"branch":"refs/heads/master"},"278fdf8b3ad9be4c87146d086b7c4da1ce76c7ee":{"kind":"REWORK","_number":3,"created":"2024-02-21 15:15:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/900913/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/13/900913/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/13/900913/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/13/900913/3"}}},"commit":{"parents":[{"commit":"fb23a40a65f1a6e06b4580ca309469f9ae2cced6","subject":"Merge \"Revert \"Dell PowerFlex password appears in plain text when creating a volume\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/fb23a40a65f1a6e06b4580ca309469f9ae2cced6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 15:48:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-21 15:15:27.000000000","tz":0},"subject":"Deprecate Windows OS support","message":"Deprecate Windows OS support\n\nUnfortunately the Winstackers project has been retired because of no\nmaintainers[1][2], and OpenStack on Windows host is no longer\nmaintained.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-November/031044.html\n[2] https://review.opendev.org/c/openstack/governance/+/886880\n\nChange-Id: I97a441700304765e6db33f3827d5146fed09a919\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/278fdf8b3ad9be4c87146d086b7c4da1ce76c7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/278fdf8b3ad9be4c87146d086b7c4da1ce76c7ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
