)]}'
{"id":"openstack%2Fos-brick~903574","triplet_id":"openstack%2Fos-brick~master~Ia22322bb8a8097900d5509b6c540355eb474f19d","project":"openstack/os-brick","branch":"master","topic":"feature/IP-ACL","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-08-13 12:04:49.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-07 08:52:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e using the hovercard menu","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"36569":{"account":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"last_update":"2024-08-13 12:04:49.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-02 18:27:25.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2024-08-13 12:04:49.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-08 20:58:19.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ia22322bb8a8097900d5509b6c540355eb474f19d","subject":"add a new funtion to return IP address(es) of a host machine","status":"MERGED","created":"2023-12-13 03:43:11.000000000","updated":"2024-08-13 21:04:15.000000000","submitted":"2024-08-13 12:04:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903574-feature/IP-ACL","meta_rev_id":"8a065c80da8f4bb35ad3a4e899393be2deb6904a","_number":903574,"virtual_id_number":903574,"owner":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-13 12:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-12 13:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-08 20:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-12 13:35:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-13 03:43:24.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-13 05:12:22.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-12-13 05:24:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-13 05:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 06:03:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-12-13 06:05:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-12-13 07:56:25.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-12-13 22:55:55.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-12-14 20:47:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-19 06:54:38.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-12-19 07:36:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-12-20 08:16:35.000000000","updated_by":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"reviewer":{"_account_id":34459,"name":"Muli Ben-Yehuda","email":"muli@lightbitslabs.com","username":"muliby"},"state":"CC"},{"updated":"2023-12-21 08:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-14 13:31:56.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-07-10 07:20:38.000000000","updated_by":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"reviewer":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"state":"CC"},{"updated":"2024-07-16 16:03:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-01 18:15:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-08-05 13:57:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-07 14:43:20.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"293671c85823e7c820f3b545fbc91c1bdc7ff43c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2023-12-13 03:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ca9605fad7f2351be2f7d338c4ac5a3be09a724","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-13 03:43:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerFlex_os-brick/8c9df73/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"19bdbc1d6dfafaacebf0e37309e12272fbf717fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-13 05:12:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/261/.","accounts_in_message":[],"_revision_number":1},{"id":"6a9dab8f259b17c3658c7eefed04129cce95199d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-13 05:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/acc759086c6f4d67838f47cb3ede88a0 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1},{"id":"81df712db50a6a91f912f027a0712dc653399b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 05:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8d19f088a4040fc99c794f539b88821\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1be3852d368e448f8ee3401888c5e56b : SUCCESS in 1h 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b16213636d84742a1b4f76e7019d651 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309d6bf4a4c1429e90bf206dd698a08a : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/548234e297b744db8a2e58a417ad08b4 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7c247fad8e40e8b68bb6a5baa5fd51 : SUCCESS in 6m 05s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/41b6ac28ff654db989d1e86d43ef2985 : SUCCESS in 3m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/12b460b79d5f450e93fee92f65424306 : SUCCESS in 1h 49m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94630b7c606f41e7aec89ca8fa1c1b49 : FAILURE in 2h 07m 15s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/279974d615a94acf818b4bc96fc7f33e : SUCCESS in 1h 26m 00s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/92d159927b424f258d75ab981ca6967d : SUCCESS in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a19f0cb2f9907404c1453acdb8f9377380d0b5d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-13 06:03:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_Unity_FC_os-brick/e3c69a4/DellEMC_Unity_FC_os-brick/127 : FAILURE in 58m 34s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_Unity_iSCSI_os-brick/d280073/DellEMC_Unity_iSCSI_os-brick/129 : FAILURE in 37m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"920fdd50f5470e7aaec48b9562805b6a2788b5d7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-13 06:05:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/74/903574/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/0e71fde/ : SUCCESS in 2h 11m 47s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/74/903574/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/32343da/ : FAILURE in 2h 01m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/74/903574/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5794e9e/ : SUCCESS in 1h 45m 40s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/74/903574/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/421b0ce/ : SUCCESS in 2h 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e1acf60b3600652e817edd3d082cfff19511c5fe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-13 07:56:25.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-1/FC/ : FAILURE in 4h 12m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-1/iSCSI/ : FAILURE in 3h 43m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b828c6b9f22aa0749a902f69ab69963393bb5cd3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-13 17:57:28.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/279/.","accounts_in_message":[],"_revision_number":1},{"id":"52c1c98e6472b3c09e61d787636b6cc1e0283280","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-13 22:55:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/1/20231214043927/NEC-V-ISCSI : SUCCESS in 2h 53m 27s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/1/20231214043927/NEC-V-FC : SUCCESS in 3h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5033417b8e81f7041af116321077c6ad7d506c23","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-14 20:47:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/1/20231214212846/NEC-FC : SUCCESS in 3h 55m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/1/20231214212846/NEC-ISCSI : SUCCESS in 8h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"032c6be249f6dc842354d11c7d7b4f224329cbbd","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2023-12-19 03:03:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4265a77f62d4a3cdae4c17fa6fec668aead4faf2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-19 04:58:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/74/903574/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/c0bcc6b/ : SUCCESS in 1h 51m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/24e63f4ea8ec4f8bad7950be5c6f4059 : RETRY_LIMIT in 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/12989415598b4d53ae440edf61e13199 : RETRY_LIMIT in 41s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/d249be6cb42a4750917c2369b7575324 : RETRY_LIMIT in 41s","accounts_in_message":[],"_revision_number":1},{"id":"21877bf3349f136a33b212311300743b2e24265f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 05:17:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8d2c93cee24779bf5ef4b8f75db1fd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b32f7fa0eaea445294e3c9c0fef36f52 : SUCCESS in 1h 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db2a3d3ca1845f0b3326ea481633d05 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a02e5fdcdb4b4b2fb5bee1ae1840c6fc : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fce70c51daa40c7a880a571380a725a : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b493716b17c41d3b6e6aaaff585a14c : SUCCESS in 5m 58s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/2abd325c6560496f920fd48a29e56d01 : SUCCESS in 4m 13s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/dc4809f0f5cc4bebbf1c9dc495373fc9 : SUCCESS in 2h 03m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df254e71cf5648e1af68edca1e1b1d86 : SUCCESS in 1h 52m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f7024cf8c36041e8b6d0eb940b44a41d : SUCCESS in 1h 19m 27s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/73e6af1fc26646bd9352e62abc020635 : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d75928035b16b0105bf69bf8ee611f3571bd769","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-19 05:18:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerFlex_os-brick/b1dc3cf/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b132bf8d468e74c4b15b4df17e871c4d3f15dad7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-19 05:23:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/20bd9d4ada6845568aea7785c94b3b6c : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0888987fc2bea3fc203fc99c9aad55d2fd5d76a7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-19 06:54:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerStore_FC_os-brick/b1a289f/DellEMC_PowerStore_FC_os-brick/75 : SUCCESS in 1h 36m 26s\n- DellEMC_PowerStore_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerStore_iSCSI_os-brick/14a42be/DellEMC_PowerStore_iSCSI_os-brick/68 : SUCCESS in 1h 28m 21s\n- DellEMC_PowerStore_NFS_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerStore_NFS_os-brick/474ee68/DellEMC_PowerStore_NFS_os-brick/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_PowerStore_NVMeTCP_os-brick/d3f8b08/DellEMC_PowerStore_NVMeTCP_os-brick/79 : SUCCESS in 1h 30m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"17a10c01bbeae3d9cfdacb21e4ec850bcac9e3c1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-19 07:13:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_Unity_FC_os-brick/02070b3/DellEMC_Unity_FC_os-brick/131 : FAILURE in 1h 27m 03s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/1/check/DellEMC_Unity_iSCSI_os-brick/a4373e3/DellEMC_Unity_iSCSI_os-brick/130 : FAILURE in 1h 55m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c13a1f2a324138b84c705517021499992f2191b2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-19 07:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/1/upstream-check/cinder-tempest-plugin-ontap-fcp/2bd6d1c/ : FAILURE in 1h 28m 49s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/b308281/ : SUCCESS in 43m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/1/upstream-check/cinder-tempest-plugin-ontap-nfs/b82e3fb/ : SUCCESS in 2h 01m 47s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c637055/ : SUCCESS in 1h 55m 06s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f1e4071c6bd47415e9f481a33884866e2a67254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2023-12-21 03:31:11.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f52c69c42a39f9a4a2df66e2b43cdfa73b13e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2023-12-21 03:33:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bd66e385d6c7e37fb52575df04eb2c17f0542c49","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-21 04:02:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/479bb834e3064d8fa10b0db91061cf8d : RETRY_LIMIT in 42s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/8a0b23a508bb4bd282bfbf37bd76498e : RETRY_LIMIT in 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/6be2af2339eb4d02804231b17e3a97b7 : RETRY_LIMIT in 41s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/d7214b2c551d459383cbcece2c95aa59 : RETRY_LIMIT in 45s","accounts_in_message":[],"_revision_number":3},{"id":"b3cddae7454847cf2d3724197e3fd35944a2a454","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-21 04:48:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/51c1fc2188124731a9c9e022bb89e646 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"db418d3bb6eb4b9ea152c9fbd9cab719c9e24596","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-21 05:00:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/297/.","accounts_in_message":[],"_revision_number":3},{"id":"c62f95cebf8856b50433b90903e77ec4dd3d1856","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-21 05:22:13.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerFlex_os-brick/4356aa5/DellEMC_PowerFlex_os-brick/86 : SUCCESS in 1h 45m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"34825409993031911ac0be8caee33a64642c4d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 06:01:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcf7ad6ff6ad4a22b9d375c63ea233d0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02b7bef54ee844068d684c740e27d73f : SUCCESS in 1h 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa6fda8e222436689f15136235332b1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a22fb94d2a794d4c90e134e272795a1c : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/493bc840531948938accb3c22b990cf9 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfaf6c3c4fe4092bcdad059eb653eda : SUCCESS in 7m 04s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/633da2c477df46d287aafa545fa0c6c4 : SUCCESS in 4m 03s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/3592eed056d94a86b25253aeedf3c32d : SUCCESS in 2h 21m 42s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/01c21ec9f0b741539997aab63f6cf1a8 : FAILURE in 2h 09m 58s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/be584d7f426748f7a6655ee21ae82e56 : SUCCESS in 1h 13m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/bb832d3c6f5b40478e729c2f32e66821 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"23bfe121528071093142bde9d2290e48f16d6b6f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-21 06:03:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_Unity_FC_os-brick/6b50df7/DellEMC_Unity_FC_os-brick/137 : FAILURE in 1h 15m 56s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_Unity_iSCSI_os-brick/905368d/DellEMC_Unity_iSCSI_os-brick/135 : FAILURE in 2h 04m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b2889dbc14dce100776ef9f62dccf962b1ef0b5e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-21 06:14:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-3/FC/ : SUCCESS in 2h 12m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-3/iSCSI/ : SUCCESS in 1h 41m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"74e87df2ea399be2bb9dd07efe7fbd8ab29596d9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-21 06:28:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-2/FC/ : SUCCESS in 2h 26m 55s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-2/iSCSI/ : SUCCESS in 2h 57m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0994b6686e8ea0170e279240d3e582b3e659e725","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-21 07:14:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/317/.","accounts_in_message":[],"_revision_number":3},{"id":"d66dc530dcfd7f0acdace998dc3c1ae904bcf1f4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-12-21 08:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/903574/3/fc : SUCCESS in 0h 58m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/903574/3/iscsi : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":3},{"id":"85b061c2704ead296441b56a5cbe8123e1e6b387","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-21 10:43:20.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/3/20231221161205/NEC-V-ISCSI : SUCCESS in 2h 59m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/3/20231221161205/NEC-V-FC : SUCCESS in 3h 31m 11s","accounts_in_message":[],"_revision_number":3},{"id":"26da3c9b1db51bf6169edc231049dbc505e88564","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-22 10:04:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/3/20231222144408/NEC-ISCSI : SUCCESS in 4h 09m 08s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/3/20231222144408/NEC-FC : SUCCESS in 4h 20m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d64f72c62e28eb593e73f8ce54ee7754e1c7b912","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-01-02 22:17:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f95e0568795cfda0bf3cd1eeef5df6fb75a1fa8","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-01-02 22:18:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9641394e932a96ff94a83226e1b84a66c28f45c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-01-03 00:09:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/74/903574/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/84d4e8c/ : SUCCESS in 1h 51m 01s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/74/903574/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/cda9aad/ : SUCCESS in 1h 45m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/74/903574/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/cf4e9d8/ : SUCCESS in 1h 49m 07s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/56ecf0874e404f0ab36a1d84f2f2ad3a : RETRY_LIMIT in 41s","accounts_in_message":[],"_revision_number":3},{"id":"ba9abe791a7ea6e0d22fdc5cfed113c7fc412554","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-03 00:10:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ed60583f588b41fe9ec3292ec6d41e66 : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":3},{"id":"8b5716c409159101b902b795f31a7760b8f7805c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 00:21:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0b280752fc47e8afc7f9a46997ec63\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9601e486e9304dbb93ba533176e049e7 : SUCCESS in 1h 42m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5453224c0c474b8bcd48e0b317f79b : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9780b7cbb2e74c13aabedf65dbd89bfd : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4e440aa167a403990f2734461443d60 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70952add646b48a88444ebca170d7ab8 : SUCCESS in 5m 16s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/e202456bb6a14b2489c967b180bcd569 : SUCCESS in 3m 56s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/d4474e8f6c6b4e0589fe9342470c3443 : SUCCESS in 1h 58m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/469bf1b8ee7d4ffc831f079f7eb74480 : SUCCESS in 1h 57m 51s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0ab73390c6844f8ab00260e9e4cc1524 : SUCCESS in 1h 18m 36s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/66542638b6c64ef9a28fddebc64e14f9 : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d35c5a12c1eeade1316d3b79e2f1a3d1de33ffa6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-01-03 01:47:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_Unity_FC_os-brick/2c04aa8/DellEMC_Unity_FC_os-brick/138 : FAILURE in 1h 02m 33s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_Unity_iSCSI_os-brick/b75e9d1/DellEMC_Unity_iSCSI_os-brick/136 : ABORTED in 30m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5cae91ff7b2a62911b54448dd25f093fb801e325","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-03 01:58:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/3/upstream-check/cinder-tempest-plugin-ontap-fcp/f326eeb/ : SUCCESS in 1h 13m 36s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/b4568eb/ : SUCCESS in 43m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/3/upstream-check/cinder-tempest-plugin-ontap-nfs/1a8917f/ : SUCCESS in 1h 29m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a9e4903/ : SUCCESS in 46m 08s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9ec1710eec4659a18d2c438ed5170844fce13944","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-03 02:21:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerFlex_os-brick/30ded3a/DellEMC_PowerFlex_os-brick/88 : FAILURE in 1h 59m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"762cd2e213b6d4be381c69f0089efe492ba84613","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-03 02:59:18.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerStore_FC-ob/e110af1/DellEMC_PowerStore_FC-ob/195 : SUCCESS in 1h 31m 52s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerStore_iSCSI-ob/af388b2/DellEMC_PowerStore_iSCSI-ob/116 : SUCCESS in 1h 36m 35s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerStore_NFS-ob/a38f063/DellEMC_PowerStore_NFS-ob/91 : SUCCESS in 56m 55s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/74/903574/3/check/DellEMC_PowerStore_NVMeTCP-ob/72b3fb4/DellEMC_PowerStore_NVMeTCP-ob/71 : SUCCESS in 1h 36m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"60ace4ea6b316073eedcf4a28cbdfe8032e79e76","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-02-14 13:31:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c781156f70e27975a30ecc80352a2b46538bacc7","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-02-15 21:15:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e3304e6fb9ea0509bfdefebad47be5aeaaf7ea24","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-05 21:20:47.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"1cba04f0e0a9377be82211fc116e94de78f017c7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-03-05 21:22:34.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11b50e45d5ff5c4953058c96dfd9facf659406d2","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-03-05 21:40:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42c28e9eb10b21027731b1b334cc5d12035765bb","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-03-06 03:49:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"442cea070ff7c4f2f2135f4bc5849d73cd52c7fb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-17 14:37:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"44118cc216f85c69d718e79587ec312755b68b95","author":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"date":"2024-04-20 00:14:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"114ca607b26250f6960a10fbef346ac52fd40cc8","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-07-10 07:20:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0c036ecbcdc3a3b5d70183107dd1ecfff77705e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-07-16 14:09:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ff12104ebaee2fa41d6785945f8f577f914a0e62","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-16 15:00:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/256/.","accounts_in_message":[],"_revision_number":4},{"id":"ed18c72c1e28bee25889f2e9ac61c31698c62518","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-16 15:14:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/16cafa3185c44fa88e627f344e62f683 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":4},{"id":"3ee804cc4294ec06f1dc54acd9deba0714111f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 15:57:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb5b457cad646ad92e4d3bc995df060\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e6ae64c52c846b8b45210d20083e91b : SUCCESS in 1h 26m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216394acccaf4a71822bd3b29935704a : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bea581aa30b544c5a0a508ba29488a10 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bdc34f171f749c5bcdff115de159ecb : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11a4a51bf1c74a0cbf4d46e3d0d35b9f : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af320a3b99b449b78872a0666ac7d6b3 : SUCCESS in 7m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/420c79e9581a4f0bb5fef041ee2106c9 : SUCCESS in 4m 04s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/933761c699b64ef091893b1955d08343 : SUCCESS in 1h 46m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ffa13fa66b84cdb82067f6bba26f02e : SUCCESS in 1h 39m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/02d38e3126b24b28b1b750a1662649cb : SUCCESS in 1h 16m 32s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/ec2b4e2521eb4266933f710c872bc7f2 : SUCCESS in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f2439c26b5359e54f2917a7bba5a199f80eb59a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-16 16:03:01.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_07_2024_15_57_20\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"c9a3a75b01cbc0592336fb100889b963f1739e9a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-16 16:14:38.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/741/.","accounts_in_message":[],"_revision_number":4},{"id":"f4d4dcd6855f17939a5b45b1665b0af46fbefb11","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-16 16:36:10.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-4/FC/ : SUCCESS in 2h 26m 26s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-74-903574-4/iSCSI/ : SUCCESS in 1h 45m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"dd41ab895bb17b056377ef017ac5853d6c9836d7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-16 17:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/903574/4/fc : SUCCESS in 0h 50m 54s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/74/903574/4/iscsi : SUCCESS in 0h 54m 46s","accounts_in_message":[],"_revision_number":4},{"id":"bf6020b732519fab6e42a79040b1c076d3a200fa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-16 17:21:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/8ee3c8b/ : RETRY_LIMIT in 8m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/4/upstream-check/cinder-tempest-plugin-ontap-nfs/5945a55/ : RETRY_LIMIT in 8m 41s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/61e42ee/ : SUCCESS in 1h 23m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"86ab427bd137de6a19176336501c4752a8eb43f8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-16 18:31:38.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerFlex_os-brick/01f6c80/DellEMC_PowerFlex_os-brick/108 : SUCCESS in 1h 57m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ce2b73bc2bbd195a44fe9bf741f7d62e38fcea8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-16 18:41:28.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_Unity_FC_os-brick/d1869e4/DellEMC_Unity_FC_os-brick/188 : SUCCESS in 1h 24m 26s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_Unity_iSCSI_os-brick/a35d795/DellEMC_Unity_iSCSI_os-brick/185 : SUCCESS in 1h 57m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ec24660f2db4372c9828dcd2456dda2ad679934d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-16 18:50:12.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerStore_FC-ob/7e26838/DellEMC_PowerStore_FC-ob/242 : FAILURE in 1h 04m 48s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerStore_iSCSI-ob/e62c0a1/DellEMC_PowerStore_iSCSI-ob/161 : SUCCESS in 2h 06m 17s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerStore_NFS-ob/4f25986/DellEMC_PowerStore_NFS-ob/135 : FAILURE in 1h 09m 02s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerStore_NVMeTCP-ob/b62744a/DellEMC_PowerStore_NVMeTCP-ob/114 : SUCCESS in 2h 06m 19s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/74/903574/4/check/DellEMC_PowerStore_NVMeFC_ob/43b6e09/DellEMC_PowerStore_NVMeFC_ob/5 : FAILURE in 2h 52m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"0e898b79f7c7a02abf96d97c271e2ed12535ccaf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-17 17:04:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/4/20240717215809/NEC-V-FC : SUCCESS in 3h 57m 27s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/4/20240717215809/NEC-V-ISCSI : SUCCESS in 4h 06m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b393423b24a880d8d867a52eff78d42534b3c2f1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-17 23:38:09.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/4/20240718024310/NEC-FC : SUCCESS in 5h 46m 16s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/4/20240718024310/NEC-ISCSI : FAILURE in 5h 54m 55s","accounts_in_message":[],"_revision_number":4},{"id":"2725db35ed7b356ad8d02bf7d10972c15cbaac05","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-07-22 07:54:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9512fd7737a3933024e16a1728e697e2a5d8ce0f","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-01 18:15:51.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"28df3e3c52a9ef8eb992ad4dc34d3c54e82c3fb7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-02 18:27:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b6ebdb8f620bbe5e77a68824c457a702828a076b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-05 13:57:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32f2ef63d6cd47e2ea09579f0edebe424c89ad64","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-05 13:59:49.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"d0e08bf511d4cf733e2f1a01e9298f99c1c76bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-07 08:38:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5fb2f089809a42f041f9a44b6607dfdd182ad7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 10:13:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0798bebe2e346258f5a39da0cdf54e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4fab4bf3614447ba46ffca948d34e65 : SUCCESS in 1h 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01d2638629b4226bce579e777a0bfb3 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b45a40245c3046efaccb0ba7dfe31abf : FAILURE in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9d4a308f83d4a5785b5dc02af8e0788 : FAILURE in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/228b35a17d5d4fc98c7eeed41966696d : FAILURE in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a114cfb1467d43b2aeeaf8d02f9f572e : FAILURE in 5m 03s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/7a3bfeb426bb460797aeca1c3e3f5e79 : FAILURE in 3m 15s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/0941e8848c30406ba773dc5e8df976db : SUCCESS in 1h 33m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/92f538706fca448aa0002769a437d886 : SUCCESS in 1h 24m 03s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/62251e35a973452f84b586a2e4836743 : SUCCESS in 1h 22m 57s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d5836d62a6134ae89e450acbe78408c2 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db4fdbba146b8c053c4bde85643169eaa262476c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-07 11:24:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"034b367367b5518fb771e131e91f6a443d261782","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-07 13:03:00.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1cabd9048e21479d947cdee9c95118b2 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":6},{"id":"2c6445841351615ca5e42fa7f7a6b3d2a826c179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-07 13:06:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fa5e93707d4a2eb1c8db68323a086f2e217e17fa","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-07 14:20:57.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/b3606f5d58ed40b4bcf7c305bc2c8fe5 : FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":7},{"id":"e16ee75818b3564cf5fd04755876775103857a3a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-07 14:43:20.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d7dd01b4b415aa2830354c24f7888bf49b283071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:56:38.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91aaee0c8b304bcba2b49bc37de520c5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd2586bfe53947529f8c7a637cb32dc6 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/553a8cd6b0364afebfe14aecbbcd00a5 : SUCCESS in 1h 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc02b60d41a4424fbaaaf2c471cb8cdc : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb4c1fd994e54c0babdaffadd776945f : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37627ecd7dbc43369bebdc5e16ecfd03 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2782fb7256614eac9e2e9d72c50c4bac : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba8bca537754c91b13af49f2773c9a4 : SUCCESS in 4m 48s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/96c7dc8fb6c643e0bcc3546909b61df0 : SUCCESS in 3m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/94c9b128a5ba439581744bca01340445 : SUCCESS in 1h 42m 51s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2839db5c4e2b44db9bd9a774def6aab4 : SUCCESS in 1h 44m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/71744c4e41e947178bc5240bbfca5c1d : SUCCESS in 1h 17m 47s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/fda6d96fbfa54d03b6fcd02942dfc640 : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"27fb44daba33c83dd53b862b0b65bddad0962b73","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-07 22:08:24.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/7/check/DellEMC_PowerFlex_os-brick/ac82b65/DellEMC_PowerFlex_os-brick/115 : SUCCESS in 1h 54m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"2e30bc249c65e962a58b56f4ff4259ad2997c9cb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-07 22:24:02.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/7/check/DellEMC_Unity_FC_os-brick/c7f50f7/DellEMC_Unity_FC_os-brick/193 : SUCCESS in 1h 56m 24s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/7/check/DellEMC_Unity_iSCSI_os-brick/5feacfd/DellEMC_Unity_iSCSI_os-brick/190 : SUCCESS in 1h 59m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"ec630d356d4b31475ea77fdb93a3a9c17e47ac53","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-08 08:21:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/7/20240808132317/NEC-V-ISCSI : SUCCESS in 3h 49m 30s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/7/20240808132317/NEC-V-FC : SUCCESS in 3h 58m 00s","accounts_in_message":[],"_revision_number":7},{"id":"7681e6ada8b24598beefe2ce7d510a3557859f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-08 10:35:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6fc1814ad76f26ef98d30e9ad3ccaa2ec2610661","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-08 10:36:51.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4aa50e25959bf104f75573cd2a202c14fdc1eab7","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-08 10:37:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a58f20bee006dcaffd82e5831b027dc005e63c7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-08 11:55:33.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/34569e54127649089a8019f6140fb1eb : SUCCESS in 1h 17m 39s","accounts_in_message":[],"_revision_number":8},{"id":"dd0555030f4baaee079695c9d4657d42435e0a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 12:29:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d60270754d1b428a97595a53359873c1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9297f5775ab24d04b59a050cc7ba4267 : SUCCESS in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d254559cea54fd7af9e1317173a667d : SUCCESS in 1h 32m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f63ff3ba7745c08e1d5eee7c7bbd81 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3a08691f58a41d1a59da69ede014028 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d578ca989a7466bbc05a0eab39821e4 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4072d09dc6494787a01733ac8858bed5 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcdd8dde72df4d6584d72812fc54d420 : SUCCESS in 5m 37s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d863cfc2beb5400e9a9c6e8a691b25be : SUCCESS in 4m 08s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/b966eed786bf460288013c78a9ff000f : SUCCESS in 1h 53m 04s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf213bfc45a24a08a6ec61a57ee44bdf : SUCCESS in 1h 24m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d1cb01094db74163b7053e1d86557d6a : SUCCESS in 1h 04m 43s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/57859ad165dc4c449502844362410630 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b7bcbcb21953e24fe37d38cdbb7ceb69e8efeba0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-08 12:35:23.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/08_08_2024_12_29_59\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"ed5b009e921762aecafd7cfc86ab37d13b493f73","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-08 13:00:21.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/6ff907b/ : RETRY_LIMIT in 3m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/8/upstream-check/cinder-tempest-plugin-ontap-nfs/afefa4b/ : RETRY_LIMIT in 3m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1e12996/ : RETRY_LIMIT in 3m 09s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"bce7a2e82fd0492c94c8f826c950a1d232fce9fc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-08 16:17:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_FC-ob/d8b9fda/DellEMC_PowerStore_FC-ob/248 : SUCCESS in 1h 05m 39s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_iSCSI-ob/0d18a4b/DellEMC_PowerStore_iSCSI-ob/167 : SUCCESS in 2h 06m 22s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NFS-ob/187507f/DellEMC_PowerStore_NFS-ob/141 : FAILURE in 1h 10m 11s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NVMeTCP-ob/b607009/DellEMC_PowerStore_NVMeTCP-ob/121 : SUCCESS in 2h 04m 01s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NVMeFC_ob/bb5bf9f/DellEMC_PowerStore_NVMeFC_ob/11 : SUCCESS in 1h 09m 25s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"eec5b46be1f440cf6a99fd1e8e7e098c07db18b3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-08 20:58:19.000000000","message":"Patch Set 8: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b05e3c9c3df355ba4db8990f631f11b80b81037f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-09 00:45:35.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_Unity_FC_os-brick/1e66b4b/DellEMC_Unity_FC_os-brick/194 : SUCCESS in 1h 43m 56s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_Unity_iSCSI_os-brick/b601ea6/DellEMC_Unity_iSCSI_os-brick/191 : SUCCESS in 2h 06m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"43fa06396abdbde88583b1ad98c487657238bdf8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 02:32:26.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerFlex_os-brick/98b5f0c/DellEMC_PowerFlex_os-brick/117 : SUCCESS in 1h 48m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"985c5abf39136781b5efa9ccf84e37edec9d9cc8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-11 00:41:56.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/8/20240811011439/NEC-FC : SUCCESS in 5h 48m 30s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/74/903574/8/20240811011439/NEC-ISCSI : SUCCESS in 8h 27m 13s","accounts_in_message":[],"_revision_number":8},{"id":"66458f4cf8c3131d3886014ac4fa5b80431d7915","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-12 13:35:10.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6238302aac1e7864c100a9318b213a249804370d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 13:35:40.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cdc10b71e6c67b81fceadac98e383a8227a5feff","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-12 15:04:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"77f2eca38ad82eaa0b6d83c1e245857890e5619b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 15:39:15.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09943fdfe2da4766bbefd3a78d90ddd2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/38032d18dbd3494b8d8f80fcc9581403 : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0076556c9ea42b4bdc816abb820597d : SUCCESS in 1h 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9be95ec70954c6b957497a506eb3650 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db8c396ed1347fe9ef8f8c7cbd1eff8 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/909bc67b7f93496fb914515a74b384b7 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/779bc1729dcc4a1a81d17b7c97fe2529 : SUCCESS in 6m 03s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91a34a8003374d6687c46b39341917e4 : FAILURE in 1h 55m 50s","accounts_in_message":[],"_revision_number":8},{"id":"3bc671c5bb47499c2fbf7958b55f5e990fb1b3f9","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-12 18:05:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d8184ae1fcb712c7af4aa61c62d5b63e274708be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 19:57:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/651d882f190b400cbe8228b10bda2e5a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fb19b3777cc482bb55ab6ac6ba746ee : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11282b879f134132995e6c6d4215685c : SUCCESS in 1h 05m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cfde3b71c246d3a9fde23d04caeb3e : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08b68f8cc8fa4e8d83f756c64a858d68 : FAILURE in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17daf8606c57431ebac652b9c29670c3 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/126fe41c515d4efc8b4c88ea7512c44f : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fbd7c844d44defa086d96da58a7fee : SUCCESS in 4m 42s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/04bc3b8ce27a4179893f786c60f51e87 : SUCCESS in 3m 33s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/da33e1da39894b2ebb7b3b77deaae0b0 : SUCCESS in 1h 20m 01s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/10327b8f28cb4e578d9ea780b9714e7c : SUCCESS in 1h 46m 08s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/413c78bff7744079ab81a81682421be0 : SUCCESS in 1h 08m 48s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/39ec7e34601d4d34814bcbbe0f2d2105 : SUCCESS in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"59562e5b434a162d5075d5fbf54a912a8766da95","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-12 20:02:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/98809d409f3146be9a2a118a28c6a4b7 : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":8},{"id":"b7d9bdf9230c108994f8c884cdafe2991a558bfc","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2024-08-13 08:41:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bb664ab5b01f3476ddf44bebefaef9c4db818d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 10:32:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74cd3dd00d8d46dc91de3b17cbcd5ebf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e462b4fdb3a4dbf993c383ea3aca3e6 : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ede06e82c8e462e8c4c943a25448e5b : SUCCESS in 1h 26m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5df717b1dd744d58d0fa1f05cde196c : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f3458d675043058836cd1e714ca199 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c86eb0e9e6d4f59aad9d1c9cbb7c62a : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6354c7d9f2044606b81fd78342cea6de : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca46b08e60ee42939bec2e5a505c85c3 : SUCCESS in 6m 26s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f5bc930b748447fa83375d92617c3429 : SUCCESS in 4m 48s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/cfe29e675595495dac5594c25c434172 : FAILURE in 1h 49m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8133197281c34f58acf8cc69612e9257 : SUCCESS in 1h 36m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/9f7f539ff6504d6e9afd5899e4d159a4 : SUCCESS in 1h 20m 11s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/185bf844c3d04fdba4e778ed6f77f04e : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"88f30475ab27a3939ed9073c99384fac14769181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 10:32:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b361ac3382a542b24582a61fac800e50a8a9c71f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-13 10:37:59.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_08_2024_10_32_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"7f7081292809014b7070afba47a2eacbf6c63e12","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-13 10:59:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_FC-ob/bbc2ca0/DellEMC_PowerStore_FC-ob/249 : FAILURE in 20m 21s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_iSCSI-ob/b7e1fcf/DellEMC_PowerStore_iSCSI-ob/168 : FAILURE in 26m 53s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NFS-ob/3da00d8/DellEMC_PowerStore_NFS-ob/142 : FAILURE in 25m 13s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NVMeTCP-ob/00cddbe/DellEMC_PowerStore_NVMeTCP-ob/122 : FAILURE in 25m 03s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerStore_NVMeFC_ob/2619be4/DellEMC_PowerStore_NVMeFC_ob/12 : FAILURE in 26m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"4e8fe90e5562721ffaccd8d0e12621561b3ce313","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-13 11:00:17.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_Unity_FC_os-brick/7560b87/DellEMC_Unity_FC_os-brick/195 : FAILURE in 19m 18s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_Unity_iSCSI_os-brick/329a37e/DellEMC_Unity_iSCSI_os-brick/192 : FAILURE in 27m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"956a7b42b5346d1c3999269fb693e61c7d8acfb8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-13 11:06:49.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/74/903574/8/check/DellEMC_PowerFlex_os-brick/a7d791f/DellEMC_PowerFlex_os-brick/119 : FAILURE in 34m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"06558112d8cdcbd5985506362f900719617d4bc1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-13 11:19:21.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/b0c4c73/ : RETRY_LIMIT in 3m 14s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/74/903574/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/66b2148/ : SUCCESS in 46m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"30394639bedc1af408c064d020615234b28d551c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 12:04:48.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b053041f27464e94e2e9f3911ef730\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2576feadd74b4e0bb3fdba2419c857fb : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7409e57f21f94971ba9de952249a5a99 : SUCCESS in 1h 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea898f71ba49469e850bb1d34f619b37 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60fdb2a68a34a41b074ab6f5a069122 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2156f05d8c154880a95d48985e23c67f : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868cc727de544653949aad900b834673 : SUCCESS in 5m 25s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fa07891864c4e2191212621533f1688 : SUCCESS in 1h 32m 12s","accounts_in_message":[],"_revision_number":8},{"id":"3b8a4584878c7b53dd804d7915cc49224a18b637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 12:04:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d119a3ff597ed9cd41ccae27f6f926b62b77e0e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 12:05:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae22e13c8ce4369af26b3897d510b67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98f8a5ae1fd46409e36e5e1e581bc13 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":8},{"id":"8a065c80da8f4bb35ad3a4e899393be2deb6904a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-13 21:04:15.000000000","message":"Patch Set 8:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/874/.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"837bd7188faa13668c25e40e17085ac9f6168222","revisions":{"22588c8666910abed746e4c99e8b30b4e6171244":{"kind":"REWORK","_number":1,"created":"2023-12-13 03:43:11.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/74/903574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/1"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"rahman-lb","email":"rahman@lightbitslabs.com","date":"2023-12-13 03:43:09.000000000","tz":-480},"subject":" Adding a new function to return IP address(es)  of a host machine to be used for IP-ACL.","message":" Adding a new function to return IP address(es)\n of a host machine to be used for IP-ACL.\n\nUnit test, added property for host IPs to Connection\n\nmade code changes to satisfy indentation requirements\n\nsatisfy test indentation requirements\n\nfixed code indentations\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/22588c8666910abed746e4c99e8b30b4e6171244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/22588c8666910abed746e4c99e8b30b4e6171244"}]},"branch":"refs/heads/master"},"9908876fe8ae7bd7718244332fcf81b92c3cd334":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-12-21 03:31:11.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/74/903574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/2"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2023-12-21 03:31:11.000000000","tz":0},"subject":"add IPACL support to the Lightbits Cinder driver","message":"add IPACL support to the Lightbits Cinder driver\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. This patch adds support for Lightbits IPACL to the\nLightbits Cinder driver. We get the IPs to set from the os_brick\nrunning on the nova node that will be accessing the volumes (see the\ncorresponding Lightbits connector os_brick patch) and set them when\nupdating the volume\u0027s ACLs on creation and on volume update due to\nattachment and disconnection.\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9908876fe8ae7bd7718244332fcf81b92c3cd334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9908876fe8ae7bd7718244332fcf81b92c3cd334"}]},"branch":"refs/heads/master","description":"Edit commit message"},"af0ee63ac1d2f8f5cec24fe21f5d0fc487c06f7e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-12-21 03:33:27.000000000","uploader":{"_account_id":36569,"name":"Rahman Muhammad","display_name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","username":"rahman-lb","status":"Lightbits delivers NVMe/TCP based storage products."},"ref":"refs/changes/74/903574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/3"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"Rahman Muhammad","email":"rahman@lightbitslabs.com","date":"2023-12-21 03:33:27.000000000","tz":0},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/af0ee63ac1d2f8f5cec24fe21f5d0fc487c06f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/af0ee63ac1d2f8f5cec24fe21f5d0fc487c06f7e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0533df4831419661e551ea59acf3a44eea6e7935":{"kind":"REWORK","_number":4,"created":"2024-07-16 14:09:21.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/74/903574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/4"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"yuval brave","email":"yuval@lightbitslabs.com","date":"2024-07-16 14:09:19.000000000","tz":180},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0533df4831419661e551ea59acf3a44eea6e7935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0533df4831419661e551ea59acf3a44eea6e7935"}]},"branch":"refs/heads/master"},"c953a50d131cdfbc05cafbb90035a4a16d812801":{"kind":"REWORK","_number":5,"created":"2024-08-07 08:38:17.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/74/903574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/5"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2024-08-07 08:38:13.000000000","tz":180},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c953a50d131cdfbc05cafbb90035a4a16d812801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c953a50d131cdfbc05cafbb90035a4a16d812801"}]},"branch":"refs/heads/master"},"e5e304bed2a640aeb16911d183a7440e76b86c3a":{"kind":"REWORK","_number":6,"created":"2024-08-07 11:24:47.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/74/903574/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/6"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2024-08-07 11:23:42.000000000","tz":180},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nadded requirement: \"psutil\u003e\u003d5.7.2 # BSD\" same as in cinder\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e5e304bed2a640aeb16911d183a7440e76b86c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e5e304bed2a640aeb16911d183a7440e76b86c3a"}]},"branch":"refs/heads/master"},"5a51c2da72f7c9114249238aad86dbe1c9f768f5":{"kind":"REWORK","_number":7,"created":"2024-08-07 13:06:00.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/74/903574/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/7"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2024-08-07 13:05:49.000000000","tz":180},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nadded requirement: \"psutil\u003e\u003d5.7.2 # BSD\" same as in cinder\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5a51c2da72f7c9114249238aad86dbe1c9f768f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5a51c2da72f7c9114249238aad86dbe1c9f768f5"}]},"branch":"refs/heads/master"},"837bd7188faa13668c25e40e17085ac9f6168222":{"kind":"REWORK","_number":8,"created":"2024-08-08 10:35:54.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/74/903574/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/74/903574/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/74/903574/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/74/903574/8"}}},"commit":{"parents":[{"commit":"8daafd376bdca8daa2d285923e61cdc26119f290","subject":"Merge \"Support force disconnect for FC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8daafd376bdca8daa2d285923e61cdc26119f290"}]}],"author":{"name":"Rahman LBL","email":"rahman@build02.lbits","date":"2023-12-07 14:21:27.000000000","tz":120},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2024-08-08 10:35:43.000000000","tz":180},"subject":"add a new funtion to return IP address(es) of a host machine","message":"add a new funtion to return IP address(es) of a host machine\n\nLightbits storage clusters have the ability to block access to a\nvolume unless it comes from a specific IP, via a non-standard\nextension to NVMe/TCP called \"IPACL\", which augments the standard\nstring-based ACLs. To know which IP(s) to set in a volume\u0027s IPACL, we\nneed the os-brick to send as part of the connector information the\nrelevant IPs.\n\nadded requirement: \"psutil\u003e\u003d5.7.2 # BSD\" same as in cinder\n\nChange-Id: Ia22322bb8a8097900d5509b6c540355eb474f19d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/837bd7188faa13668c25e40e17085ac9f6168222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/837bd7188faa13668c25e40e17085ac9f6168222"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
