)]}'
{"id":"openstack%2Fos-brick~905231","triplet_id":"openstack%2Fos-brick~stable%2F2023.2~I7a33f38fb1b215dd23e2cff3ffa79025cf19def7","project":"openstack/os-brick","branch":"stable/2023.2","topic":"nvmeof-v2-2023.2","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-01-16 17:22:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a33f38fb1b215dd23e2cff3ffa79025cf19def7","subject":"NVME-oF: Fix to support new \"address\" in sysfs","status":"MERGED","created":"2024-01-10 14:50:14.000000000","updated":"2024-01-16 17:23:43.000000000","submitted":"2024-01-16 17:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905231-nvmeof-v2-2023.2","meta_rev_id":"ff38c8e2d7897734eb95eddb52a0ae683e5c42e6","_number":905231,"virtual_id_number":905231,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 17:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-01-11 20:25:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-16 15:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-16 15:06:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 16:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-10 19:36:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-01-11 20:25:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-01-16 15:06:33.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"b438f3a637436b58332484a6f21b76dceaefe0af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-10 14:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46e41ebac757335cefa864721777c5406c4d19c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 16:51:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa0320b0c9764ee2a272b102d1085d30\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b520939c3ab4cde83369550318745ed : SUCCESS in 1h 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f11df9b548b443ab9361505f55637ca : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f9af50e64064013bf4e324a5745eb1b : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50fb5b0429404b15848ca7c16324c413 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f603eee295264538a6d0a59605c68c27 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6ef85b3d57846fa86dd0d4ca2e357d9 : SUCCESS in 4m 41s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a4491246b75c494086b9b7ec6520792b : SUCCESS in 4m 45s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/f12ca8102f3842a683a1d65dcdee0151 : SUCCESS in 1h 55m 34s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9f28e3a54494bb497fd7a07f6bf34f1 : SUCCESS in 1h 53m 11s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7225690829814e119b64118f75d51ec6 : SUCCESS in 1h 08m 31s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/131d2c71ed2b44349ee4b27493cd4c86 : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39e8c4ab25eae8840913b1ebe67bf93da7bfe52e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-10 19:36:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/905231/1/upstream-check/cinder-tempest-plugin-ontap-fcp/4dfe5f3/ : SUCCESS in 2h 43m 50s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/905231/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/b0b8b13/ : SUCCESS in 2h 18m 59s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/905231/1/upstream-check/cinder-tempest-plugin-ontap-nfs/0889e6a/ : SUCCESS in 51m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"46b0872ee4a6a8e2d6e9d08fe950cabfa84fafe7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-11 20:25:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d469ea4a6830a0155fe13a7a0874f11371728770","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-16 15:06:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9d9a389ca0008c63dbc0ff814d96478f15e8188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:07:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ccbf84e5d65c3a055031b8f2ef8bbe6c05bb510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:22:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbcae2d160b244ca871cb71f26b77746\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f06598dd05e44e379b465a07627901d8 : SUCCESS in 1h 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a368d71ae6947f68c2cf05079859d8b : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/305fdc0421594a43a0f7eda11b13bfca : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e822288784f342f796c941f40f490f05 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21ddc4aeeec6401e8c5250323446230c : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e41fdc3e8e84cfab9da1e4631bbf90e : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6110cf3bffe0415fb4e4d127726c47d2 : SUCCESS in 5m 22s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/17ed7de817514c99b9b75cae00f23baa : SUCCESS in 1h 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f894a5dfd32b11a59c743008c0d723d2d82867de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:22:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff38c8e2d7897734eb95eddb52a0ae683e5c42e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:23:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f033b0b01fc64d8fab20bf3d9d773ea2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa34c4f86434187af8e3e5f97e2685d : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8573e0430bd0427297e61cc6fb4fd20f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05bcadc935fce5a199ccd30ccfe6ed795900112e","revisions":{"05bcadc935fce5a199ccd30ccfe6ed795900112e":{"kind":"REWORK","_number":1,"created":"2024-01-10 14:50:14.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/31/905231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/31/905231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/31/905231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/31/905231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/31/905231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/31/905231/1"}}},"commit":{"parents":[{"commit":"7419306d2669568b8ef1aac6283e680da841d82f","subject":"NVMe-oF: Fix attach when reconnecting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7419306d2669568b8ef1aac6283e680da841d82f"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-09-12 13:12:12.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-10 14:49:04.000000000","tz":60},"subject":"NVME-oF: Fix to support new \"address\" in sysfs","message":"NVME-oF: Fix to support new \"address\" in sysfs\n\nAttaching NVMe-oF no longer works in CentosOS 9 stream using nvme 2.4\nand libnvme 1.4.\n\nThe reason is that the \u0027address\u0027 file in sysfs now has the \u0027src_addr\u0027\ninformation.\n\nBefore we had:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420 After:\n\nNow we have:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420,src_addr\u003d127.0.0.1\n\nThis patch fixes this issue and future proofs for any additional\ninformation that may be added by parsing the contents and searching for\nthe parts we care: destination address and port.\n\nCloses-Bug: #2035811\nChange-Id: I7a33f38fb1b215dd23e2cff3ffa79025cf19def7\n(cherry picked from commit 268b1261ecf142c9ef603af2ff62815f9670e5d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/05bcadc935fce5a199ccd30ccfe6ed795900112e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/05bcadc935fce5a199ccd30ccfe6ed795900112e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
