)]}'
{"id":"openstack%2Fos-brick~905992","triplet_id":"openstack%2Fos-brick~stable%2F2023.1~I7a33f38fb1b215dd23e2cff3ffa79025cf19def7","project":"openstack/os-brick","branch":"stable/2023.1","topic":"nvmeof-v2-2023.1","attention_set":{},"removed_from_attention_set":{"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-01-18 15:39:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a33f38fb1b215dd23e2cff3ffa79025cf19def7","subject":"NVME-oF: Fix to support new \"address\" in sysfs","status":"MERGED","created":"2024-01-17 18:40:11.000000000","updated":"2024-01-19 04:45:28.000000000","submitted":"2024-01-18 15:39:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905992-nvmeof-v2-2023.1","meta_rev_id":"1761830c1377898edc1b026c136386f8990c81a2","_number":905992,"virtual_id_number":905992,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 15:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-01-18 08:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 09:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 09:28:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 20:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 23:07:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-01-18 09:28:36.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-01-19 04:45:28.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"bd62d1f43e953df3ffdfefafa74ddd3ac52adab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-17 18:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae8bcf13d7cd3f9c0bfd144e0c1eca8d9986b65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:39:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23be01d68390435c8238c543ab3ec04a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b0a0ba607534e168b11c6771537e2e5 : SUCCESS in 1h 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2081f2721a0a4f8faac72f61f355d515 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4026bfde204bee9309968021a31479 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8058c07021224401bd9495eeb188304d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7140e3edca914213af75a09c08c42faa : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d02a37e6be3144b79413ce0fc888fce8 : SUCCESS in 4m 15s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/49294fac77404d5b8e2af9aca6dc70f7 : SUCCESS in 4m 04s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/f3e8934442f3450b98ee08ba13efdf7c : SUCCESS in 1h 33m 06s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/65fd95e55172483296b5519bee013758 : SUCCESS in 1h 28m 44s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8194b0dc07c147c9981f26e7db077035 : SUCCESS in 1h 12m 06s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/e15f6f021f964c9a84ea4746bf2bf49d : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"792a2927191a3567916baa25c9f1d1a82710e322","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-01-17 23:07:14.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"872db81106c9f79a816849d2fda5703b5810ecf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-18 08:55:03.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c325be1be88dc7613826271e3c88cf78f029142","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-18 09:28:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af28b9d40c9b53f3643f4e59cc8c58d93b1a7c98","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-18 09:28:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0446bde4663750acd267dd2d16689dfd5b0fe795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:35:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/631d51d35eb14573a30d406498e870bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55b14045e2954538a32d0bb8d9d461c1 : SUCCESS in 1h 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4925773c2fc04faca409637d15c68515 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2431e208d910445ea02a0ca2bf82f5bf : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/addd96697adb43d385a62a3257bcab78 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a046aefbfa14ac4a32dff0d20d0dc43 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e981926667047998cf874f5ea0790c3 : SUCCESS in 4m 32s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/63aa577cf2404c599ba11a643d387bb9 : SUCCESS in 4m 18s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/60f1fc54c5954fddb04c324f01fa2de4 : SUCCESS in 1h 23m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9593e16b404a4bcbaa0bf978a2348dd3 : SUCCESS in 1h 35m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0ed7267b90044609a732cdd9234e2049 : SUCCESS in 1h 30m 32s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f1479d10c57a4546888e686ddb611840 : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82c42574aa947bd9ffc6fd2e2fa8b0d9115f35c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:23:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d76c6439f3569a8eef5ad9ee350ac7a5550f7429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:39:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18fed8d9934542cc913caf3ecee1a489\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bc0d84a89fa497c81541d5623dfdf50 : SUCCESS in 1h 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3811c87bf5dd4734be1537f7fba22c1c : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3dda199dbd7400ebb3b65fa1fa3a582 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/637247ff320745fa883581a79cf2fc92 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043dbdc3f53b4e4aa379c060c51dba87 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/beff7d73ea1c415aa95ec4af32351fe6 : SUCCESS in 4m 15s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7761d1ce33f646c28713b8a5b62635c6 : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7b74252f139b41bb76c5effd496114a2f5c2c4c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:39:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1761830c1377898edc1b026c136386f8990c81a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 04:45:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/454/.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1363ca36a34225f4ae6e4e835f504249479dd16","revisions":{"4ab8334065039a3684f6a95b9e47897fddf7dde5":{"kind":"REWORK","_number":1,"created":"2024-01-17 18:40:11.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/92/905992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/92/905992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/92/905992/1"}}},"commit":{"parents":[{"commit":"936d1e5c56ad755c2691cc4d6e52887dfef45121","subject":"NVMe-oF: Fix attach when reconnecting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/936d1e5c56ad755c2691cc4d6e52887dfef45121"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-09-12 13:12:12.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-17 18:37:14.000000000","tz":60},"subject":"NVME-oF: Fix to support new \"address\" in sysfs","message":"NVME-oF: Fix to support new \"address\" in sysfs\n\nAttaching NVMe-oF no longer works in CentosOS 9 stream using nvme 2.4\nand libnvme 1.4.\n\nThe reason is that the \u0027address\u0027 file in sysfs now has the \u0027src_addr\u0027\ninformation.\n\nBefore we had:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420 After:\n\nNow we have:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420,src_addr\u003d127.0.0.1\n\nThis patch fixes this issue and future proofs for any additional\ninformation that may be added by parsing the contents and searching for\nthe parts we care: destination address and port.\n\nCloses-Bug: #2035811\nChange-Id: I7a33f38fb1b215dd23e2cff3ffa79025cf19def7\n(cherry picked from commit 268b1261ecf142c9ef603af2ff62815f9670e5d0)\n(cherry picked from commit 05bcadc935fce5a199ccd30ccfe6ed795900112e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ab8334065039a3684f6a95b9e47897fddf7dde5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4ab8334065039a3684f6a95b9e47897fddf7dde5"}]},"branch":"refs/heads/stable/2023.1"},"b1363ca36a34225f4ae6e4e835f504249479dd16":{"kind":"NO_CHANGE","_number":2,"created":"2024-01-18 08:55:03.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/92/905992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/92/905992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/92/905992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/92/905992/2"}}},"commit":{"parents":[{"commit":"c0fded9fcd6bd58f883868840df5d932a68b6bad","subject":"NVMe-oF: Fix attach when reconnecting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c0fded9fcd6bd58f883868840df5d932a68b6bad"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-09-12 13:12:12.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2024-01-18 08:54:49.000000000","tz":60},"subject":"NVME-oF: Fix to support new \"address\" in sysfs","message":"NVME-oF: Fix to support new \"address\" in sysfs\n\nAttaching NVMe-oF no longer works in CentosOS 9 stream using nvme 2.4\nand libnvme 1.4.\n\nThe reason is that the \u0027address\u0027 file in sysfs now has the \u0027src_addr\u0027\ninformation.\n\nBefore we had:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420 After:\n\nNow we have:\n   traddr\u003d127.0.0.1,trsvcid\u003d4420,src_addr\u003d127.0.0.1\n\nThis patch fixes this issue and future proofs for any additional\ninformation that may be added by parsing the contents and searching for\nthe parts we care: destination address and port.\n\nCloses-Bug: #2035811\nChange-Id: I7a33f38fb1b215dd23e2cff3ffa79025cf19def7\n(cherry picked from commit 268b1261ecf142c9ef603af2ff62815f9670e5d0)\n(cherry picked from commit 05bcadc935fce5a199ccd30ccfe6ed795900112e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b1363ca36a34225f4ae6e4e835f504249479dd16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b1363ca36a34225f4ae6e4e835f504249479dd16"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
