)]}'
{"id":"openstack%2Fos-brick~906743","triplet_id":"openstack%2Fos-brick~master~Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-14 17:24:29.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-02-15 02:09:12.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-02-14 16:20:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9535\u003e replied on the change","reason_account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}}},"hashtags":[],"change_id":"Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16","subject":"Fix: FC partial target scan","status":"MERGED","created":"2024-01-25 16:37:25.000000000","updated":"2024-02-15 07:46:27.000000000","submitted":"2024-02-15 02:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906743","meta_rev_id":"1710855360fa4d5fae865a4276d9daae90386deb","_number":906743,"virtual_id_number":906743,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-15 02:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-14 17:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-14 16:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-14 17:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 18:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-25 19:34:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-01-25 19:41:32.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-25 19:58:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-01-25 20:17:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-25 20:37:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-01-29 21:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-01-30 11:20:50.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2024-01-31 22:53:44.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-02-05 21:48:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-02-06 13:49:53.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-02-12 00:16:14.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-02-13 03:50:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"09cec847cbdf75826f31b80cc5b6f5db0deef0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-25 16:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0133d66cfa6e91fcc4756b81749b455086f9dfdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 18:44:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9d176bc78a4475805edab9ccb6be2b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c8ebd09e4b246e99cb0358bf2746fed : SUCCESS in 1h 07m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87cdb6b1e9de46f69fc41c338612bb66 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24d90cbc7634d3785f6bfadeebc358c : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/134748c98b4145ceb8fc208c694e5494 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52400efbac884a149d18c860f38d5c7b : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66336688df3f4859a11762d6397308e7 : SUCCESS in 4m 06s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/054f0dac22a240c1b1c85f997d066db6 : SUCCESS in 3m 53s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/54bdd7deb0be41269f948bd3b3c40425 : FAILURE in 2h 03m 49s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/616fedbab3f24356ae9542402fe8dc66 : SUCCESS in 1h 53m 51s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ab21055cb2b9461eb2208bff5b4389ae : SUCCESS in 1h 15m 15s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/da677cc64e21460eb2eaf34b29af9234 : SUCCESS in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5e82ad7a60af0041488ab74482e9c838eb534e7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-25 19:34:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/485/.","accounts_in_message":[],"_revision_number":1},{"id":"689d8df15ef1c988d72179bafc1f9ef4890cc309","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-25 19:41:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/541a79f1a7194137959b19619762bdc8 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cbd2e3221a1a4399760f11c2e2fce8030793e5df","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-25 19:58:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/1/upstream-check/cinder-tempest-plugin-ontap-fcp/5a3c8f9/ : SUCCESS in 1h 14m 15s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/ce9cb3e/ : SUCCESS in 40m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/1/upstream-check/cinder-tempest-plugin-ontap-nfs/3e15528/ : RETRY_LIMIT in 8m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c4b0d2f/ : SUCCESS in 47m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b06cc2f23e3e228565ddad1bb540c264250deab8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-25 20:17:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/43/906743/1/check/DellEMC_PowerFlex_os-brick/2a15c54/DellEMC_PowerFlex_os-brick/97 : SUCCESS in 1h 57m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a103ded6a23a7fa5de0c31ca6d300738661527c3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-25 20:37:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/43/906743/1/check/DellEMC_PowerStore_FC-ob/0ba7d4b/DellEMC_PowerStore_FC-ob/203 : FAILURE in 1h 34m 35s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/43/906743/1/check/DellEMC_PowerStore_iSCSI-ob/627a13d/DellEMC_PowerStore_iSCSI-ob/124 : SUCCESS in 1h 41m 50s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/43/906743/1/check/DellEMC_PowerStore_NFS-ob/3cd6587/DellEMC_PowerStore_NFS-ob/99 : FAILURE in 26m 34s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/43/906743/1/check/DellEMC_PowerStore_NVMeTCP-ob/9fb38da/DellEMC_PowerStore_NVMeTCP-ob/79 : FAILURE in 1h 52m 59s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"0866427b284d5fce0bf0ba77927d055145d32bf7","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-26 12:03:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/495/.","accounts_in_message":[],"_revision_number":1},{"id":"f5860ef9a538cb2aff907a10c42dcda527f069d4","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-29 21:00:06.000000000","message":"Patch Set 1:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/1/fc : SUCCESS in 0h 42m 26s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/1/iscsi : SUCCESS in 0h 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f414fae7a37f23d0231565bd53ab010a46feb33d","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-01-30 11:20:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3ae2ed127646aff89b9df43ef5c642787624fac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-31 20:32:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"de035a80f88ed8c4516e24cedcc9823f529a5ce2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-31 20:40:51.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a3407a61377611a0a262f5ed8757e202152fb231","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-01-31 21:09:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2345d3268390a25edcff4761dbf53a1b3f21a2a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-31 21:25:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8105cbe99bb64226aa4c4e310879ee6f : FAILURE in 50m 49s","accounts_in_message":[],"_revision_number":2},{"id":"98d10d915d0759c7c6c3da6df4b77ed2490d3851","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-31 22:15:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/516/.","accounts_in_message":[],"_revision_number":2},{"id":"aa75319bba2a3977b6e1d14664503086ae88d652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 22:42:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3faea812dcd842079de6749909200238\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36faa14889a24e34958920651741a593 : SUCCESS in 1h 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07944dea16d4be98bd96e35ba341aba : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e77eabf2558547418734589df87531ca : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e44248024fce41c2b5d6babbc86840aa : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeed01d0dc394efb9f615d1d48972d64 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e885dbd6c22c43fc822c973efd32a711 : SUCCESS in 4m 12s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f6047785a8d74a81b288a9046175cae7 : SUCCESS in 5m 08s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/6cbd6408c1a44fd598975557305f90fc : SUCCESS in 2h 07m 41s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b9a448e2b11e40cb942cfa8fe5347c01 : SUCCESS in 1h 47m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/f1f67e6f2e1f45f898efc9d0b6e41c16 : SUCCESS in 1h 28m 22s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c0c32dbe269e4d81890981fa5bd81fda : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d8f99407d07389f7301dd7ec5292c3328f7ef7b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-31 22:53:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-2/FC/ : SUCCESS in 2h 20m 38s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-2/iSCSI/ : SUCCESS in 1h 51m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"7a4cef36d581c6fa7b9c2fa773a1776a810ee883","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-01 00:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/2/fc : SUCCESS in 1h 00m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/2/iscsi : SUCCESS in 0h 52m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3d41b6689069d9e5d815323f616f5acf39e95906","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-01 00:24:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/2/upstream-check/cinder-tempest-plugin-ontap-fcp/4580955/ : SUCCESS in 1h 03m 46s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/12aef64/ : SUCCESS in 45m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/2/upstream-check/cinder-tempest-plugin-ontap-nfs/47e1e8c/ : SUCCESS in 45m 25s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ad0573c/ : SUCCESS in 1h 26m 05s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d43b578978f963af4fb1560be8d14ca1b90b63c9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-01 01:00:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/527/.","accounts_in_message":[],"_revision_number":2},{"id":"5489405afe590ab342328fe8736a7eabee94f56f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-01 04:06:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/43/906743/2/check/DellEMC_PowerFlex_os-brick/0e903cd/DellEMC_PowerFlex_os-brick/98 : SUCCESS in 2h 05m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"1e041c726d84db48e286456a19fe2a6535cabe2b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-01 04:09:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/43/906743/2/check/DellEMC_PowerStore_FC-ob/316c617/DellEMC_PowerStore_FC-ob/204 : SUCCESS in 1h 27m 33s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/43/906743/2/check/DellEMC_PowerStore_iSCSI-ob/53784bd/DellEMC_PowerStore_iSCSI-ob/125 : SUCCESS in 1h 32m 16s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/43/906743/2/check/DellEMC_PowerStore_NFS-ob/d13eae9/DellEMC_PowerStore_NFS-ob/100 : SUCCESS in 1h 02m 30s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/43/906743/2/check/DellEMC_PowerStore_NVMeTCP-ob/fc7d8b7/DellEMC_PowerStore_NVMeTCP-ob/80 : FAILURE in 1h 55m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c303630d37503194d864cff1a78a57e360e8859","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-02 09:46:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce7fffc3581a626ef085699ad85d4500c664878d","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-02-05 10:52:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3a4addba8be11f1a52874387f14dd443446ff850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-05 19:56:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c79f3a77a7166af110c0fde01a19a26bbfc6c223","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-05 19:57:08.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"16128b79c910b6355144d1c81574160752305460","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 21:18:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/602/.","accounts_in_message":[],"_revision_number":3},{"id":"4e7e5992ce8859f3c96136026977fb9fb5598068","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-05 21:33:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/613/.","accounts_in_message":[],"_revision_number":3},{"id":"7088d32aad4bf95318354ed7db131aaedcfa1cae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-05 21:48:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_Unity_FC_os-brick/955c1a0/DellEMC_Unity_FC_os-brick/139 : SUCCESS in 1h 19m 14s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_Unity_iSCSI_os-brick/d7b7ff3/DellEMC_Unity_iSCSI_os-brick/137 : SUCCESS in 1h 28m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4fdba225ea840deab4be4175916cff297bd7967b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 21:53:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5666d12727ee4f40a6542b36ebe4e338\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52aa997456694656a96091ba20e0f9d7 : SUCCESS in 1h 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096cd4aa2cc54ba89bb6646d3879fde5 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e042746ffe4433a89c4c6f6836e7a8b : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52f4fc2b49dc4c98b27d81e0391c01aa : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15f4d60b4c2478c83e997d4318d5d9a : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19b7b04e90904545bcc095902d1386e8 : SUCCESS in 3m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f983762dda714aacaa4542aeca256604 : SUCCESS in 3m 55s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/321e382aeac64340b037b0595bd99f29 : SUCCESS in 1h 55m 37s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4d2545f5ffe84c2c80fcd2c81be950c1 : SUCCESS in 1h 45m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/b4fc90b887bf40be92a0903df4b2a8f7 : SUCCESS in 1h 29m 27s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1068eeecd4164afe9107138fc1648837 : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"48b3d9e2f79ef17732f390e70ed2c423876e58b2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-05 23:30:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_Unity_FC_os-brick/15519c1/DellEMC_Unity_FC_os-brick/140 : SUCCESS in 1h 18m 04s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_Unity_iSCSI_os-brick/cc62436/DellEMC_Unity_iSCSI_os-brick/138 : SUCCESS in 1h 36m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bb4fdd69e29ab4fdf2d3dbc103274ab473280bcb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-05 23:31:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/3/upstream-check/cinder-tempest-plugin-ontap-fcp/859d0eb/ : SUCCESS in 1h 09m 33s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/d5fa4a7/ : SUCCESS in 1h 31m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/3/upstream-check/cinder-tempest-plugin-ontap-nfs/5b5379a/ : SUCCESS in 45m 00s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3e9c805/ : RETRY_LIMIT in 7m 16s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ebd9a020378c86adff65f81072f848048746bce3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-05 23:48:28.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_PowerStore_FC-ob/29e77d5/DellEMC_PowerStore_FC-ob/206 : SUCCESS in 1h 27m 29s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_PowerStore_iSCSI-ob/8ad6e68/DellEMC_PowerStore_iSCSI-ob/127 : SUCCESS in 1h 33m 53s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_PowerStore_NFS-ob/4ce7e0c/DellEMC_PowerStore_NFS-ob/102 : SUCCESS in 1h 12m 37s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_PowerStore_NVMeTCP-ob/6772ce7/DellEMC_PowerStore_NVMeTCP-ob/82 : SUCCESS in 1h 54m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"4468680e9eb638928ce3898d49d27bf2cfc67c73","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-06 00:00:07.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/3/fc : SUCCESS in 0h 47m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/3/iscsi : SUCCESS in 0h 57m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a1c77af639c017827c828fa6b9dcff8e112d0e3f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-06 01:10:12.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-3/FC/ : SUCCESS in 3h 51m 3s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-3/iSCSI/ : FAILURE in 4h 46m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"2a994202aab39a4bcd15c97f5510257b3da5ed5b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-06 03:03:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/37e1eccfe24b4fc8a795577735af878b : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":3},{"id":"666dad78f68858b48347ce73df91fcd55db7973a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-06 09:45:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/43/906743/3/check/DellEMC_PowerFlex_os-brick/c0aff25/DellEMC_PowerFlex_os-brick/100 : FAILURE in 1h 55m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"70c85ef184a4bf4258113be968883a147992c1cb","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-02-12 00:16:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/43/906743/3/20240212054953/NEC-V-FC : SUCCESS in 3h 25m 56s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/43/906743/3/20240212054953/NEC-V-ISCSI : SUCCESS in 3h 26m 17s","accounts_in_message":[],"_revision_number":3},{"id":"35b27da50fe45384d99f2c09963de989a8fe0729","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-02-12 14:51:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2707bbd268e3fb619e50dfe452b8dbf77e8c1857","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-13 03:50:43.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b311501fcbc85f62e7379c364244c559a6234ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-13 16:44:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ee2064d8a9e83b06d6689534e2c5828d863aa7bf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-13 16:45:32.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c421737df3ec3d8c3a8f75990027db328b3b8323","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-13 17:12:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/39593c80dc7e4973a634faa1ea8f9aad : RETRY_LIMIT in 6m 42s","accounts_in_message":[],"_revision_number":4},{"id":"6d8673d9f764ab632646d52a1e53c6eb36995876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:41:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cfa0f76ad7e4c76b1200cd32e760572\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df3e70fbcb8b4ef6b0974609e342b2ec : SUCCESS in 1h 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a300f33432554208b1ce885e84460d4d : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a03a4da65bd4ca78d70ee489e56ca6e : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/812305e2868e4097894fe06753a4dd6f : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2843b146f443c993f6e23e64ab8e96 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc6b0b51a78344c6a5facc6ca3ee6131 : SUCCESS in 3m 49s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/1f78b0eb1f29419bbe874ffc1c4600dc : SUCCESS in 5m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2cfc8b39d9c54480b22ff9a826f9ce7e : FAILURE in 1h 49m 41s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1b57f80b5e4447c88ea1499c98c5f031 : SUCCESS in 1h 56m 06s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4562f97c2bf840ab828eab4dbb633f31 : SUCCESS in 1h 12m 39s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/0becc8bdfac54d5484ff9d7b4d869bdb : SUCCESS in 3m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"73818ddc516088a47cdeb7523433cad6e66f16f7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-13 19:07:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e7e8f8a/ : RETRY_LIMIT in 9m 19s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/35f9d48/ : RETRY_LIMIT in 8m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-nfs/4b4b1db/ : RETRY_LIMIT in 8m 55s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/60e089c/ : RETRY_LIMIT in 9m 01s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"72d1d1b246c4a1426e9db49a44fcbf7de1e0c6ea","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-02-13 19:48:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/702/.","accounts_in_message":[],"_revision_number":4},{"id":"e67f3e09d45f60f422533cb5a76d9a9c419e1e5a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-13 20:29:17.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_FC-ob/3e07cf4/DellEMC_PowerStore_FC-ob/210 : SUCCESS in 1h 42m 51s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_iSCSI-ob/244e574/DellEMC_PowerStore_iSCSI-ob/131 : SUCCESS in 1h 47m 58s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_NFS-ob/0b63697/DellEMC_PowerStore_NFS-ob/106 : SUCCESS in 1h 10m 10s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_NVMeTCP-ob/a19bc0b/DellEMC_PowerStore_NVMeTCP-ob/86 : SUCCESS in 1h 36m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"8da9ae322e8b981a02de487014dea51401c3d642","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-13 20:44:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_Unity_FC_os-brick/da66aa9/DellEMC_Unity_FC_os-brick/143 : SUCCESS in 1h 14m 25s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_Unity_iSCSI_os-brick/3cdf08f/DellEMC_Unity_iSCSI_os-brick/141 : SUCCESS in 1h 30m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"40e27ded1cbb63b89638056568730b04230a4201","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-02-13 20:53:00.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-4/FC/ : SUCCESS in 2h 33m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-906743-4/iSCSI/ : FAILURE in 4h 8m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"a58e1079185b8eb32e416db274e9fe057f4233da","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-13 23:44:37.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerFlex_os-brick/376a012/DellEMC_PowerFlex_os-brick/103 : FAILURE in 2h 19m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f38495e70010a47206e2dc4459d1a685f6bb995e","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-14 06:00:06.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/4/fc : SUCCESS in 0h 38m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/43/906743/4/iscsi : SUCCESS in 0h 51m 17s","accounts_in_message":[],"_revision_number":4},{"id":"9ad84c660a0c84029f14975042c6cb5a539c7d73","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-02-14 16:20:31.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d488c0eb817a13240f6e519118c128fb96de6fdc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-14 17:24:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c0ecd3faa240d1ac4ed99e23fefdee549e97081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 17:25:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c90d18545f3d533341a732110a6baf37635baa46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 19:15:29.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d54a7167e6824fbdacdf8b357053cc66\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cd9b28fde324157a0d9ce123b6e30fc : SUCCESS in 1h 44m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd767dd7cdcc492ab13226a5135aad04 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e34117935b9e4c62af145e5b5615f4f4 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fdc803f28f7a405aa3b97281d5c0355e : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937f789509c74a3297db8ee8d66c8d60 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632e432593ba4bdabeb198b2fdaf03ef : SUCCESS in 3m 59s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f23b4af3777b4b32873a15439e8e78ba : FAILURE in 1h 34m 05s","accounts_in_message":[],"_revision_number":4},{"id":"a5e93eef22f2ea7ff1bde4da3c36e757015e0169","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-14 22:04:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eddbdfb1084f2a6ac74cb88a7a7765e2f321ff60","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-02-14 22:27:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d4a48e997d314f11838b2269885fb83c : RETRY_LIMIT in 6m 19s","accounts_in_message":[],"_revision_number":4},{"id":"c096e9e143f13a809abba49c92a6a54e3d9c37f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 23:56:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c300ceea1d34035a5a31ff918a3d155\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19eeff26f54d4d558da17084fa5c96ff : SUCCESS in 1h 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85141cee4eb94e55974dd3abb4a4f704 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5bfeed53658484fae822f6e06bbb5a8 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e546b90468314375b5470c82dd03a901 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e401e954441147fdb08c30803ed0a0a7 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37663958f969491bbb2b3f81c06dca45 : SUCCESS in 3m 32s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0677d18e61ff46feb4896d142eb9f1f4 : SUCCESS in 3m 18s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/cbc1328ad4794a4ab0875f92e2078eca : SUCCESS in 1h 34m 02s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/39e8aa78d5f446a6967779a5de3e8526 : SUCCESS in 1h 51m 42s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/09044af9f83342b68f00108e2540f3c2 : SUCCESS in 1h 45m 07s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d9c28d9f72504b3bb0746c8b17555cf0 : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16b595abed8f2a67ca596a464b50765537f20755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 23:56:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"27f31908d34ddfbaf3e25e08fdcc2ae8f58187a4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-15 00:22:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-fcp/0c20fe8/ : RETRY_LIMIT in 9m 06s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/05415c9/ : RETRY_LIMIT in 8m 48s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-nfs/d1134ab/ : RETRY_LIMIT in 8m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/906743/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/e7e060e/ : RETRY_LIMIT in 9m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd3b1859b0170186441fab2f4f1ec7cdd5a44db2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-02-15 01:39:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_FC-ob/81922c4/DellEMC_PowerStore_FC-ob/211 : FAILURE in 1h 25m 35s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_iSCSI-ob/85dcf68/DellEMC_PowerStore_iSCSI-ob/132 : SUCCESS in 1h 42m 11s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_NFS-ob/89fcea7/DellEMC_PowerStore_NFS-ob/107 : SUCCESS in 1h 01m 10s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerStore_NVMeTCP-ob/d722241/DellEMC_PowerStore_NVMeTCP-ob/87 : SUCCESS in 1h 42m 30s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"e89b18144b986db6908516ad8cdb83381ef797b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 02:09:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c70da5bf8244748bee029530fcc54d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d29877507ac415ab3240e2eb69912cf : SUCCESS in 1h 50m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7833d056499c47fcbcccf062d66acff0 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3241c3be309494b9495da68dc0b8a43 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/deefc819ed9c44f89266037756513b91 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6222a6143ca64d818cf7eb27fd1fecec : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfa4eb80a4434bae90561b0ebe83dfdf : SUCCESS in 3m 30s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28629201337e48eba3c9e79ba8d4954e : SUCCESS in 2h 06m 24s","accounts_in_message":[],"_revision_number":4},{"id":"498bb57256c1f3807719d7e075a406c2be39073d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 02:09:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cd1b351fbf50adfd28949e9a554c659be4cf90a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 02:14:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d9a824d13184f549c1bd3837ebe398e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7bf4fcbad048f5844b1d2e58732e0f : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e558566753534f57aae88cdd0a45ec14 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4},{"id":"ad417c0947a44a7ba243659ef7defbe51661f652","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-15 03:17:00.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_Unity_FC_os-brick/8912dc0/DellEMC_Unity_FC_os-brick/144 : SUCCESS in 1h 16m 16s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_Unity_iSCSI_os-brick/2a1fd69/DellEMC_Unity_iSCSI_os-brick/142 : SUCCESS in 1h 34m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1710855360fa4d5fae865a4276d9daae90386deb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-15 07:46:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/43/906743/4/check/DellEMC_PowerFlex_os-brick/0494642/DellEMC_PowerFlex_os-brick/104 : SUCCESS in 1h 57m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f2154eedf0d04ef960dbf1df7fb87f74f6a35dbf","revisions":{"0b8c585e3cbc705c954f59a13ede93c83043729b":{"kind":"REWORK","_number":1,"created":"2024-01-25 16:37:25.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/43/906743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/906743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/906743/1"}}},"commit":{"parents":[{"commit":"7ad6acdff509b49a0a2ff39efe0af8e6123a6144","subject":"Merge \"NVMe-oF: Support nvme cli v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ad6acdff509b49a0a2ff39efe0af8e6123a6144"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-25 14:42:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-25 14:42:55.000000000","tz":330},"subject":"Fix: FC partial target scan","message":"Fix: FC partial target scan\n\nWhen fetching the target value (T in HCTL) for the storage HBAs,\nwe use the /sys/class/fc_transport path to find available targets.\nHowever, this path only contains targets that already have a LUN\nattached from, to the host.\n\nScenario:\nIf we have 2 controllers on the backend side with 4 target HBAs each (total 8).\nFor the first LUN mapping from controller1, we will do a wildcard\nscan and find the 4 targets from controller1 which will get\npopulated in the /fc_transport path.\nIf we try mapping a LUN from controller2, we try to find targets in the\nfc_transport path but the path only contains targets from controller1 so\nwe will not be able to discover the LUN from controller2 and fail with\nNoFibreChannelVolumeDeviceFound exception.\n\nSolution:\nWe have a retry mechanism with a default of 3 retries (total 4 scans\nincluding the first attempt), we address this issue by using the\n/sys/class/fc_remote_ports/rport-* path in the last try, to find all\nthe targets accessible to the host.\nThis way, we don\u0027t alter the current \"working\" mechanism of scanning\nbut also adding an additional way of discovering targets and improving\nthe scan to avoid failure scenarios during the last attempt.\n\nCloses-Bug: #2051237\nChange-Id: Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0b8c585e3cbc705c954f59a13ede93c83043729b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0b8c585e3cbc705c954f59a13ede93c83043729b"}]},"branch":"refs/heads/master"},"e18d164ab09b2f77b223891cbb49038013b2c361":{"kind":"REWORK","_number":2,"created":"2024-01-31 20:32:44.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/43/906743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/906743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/906743/2"}}},"commit":{"parents":[{"commit":"7ad6acdff509b49a0a2ff39efe0af8e6123a6144","subject":"Merge \"NVMe-oF: Support nvme cli v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ad6acdff509b49a0a2ff39efe0af8e6123a6144"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-25 14:42:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-31 20:27:06.000000000","tz":330},"subject":"Fix: FC partial target scan","message":"Fix: FC partial target scan\n\nWhen fetching the target value (T in HCTL) for the storage HBAs,\nwe use the /sys/class/fc_transport path to find available targets.\nHowever, this path only contains targets that already have a LUN\nattached from, to the host.\n\nScenario:\nIf we have 2 controllers on the backend side with 4 target HBAs each (total 8).\nFor the first LUN mapping from controller1, we will do a wildcard\nscan and find the 4 targets from controller1 which will get\npopulated in the /fc_transport path.\nIf we try mapping a LUN from controller2, we try to find targets in the\nfc_transport path but the path only contains targets from controller1 so\nwe will not be able to discover the LUN from controller2 and fail with\nNoFibreChannelVolumeDeviceFound exception.\n\nSolution:\nIn each rescan attempt, we will use two different paths to find\nthe targets.\n1. /sys/class/fc_transport/target\u003chost\u003e*\n2. /sys/class/fc_remote_ports/rport-\u003chost\u003e*\nIf a [c,t,l] combination is found from either path, we add it to\nthe list of ctls we later use it for scanning.\n\nThis way, we don\u0027t alter the current \"working\" mechanism of scanning\nbut also add an additional way of discovering targets and improving\nthe scan to avoid failure scenarios in each rescan attempt.\n\nCloses-Bug: #2051237\nChange-Id: Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e18d164ab09b2f77b223891cbb49038013b2c361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e18d164ab09b2f77b223891cbb49038013b2c361"}]},"branch":"refs/heads/master"},"2b661f0e5862cc826533e554f7252b2c144cd131":{"kind":"REWORK","_number":3,"created":"2024-02-05 19:56:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/43/906743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/906743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/906743/3"}}},"commit":{"parents":[{"commit":"7ad6acdff509b49a0a2ff39efe0af8e6123a6144","subject":"Merge \"NVMe-oF: Support nvme cli v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ad6acdff509b49a0a2ff39efe0af8e6123a6144"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-25 14:42:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-05 19:56:25.000000000","tz":330},"subject":"Fix: FC partial target scan","message":"Fix: FC partial target scan\n\nWhen fetching the target value (T in HCTL) for the storage HBAs,\nwe use the /sys/class/fc_transport path to find available targets.\nHowever, this path only contains targets that already have a LUN\nattached from, to the host.\n\nScenario:\nIf we have 2 controllers on the backend side with 4 target HBAs each (total 8).\nFor the first LUN mapping from controller1, we will do a wildcard\nscan and find the 4 targets from controller1 which will get\npopulated in the /fc_transport path.\nIf we try mapping a LUN from controller2, we try to find targets in the\nfc_transport path but the path only contains targets from controller1 so\nwe will not be able to discover the LUN from controller2 and fail with\nNoFibreChannelVolumeDeviceFound exception.\n\nSolution:\nIn each rescan attempt, we will first search for targets in the\nfc_transport path: \"/sys/class/fc_transport/target\u003chost\u003e*\".\nIf the target in not found then we will search in the fc_remote_ports\npath: \"/sys/class/fc_remote_ports/rport-\u003chost\u003e*\"\n\nIf a [c,t,l] combination is found from either path, we add it to\nthe list of ctls we later use it for scanning.\n\nThis way, we don\u0027t alter the current \"working\" mechanism of scanning\nbut also add an additional way of discovering targets and improving\nthe scan to avoid failure scenarios in each rescan attempt.\n\nCloses-Bug: #2051237\nChange-Id: Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2b661f0e5862cc826533e554f7252b2c144cd131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2b661f0e5862cc826533e554f7252b2c144cd131"}]},"branch":"refs/heads/master"},"f2154eedf0d04ef960dbf1df7fb87f74f6a35dbf":{"kind":"REWORK","_number":4,"created":"2024-02-13 16:44:16.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/43/906743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/43/906743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/43/906743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/43/906743/4"}}},"commit":{"parents":[{"commit":"7ad6acdff509b49a0a2ff39efe0af8e6123a6144","subject":"Merge \"NVMe-oF: Support nvme cli v2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/7ad6acdff509b49a0a2ff39efe0af8e6123a6144"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-01-25 14:42:55.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-02-13 14:46:22.000000000","tz":330},"subject":"Fix: FC partial target scan","message":"Fix: FC partial target scan\n\nWhen fetching the target value (T in HCTL) for the storage HBAs,\nwe use the /sys/class/fc_transport path to find available targets.\nHowever, this path only contains targets that already have a LUN\nattached from, to the host.\n\nScenario:\nIf we have 2 controllers on the backend side with 4 target HBAs each (total 8).\nFor the first LUN mapping from controller1, we will do a wildcard\nscan and find the 4 targets from controller1 which will get\npopulated in the /fc_transport path.\nIf we try mapping a LUN from controller2, we try to find targets in the\nfc_transport path but the path only contains targets from controller1 so\nwe will not be able to discover the LUN from controller2 and fail with\nNoFibreChannelVolumeDeviceFound exception.\n\nSolution:\nIn each rescan attempt, we will first search for targets in the\nfc_transport path: \"/sys/class/fc_transport/target\u003chost\u003e*\".\nIf the target in not found then we will search in the fc_remote_ports\npath: \"/sys/class/fc_remote_ports/rport-\u003chost\u003e*\"\n\nIf a [c,t,l] combination is found from either path, we add it to\nthe list of ctls we later use it for scanning.\n\nThis way, we don\u0027t alter the current \"working\" mechanism of scanning\nbut also add an additional way of discovering targets and improving\nthe scan to avoid failure scenarios in each rescan attempt.\n\nCloses-Bug: #2051237\nChange-Id: Ia74b0fc24e0cf92453e65d15b4a76e565ed04d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2154eedf0d04ef960dbf1df7fb87f74f6a35dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f2154eedf0d04ef960dbf1df7fb87f74f6a35dbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
