)]}'
{"id":"openstack%2Fos-brick~916062","triplet_id":"openstack%2Fos-brick~master~Iae1fd01f85bc429c7976e71fb9eeb8dbda795789","project":"openstack/os-brick","branch":"master","topic":"bug/2034685","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-08-21 20:03:48.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-08-20 11:39:25.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_4523\u003e using the hovercard menu","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2024-08-22 15:16:37.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-22 09:43:15.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Iae1fd01f85bc429c7976e71fb9eeb8dbda795789","subject":"Dell PowerFlex: Added retry after disconnect volume","status":"MERGED","created":"2024-04-18 11:51:53.000000000","updated":"2024-08-23 00:27:14.000000000","submitted":"2024-08-22 15:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"916062-bug/2034685","meta_rev_id":"8491e1e5542803efdb18de946423dc165ef93381","_number":916062,"virtual_id_number":916062,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 15:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-22 03:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-22 09:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-22 09:43:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-21 20:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 11:52:06.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-18 13:28:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-18 13:29:30.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-04-18 14:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-18 14:22:02.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-18 15:47:30.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-04-18 19:50:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-18 23:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-04-19 03:59:41.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-04-20 13:33:07.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-20 17:22:27.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-09 02:11:37.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2024-07-05 09:19:40.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"state":"CC"},{"updated":"2024-07-08 13:02:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-13 09:30:13.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2024-08-15 12:34:42.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-19 05:01:09.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-08-21 23:32:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"e6637bedfca53d9c787d14f97fbe97e04d9164ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-18 11:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d16e87727ec596246e48734bc722c63461da4673","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-18 11:52:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/1/check/DellEMC_PowerFlex_os-brick/7cb31b6/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"22dc80e33a7ba6e664e07e4d3cd5fa87b048ae69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-18 11:55:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4586952fccd23f8bcb62fe4433ddc88b2d1d1ee3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-18 11:55:37.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/2/check/DellEMC_PowerFlex_os-brick/732474a/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"9fc9955b3bf40b9580ce23a318afeea362477d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-18 11:56:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94d4a41ac89224d67d7f8e79a6f1597a95f3b211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-18 11:57:02.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5c6bb28d53b6fe8defe9db6235019699a141e05d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-18 11:57:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/3/check/DellEMC_PowerFlex_os-brick/884588b/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"52afe54ced044c189f876712658af60a64993347","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-18 11:57:15.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerFlex_os-brick/cac580f/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"44420f6223886977c46512a8e1490fe0752538f5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-18 13:28:59.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_Unity_FC_os-brick/b5454e7/DellEMC_Unity_FC_os-brick/164 : SUCCESS in 1h 18m 55s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_Unity_iSCSI_os-brick/ecf1f84/DellEMC_Unity_iSCSI_os-brick/161 : SUCCESS in 1h 31m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2c296341d250983a8d3bdfeda3db962010e7f131","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-18 13:29:30.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/235/.","accounts_in_message":[],"_revision_number":1},{"id":"5c7218014a8539074c41c162559330c5005dd4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 14:01:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ba408b16fb4455ab4ce65d463e9f1e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad01d3fd8d0f4ec49647541b3536276d : SUCCESS in 1h 08m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0228cd38c43c4ddc91c188dcadad2aa4 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c3a4d00fb24acb8c3db32a926489ba : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/566d267b2c7843eba45fcb7a9a8400a2 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b255a72a4f2142b8b9d0b2b6069e5fec : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9158c0764a67430bb464a9cc28a27a37 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f34f396fc214db48818a6fafacb5674 : SUCCESS in 3m 38s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a5f348ec0d9048f095b8c7cb6d557469 : SUCCESS in 3m 37s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ec4e16d6cced4ea99a32424be1b2801a : SUCCESS in 2h 04m 01s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f82428b40eda485e9a684dafe2626b0d : SUCCESS in 1h 36m 48s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/56f939eb2f34453cb2d5b936ba254804 : SUCCESS in 1h 22m 44s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/2b20e7fbe5e34bde84f07ea56bf7cd7d : SUCCESS in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"95bdcc9e760bf154bac831f6d21641305f35cf1c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-18 14:01:53.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerFlex_os-brick/259bf67/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"8bb20b10eb7798bf9c1700fc33b7d1d77d5bfb4c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-18 14:22:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/4/upstream-check/cinder-tempest-plugin-ontap-fcp/c5fdb88/ : RETRY_LIMIT in 3m 26s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/9f85e04/ : RETRY_LIMIT in 3m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/4/upstream-check/cinder-tempest-plugin-ontap-nfs/9619dc8/ : RETRY_LIMIT in 3m 49s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/03a8046/ : RETRY_LIMIT in 3m 23s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"011ea154af1b860bce2bf58ba04dc81f1b431333","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-04-18 14:55:36.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/236/.","accounts_in_message":[],"_revision_number":4},{"id":"9596de50051052189472908e925b2156a97d4893","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-18 15:42:35.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_Unity_FC_os-brick/aa0d6b4/DellEMC_Unity_FC_os-brick/165 : FAILURE in 1h 36m 15s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_Unity_iSCSI_os-brick/2c90868/DellEMC_Unity_iSCSI_os-brick/162 : SUCCESS in 1h 40m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e2c783843c203079c38ad241ed194c7d350bb63d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-18 15:47:30.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerStore_FC-ob/c6e3429/DellEMC_PowerStore_FC-ob/225 : SUCCESS in 53m 26s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerStore_iSCSI-ob/76b21e4/DellEMC_PowerStore_iSCSI-ob/145 : SUCCESS in 1h 39m 00s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerStore_NFS-ob/d82a669/DellEMC_PowerStore_NFS-ob/120 : SUCCESS in 1h 03m 07s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerStore_NVMeTCP-ob/3f95df2/DellEMC_PowerStore_NVMeTCP-ob/99 : SUCCESS in 1h 45m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"5a44444a866616ca6c30be171dabbc01f2b6c8ba","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-18 19:50:07.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-2/FC/ : FAILURE in 7h 28m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-2/iSCSI/ : FAILURE in 4h 12m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"45b3576de97d9de4af128fb8c709e8182990b7b2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-18 20:08:20.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-1/FC/ : FAILURE in 8h 16m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-1/iSCSI/ : SUCCESS in 2h 23m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"b98dd452f88f7c664cc499da0fc2a13b09fb59ea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-18 20:33:42.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-3/FC/ : FAILURE in 7h 41m 50s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-3/iSCSI/ : SUCCESS in 4h 6m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"1ccbb6c75a361eefa54edb7788ccfcc2cf6950f4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-18 20:56:54.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-4/FC/ : FAILURE in 7h 35m 2s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-4/iSCSI/ : FAILURE in 6h 3m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"13d7403210524dfa7efbb1004f367b5d99a2d769","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-04-18 23:00:08.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/916062/3/fc : SUCCESS in 0h 42m 41s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/916062/3/iscsi : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1c14f918c90a0d0b8a138de59ec2fe4d0be08dc0","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-19 03:59:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de8b1bd6d5cf2a986a352db9ca0d6ad7b6ac720d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-04-20 13:33:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/4/20240420190652/NEC-V-ISCSI : SUCCESS in 3h 04m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/4/20240420190652/NEC-V-FC : SUCCESS in 3h 26m 12s","accounts_in_message":[],"_revision_number":4},{"id":"9c915bf348d31a1824a87522960e8279a915afcf","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-04-20 17:22:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/4/20240420172735/NEC-ISCSI : SUCCESS in 4h 14m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/4/20240420172735/NEC-FC : SUCCESS in 8h 54m 48s","accounts_in_message":[],"_revision_number":4},{"id":"edf3af7fb38f0ac27af59944db3552631f2b19b8","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-22 08:42:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb6c2157773670cf8eafd1ecafec9a00315a86f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-22 08:42:23.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerFlex_os-brick/cb7c996/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"4a6a7abb4db0048fb097767179bec033e2436dcf","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-04-23 06:40:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e463f64a5fba5a949466937bab3d1391c13609e","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-05-09 02:11:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a9b5f19b4560465a5d7b0f7a2cd6c38cb2f092e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-09 04:01:18.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/4/check/DellEMC_PowerFlex_os-brick/7da1108/DellEMC_PowerFlex_os-brick/84 : SUCCESS in 1h 49m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f875bb11c93df7c0559f578319faf2739b7a3442","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-05 09:19:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"779afdd1612a67d2744baea80988ad4a752f8ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-07-08 09:28:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"95828e3b62a058f8e4445e2541299c4ada02f8ca","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-07-08 09:30:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8aba9f6f59f5d5a12389886bd6fc3e65415c8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 11:25:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef574394359423184b7806e0ffdfedf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0671a3e7879d44aebba57195913d96da : SUCCESS in 1h 40m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8579c10cd9e6470a87a922fed20ee009 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/872fd31f332a4177902ec59a13dc10fd : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8d3c89bf7984b30a4851e6d680bb3b2 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7db67bee0b94cd3bdb28c10cefd91cb : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288ece81cecb4f1096726f27da287329 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/691ad6ab9d4f4666971e1f3389b2e76a : SUCCESS in 3m 31s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8edd786cbc7040ac9b56851d962fc9b2 : SUCCESS in 3m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/0ec89b4ca815459dbfaced98dd57bfc3 : SUCCESS in 1h 51m 11s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/87e2a71e4d6d4aab96ef8ac8e00e1e29 : SUCCESS in 1h 47m 00s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1243f6b9bb9f4334a03dd79a92d5d677 : SUCCESS in 1h 26m 35s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/62e52ac572a04593b3cf9c83a4ed968d : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7a555fe41e9e322d447ecd5afad52dbc94e90f42","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-08 12:21:35.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerFlex_os-brick/34e06d1/DellEMC_PowerFlex_os-brick/105 : FAILURE in 2h 52m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"698975036a5ee722579f11c420b50b106f1ab80f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-08 13:02:57.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8f83b231d4664937b29803a5b40bf95d : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":5},{"id":"05a82b8143947ce156e4f4ba7fbe6cf4a0953ad3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-08 13:27:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/880841a/ : SUCCESS in 1h 42m 59s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/5/upstream-check/cinder-tempest-plugin-ontap-nfs/bdba160/ : SUCCESS in 1h 49m 40s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c76b703/ : SUCCESS in 1h 52m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"534874eea73e80eac615f3c60f906475979969bb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-08 14:08:59.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerStore_FC-ob/0f56793/DellEMC_PowerStore_FC-ob/237 : SUCCESS in 1h 02m 22s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerStore_iSCSI-ob/3b78a83/DellEMC_PowerStore_iSCSI-ob/157 : SUCCESS in 1h 51m 03s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerStore_NFS-ob/e1b7c5d/DellEMC_PowerStore_NFS-ob/132 : FAILURE in 1m 35s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerStore_NVMeTCP-ob/d56f186/DellEMC_PowerStore_NVMeTCP-ob/111 : SUCCESS in 1h 49m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"06fe68c4909ddf733d5a5da633305b21d6d49ecb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-08 14:26:42.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-5/FC/ : SUCCESS in 2h 26m 21s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-916062-5/iSCSI/ : SUCCESS in 4h 20m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"b8453fd73f77d853f670de8c52f48d782ff7b861","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-08 14:31:19.000000000","message":"Patch Set 5:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/196/.","accounts_in_message":[],"_revision_number":5},{"id":"60420e15059b1f80ac93955fe5424fc794c3e0e8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-08 15:03:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/692/.","accounts_in_message":[],"_revision_number":5},{"id":"bae41780c5059dad47e0ce48298e0f2648bdb976","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-08 17:54:05.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_Unity_FC_os-brick/71eee98/DellEMC_Unity_FC_os-brick/186 : SUCCESS in 1h 34m 39s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_Unity_iSCSI_os-brick/2eeaa20/DellEMC_Unity_iSCSI_os-brick/183 : SUCCESS in 1h 57m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"db656d8500e81c1eb1d3f45943d5247250c08b5a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-09 18:14:03.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/5/20240709185744/NEC-ISCSI : SUCCESS in 5h 15m 53s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/5/20240709185744/NEC-FC : SUCCESS in 8h 16m 15s","accounts_in_message":[],"_revision_number":5},{"id":"a5efcdc7440a941c397159e51ddf3c1b1fd8c40f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-10 22:00:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/5/20240711031939/NEC-V-ISCSI : SUCCESS in 3h 32m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/916062/5/20240711031939/NEC-V-FC : SUCCESS in 3h 41m 11s","accounts_in_message":[],"_revision_number":5},{"id":"bc4d4552413d1c97adb32799a1ff10c3660d0d95","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-01 17:43:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff24f1b1bc636076728330206dca85d5e46e22d7","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-07 18:11:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"647cbd7d86b6ae8ba0c8dfbda596fed38fb9002d","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-08 04:17:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bcf86d6bc5f14393796ee7289d0b5b6d0440f11e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-08 16:28:14.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerFlex_os-brick/e19d221/DellEMC_PowerFlex_os-brick/116 : FAILURE in 2h 01m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"f834dcac1c2b3daa0887f004fd7a25cf839dd1ce","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-08 18:39:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f7a87eab9129a78aae62f5846ec34c56a7a36a4d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 08:45:16.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/5/check/DellEMC_PowerFlex_os-brick/09fd2a8/DellEMC_PowerFlex_os-brick/118 : SUCCESS in 1h 44m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"5040a259114cabc24041e408e58df9c5592b83eb","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-19 05:01:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4d3a0e5fbbbb462e819bb4133207292cbf2edb12","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 14:46:29.000000000","message":"Patch Set 5: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3274d3a458b917c26ccc3d95e843e6874cebad67","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-21 15:10:59.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ff0b777bf6f39de9c12fd3cc5da831ef0df2f1cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 16:00:33.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e68820b838482f9d0a39a4f478299ec20213f3b4","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 16:01:54.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a5608bdd3244fcb4fe330281c8c3af91420822e6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 16:58:09.000000000","message":"Patch Set 6: Code-Review-1 Review-Priority+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8b6a29afe3c7561d08b11985fd3a3546f33e3318","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 17:12:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/365/.","accounts_in_message":[],"_revision_number":6},{"id":"f6787310f26beb8245e9dcbf1e31e701a9c72dc9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 17:27:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7395e7014422bbbec38f0c8a116d6bcdbb003dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 17:54:03.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"aaf869bb7aa1627be1527d7ecab4197e078d59a3","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 17:57:40.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d26c52775ce0c75bc1cf1296958243c48ab49577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 18:02:25.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0ef97f9ad6eac4a7f8ce270bfce4d901955a2793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 18:04:38.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"8cdbb61330b48a8aa5c0f03c260b57a7e3b8d787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 18:16:59.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"46d97a1b39081fddaf21641b29ddff1003e413e6","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 18:24:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dd41164fe01745199287a180cc0dd958525d6490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-21 18:44:31.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"2b5af22af003b6d4613073f6e0a448c1ec219661","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 19:35:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/369/.","accounts_in_message":[],"_revision_number":11},{"id":"42a80775433c42248956aaf136659d25a391ce8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 20:03:48.000000000","message":"Patch Set 11: Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"457f27ffab099752e060c4fde5d46ed49392198d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 20:24:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8b7b2e1b9b74b24848ba4f907063da8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c2bafe6a3174b4ead22fd9e34f5dc45 : SUCCESS in 1h 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aefabfc56a34fab8b35172f0ce3bc5a : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5bea53331204c55899348e79f7f287a : FAILURE in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf4e093870a4c37af7d592348992b86 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9c981787108465e9f60d7d2f20abaca : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41314c97807c4ff890dc8a7959cd35f3 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73abcc167c704e95872e9ccf60254a9c : SUCCESS in 4m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f22d65b412654c568be4aed32b8930d0 : SUCCESS in 4m 52s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/b2664576530f4216903cbdb6005eb1fc : SUCCESS in 1h 33m 16s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5a161cb5459b4d31a167c48cc6275ecb : SUCCESS in 1h 38m 52s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/91b45ee36ccb494491449fdbed1fa47c : SUCCESS in 1h 09m 16s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5a54b3700cfc45c688838d5c4a1f3386 : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0758ab3b5d6baca5d066ef8ec2b8c514d3522eea","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-21 20:54:17.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d6ddeaf7b6a43238330680c51ae37ac : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":11},{"id":"cdfd6c3a305b0d62e72cbdc11175e83433ba219b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 21:19:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4ad29e5b0425f50fe52106e811bf86749a0f57fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 23:26:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/506627d750a542b28e28cbefe095dae9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/643c78bab018411ea2d9c16074b87450 : SUCCESS in 1h 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227a881b4f20417baba65d586591f55b : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/420889b8669a481ca4cc3055bbf70e92 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bff3e7966d64f3fbadc77689546ef47 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b35f4b26e5dd443183ababb5b3f45901 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec42b848d1f846d586723d5162e9e6bc : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d1a5b1b59d45cdacc583b65b1585a7 : SUCCESS in 5m 05s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/df3194ca60804828a6a2145332f134cd : SUCCESS in 3m 54s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7dcf457a2bf84e3185d198f4578d2200 : SUCCESS in 1h 41m 13s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f7292a7e35394a319f2f6f3004f59d93 : SUCCESS in 2h 01m 28s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8a4ca56fa1d5408a9fd4386e14b07d68 : SUCCESS in 1h 34m 40s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/8082c05516524ac0994684a7874ea963 : SUCCESS in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"af51dbc0da36e755e5d6cf2e4ae69a9c9c1d6e07","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 23:32:31.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_23_26_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"706e84f3650242c96bab0fd630203bfbc98996ee","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-21 23:36:41.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f48d87cda27046698d9d60354bdeaf0f : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":11},{"id":"cb33b5c5d3bd2b47c7889d91c25162a2d8ffacad","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 23:57:37.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/6238181/ : RETRY_LIMIT in 3m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-nfs/5141d1b/ : RETRY_LIMIT in 7m 10s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d18f2bf/ : RETRY_LIMIT in 3m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"3e31fc28a7fd355d71e010e6a70859ef704293ad","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 00:48:41.000000000","message":"Patch Set 11:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/957/.","accounts_in_message":[],"_revision_number":11},{"id":"73f55c7e74c3b61798ea67e9cc2577c8ec5ae540","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-22 01:08:34.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_Unity_FC_os-brick/4be7136/DellEMC_Unity_FC_os-brick/200 : SUCCESS in 1h 57m 44s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_Unity_iSCSI_os-brick/8132265/DellEMC_Unity_iSCSI_os-brick/197 : SUCCESS in 2h 09m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"ac89726580c33ef59d62c29404c469b3fa28b179","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 03:18:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d0bd80924dca70cbcc118a339e60a02988ec76b3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-22 03:31:43.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"abd6331cd74291bd7b7f626b058abab655c00e31","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-22 03:55:20.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f129fddd0f38f304072a6d08ab99bb6a951b18cd","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 03:56:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"dc06d0e88216ab9884d8c0c51d1f720ca4601a3e","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 04:03:14.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"cca621a392d5559c8752d2b6de3a631471ccd2bd","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 04:17:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1330470bac05d235cebbdbccd9cf987652b427e7","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 04:29:35.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"b5e5f818fb33bb79b3adb33a2e7c99f841551c07","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 05:55:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"39a88f53c4896cf862a2f419d53c32298aad6a0f","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 08:09:23.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c2f206b76ccf2ed17ac3b0abe669b0d4ba48b597","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 09:13:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3d070159f7face9cd6f5cf8d79c14dd326a84547","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-22 09:18:55.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_FC-ob/c488200/DellEMC_PowerStore_FC-ob/251 : FAILURE in 1h 25m 44s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_iSCSI-ob/00124b5/DellEMC_PowerStore_iSCSI-ob/170 : SUCCESS in 1h 58m 43s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NFS-ob/cec1e6b/DellEMC_PowerStore_NFS-ob/144 : FAILURE in 1h 17m 30s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NVMeTCP-ob/c68474b/DellEMC_PowerStore_NVMeTCP-ob/124 : FAILURE in 2h 06m 57s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NVMeFC_ob/e3e04f3/DellEMC_PowerStore_NVMeFC_ob/14 : SUCCESS in 1h 09m 16s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"a1585beac8ff0a2458ed0ce21e187d7f3ad9a288","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-22 09:43:15.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"57346fb0d6bd424b4560e0169ebb5b78ae6a7551","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-22 09:43:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"89c6ce8f6496abea5cccd76c8303b283f24480bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 09:43:46.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d15224cc4772b064f4ffb98c581a9ba7203eaab6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 11:01:11.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerFlex_os-brick/f9252a4/DellEMC_PowerFlex_os-brick/125 : SUCCESS in 1h 46m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"e8438422339ac1c36d1744251461adf63ab3f74b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:33:56.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b97cfa2149e4145be79836e09aae71e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cef5a983f71e4eb78ab0d9190e8794cd : FAILURE in 1h 21m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5770bc1e51ef4157ace715838691b26d : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/895c4a066dee45aea78c1317d334a1cf : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ddb222bebff4567954d0627740ad9fe : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219b6a6db7df40ea9926d01680e530d4 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/247089264ddb4c2594adbc8a9efbced8 : SUCCESS in 3m 32s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52d80b5a536d4e30bc71f4c94b317ae9 : SUCCESS in 1h 44m 50s","accounts_in_message":[],"_revision_number":11},{"id":"21f40f57838af04e5d52d1b123954bd0d1eae12c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 11:35:52.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d9fa1e06c031fe9a9b9bfbf951e33c36582c5dda","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2024-08-22 11:38:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"27a7dcd96274686fe037f2531aa2613bdf2696a3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 12:54:14.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0f87249cc3494b3091ed6e5b016337ea : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":11},{"id":"aeddfec3593d604e12e92604fb97b9a6841fe23f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 13:18:34.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f0ea3d29e8b4e569ddfe1a99457bacc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3e47c029d45491b86c2592c85452fa9 : SUCCESS in 1h 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee428edde1c4b0293b0e65464b54161 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5b6a9bc10124271a2c1a19c0a196a8d : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a2edb5ca59e477a9e6478e35dc38a65 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11d70fdc5e2d4f84b0754e85932d81fc : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81a1372c605453cb64e0d6effa58167 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebcd7fc00f184874b9675bdf1247fe8a : SUCCESS in 6m 24s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/90f89aeb04f643eea64d14b1f85014d1 : SUCCESS in 3m 49s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/452d808e1be34e4981503d31dde323c5 : SUCCESS in 1h 37m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1241252cf1b44700a427f9a9df81e0c5 : SUCCESS in 1h 26m 44s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/34b0df120d1244edaec0fdf5bfd58ec6 : SUCCESS in 1h 28m 59s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/862b0b51129542b09d37eb0a7e945665 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bfd6372d3f5c9174f19f6b8aad56be68f2b733d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 13:20:43.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"8f18f3b9c68e31e682290bd05a0ac1aae7da3f4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-22 13:24:25.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2024_13_18_44\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"16aac9dae7fbeed60ef235606caa7ac2c2d94acd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-22 13:48:46.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-iscsi/0144815/ : RETRY_LIMIT in 8m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-nfs/d3213c3/ : RETRY_LIMIT in 8m 00s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/916062/11/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1bba5bb/ : RETRY_LIMIT in 3m 14s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"be0d3a6c6c1f8adc3bf576155bd00324cf3a01f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 15:16:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/487349ca4df5426a8b46961edc0c589d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8eb858510351474886e8910f4a7726d8 : SUCCESS in 1h 52m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3976a3838e83449394b3bb0d37cf95f3 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf133055afa4bfaab338bac9ce72282 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f73945622d74402b8f201b40e5a5f678 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93b4fec9b984898911b7eb14575f4a9 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6db5d5186424f66b13caf7126583208 : SUCCESS in 3m 28s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d08d24c749e6454eaa546db6710df3fc : SUCCESS in 1h 50m 22s","accounts_in_message":[],"_revision_number":11},{"id":"956c11774245775d3c7166d8b95124f579e7eb8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 15:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"847c4bdad27ed4b40a817afc1d09e202c62c72bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 15:18:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264831e7732f4f56badecc9ead3f7a03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35b75c3c3dd4410b4558db7bb181372 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b41e4a98ef5442ea8828d5b8a0c801fd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":11},{"id":"9ba72a033f97f1b26d6a12407d0c19c7e15305c6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-22 16:29:49.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_FC-ob/2f1b6ea/DellEMC_PowerStore_FC-ob/253 : FAILURE in 1h 04m 02s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_iSCSI-ob/d7e2975/DellEMC_PowerStore_iSCSI-ob/172 : SUCCESS in 2h 03m 05s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NFS-ob/a897faa/DellEMC_PowerStore_NFS-ob/147 : SUCCESS in 1h 06m 16s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NVMeTCP-ob/446018c/DellEMC_PowerStore_NVMeTCP-ob/126 : FAILURE in 2h 05m 16s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerStore_NVMeFC_ob/2b42f68/DellEMC_PowerStore_NVMeFC_ob/16 : SUCCESS in 1h 16m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"dd8ac5c1023c9e93f96487111813fee928be1635","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 20:43:57.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_PowerFlex_os-brick/551707c/DellEMC_PowerFlex_os-brick/127 : SUCCESS in 2h 02m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"8491e1e5542803efdb18de946423dc165ef93381","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-23 00:27:14.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_Unity_FC_os-brick/7cf1ab1/DellEMC_Unity_FC_os-brick/202 : SUCCESS in 1h 56m 33s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/62/916062/11/check/DellEMC_Unity_iSCSI_os-brick/f711e46/DellEMC_Unity_iSCSI_os-brick/199 : SUCCESS in 2h 04m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"267895ce071abc246f9d5b484734eb3b7b01abfb","revisions":{"80882351f72ff9236432ea33c22ac3145855e5de":{"kind":"REWORK","_number":1,"created":"2024-04-18 11:51:53.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/1"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:50:41.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2055022\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80882351f72ff9236432ea33c22ac3145855e5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80882351f72ff9236432ea33c22ac3145855e5de"}]},"branch":"refs/heads/master"},"88c04ebea0a41943e673cae0362c8f3095420ea9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-18 11:55:24.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/2"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:55:24.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88c04ebea0a41943e673cae0362c8f3095420ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/88c04ebea0a41943e673cae0362c8f3095420ea9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"33f74f4dc5e033b10e16c3379e93cf59d542a4f9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-04-18 11:56:49.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/3"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:56:49.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnecting volume","message":"Dell PowerFlex: Added retry after disconnecting volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/33f74f4dc5e033b10e16c3379e93cf59d542a4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/33f74f4dc5e033b10e16c3379e93cf59d542a4f9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f56c3961a9126ee54f6e8562901a64b4b378d1b6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-04-18 11:57:02.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/4"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:57:02.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f56c3961a9126ee54f6e8562901a64b4b378d1b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f56c3961a9126ee54f6e8562901a64b4b378d1b6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d4b1d58f9ade8549fce94bc1ab0c1d652e20f67c":{"kind":"REWORK","_number":5,"created":"2024-07-08 09:28:54.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/5"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-07-08 09:28:35.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d4b1d58f9ade8549fce94bc1ab0c1d652e20f67c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d4b1d58f9ade8549fce94bc1ab0c1d652e20f67c"}]},"branch":"refs/heads/master"},"9496bd242498b94ecdcd587d39cd04e43ed5bfbd":{"kind":"REWORK","_number":6,"created":"2024-08-21 16:00:33.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/6"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 15:59:21.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9496bd242498b94ecdcd587d39cd04e43ed5bfbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9496bd242498b94ecdcd587d39cd04e43ed5bfbd"}]},"branch":"refs/heads/master"},"155ee73df4a1e9ec2d261b01ea2eddf493e2a723":{"kind":"REWORK","_number":7,"created":"2024-08-21 17:54:03.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/7"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 17:53:03.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/155ee73df4a1e9ec2d261b01ea2eddf493e2a723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/155ee73df4a1e9ec2d261b01ea2eddf493e2a723"}]},"branch":"refs/heads/master"},"f785a850f49feb94c9d6e325318a6bda94cbe658":{"kind":"REWORK","_number":8,"created":"2024-08-21 18:02:25.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/8"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 18:01:53.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f785a850f49feb94c9d6e325318a6bda94cbe658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f785a850f49feb94c9d6e325318a6bda94cbe658"}]},"branch":"refs/heads/master"},"5541195bc4206c6a6b851a31e2067034dc3a3579":{"kind":"REWORK","_number":9,"created":"2024-08-21 18:04:38.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/9"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 18:04:20.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5541195bc4206c6a6b851a31e2067034dc3a3579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5541195bc4206c6a6b851a31e2067034dc3a3579"}]},"branch":"refs/heads/master"},"e16910728a9492bfb067fe5e89fcd1276352dc5e":{"kind":"REWORK","_number":10,"created":"2024-08-21 18:16:59.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/10"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 18:16:35.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e16910728a9492bfb067fe5e89fcd1276352dc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e16910728a9492bfb067fe5e89fcd1276352dc5e"}]},"branch":"refs/heads/master"},"267895ce071abc246f9d5b484734eb3b7b01abfb":{"kind":"REWORK","_number":11,"created":"2024-08-21 18:44:31.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/62/916062/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/62/916062/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/62/916062/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/62/916062/11"}}},"commit":{"parents":[{"commit":"3f17bca07d4cf3d7cf9d591c13124fc53eedbf47","subject":"Merge \"Update master for stable/2024.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3f17bca07d4cf3d7cf9d591c13124fc53eedbf47"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-04-18 11:32:20.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2024-08-21 18:44:05.000000000","tz":0},"subject":"Dell PowerFlex: Added retry after disconnect volume","message":"Dell PowerFlex: Added retry after disconnect volume\n\nImplemented a retry mechanism post-volume\ndisconnection to ensure reliability.\nIntegrated a function execution to validate\nthe existence of the volume, safeguarding\nagainst connectivity issues arising from the\nremoval of the old path during a hard reboot.\n\nCloses-Bug: #2034685\nChange-Id: Iae1fd01f85bc429c7976e71fb9eeb8dbda795789\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/267895ce071abc246f9d5b484734eb3b7b01abfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/267895ce071abc246f9d5b484734eb3b7b01abfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
