)]}'
{"id":"openstack%2Fos-brick~923345","triplet_id":"openstack%2Fos-brick~master~Iacc65567941a7eb9c1d6a799cd3b877624c26d9d","project":"openstack/os-brick","branch":"master","topic":"bp/dell-powerstore-nvmefc","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2024-11-20 14:48:14.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"35759":{"account":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"last_update":"2026-09-23 13:07:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iacc65567941a7eb9c1d6a799cd3b877624c26d9d","subject":"Add  NVMe-FC support in NVMeOF connector","status":"ABANDONED","created":"2024-07-03 01:59:40.000000000","updated":"2026-09-23 13:07:01.000000000","total_comment_count":25,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"27e020fbc587350d378b589a44f5ce70f5796383","_number":923345,"virtual_id_number":923345,"owner":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-10 04:13:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":-1,"date":"2024-07-31 15:51:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-01-08 18:57:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 04:37:16.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-07-03 05:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 05:21:11.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-07-03 05:47:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-03 06:01:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-03 06:14:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-07-03 06:17:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-03 07:06:57.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-07-03 07:29:15.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-07-04 04:55:40.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-07-04 04:55:40.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-07-04 04:55:40.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2024-07-11 02:31:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 08:19:02.000000000","updated_by":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"reviewer":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"state":"CC"},{"updated":"2024-07-12 02:50:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-07-23 16:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-07-24 05:55:53.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2024-07-24 05:55:53.000000000","updated_by":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2024-07-31 15:50:39.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2024-07-31 15:51:08.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2024-08-22 15:01:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-01-08 15:10:07.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"CC"},{"updated":"2025-01-08 18:57:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-07-30 07:14:08.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"},{"updated":"2026-09-23 07:05:05.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"}],"messages":[{"id":"289d969fc4e61ff9751f6cab4feb6660cea8d31d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-03 01:59:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"846ac044197b4e46e184b41a4d2da116eafc066c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-03 02:27:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"897a46dc4292085171851fac34f0c52d99ed6a24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-03 02:28:17.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4a91336a928b7c577121197ac9d4442cb6362084","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-03 04:37:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-1/FC/ : SUCCESS in 2h 37m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-1/iSCSI/ : SUCCESS in 2h 12m 9s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"a3e5b0f4528d1209a39f44416f003c1ae60d235b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 05:15:00.000000000","message":"Patch Set 3: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abeebf5de1024a10b12b0a22204449fc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c81024bd8b48329b8c17d5092790fd : SUCCESS in 1h 50m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68af0b063115466c93843cabdd545384 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7a5d28e0dad4696bd151495f7e5813d : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e61e1e9cf3e74694bab35989e614c477 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d38d6cf306a44859a3ac42f7ab0c8ec5 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0852f77d335247d9b33d93b484511363 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/000438fd881c4d62b63703c95824dd09 : SUCCESS in 3m 44s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/cba37045c37f470b9e0310e1ff5c2736 : SUCCESS in 3m 53s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/e954cc21336e40038f803c03ecd8b54c : SUCCESS in 1h 44m 15s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c8de614833854e7099cec19c3a86c053 : SUCCESS in 1h 36m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/06edace361394c41bece9bbf8de868a8 : SUCCESS in 1h 24m 00s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4fff1910a20f49d4b2881d7af0e3eb61 : FAILURE in 3m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4e303f9f4f2a1496eae373d8570ed5698ec9bb5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-03 05:21:11.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_07_2024_05_15_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"5e3875e673551a46dec49fc07b3ec031102bd261","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-03 05:47:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/671/.","accounts_in_message":[],"_revision_number":3},{"id":"b642ab780773d8d72e3fcbe9cc0dc009def97521","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-03 06:01:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/928b88b47d8b4051916c459bdd9eceba : SUCCESS in 1h 15m 08s","accounts_in_message":[],"_revision_number":3},{"id":"38ea67453c0499ed49223cf1907272634a0d25b0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-03 06:14:48.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_Unity_FC_os-brick/82e10a0/DellEMC_Unity_FC_os-brick/183 : SUCCESS in 1h 20m 26s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_Unity_iSCSI_os-brick/40b520c/DellEMC_Unity_iSCSI_os-brick/180 : SUCCESS in 1h 51m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c49b39dd012778ef4988d10e3dd74f49ab1fb39","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-03 06:17:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_PowerFlex_os-brick/b2863d2/DellEMC_PowerFlex_os-brick/102 : SUCCESS in 2h 02m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"f0acb52ff5cc5ceed2b60b8209c74539c4b5900c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-03 07:06:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/31976ec/ : SUCCESS in 1h 34m 13s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/3/upstream-check/cinder-tempest-plugin-ontap-nfs/ec03fde/ : SUCCESS in 1h 40m 10s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/8b095a0/ : SUCCESS in 46m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"6a571ae6bb4373e860f2ebc23f1098a916dea6b1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-03 07:29:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_PowerStore_FC-ob/a9f513a/DellEMC_PowerStore_FC-ob/235 : SUCCESS in 1h 02m 31s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_PowerStore_iSCSI-ob/3771f9b/DellEMC_PowerStore_iSCSI-ob/155 : SUCCESS in 1h 59m 41s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_PowerStore_NFS-ob/94536aa/DellEMC_PowerStore_NFS-ob/130 : FAILURE in 1m 37s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/45/923345/3/check/DellEMC_PowerStore_NVMeTCP-ob/5546b23/DellEMC_PowerStore_NVMeTCP-ob/109 : SUCCESS in 2h 14m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"2321c4b77e9c38b01d4dc6a9cbd864c4388d34ae","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-03 10:39:13.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-2/FC/ : FAILURE in 7h 39m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-2/iSCSI/ : SUCCESS in 1h 54m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"0b14eed85914535b61072e7162098b9b1e949b8d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-03 11:25:44.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-3/FC/ : FAILURE in 7h 55m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-3/iSCSI/ : SUCCESS in 2h 27m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"f733c95735cc1389a9825760bc0b27c5bdd334d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-07-10 02:27:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a44cd4619ba36337e2b8364150c56ab135c8e59","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-07-10 03:40:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bc832c3922c048d5b8da0a13c58865d7 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":4},{"id":"f6b2da9e693864aec18e64be0753f87fcf987da9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-10 04:07:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/708/.","accounts_in_message":[],"_revision_number":4},{"id":"af8a75e924b718ca469766f485dfae008dcc7059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 04:13:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30367ec5b57d4851a65f95ea69c4cfe1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f394eea0e742437d82277f302695df4f : SUCCESS in 1h 40m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2bd25f7af14b6994c478d1c070881b : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5281cfa8b0bc4ed28ba745a3333948a9 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1588ea9d00b14370bf49504116a913cf : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/179a2e962a9140dca644fdc7c591932f : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5642c0fe884cd58d1bcaaa131da635 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c93f593fa9434a37ba21e2a5b6951a5b : SUCCESS in 4m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/2dcd8e64e61944508eb66d2d6f16619b : SUCCESS in 4m 13s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/4b1ef61d94814290b01d603f45047623 : SUCCESS in 1h 25m 49s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64d8e96eb056430581080d6f54c9165c : SUCCESS in 1h 29m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0cfbbf28cdd64b22b7d93fbe75a901e1 : SUCCESS in 1h 26m 39s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/6b2bc7e65bd84c34913404ebce86a641 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f08d04b7754cf5dd721432d683b361dc33f35953","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-10 04:16:33.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_Unity_FC_os-brick/6567905/DellEMC_Unity_FC_os-brick/187 : SUCCESS in 1h 30m 21s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_Unity_iSCSI_os-brick/f5d5dde/DellEMC_Unity_iSCSI_os-brick/184 : SUCCESS in 1h 48m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8bfbb583a92ac1e4875e4145c55cd9df2fd5cc76","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-10 04:16:41.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/211/.","accounts_in_message":[],"_revision_number":4},{"id":"515bd626dc9ceba25461eec9a45f640efd63782d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-10 05:20:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerFlex_os-brick/9a41c0b/DellEMC_PowerFlex_os-brick/106 : FAILURE in 2h 52m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f0569fab1aecb64bc19472e0bbb349dd0a3bf611","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-10 05:50:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/83f06fd/ : SUCCESS in 1h 19m 44s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/4/upstream-check/cinder-tempest-plugin-ontap-nfs/f85253c/ : SUCCESS in 52m 14s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/45/923345/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/65fa569/ : SUCCESS in 46m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"24ce0e6ad02fd0aeeaf356aba18a7e83fde0f799","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-10 06:24:59.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-4/FC/ : SUCCESS in 2h 21m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-45-923345-4/iSCSI/ : SUCCESS in 3h 56m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"d48ce3eda3ec7aa72c23693524eae4a83e66116c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-10 06:58:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_FC-ob/ec06b80/DellEMC_PowerStore_FC-ob/239 : SUCCESS in 1h 08m 00s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_iSCSI-ob/ab983df/DellEMC_PowerStore_iSCSI-ob/158 : SUCCESS in 2h 05m 35s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NFS-ob/8e53c13/DellEMC_PowerStore_NFS-ob/133 : FAILURE in 1h 18m 41s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeTCP-ob/190cf7d/DellEMC_PowerStore_NVMeTCP-ob/112 : SUCCESS in 1h 51m 57s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeFC_ob/e988c5a/DellEMC_PowerStore_NVMeFC_ob/1 : FAILURE in 4m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"496a8841b68c6eb46ea7523aee27a6c8ae0475bc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-11 02:31:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/45/923345/4/20240711054651/NEC-ISCSI : SUCCESS in 5h 16m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/45/923345/4/20240711054651/NEC-FC : SUCCESS in 5h 44m 49s","accounts_in_message":[],"_revision_number":4},{"id":"50cd266421be7822ea9103f43c7c223330b06233","author":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"date":"2024-07-11 08:19:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d4a766d66cee0330a6387ec357e9774eb141e90","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-11 10:13:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_FC-ob/40baf3b/DellEMC_PowerStore_FC-ob/240 : FAILURE in 35m 44s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_iSCSI-ob/9fddc4a/DellEMC_PowerStore_iSCSI-ob/159 : FAILURE in 1h 54m 10s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NFS-ob/885c1dc/DellEMC_PowerStore_NFS-ob/134 : FAILURE in 1h 11m 18s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeTCP-ob/091a67f/DellEMC_PowerStore_NVMeTCP-ob/113 : SUCCESS in 1h 48m 18s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeFC_ob/3a1bd0f/DellEMC_PowerStore_NVMeFC_ob/2 : FAILURE in 4m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"3f13ec13f3beac2c2122e7fcf5aaf548180d71b2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-12 02:50:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/45/923345/4/20240712075040/NEC-V-FC : SUCCESS in 3h 41m 51s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/45/923345/4/20240712075040/NEC-V-ISCSI : SUCCESS in 3h 59m 25s","accounts_in_message":[],"_revision_number":4},{"id":"51bf24d81f193576b91fb07e79eddc1f23705aea","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-15 08:41:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_07_2024_08_35_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"d14f59ea17fdd93036f5e3c84913ec69914b7030","author":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"date":"2024-07-23 05:55:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e3704e4a14cce96e3eac36c861e921929f0160e9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-23 09:30:50.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_FC-ob/4373141/DellEMC_PowerStore_FC-ob/243 : SUCCESS in 1h 03m 00s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_iSCSI-ob/5def7dc/DellEMC_PowerStore_iSCSI-ob/162 : FAILURE in 2h 09m 23s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NFS-ob/e9b001c/DellEMC_PowerStore_NFS-ob/136 : FAILURE in 1h 17m 05s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeTCP-ob/1d5005c/DellEMC_PowerStore_NVMeTCP-ob/115 : SUCCESS in 1h 53m 02s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeFC_ob/0411a2c/DellEMC_PowerStore_NVMeFC_ob/6 : SUCCESS in 1h 13m 08s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"3b81de89a0b6c194576b08e1be7501aed42fcdd9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-23 16:00:07.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/45/923345/4/fc : SUCCESS in 1h 05m 02s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/45/923345/4/iscsi : SUCCESS in 0h 56m 14s","accounts_in_message":[],"_revision_number":4},{"id":"8b473d8a62e2dee505d5bf9228bea7b45d3b9f8a","author":{"_account_id":36625,"name":"Rick Liu","email":"rick.liu@dell.com"},"date":"2024-07-24 00:50:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a08ba2f918ed300cd3671f865abb4c0d085975ef","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-24 02:56:02.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_FC-ob/d4987a2/DellEMC_PowerStore_FC-ob/245 : SUCCESS in 59m 36s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_iSCSI-ob/be5971e/DellEMC_PowerStore_iSCSI-ob/164 : SUCCESS in 2h 05m 42s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NFS-ob/576d7b5/DellEMC_PowerStore_NFS-ob/138 : FAILURE in 1h 11m 36s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeTCP-ob/7dbc51f/DellEMC_PowerStore_NVMeTCP-ob/117 : SUCCESS in 1h 47m 37s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/45/923345/4/check/DellEMC_PowerStore_NVMeFC_ob/558d4b6/DellEMC_PowerStore_NVMeFC_ob/8 : SUCCESS in 1h 19m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"438d7e576dc8315e462e974f613dbd9d13e5144e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-07-31 15:50:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fb883c8d4edf9dbbd6390815a565cf17238243ea","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-07-31 15:51:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4eaf842b384d7e756c57018dfbf2db0c3e7fbfcc","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-01 04:45:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"69faa8ca40d252840fc0f88b90a8ac680dadc51e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-12 21:43:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26067c3e65e6035ab755082ea7883d795447060d","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-22 06:50:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea70db99e71efad4c7a58eafa2ec5e23ff0cc4be","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-08-22 13:28:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"445ba8a3d19af2ced8ec48e079d467c2df691040","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-22 15:01:43.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/0f6118d035e441e585a954de35cfdeb7\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/33dd3a94c485445baba966abbad203b1 : SUCCESS in 37m 35s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a3fecd8a1a114225bd9c4fb07141fac0 : FAILURE in 48m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/93e0ef14ac36459fb389b3aa7521234f : FAILURE in 1h 33m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/846d1d0c3b6c4116ab58d3acce5970cb : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":4},{"id":"af2cea6de4eb39b7fe973022093fd326b8327fb2","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-11-13 14:06:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d06c31d1faea0274f718d27f81099e9e36366e4d","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-11-13 14:07:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"64760366d49d23b1f20be41008f9705bed33722b","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-11-19 08:32:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac2180582d1120b013e0343c447d48b8e1f9ae11","author":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"date":"2024-11-19 15:35:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"494cf4cc1ef43d4a20373401054903f61439d066","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-11-20 14:48:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85554888a7f582d62ba7312db3ffdb9381195a55","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-08 14:33:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3e5142546ab6cb2d86e00368328b827ad76fac62","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-01-08 15:10:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a9d37e62568048a83f30b706ce11e0a495c48f2d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-01-08 18:57:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"016bb9606734da3a8f534aaa96a4bf32295c3d48","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-01-09 15:30:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c83926a354d28389e553d4ed3c40d280be9e92e","author":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"date":"2025-07-30 07:14:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c98b59623226e492fbeaba92e000067e23332a64","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2026-07-18 04:47:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e023945249902fdef6c5e0049e934bac50eeb28","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-23 07:05:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c88656adf04f5f7ab88620c6dca4ca4e667e8cf3","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-23 07:07:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"27e020fbc587350d378b589a44f5ce70f5796383","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-23 13:07:01.000000000","message":"Abandoned\n\nsuperseded by changes 997815 and 997816","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d0eea3d2c9db1c7ef642ad97c41340d9744b6928","revisions":{"f6007a7f739ca729d6bd445ddfd61a656ca1895b":{"kind":"REWORK","_number":1,"created":"2024-07-03 01:59:40.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/45/923345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/923345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/923345/1"}}},"commit":{"parents":[{"commit":"383edfddb8de9f5186516b1123b013e1965ac4b0","subject":"Merge \"Fix fast8 tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/383edfddb8de9f5186516b1123b013e1965ac4b0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-06-20 07:03:35.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-07-03 01:59:23.000000000","tz":0},"subject":"Dell PowerStore NVMe-FC","message":"Dell PowerStore NVMe-FC\n\nAdd NVME-FC support for Dell PowerStore driver\n\nImplements: blueprint dell-powerstore-nvmefc\nChange-Id: Iacc65567941a7eb9c1d6a799cd3b877624c26d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f6007a7f739ca729d6bd445ddfd61a656ca1895b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f6007a7f739ca729d6bd445ddfd61a656ca1895b"}]},"branch":"refs/heads/master"},"f393ae2f44ff0bb5395e7e84525fc9cd3aa6ec9d":{"kind":"REWORK","_number":2,"created":"2024-07-03 02:27:54.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/45/923345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/923345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/923345/2"}}},"commit":{"parents":[{"commit":"383edfddb8de9f5186516b1123b013e1965ac4b0","subject":"Merge \"Fix fast8 tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/383edfddb8de9f5186516b1123b013e1965ac4b0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-06-20 07:03:35.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-07-03 02:22:27.000000000","tz":0},"subject":"Add  NVMe-FC support in NVMeOF connector","message":"Add  NVMe-FC support in NVMeOF connector\n\nAdd NVME-FC support in NVMeOF connectore.\nThis patch is tested with Dell PowerStore driver:\nhttps://review.opendev.org/c/openstack/cinder/+/923346.\n\nImplements: blueprint dell-powerstore-nvmefc\nChange-Id: Iacc65567941a7eb9c1d6a799cd3b877624c26d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f393ae2f44ff0bb5395e7e84525fc9cd3aa6ec9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f393ae2f44ff0bb5395e7e84525fc9cd3aa6ec9d"}]},"branch":"refs/heads/master"},"004af6209f30402cbcd2391055ba5133b653508d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-03 02:28:17.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/45/923345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/923345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/923345/3"}}},"commit":{"parents":[{"commit":"383edfddb8de9f5186516b1123b013e1965ac4b0","subject":"Merge \"Fix fast8 tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/383edfddb8de9f5186516b1123b013e1965ac4b0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-06-20 07:03:35.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-07-03 02:28:17.000000000","tz":0},"subject":"Add NVMe-FC support in NVMeOF connector","message":"Add NVMe-FC support in NVMeOF connector\n\nAdd NVME-FC support in NVMeOF connectore.\nThis patch is tested with Dell PowerStore driver:\nhttps://review.opendev.org/c/openstack/cinder/+/923346.\n\nImplements: blueprint dell-powerstore-nvmefc\nChange-Id: Iacc65567941a7eb9c1d6a799cd3b877624c26d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/004af6209f30402cbcd2391055ba5133b653508d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/004af6209f30402cbcd2391055ba5133b653508d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d0eea3d2c9db1c7ef642ad97c41340d9744b6928":{"kind":"REWORK","_number":4,"created":"2024-07-10 02:27:40.000000000","uploader":{"_account_id":35759,"name":"Yian Zong","display_name":"Yian Zong","email":"yian.zong@dell.com","username":"yianzong"},"ref":"refs/changes/45/923345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/923345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/923345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/923345/4"}}},"commit":{"parents":[{"commit":"383edfddb8de9f5186516b1123b013e1965ac4b0","subject":"Merge \"Fix fast8 tox env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/383edfddb8de9f5186516b1123b013e1965ac4b0"}]}],"author":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-06-20 07:03:35.000000000","tz":0},"committer":{"name":"Yian Zong","email":"yian.zong@dell.com","date":"2024-07-10 02:23:54.000000000","tz":0},"subject":"Add  NVMe-FC support in NVMeOF connector","message":"Add  NVMe-FC support in NVMeOF connector\n\nAdd NVME-FC support in NVMeOF connectore.\nThis patch is tested with Dell PowerStore driver:\nhttps://review.opendev.org/c/openstack/cinder/+/923346.\n\nImplements: blueprint dell-powerstore-nvmefc\nChange-Id: Iacc65567941a7eb9c1d6a799cd3b877624c26d9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d0eea3d2c9db1c7ef642ad97c41340d9744b6928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d0eea3d2c9db1c7ef642ad97c41340d9744b6928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
