)]}'
{"id":"openstack%2Fos-brick~925905","triplet_id":"openstack%2Fos-brick~stable%2F2024.1~Ib075ec62a2bf993615c5c802f34acd7838bfa2af","project":"openstack/os-brick","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-14 16:54:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib075ec62a2bf993615c5c802f34acd7838bfa2af","subject":"Wait for multipath device to be ready for I/O","status":"MERGED","created":"2024-08-07 16:53:33.000000000","updated":"2024-08-14 16:55:21.000000000","submitted":"2024-08-14 16:54:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925905","cherry_pick_of_change":920516,"cherry_pick_of_patch_set":10,"meta_rev_id":"aacf92c3fe59a2e2d28a440603c137c0554a418e","_number":925905,"virtual_id_number":925905,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-14 16:54:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-14 14:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-14 14:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2024-08-14 15:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2024-08-14 15:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 19:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-09 07:45:11.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-09 08:17:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-13 12:58:10.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-14 14:45:46.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2024-08-14 14:49:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-14 15:11:29.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"}],"messages":[{"id":"3ee0cbcca45dcfbcefb722db06b885cf50247532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-07 16:53:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"493f7238503e628f84a0ff955627369964fed9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 19:21:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85bec447dab84e4499aa7b118ddf96fe\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d0cfcd4bc954b7fad34a41705bf2743 : SUCCESS in 1h 31m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a2c0ede0294dc99472e9171c7e25f5 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1f8bd957d834f629eb1b48648f8aaf4 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a57adfba4694d6ebf12cadabe1ff8c2 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e82c1c88c874636bef9e65d6703040f : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cabec77c91743e4915a2b7486c41548 : SUCCESS in 4m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6786a703c9bd4e07bfcf566c9b77d3f4 : SUCCESS in 3m 57s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a9e09c50defb47a0b43a9837e7e5687a : SUCCESS in 2h 00m 45s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9f65d49fafe9490487a33647685490c9 : FAILURE in 2h 19m 03s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/34fa04f7537e4a6daf81b423dd4a5546 : SUCCESS in 1h 26m 31s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/c5562aef6445492aa5cfc2a63669ec1f : SUCCESS in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d8c410d8608fe16c615bfbd5c31866a245c6dda","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-09 05:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ee8994fba7bdb8766e62b6aaa09a447b3fe5e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:39:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c862da8b6a845b6aececf0fd5d38b2a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6edebc0e91d4cf4bc5205c11c3f5e1f : SUCCESS in 1h 39m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/608a68ba3b174daca78d3d256bffad40 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c464912e3414ab09491553521cdd755 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7899bed66bbb46629f59bbcdd8a6e17f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e76dffca0fe4bb59394b54fdfb82ca4 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b93475cb169e4a2683b03eb840512311 : SUCCESS in 3m 40s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9e74d3e1fae6461b9e03ef4b51c60ed7 : SUCCESS in 4m 50s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/12963460cfe147f19e8c2437fea1ccd7 : SUCCESS in 1h 33m 49s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78a3aac6b0b149948b13b20f17c5782d : SUCCESS in 1h 59m 36s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/52cb814e52cb4e50927709546124389a : SUCCESS in 1h 17m 22s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/3fc76d54f6ac4c8b8d44337b7c7a5a44 : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6ca559dadf4db9a5e0464c5f2a04c20cfb3d53d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-09 07:45:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/09_08_2024_07_39_31\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"835263aadc85b06420af16a5ba7cf43ccbe88b4a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-09 08:17:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/925905/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/115a07e/ : RETRY_LIMIT in 3m 38s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e7a33d23de5f36cbe4d3acaa60017989b2b10f5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-13 12:58:10.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/870/.","accounts_in_message":[],"_revision_number":1},{"id":"34c2186e847b8df0c76732007c5a6be2689d8636","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2024-08-14 14:45:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ec1553af2070f1d67a53ebd3d6fbf97ef80ae8b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-14 14:49:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a9238bc8b085a04f937dc63009a9022411b3f9a","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2024-08-14 15:11:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"699fc33d7bf33e3dc5daf17ce44166e5a5e63908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 15:12:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f42ff0ea62cc48f14e42d00031a00306677ef1d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:54:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/644ac62a96a44c53a164cf16d1174da5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5346a6a82ace4f2eac6df5c02983f64e : SUCCESS in 1h 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e277cb4380e749989fb0bf0100774cd8 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8af3ba92aafb4d64a6631e9b08915f48 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f59b2b8f8198458da1db5b7447b27d23 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f75e275c2e478bacbf2731b2b6c52d : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac000a450eca400d929bde51d9aaf0d3 : SUCCESS in 3m 44s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/20bad68778334ab3ad678e1f5b89d20a : SUCCESS in 1h 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"99b38f4269033362cdc38a608474779fc2517822","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:54:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aacf92c3fe59a2e2d28a440603c137c0554a418e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:55:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17596d2d9cb948e6b12e6bf9052ae1f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4dbb3106f7d4f029c300296c0ed66ef : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/563487c861d446dab1377226931fed8b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80da091dd3f949b3f06deef80b92061c7a487c58","revisions":{"80da091dd3f949b3f06deef80b92061c7a487c58":{"kind":"REWORK","_number":1,"created":"2024-08-07 16:53:33.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/05/925905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/05/925905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/925905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/925905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/05/925905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/05/925905/1"}}},"commit":{"parents":[{"commit":"823cda4b8a2102281facfcd666514fc6147d770a","subject":"Silence warning when running in a container with overlayfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/823cda4b8a2102281facfcd666514fc6147d770a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-05-27 13:03:54.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-08-07 16:53:33.000000000","tz":0},"subject":"Wait for multipath device to be ready for I/O","message":"Wait for multipath device to be ready for I/O\n\nThe \"multipath -C \u003cmpath_name\u003e\" command waits for the multipath\ndevice map to be ready for I/O. This is useful in preventing race\nconditions when we are trying to write to the multipath device\nbefore it is ready for I/O.\n\nWe added 2 new config options to make the wait time configurable,\n\n1. wait_mpath_device_attempts - defaults to 4 attempts\n2. wait_mpath_device_interval - defaults to 1 second\n\nCloses-Bug: #2067949\n\nChange-Id: Ib075ec62a2bf993615c5c802f34acd7838bfa2af\n(cherry picked from commit 639f953194cdc07b1e6aff1fc1ca2a7bc9d28536)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80da091dd3f949b3f06deef80b92061c7a487c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/80da091dd3f949b3f06deef80b92061c7a487c58"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
