)]}'
{"id":"openstack%2Fos-brick~926293","triplet_id":"openstack%2Fos-brick~master~I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515","project":"openstack/os-brick","branch":"master","topic":"LUKS-image-encryption","attention_set":{},"removed_from_attention_set":{"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2024-08-22 21:07:18.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-08-22 12:28:56.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-22 21:07:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515","subject":"Add get_passphrase_from_secret utility function","status":"MERGED","created":"2024-08-14 15:42:54.000000000","updated":"2024-08-22 21:08:23.000000000","submitted":"2024-08-22 21:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926293-LUKS-image-encryption","meta_rev_id":"b38b77a145a02e012413e27e4304c4e4b86d39fe","_number":926293,"virtual_id_number":926293,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 21:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-22 12:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-08-22 19:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-08-22 19:14:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-22 12:28:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 16:12:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-14 16:22:46.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2024-08-14 16:30:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 18:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 19:59:42.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-15 12:27:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-08-15 16:32:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-16 07:26:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-08-16 20:54:58.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-19 18:26:35.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-08-21 11:15:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-08-21 11:32:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-21 13:38:16.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-21 14:47:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-08-22 19:14:09.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"b196e59644d5366b0a6ba86a80dca15a18cb1b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dc5fbc52e407ed9804ff23c48851d38fa23968d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:44:41.000000000","message":"Topic set to LUKS-image-encryption","accounts_in_message":[],"_revision_number":1},{"id":"6ba11d7ecde392b8c22f048e9899eda10f899703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:45:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"59ebb1f3f2cace6e614feea136439c5082bff4b9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-14 16:12:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/926293/2/check/DellEMC_Unity_FC_os-brick/7639130/DellEMC_Unity_FC_os-brick/197 : FAILURE in 18m 12s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/926293/2/check/DellEMC_Unity_iSCSI_os-brick/4f69ed9/DellEMC_Unity_iSCSI_os-brick/194 : FAILURE in 26m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a44b2ec09f187073eb523b873fa676a2d775b624","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-14 16:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23f1c411dda4d08452fcf904cee55fbef6c0da19","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-14 16:30:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/926293/2/check/DellEMC_PowerFlex_os-brick/55097fd/DellEMC_PowerFlex_os-brick/121 : FAILURE in 44m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"9423c3833682c92ed11d3674ae28a1682b999c6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:51:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dea0802c6d10443cb8cb171754e40538\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2209c74ae4ad41a08ecc254ae040b247 : SUCCESS in 1h 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e05194b2b699495eacff4fc86fdf603c : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99fdbc4ded2d4fedb33e73d85dd47230 : FAILURE in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c92504a262c420496db0c359d93f9a6 : FAILURE in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e8597827f584e419504163ba14d09f0 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/586f3ff60e4549a19f080104a34e0f7a : SUCCESS in 7m 26s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/5d5b712324d547d3a4e8f619a89f0be2 : FAILURE in 3m 23s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/a25a63212b684b2ebcc10054eb25a5a8 : TIMED_OUT in 3h 04m 26s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/0b67290d44604a2a805d486be14f94d4 : FAILURE in 1h 40m 30s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e2d55c1256948129ffe861820fd889b : SUCCESS in 1h 47m 29s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/4add7081664a4727878ffd4ebdf9e734 : SUCCESS in 1h 23m 37s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/00c13a45828c482e82f536a99de8062f : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bca598854caa0e237f2a89bbae978254deb54526","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-14 19:59:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4ddebd3c3b7a409a8a8d8740c376ec4a : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2472fc2cf8f44bcba4502b753d6332fc45bd5009","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-15 12:27:57.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/894/.","accounts_in_message":[],"_revision_number":2},{"id":"0e4c649aa8285802b8cecc7e9d71162c78fe46e4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-15 14:24:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cc53f539b8e8b2841263a29038d856e51f880df","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-15 14:37:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6827c1e35e541a7dc66e634442781d4225256275","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-15 16:32:24.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/c765541a2b954b66bd69218af7863f15\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/eaecf8b1b45547e7b6a29f7805044182 : FAILURE in 1h 15m 56s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/412658f589354e1b8c3fa128612b60bb : DISK_FULL in 1h 54m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d08d84284f44c5189536e9bc0e40ac1 : RETRY_LIMIT Ansible setup timeout in 1m 15s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d4ae8999e802432ba1d7b35cd3b82da8 : DISK_FULL in 1h 27m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4d93689cbfffdc70eaea7d4833d8aaef135845be","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2024-08-15 21:46:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f27730eb0416a914d2bdf4714e3fc3bb3bc871dd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-15 23:56:43.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/8098e0f1253c4b2f849ce61bf3a2a8f3\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e1036c5c1714f90a68e8ddf4bc02e77 : FAILURE in 59m 09s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/09b93bbc432441b68dd691fa6534a778 : DISK_FULL in 1h 30m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3145a693bd7a4a5794a4b1a6a810a88b : RETRY_LIMIT Ansible variable freeze timeout in 1m 16s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/af5a291c493448329ea778da48b4fa8e : DISK_FULL in 2h 04m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2780b6836a5538199921d9307af17eb00bbb56d6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-08-16 07:26:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/926293/2/20240816120224/NEC-V-ISCSI : SUCCESS in 3h 35m 50s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/926293/2/20240816120224/NEC-V-FC : SUCCESS in 4h 23m 42s","accounts_in_message":[],"_revision_number":2},{"id":"58ab6ae1ea5052f583bcf837ffc637a71e8cf1c2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-08-16 20:54:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/926293/2/20240816205734/NEC-ISCSI : SUCCESS in 5h 42m 46s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/93/926293/2/20240816205734/NEC-FC : SUCCESS in 8h 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"daaf0e5ee9346584fd9ba90226c618fa2d5270b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-20 15:10:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fae16ba8238dfa73a5db3608b80a5ff0051a0884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:59:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62c9b693a9dd41dda198c6a9a8b515a5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5325f8c9a5b74fc8a5d78b623b60e6cb : SUCCESS in 1h 48m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1bf338c0e04a9ab130d37f7abc0a4d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c95e58758804b8f9eff0390052d9e3e : FAILURE in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80c7f366148b49b59c1652b4112ea790 : FAILURE in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23dda6fb166a4efea5c8328be69e0c6e : FAILURE in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65784f42f87641d78ce19db975511130 : SUCCESS in 5m 40s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d0c829873e344ab0b2ba37f383373970 : FAILURE in 4m 20s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/7688f9c6bdad4564b78a1662ab76b2ce : SUCCESS in 1h 44m 11s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/bfaaa868de70449ca6a5dc2ed917b48b : SUCCESS in 1h 41m 14s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91dcff7e39ec4996bb66c3d174b88392 : SUCCESS in 1h 34m 04s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/86d222bfbdd3476e9046f41250b33a72 : SUCCESS in 1h 22m 20s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/0701f40c45034de68d56240f083fbead : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17c3610854624f7bb2898f1c77dc9e5a213031ce","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-20 17:10:05.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f454cf4af69445119d7995bed5eac1f6\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0d9433e9ff7541c8b7f41685adabcdbd : FAILURE in 38m 14s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b3ca1fae01e94678a092c77f19555a21 : FAILURE in 1h 52m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a8aebc5f8cf24ac38ccedcb064e29ea1 : RETRY_LIMIT Ansible setup timeout in 1m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0135f17e397e4be1b1982afee2413612 : FAILURE in 1h 41m 40s","accounts_in_message":[],"_revision_number":3},{"id":"d4a99b24b6124a657a8dc4e91eb7d436aca8a5fa","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 17:34:08.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/345/.","accounts_in_message":[],"_revision_number":3},{"id":"7f28bb22aa31a4dff7f4a736c9e7df95e28ae9f2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-20 18:47:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3a95d1a99ffe496cae33fd04feca7854 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b0498649fd04047f8d05ca2c0c650d9e3cd17a93","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-20 20:32:49.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/930/.","accounts_in_message":[],"_revision_number":3},{"id":"71ef40caf8e22ccd18e07234bc5736b733d0ef9e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-20 21:03:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/926293/3/check/DellEMC_Unity_FC_os-brick/85361d7/DellEMC_Unity_FC_os-brick/198 : SUCCESS in 2h 01m 57s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/926293/3/check/DellEMC_Unity_iSCSI_os-brick/f9990a5/DellEMC_Unity_iSCSI_os-brick/195 : SUCCESS in 1h 53m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2dc6bed9b7570238431272309c0910b72e7d0fb8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 21:56:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/926293/3/check/DellEMC_PowerFlex_os-brick/103cdc0/DellEMC_PowerFlex_os-brick/123 : SUCCESS in 1h 57m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b0d772fba01499ae8a01e3d162fdc604b4865efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-21 09:16:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"86eea4700e18c6cb42c09813576514ce69c7ac5d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-21 09:16:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerFlex_os-brick/f3b9cd2/DellEMC_PowerFlex_os-brick/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f25a545faef9067f5af787d1b9c8d549c9e598d6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 10:50:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/353/.","accounts_in_message":[],"_revision_number":4},{"id":"bcf7933917b61e1c8936e9953008b3313b6a55f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-21 10:58:36.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/59074fb9bf664f9faa0ac1e9d33d615d\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1ac159f0f5d84be4a6ea59c0348c47dc : FAILURE in 36m 16s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/8496296ebca14028ac589a837d737323 : FAILURE in 1h 40m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3a38c453268a4f078c48475831ac8651 : RETRY_LIMIT Ansible setup timeout in 1m 15s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/efaa7d955e704d03a6f0fd87d5b469af : FAILURE in 1h 42m 06s","accounts_in_message":[],"_revision_number":4},{"id":"2eb934adb3d9ff443abc1b6dd7c000508faf69b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:10:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f48d550fa29407c9b82ac89ba969b42\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73fc0e5892d048168ccccc49c19575e6 : SUCCESS in 1h 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5c2f31d2bd429a921dc997ee31a4bd : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/011e8c221f7f41ab97db1d5f03d7893a : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff4f22e5ad914e2f80f206cc734f2700 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5a216e4b491416f99dbf197253eae3f : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a92746fe4a418393dfe684fc4ae7c5 : SUCCESS in 5m 23s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f1caf3f6e7914ae2948f15cd04f386cf : SUCCESS in 4m 00s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/9707d8bdcbb448d2b618c66bfe8fa0c6 : SUCCESS in 1h 44m 28s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/706e0f61c80442c5a84fd0267c3d75a5 : SUCCESS in 1h 37m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90eb23d661a34e7f99593dbbd9da850a : SUCCESS in 1h 48m 50s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/bb641ba2bc754cf7a60f23ad2e2b1ea7 : SUCCESS in 1h 28m 57s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/8b48456e95e34882887e62a0b04d03de : SUCCESS in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0a026303ca089edd418bfecb217cdedd7783bd68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-21 11:15:44.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/21_08_2024_11_10_14\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"ade412df138ec2def59e330e42cead4c260f2eb4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 11:32:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/53a7a02/ : RETRY_LIMIT in 3m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/4/upstream-check/cinder-tempest-plugin-ontap-nfs/d0945f6/ : RETRY_LIMIT in 16m 10s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f0b3b88/ : RETRY_LIMIT in 5m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"2f5ca6b588b7d066e15d3fa78f6ff37d686f7696","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-21 12:21:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/05357c12d48f4290ace27d91ad4f2e83 : FAILURE in 1h 10m 54s","accounts_in_message":[],"_revision_number":4},{"id":"226d22c4f9dc4beb62969582111c3ab4899affa4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-21 13:18:03.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/938/.","accounts_in_message":[],"_revision_number":4},{"id":"b3ebf90168c9d4ea5e6acc7163287a924f9a481f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-21 13:38:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerStore_FC-ob/2b123ba/DellEMC_PowerStore_FC-ob/250 : FAILURE in 1h 15m 21s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerStore_iSCSI-ob/f55d823/DellEMC_PowerStore_iSCSI-ob/169 : SUCCESS in 2h 05m 16s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerStore_NFS-ob/53b2c66/DellEMC_PowerStore_NFS-ob/143 : FAILURE in 1h 07m 53s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerStore_NVMeTCP-ob/84638c9/DellEMC_PowerStore_NVMeTCP-ob/123 : FAILURE in 2h 01m 52s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_PowerStore_NVMeFC_ob/f14a6fa/DellEMC_PowerStore_NVMeFC_ob/13 : SUCCESS in 1h 10m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"7ead8f3eedeac216974f0837db6df8c5c1c0248a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 14:47:15.000000000","message":"Patch Set 4: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5dac5b982cc407cfea380d0ac5aa214ebe198a0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-21 15:15:11.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_Unity_FC_os-brick/7860ba6/DellEMC_Unity_FC_os-brick/199 : SUCCESS in 1h 53m 24s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/926293/4/check/DellEMC_Unity_iSCSI_os-brick/38b3345/DellEMC_Unity_iSCSI_os-brick/196 : SUCCESS in 2h 05m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a099923475d4f17d08ea628fa4746e2c828383d4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-21 22:23:23.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"27929560630d491ae670f042cd695526752147aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-22 07:59:26.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"83750cae727032a32c901c834ab18db0a4502f54","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-22 08:01:42.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"768d3056c010a56fdccb2bd2817ca7a08bdf72ef","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 08:49:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/376/.","accounts_in_message":[],"_revision_number":5},{"id":"357fe159756824efff9dfafd674628e8fde5fe47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-08-22 09:18:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c4793b27cc4c4f079dcae5d0bb8092d6 : FAILURE in 1h 17m 30s","accounts_in_message":[],"_revision_number":5},{"id":"68c0f5dda4fc4fb32302ba2f1e7a63309ba704de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-22 09:33:47.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f16be57394e9444cab1c662e1dbc9e8e\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/03a92a02200c46348e1507c91b32807b : SUCCESS in 41m 14s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ec9722099e1c486faeda54fd77f770f1 : FAILURE in 1h 05m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e6b07ce345dd4c7787c124a124e387c9 : FAILURE in 1h 33m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6e30ea97fa4449c78268dd95dec2cd4d : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":5},{"id":"5c30ee8a43020315dfecd086dd650ae937133169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 09:57:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1265a3b3c93c47369249ab1f3ee3260e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8804a1f8f8254847b2762fda47bf72aa : SUCCESS in 1h 43m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a2b707a6d24bd29e7b29484184cc2d : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b948aae2938d454298c9b7e06f04f55f : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a07e41136f6a480f8ce032c28bd38b06 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22df5d8f0d974bb49f9d8ab6d1c96123 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05ed93bd1154577a97cc2c177fee506 : SUCCESS in 6m 47s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/306aca972ba7470a92711343578a4342 : SUCCESS in 3m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/590faa0172fb4591b0a073c993eca92e : SUCCESS in 1h 52m 41s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ed24643713734035bf66f91627410d3a : SUCCESS in 1h 45m 17s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5f39ced7e71a43f580379fb087a45b76 : SUCCESS in 1h 49m 40s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/cbb0f922594249c59e770ccb473e666e : SUCCESS in 1h 23m 59s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b0cfdac69a974f0787e719825fa17cac : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d9f9b01a8e01bca73f3f51e9c390cea72eaaa043","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-22 10:07:17.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2024_10_00_40\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"372a950d3b07c236fc1140bcfd2f3b733848b74b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-22 11:13:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/577ab4d/ : RETRY_LIMIT in 9m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/5/upstream-check/cinder-tempest-plugin-ontap-nfs/1f69d34/ : RETRY_LIMIT in 9m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/93/926293/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/263fb3d/ : SUCCESS in 54m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"045a6b0b925e784b5e41784cd6683badf4d5078d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-08-22 11:36:46.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/964/.","accounts_in_message":[],"_revision_number":5},{"id":"eeb73d58849103f6c0a7accd5a43b998e8f37da2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-22 12:01:49.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC-ob https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerStore_FC-ob/21a80dd/DellEMC_PowerStore_FC-ob/252 : FAILURE in 1h 07m 58s\n- DellEMC_PowerStore_iSCSI-ob https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerStore_iSCSI-ob/69de59e/DellEMC_PowerStore_iSCSI-ob/171 : SUCCESS in 1h 54m 56s\n- DellEMC_PowerStore_NFS-ob https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerStore_NFS-ob/9b5190a/DellEMC_PowerStore_NFS-ob/146 : SUCCESS in 1h 08m 21s\n- DellEMC_PowerStore_NVMeTCP-ob https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerStore_NVMeTCP-ob/f8c9b98/DellEMC_PowerStore_NVMeTCP-ob/125 : FAILURE in 2h 04m 07s\n- DellEMC_PowerStore_NVMeFC_ob https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerStore_NVMeFC_ob/9531298/DellEMC_PowerStore_NVMeFC_ob/15 : SUCCESS in 1h 24m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"b069caf96a0b75a0e85d4a6fd910a3a91ebbb548","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-22 12:28:56.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"92ca3fa09dc5da673aa7667761a579d69ccf701c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 14:50:12.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_PowerFlex_os-brick/e3d5a1f/DellEMC_PowerFlex_os-brick/126 : SUCCESS in 2h 02m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"61fbe82f63cc8409e6460deb473048c2c2a72b4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-22 18:00:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC_os-brick https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_Unity_FC_os-brick/1363071/DellEMC_Unity_FC_os-brick/201 : FAILURE in 2h 37m 21s\n- DellEMC_Unity_iSCSI_os-brick https://elab-os-logsrv.delllabs.net/93/926293/5/check/DellEMC_Unity_iSCSI_os-brick/e388c35/DellEMC_Unity_iSCSI_os-brick/198 : SUCCESS in 2h 13m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"8f38f298a628237113a20754e9c517a49330d3b1","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-22 19:14:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"674e526725a8794614aaeae4d0173d4be7f588a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 19:14:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"25a80eb0301434f3bea4f3b6f6557722907124d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:07:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a99751e661b24ce58cea7b31c090a630\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f681a0a608b4f838484667bc9d16ce8 : SUCCESS in 1h 39m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc816aef89f4f009d4225fa0d2a0185 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00503e8c646b4763bb05e14ca4222ff8 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c60e277381d5467388685a59d33d5adb : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af41344813e46b797c333ee3925d06b : SUCCESS in 5m 27s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf4cda0623f441dc9dea36d7633cc392 : SUCCESS in 1h 47m 46s","accounts_in_message":[],"_revision_number":5},{"id":"ba2904dc853f07e46a6e61c9f6d976307b8a276a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:07:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b38b77a145a02e012413e27e4304c4e4b86d39fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:08:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2ceb4b3ca14f2a8dee4de4ec365bfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efbeaf9c74749b8bee17c499c3a698f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2b42257eff03615816dc66d3171b87c22ad67eb1","revisions":{"1dd631a93d80d482d40eebe9e315190840a21ad5":{"kind":"REWORK","_number":1,"created":"2024-08-14 15:42:54.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/93/926293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/926293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/926293/1"}}},"commit":{"parents":[{"commit":"0ca33235d50e57248d16ce2da29e7f8e98314055","subject":"Merge \"Wait for multipath device to be ready for I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ca33235d50e57248d16ce2da29e7f8e98314055"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:40:52.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:42:52.000000000","tz":120},"subject":"Add get_passphrase_from_secret utility function","message":"Add get_passphrase_from_secret utility function\n\nThis function outsources Cinder\u0027s encryption passphrase handling based on Key\nManager secrets. It extends Cinder\u0027s existing implementation to handle secret\nobjects of type \u0027passphrase\u0027 additionally, for which the binascii.hexlify\nconversion is now skipped similar to Nova\u0027s implementation of handling\nqcow2+LUKS images. Original Cinder behavior for \u0027symmetric\u0027 secret objects is\nretained.\nProvides the basis for image encryption standardization as per\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1dd631a93d80d482d40eebe9e315190840a21ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1dd631a93d80d482d40eebe9e315190840a21ad5"}]},"branch":"refs/heads/master"},"16657fd3759bb5f1afbc3b116104d77ed04b0466":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-14 15:45:55.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/93/926293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/926293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/926293/2"}}},"commit":{"parents":[{"commit":"0ca33235d50e57248d16ce2da29e7f8e98314055","subject":"Merge \"Wait for multipath device to be ready for I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ca33235d50e57248d16ce2da29e7f8e98314055"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:40:52.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:45:31.000000000","tz":120},"subject":"Add get_passphrase_from_secret utility function","message":"Add get_passphrase_from_secret utility function\n\nThis function outsources Cinder\u0027s encryption passphrase handling based\non Key Manager secrets. It extends Cinder\u0027s existing implementation to\nhandle secret objects of type \u0027passphrase\u0027 additionally, for which the\nbinascii.hexlify conversion is now skipped similar to Nova\u0027s\nimplementation of handling qcow2+LUKS images. Original Cinder behavior\nfor \u0027symmetric\u0027 secret objects is retained.\nProvides the basis for image encryption standardization as per\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/16657fd3759bb5f1afbc3b116104d77ed04b0466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/16657fd3759bb5f1afbc3b116104d77ed04b0466"}]},"branch":"refs/heads/master"},"a220c4fbf4c7328ad75180deaf02791b4d1add66":{"kind":"REWORK","_number":3,"created":"2024-08-20 15:10:23.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/93/926293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/926293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/926293/3"}}},"commit":{"parents":[{"commit":"0ca33235d50e57248d16ce2da29e7f8e98314055","subject":"Merge \"Wait for multipath device to be ready for I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ca33235d50e57248d16ce2da29e7f8e98314055"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:40:52.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-20 15:10:13.000000000","tz":120},"subject":"Add get_passphrase_from_secret utility function","message":"Add get_passphrase_from_secret utility function\n\nThis function outsources Cinder\u0027s encryption passphrase handling based\non Key Manager secrets. It extends Cinder\u0027s existing implementation to\nhandle secret objects of type \u0027passphrase\u0027 additionally, for which the\nbinascii.hexlify conversion is now skipped similar to Nova\u0027s\nimplementation of handling qcow2+LUKS images. Original Cinder behavior\nfor \u0027symmetric\u0027 secret objects is retained.\nProvides the basis for image encryption standardization as per\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a220c4fbf4c7328ad75180deaf02791b4d1add66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a220c4fbf4c7328ad75180deaf02791b4d1add66"}]},"branch":"refs/heads/master"},"d436d39b6e3ac6cbc7c1c972446075b8b6c6f185":{"kind":"REWORK","_number":4,"created":"2024-08-21 09:16:00.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/93/926293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/926293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/926293/4"}}},"commit":{"parents":[{"commit":"0ca33235d50e57248d16ce2da29e7f8e98314055","subject":"Merge \"Wait for multipath device to be ready for I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ca33235d50e57248d16ce2da29e7f8e98314055"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:40:52.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-21 09:15:49.000000000","tz":120},"subject":"Add get_passphrase_from_secret utility function","message":"Add get_passphrase_from_secret utility function\n\nThis function outsources Cinder\u0027s encryption passphrase handling based\non Key Manager secrets. It extends Cinder\u0027s existing implementation to\nhandle secret objects of type \u0027passphrase\u0027 additionally, for which the\nbinascii.hexlify conversion is now skipped similar to Nova\u0027s\nimplementation of handling qcow2+LUKS images. Original Cinder behavior\nfor \u0027symmetric\u0027 secret objects is retained.\nProvides the basis for image encryption standardization as per\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d436d39b6e3ac6cbc7c1c972446075b8b6c6f185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d436d39b6e3ac6cbc7c1c972446075b8b6c6f185"}]},"branch":"refs/heads/master"},"2b42257eff03615816dc66d3171b87c22ad67eb1":{"kind":"REWORK","_number":5,"created":"2024-08-22 07:59:26.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/93/926293/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/93/926293/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/93/926293/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/93/926293/5"}}},"commit":{"parents":[{"commit":"0ca33235d50e57248d16ce2da29e7f8e98314055","subject":"Merge \"Wait for multipath device to be ready for I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ca33235d50e57248d16ce2da29e7f8e98314055"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:40:52.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-22 07:59:07.000000000","tz":120},"subject":"Add get_passphrase_from_secret utility function","message":"Add get_passphrase_from_secret utility function\n\nThis function outsources Cinder\u0027s encryption passphrase handling based\non Key Manager secrets. It extends Cinder\u0027s existing implementation to\nhandle secret objects of type \u0027passphrase\u0027 additionally, for which the\nbinascii.hexlify conversion is now skipped similar to Nova\u0027s\nimplementation of handling qcow2+LUKS images. Original Cinder behavior\nfor \u0027symmetric\u0027 secret objects is retained.\nProvides the basis for image encryption standardization as per\nhttps://specs.openstack.org/openstack/cinder-specs/specs/2024.2/LUKS-image-encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I8836fe3c2af5b61ba33a7fe3c8eb5a7f4961c515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2b42257eff03615816dc66d3171b87c22ad67eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/2b42257eff03615816dc66d3171b87c22ad67eb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
